)]}'
{"id":"openstack%2Fmanila-tempest-plugin~993836","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I2f5966643b754619abb30bdb617689243405a953","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-07-23 18:05:05.000000000","reason":"Change was abandoned"}},"hashtags":["manila","remove-enforce-scope-flag"],"change_id":"I2f5966643b754619abb30bdb617689243405a953","subject":"Use the new variable to enable the RBAC new defaults","status":"ABANDONED","created":"2026-06-17 20:22:10.000000000","updated":"2026-07-23 18:05:05.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f419e211dd53b4c7a39135bfa10a9decd445d435","_number":993836,"virtual_id_number":993836,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-17 23:07:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2026-07-17 16:59:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-17 23:07:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-18 00:47:12.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-07-17 16:59:44.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"c83838e640cd1cd7cf455cfd8a0b8c607d1971a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-17 20:22:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7acf662bab8dcadd3400e1942b8dd1aba8a3787","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-17 20:26:34.000000000","message":"Hashtag added: remove-enforce-scope-flag","accounts_in_message":[],"_revision_number":1},{"id":"0b5a20ed335a47c0a9122cee54a7c0b80bcdc101","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 23:07:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b4ffdc785d746fc88c5f5b5d3329ac3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d9052ebcb44da5987ad26d1e801254 : SUCCESS in 2m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8c5264794e2f47559d21b4c907ee437c : SUCCESS in 1h 05m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/af71fea5d4a8487b866c74e7e7c040d7 : FAILURE in 2h 11m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8c9f1ce31a7a4a5795c971f4d564a813 : SUCCESS in 2h 34m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b478c634efec47349aa9b54af98df8b3 : SUCCESS in 1h 02m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a9b41f7e2aaf4d8192bfab93c78cc761 : SUCCESS in 41m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cf241a41cb7849ef972acc63267b6f62 : SUCCESS in 53m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e3854f729c4f41c59dd5e64db5630136 : SUCCESS in 33m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/565d496a504a4062a000628e8142c292 : SUCCESS in 37m 49s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/f09c50aaaa244c23b9b9cec4e39fca3e : SUCCESS in 36m 24s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/ac21b46963364112869af8d6ce51454d : SUCCESS in 28m 54s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/094cba91a76c459d9865dddfdc7e6f93 : SUCCESS in 1h 09m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/824de83cc3ca45a89ae82c70dbe4ee19 : SUCCESS in 20m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3fb40629968e4a4fbbfdfa89a3ddba71 : FAILURE in 16m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"beceb4d71676f4afd5f544ca808c5becd4decd95","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-18 00:47:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/993836/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/f7be751/ : SUCCESS in 1h 19m 47s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"81de151b94dcbffd8301b696339847d7fc2717d7","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-17 16:44:24.000000000","message":"Hashtag added: manila","accounts_in_message":[],"_revision_number":1},{"id":"bf8c517fd8d3b43e5c2592cbde071bcc49747c9d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-07-17 16:59:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f419e211dd53b4c7a39135bfa10a9decd445d435","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-23 18:05:05.000000000","message":"Abandoned\n\nthis is not needed as we are enabling SRBAC by defaults and remove the glance SRBAC disabling in https://review.opendev.org/c/openstack/manila-tempest-plugin/+/997777","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8bdda9ba302a7ffd042b1550e0cabe7e473612fd","revisions":{"8bdda9ba302a7ffd042b1550e0cabe7e473612fd":{"kind":"REWORK","_number":1,"created":"2026-06-17 20:22:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/36/993836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/36/993836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/993836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/993836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/993836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/36/993836/1"}}},"commit":{"parents":[{"commit":"fc809ff749625f28a276d2f5ab7c1cb12f992569","subject":"Merge \"Replication: verify access rules in detail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc809ff749625f28a276d2f5ab7c1cb12f992569"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-17 20:17:44.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-06-17 20:22:09.000000000","tz":0},"subject":"Use the new variable to enable the RBAC new defaults","message":"Use the new variable to enable the RBAC new defaults\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/993821\nDepends-On: https://review.opendev.org/c/openstack/manila/+/993826\n\nNeeded-By: https://review.opendev.org/c/openstack/oslo.policy/+/986475\n\nChange-Id: I2f5966643b754619abb30bdb617689243405a953\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8bdda9ba302a7ffd042b1550e0cabe7e473612fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8bdda9ba302a7ffd042b1550e0cabe7e473612fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
