)]}'
{"id":"openstack%2Fmanila-tempest-plugin~994878","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I8840e31b108f61b58361d04c434c090c81e36931","project":"openstack/manila-tempest-plugin","branch":"master","topic":"snapshotinheritshareruletests","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2026-08-31 17:41:03.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-08-31 18:58:58.000000000","reason":"Change was submitted"},"38607":{"account":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"last_update":"2026-08-31 18:58:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8840e31b108f61b58361d04c434c090c81e36931","subject":"Add tests for new snapshot_inherit_share_access share type","status":"MERGED","created":"2026-06-25 11:06:12.000000000","updated":"2026-08-31 18:58:58.000000000","submitted":"2026-08-31 18:58:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994878-snapshotinheritshareruletests","meta_rev_id":"d72f2ef571a2f452f6dc22f8b24c82a99e1d727b","_number":994878,"virtual_id_number":994878,"owner":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-31 18:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"all":[{"value":2,"date":"2026-08-28 11:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":2,"date":"2026-08-31 17:41:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":1,"date":"2026-08-31 17:41:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-25 12:21:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-26 10:42:28.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-07-08 05:14:29.000000000","updated_by":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2026-07-08 05:14:29.000000000","updated_by":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2026-08-28 11:58:54.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"}],"messages":[{"id":"3159a5103e0058357b1853719db31e90539de03a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-06-25 11:06:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"612b8d2defac1969e21cbef152b9e4b4d9296ad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 12:21:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd789e9c45174866a7c8ff9683da791b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d59370b09834c598fd5b8a4ec2f922c : SUCCESS in 2m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/da5416f785734ac58fd778266bec70a3 : SUCCESS in 36m 13s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/61001e9321364cbab00bc2841f2171b6 : SUCCESS in 57m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/73bc988ac2d14bbf97d8559f0d2c80cb : SUCCESS in 35m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ba22a0ce906d46839d9d6d7150502b86 : SUCCESS in 58m 29s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/71b7fcb531ce48edaafda75c7f0f8e2d : SUCCESS in 47m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/499f5d98aa794e84b2bdf2226fb4fda4 : SUCCESS in 46m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6c7552529ca5406c87b72f1db8d55c11 : SUCCESS in 30m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f6c7938f4f174d2188523b471e5e3ab5 : FAILURE in 16m 43s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/bd9fdfa9d1c14cdb8072e14690830834 : SUCCESS in 1h 06m 53s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/fba63e3145d64474ac6310f13be82426 : SUCCESS in 35m 23s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/db991810594346c888506ac1a6818982 : SUCCESS in 1h 09m 49s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ed20e5de3edf4f36a7ddd310d51f4723 : SUCCESS in 25m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/49dd730bca104a5c9fadc6fe14ac4fdf : FAILURE in 19m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b58ff66b70468f5664d4b582b0afd599cc2a6de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-06-25 12:42:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1e38f3266549f1eca6c93c80210ea341cb4f76f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 14:40:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00d828ab57314959bae7c47d3779709f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ecd7b7f4570432d9fa45816d246d864 : SUCCESS in 3m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/66f056d848de4bcfbdee1699f53559c4 : SUCCESS in 1h 18m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/25ba2dd09d744b22b51b7c0f2555a0d5 : SUCCESS in 1h 51m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/92e05158a37044849d79adb8536b7502 : SUCCESS in 1h 10m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/556177b928034f4dbb082f28550232cb : SUCCESS in 29m 40s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/543a85b8773447a4b7a47c542150a341 : SUCCESS in 1h 44m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d33207ef6b5e48cfa8c2b26efa3231e8 : FAILURE in 35m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e0d5c2c1cb0947138d2dd1c6d25bd902 : SUCCESS in 29m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a782555b433c4a7e878ed012f99aff95 : SUCCESS in 43m 17s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/1a136b2276e042cfbbbaf9b0b1220ee8 : SUCCESS in 33m 08s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/0c76d2c7c8a84fa0be9158b334fccc11 : SUCCESS in 1h 10m 56s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/6eaf44247afc4687ad3c0398227b1a77 : SUCCESS in 33m 30s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/df6fa539f5c34ccaa549878d2ac81e70 : SUCCESS in 23m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f25419bc49d54c97b7333bf11c60b237 : FAILURE in 13m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0633f442904ecf4fdc5dc5893d705d87ca7ea0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-06-25 15:09:29.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"215f79cad796631ddfee340daa0c66e44b51c8fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 17:34:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1abd55ae1f634ad18bee8bf3d29c2508\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb02c4ce39544758988a69a63cd1ff60 : SUCCESS in 3m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9847df235a564102830246d37ac19a7e : FAILURE in 1h 16m 14s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/35863c2707ad4aa38471b184040f44da : SUCCESS in 2h 17m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/341146efadfd4bc1b274702dcf957851 : SUCCESS in 1h 32m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/cfed11100cb54fa99446cd5c29ac31be : SUCCESS in 1h 01m 41s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ff40897815e748a087f14c578e82f47c : SUCCESS in 48m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a8a3e453ec4b4988b4d6c3064a86d1c2 : FAILURE in 1h 16m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/78cc3acb19ff40148d9db5003b2dbfd9 : SUCCESS in 27m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/14e080b631f8439183ce669d92f5bad5 : SUCCESS in 46m 28s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/360d03642a0c42f08982e67745c1e243 : SUCCESS in 1h 13m 30s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/5d09c7f7b649490b9e5e40a2c9a5bc9f : SUCCESS in 1h 09m 50s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/96cb23bed1c04699a4a0c4eff7047a8c : SUCCESS in 1h 25m 17s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8f78a5f12fcf450798b92d82b2f95e90 : SUCCESS in 32m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d358978d811f4905b6ba7360d2b30de6 : FAILURE in 20m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f7255b82689a51f17125918553995c7e85e2d79c","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-06-26 06:59:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e775dc7a35009723e52ed2f348bcacf150dccb5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-06-26 07:16:48.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"80d96ddd42fac59f8d306375bb27bdcd11915191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 09:16:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6edcb64dc1774943beda34e37b58cbac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354216589acc49faaf35e18fe24bdd6e : SUCCESS in 2m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5a51ff0cb66b43b7b81b79251e3c4ecd : SUCCESS in 1h 09m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6c6895ba32ef49768ff636a4c5d6e927 : FAILURE in 1h 56m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/13c7c43c1d794b57a8d796a9f48ed803 : SUCCESS in 38m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/88181e6f411c4f93bd1e2d2fbb14a139 : SUCCESS in 59m 23s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b9ef9340d9aa4db39a5f4d11035cd4ee : SUCCESS in 41m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/de8b086d43774d9da0b3ca5131a3fbb0 : FAILURE in 41m 59s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f8ac38008a98422d933450d81af971f9 : SUCCESS in 37m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5194af0ebb094c218e7843c01f30a42d : SUCCESS in 43m 22s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/c964e5b7d28c441fafc5035322842687 : SUCCESS in 31m 59s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/00147919e18c440f84a6e0864a0a76b8 : SUCCESS in 1h 01m 29s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/42161822a4b64257a2e12b8dc69a4b5a : SUCCESS in 1h 20m 54s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0d26ccf18c28432e89d73e14cdefa4ca : SUCCESS in 32m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/19fafabce6c24bec90a4fe5ac036bd4a : FAILURE in 20m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"510fe2153f84e4de1633580d825f483a4c8aaa41","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-26 10:42:28.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/994878/4/upstream-check-manila/manila-tempest-plugin-ontap-dhss/f76b666/ : SUCCESS in 1h 23m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"8d7a1d4e484f6222dec9767a70d81bbc205144e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-06-29 09:32:38.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"b46bda94a707d90f52678a56bf931cd802e3bf53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 11:17:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67931179472e461aba4ef237ea36289e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1884b692134b41d19de40ca067a9aa58 : SUCCESS in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/93c46937213c4e68be666c735eaf843b : SUCCESS in 1h 24m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7cbfc86aed60474b9c62342d9dc431ea : SUCCESS in 1h 43m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/00adbac7637748d0999efd4a54da5202 : SUCCESS in 1h 17m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d974de3927304ab5a14cf08b9421f6a2 : SUCCESS in 1h 12m 26s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8050d07320224ee396815d3dd63ed4dd : SUCCESS in 1h 25m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5c09277931d74757a8dcab7efcc92c1f : FAILURE in 1h 03m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c786b9d7332a4e1497046fef840d6c7a : SUCCESS in 22m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ce37ddfd780a409592ea09dbce4d9c85 : SUCCESS in 42m 19s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/a34ca89909c54541a3b39090a5e4590d : SUCCESS in 1h 09m 24s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/cca9d787f2774d468e1de3aa30792329 : SUCCESS in 1h 06m 14s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/b6db3961f01d4e89ae0691b7eb504a7a : SUCCESS in 47m 17s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/40205ef447524bab8a15bd15b4a006d2 : SUCCESS in 28m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0ba876b9abba4198b53a6ad51ebd113b : FAILURE in 16m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0e325008f3dabd85579e2a4fdd6382b2653a0ff9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-06-29 12:52:21.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/994878/5/upstream-check-manila/manila-tempest-plugin-ontap-dhss/7d3f989/ : SUCCESS in 1h 32m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"6a7e6be37a38603e10759bf73dea18782aaf7f52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-07-06 05:29:56.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"c0e992071574b5e7b14558703eeb7e4988f4b8ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 06:40:31.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/992850 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"f91693e292b64a87d924c454d233d0198ba07f57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-07-06 06:43:14.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"5310c0a943437191b4d5213383aaf0dde700e762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 09:34:59.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9b736b3c31d4d01a8e83f4c78e9e4e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e838d6b2b64ddc892e488c54629e42 : SUCCESS in 3m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b23f378f75ea4133a194863d6dfe2d71 : SUCCESS in 35m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/864d2c09408243788b085f54e3a1fa33 : FAILURE in 2h 44m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7e8a4205ba7946ee906bc181922a2e9f : FAILURE in 1h 09m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d26ec5a720e247068c819702bfeb1213 : SUCCESS in 1h 07m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8a26ce4bbb204cde9955c85acbd49862 : SUCCESS in 1h 34m 59s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d3d241b9208847cd85196493fedc8964 : FAILURE in 1h 09m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ced7c5305bac452bb00952d2a817b74b : SUCCESS in 24m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f1518c4a021b4ed4a19435696f916b47 : SUCCESS in 34m 38s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/acd864abb2c64963be3da9e78e47bd53 : SUCCESS in 34m 49s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/a19e73a1db1345b6b0afb5771df5b420 : SUCCESS in 33m 00s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/5cb92a39d7c94923a1f4f6de4987322f : SUCCESS in 57m 49s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0d5721fd988a450fa3430023e0f7c7f8 : SUCCESS in 20m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e380b704627b495bbf08f61f014ebf7a : FAILURE in 19m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"012ac468acadc3feb8f37960fc162a7ffecec2a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-07-06 10:43:54.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"0a28aa6854b3f03c85edfbf9addd0a15f209b199","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 13:30:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb985a50dd2743e7bf326427f6179d51\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1933fb306f244ce3995e3f40fdd5573e : SUCCESS in 4m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f0ff14919ac2433ba97092e0c7cc3db3 : SUCCESS in 57m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4f8aaca525584c0cbde0077ad3320657 : FAILURE in 2h 23m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c648bf67679f494c92ffb8cea2458340 : FAILURE in 56m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/33985e389a5245c0a1e6e8cb8090da2d : SUCCESS in 1h 06m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/444707ddc75e45b895ed19df18cf64d9 : SUCCESS in 1h 12m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6ab8aca690ff4bf8b09037811570dcae : FAILURE in 1h 09m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/656861da98e141ac8350e9cb88e0f0dd : SUCCESS in 34m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0111d8fd346749bd801a660e84612b4e : SUCCESS in 27m 15s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/91efb366b07046e5bc51ffa9cc8684f9 : SUCCESS in 1h 10m 04s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/5850fd1d30b540c9b8e648708454420c : SUCCESS in 36m 06s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/cdc15d650b744f60ae69ac1367a92e9d : SUCCESS in 1h 13m 49s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ee332747ef814909af0368ae37e69e3c : SUCCESS in 31m 09s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/155e67e0f931442b8b6a9aac7f2877c5 : FAILURE in 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"30cb78c794ce536bd644a7f6ff9528bb886d3841","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-07-07 05:54:24.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"a0097977be6015561505aca2f106727b8241015c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-08-07 06:58:16.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"0d674bd55501f20e7a12769590944059c5d1b66f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 09:22:40.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d935cc6265d4df785911798ddf03f69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b4470664f44ff48623c5f792c3f37b : SUCCESS in 4m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/46f1593b97514c31a9dbe2e57d9a786f : FAILURE in 1h 47m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b0f8457bc4b3454393885aedd573dbda : SUCCESS in 1h 50m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a7b93c1d51554dfba459cb4106255a20 : SUCCESS in 1h 18m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9a6c102841564c039511a49d9818fbcf : SUCCESS in 1h 02m 27s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c3d63c6e14934facaaf8a41c0ca374a5 : FAILURE in 2h 09m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/31a68690a6854980b03ed361f7604915 : SUCCESS in 1h 10m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a45ac69923304057b371f100bc74cead : SUCCESS in 35m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/11ec3cb91e474d6fba52ee8dc051b366 : SUCCESS in 35m 37s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/9bef1b5a054a41eb87bff8b36bfb87fa : SUCCESS in 51m 40s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/30872f59ed194df992f9c22c5c2f40cf : SUCCESS in 50m 03s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/35814f6415fa455f981d234cab2dd95b : SUCCESS in 52m 27s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/f4098b14e3e346dea2d85374645d28c2 : SUCCESS in 29m 13s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/45bea8439d244b4c810d9c46f5840d33 : SUCCESS in 56m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e7ce76f36aba46020ee189d604d1ac915f44be35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-08-07 11:34:32.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"dc5f3af356ec4f17c4e13e8abfdd9ee037c8b2fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 14:08:38.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f268c7d6df4482d97ad87ee0be550b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f49c4b8b4b344e0888bf99c11156d938 : SUCCESS in 4m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0dab23811c2640f9a396137d48172c7a : FAILURE in 43m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f61c8845b6584f5892d192588f2399c9 : SUCCESS in 1h 58m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/047e5542c6db47d8b02b675eb07e40da : SUCCESS in 1h 17m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/01208c7139ed41e185817e04e0446ad0 : FAILURE in 1h 10m 10s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4e2f7c99f6ba4a679c0324e655158117 : FAILURE in 2h 20m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/94aa300a4168430abbddc0f3fc826a0e : SUCCESS in 1h 06m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/de11182b35a0480bac819c3aa8cbc943 : FAILURE in 20m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/11b34b3aa5054898b3540837133622f7 : FAILURE in 26m 08s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/63fe75d9f6f546fcb89ffc5c0d55be15 : SUCCESS in 1h 14m 45s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/253f12e2fff74d728cd02b1184741cd8 : SUCCESS in 55m 30s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/9e8b8a0673df41048f109b73d906339c : SUCCESS in 1h 05m 35s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/590e5dfb88834cd58aff8b567d78c606 : SUCCESS in 19m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/aa1fe5c84fed46dbb3ba939990d9b947 : FAILURE in 44m 55s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"21643966bf8383f113be1cc1c1d99f33219d22a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-08-09 06:50:19.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"7f987b95420cf1f7e4338497ee66c891aa920699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-09 08:59:57.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7345e5a4a1534db681de36bffab4b874\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e2580d7e96f447485b89bc3cd4a37b9 : SUCCESS in 2m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/79f7ec7b1f3843d78fe0e09a6ea07a30 : SUCCESS in 1h 02m 14s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2bfd314a3cdf4463bc4772d94d3b26f4 : SUCCESS in 2h 04m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/58df3b6d0885466daf7e759c69d34f13 : SUCCESS in 1h 16m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/33396ac38341482fbcd0bed0a65326d9 : FAILURE in 1h 06m 12s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ddfd1b1d0ed84bfdbb6473343d467f87 : FAILURE in 1h 34m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7ed847a189314701ac498ff029ae605a : SUCCESS in 55m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8cc3e33099a14dda9c776cf3952be6ab : FAILURE in 24m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d00ec03bc7c5492fa91d7684f92d2fa6 : FAILURE in 25m 40s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/15cb685b2345489a9c89585210cef250 : SUCCESS in 46m 57s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/369d83a86ec7423abd6b2005216f6331 : SUCCESS in 1h 16m 19s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/4cc263c41a55451b92667ef0f4c16506 : SUCCESS in 35m 21s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/e912a21438c1404d9f7a081edade8ffc : SUCCESS in 31m 17s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bc2b5235642240fe96808cf4bb78c99a : FAILURE in 30m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a3c7317d550159746c8e38abc1cea34b634c0ad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-08-09 09:55:46.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"04195dabf071c51c28e894a6e55e6b4cec3c4ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-09 09:57:30.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/manila/+/1000068 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"0133d08a8cd26097c7646693466a5011cd8dd60e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-08-09 16:57:19.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"c02a0efedd058f4b024ded42c4ca748c0af2b4d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-09 19:09:10.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/489f56f6b0894d9d982e92c3116496fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50eec46c71f24dcd8fa378e33411d358 : SUCCESS in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ec05c02439ce40d483e415d1296b1c6a : SUCCESS in 1h 11m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2942d416a974488eac5aaf1905d10774 : FAILURE in 2h 04m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6dc526d3f1534c26b9f46d9a6869dd6f : SUCCESS in 1h 03m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b9a827f678584796866c52635707c9e4 : SUCCESS in 1h 12m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f2563d1cda424f07a7cc1458f4b5eea5 : FAILURE in 2h 05m 54s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d7e9e469839e436f98beb8942c240248 : SUCCESS in 1h 10m 03s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/874b51c816254458867e06131bc55a01 : SUCCESS in 33m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ea008091e77e4557b8500091909a63b6 : SUCCESS in 29m 25s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/fb1958d99ae842829c20a0a100fcecf2 : SUCCESS in 1h 02m 41s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/ada43474db954bf98731ae1273239150 : SUCCESS in 34m 25s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/c01c1c19810f4334933a339a2f16cce7 : SUCCESS in 1h 11m 55s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/3e110544376341d680aab0fedda36991 : SUCCESS in 21m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cc1d6eb943fd4546a6fc2e45a3b9667d : SUCCESS in 27m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"99ec0009a32f75f826dbf4d26e0f685eadab05f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-08-17 18:15:15.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"8a487a86e2888fcc86ecc4fde1f93255573a8997","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 20:55:51.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2650bd14278b4b74945067ec8407d135\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/623db326212c440f878504423e5853cc : SUCCESS in 4m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fd251fe9eb8248cd9e6b35371366c38b : SUCCESS in 51m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bd674cffe5d74f66b074bcbc243a03c3 : FAILURE in 2h 28m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5528c2c9112c4735a607777fe2b1bc6e : SUCCESS in 1h 08m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ebe1fc8eea6e47dca9821c4264e1179e : FAILURE in 54m 13s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/72eef83c71f84248aae51b32c6695958 : FAILURE in 2h 17m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ad99485da1854129a377b7897c1f43e4 : SUCCESS in 53m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/17c537418f3a4cf09b77a2f16b4a70b0 : SUCCESS in 30m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/edc386773d894675a23a0249c012aec3 : SUCCESS in 40m 58s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/0f8aef65d20b466cb510e9d667376d80 : SUCCESS in 1h 08m 10s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/3a40648ecacf45a2a6ba632d1f78fed1 : SUCCESS in 47m 19s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/977a0ab262e64842b6c4460d824979b3 : SUCCESS in 1h 01m 30s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/88370a8ea207403a82d845d6572c272a : SUCCESS in 32m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/51e8bebb92b94ee2af4018f31c4991e5 : SUCCESS in 54m 09s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9de583738076178b412c57af17b9cfb31e4128cd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-17 21:03:51.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/994878/14/upstream-check-manila/manila-tempest-plugin-ontap-dhss/a27538e/ : RETRY_LIMIT in 7m 15s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"106555f710d10cf25cb0a4b415a922da6a303600","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-08-24 14:11:04.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"9c5eaedfd26d4044bffa7b7ca552c2853cc2a6bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 16:25:32.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd496c99d4b3492f9cc3dc4f474fac20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e10ef220ecbf4bb49818e8ef4af91097 : SUCCESS in 3m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/57321ebee8d749ccb11fbc431c2e19a9 : SUCCESS in 1h 05m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dd830be293354421abe5277c03dec22a : SUCCESS in 1h 41m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a549de6601624c1393e89f061e676a47 : SUCCESS in 1h 23m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/42143dfe3b4741b48624d737a2bf0e91 : FAILURE in 42m 43s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c133dc35e9434660a9c606de3e6e2fc4 : FAILURE in 2h 06m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f773dedb4b604c7d8aa9a85541f8a339 : SUCCESS in 1h 03m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/235af56178a34885b3262a7ef3d2c641 : SUCCESS in 38m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cd1cbbffb7034bdda56a170cd7e7af71 : SUCCESS in 39m 47s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/3b21590b64984fb2b274459f5d5af1ad : SUCCESS in 1h 02m 44s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/581ef4cd87eb441d806e761b9cd77a6c : SUCCESS in 1h 05m 38s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/6f8bd573585547c19b5e5b298d1beb49 : SUCCESS in 1h 09m 36s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/127eef1ae39746e4b516a8e56bc01c83 : SUCCESS in 34m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6bf5ab95498d48ed8f2525f6661fff25 : SUCCESS in 1h 01m 23s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"343d988b612271d8751fb376c28822e4dfafc490","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-24 17:43:39.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/994878/15/upstream-check-manila/manila-tempest-plugin-ontap-dhss/9f8ea58/ : FAILURE in 1h 15m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":15},{"id":"5f6982f7e964b88a23847499f8455e98ffe87485","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"real_author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"date":"2026-08-26 14:30:28.000000000","message":"Patch Set 16: Patch Set 15 was rebased on behalf of \u003cGERRIT_ACCOUNT_38607\u003e\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"}],"_revision_number":16},{"id":"9ae49e62af886a37e635450a06e84c59f51d123e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 16:44:36.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc07b6139f0b4fb8b208a92b76935dcf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67871f0e52d54eafbb85e56440d2a0f0 : SUCCESS in 3m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e5cb36d97c334173b2b50409e0dbc434 : SUCCESS in 1h 16m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b05636a0dc9f46f5a4bb60ab251576f1 : SUCCESS in 2h 00m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e816267f54e3449ba022e2c38b968929 : SUCCESS in 50m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ac4087a071ce423ebca27990ecbbc6fd : FAILURE in 36m 34s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4a1267db1bcd4e51b8b8c22f429c03d9 : FAILURE in 2h 11m 08s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/69a573c66b75466d9cf4021674f710cd : SUCCESS in 1h 12m 17s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/f83dd73876ce40de9c15bb7260db2949 : SUCCESS in 43m 20s (non-voting)\n- manila-tempest-plugin-lustre-scenario https://zuul.opendev.org/t/openstack/build/dea2cd753d514154a83a6e3a0f381586 : FAILURE in 37m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/92b6593c04fd47ab872a5ce61d7385c7 : SUCCESS in 34m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1b85b53a2d194df0a088bfa6641221ad : FAILURE in 41m 16s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/e1b75b579c9545ffb0bc96daf11490c7 : SUCCESS in 1h 09m 32s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/0240219f7d8148f592dd26e3321c6b81 : SUCCESS in 1h 08m 51s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/5fc339ec83274ed0a83b008abb4f10c6 : SUCCESS in 49m 39s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/fe6e068798764f1cb7bc9206fddfea4c : SUCCESS in 26m 02s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5b2f6fa1e1b04b36a31cd4fe568ed08d : SUCCESS in 28m 13s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"5705becebfb4d616af82a003024715e091c5be41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-08-26 17:39:53.000000000","message":"Uploaded patch set 17: New patch set was added with same tree, parent tree, and commit message as Patch Set 16.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"38c5d73a1587efc50ab24cc46d237e124be1e990","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:22:13.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50b9dea66fb54ba28cbbb14e131d777f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d5d24c8bd704febbfec14cf1bfd17ff : SUCCESS in 3m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7ea43517ceae45aa9f0ac6ee861390ab : SUCCESS in 50m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/314ee67438344b7b9c5e9974fa2260e9 : SUCCESS in 48m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ef0cb32eee9644d8b7171a12d9c2421c : SUCCESS in 1h 35m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6ddb71f76da54b299cd4f9ed8bc0780e : FAILURE in 38m 38s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e5888c8f19ac48ce846f901b716819b6 : FAILURE in 1h 15m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3a6e497cae9b4cc697564afd32a93bc7 : SUCCESS in 1h 07m 06s (non-voting)\n- manila-tempest-plugin-lustre https://zuul.opendev.org/t/openstack/build/734ace00fb844dc49199d4c5bd8e5d20 : SUCCESS in 1h 01m 40s (non-voting)\n- manila-tempest-plugin-lustre-scenario https://zuul.opendev.org/t/openstack/build/d6cd21a0d0a74927beb23c19b6168373 : FAILURE in 1h 04m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7515b713280b46d89e28067ed0e5cdcd : SUCCESS in 38m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/086d888a306548df9e3d987966555901 : SUCCESS in 31m 26s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/7b5843b0a7f24b96b6ec8ee60d8d9abc : SUCCESS in 50m 26s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/5d80eb8b08c74e79a94c6e12445c1913 : SUCCESS in 1h 01m 25s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/6831b670e7fb4488ad46ef97f0d22995 : SUCCESS in 51m 47s\n- manila-tempest-plugin-dummy-no-dhss-old-rbac https://zuul.opendev.org/t/openstack/build/883059d6fd114e85b4eef0b0e041deec : SUCCESS in 23m 13s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1f17a2e3fd714cadb5c02f992d723385 : SUCCESS in 54m 26s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"32ab48cb8d83309ba7fdcc29f3f62e67dcc2d9e7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-08-26 20:33:22.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/78/994878/17/upstream-check-manila/manila-tempest-plugin-ontap-dhss/e427792/ : SUCCESS in 1h 08m 35s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":17},{"id":"32061471704797a0824da422c90dcedb9c85b5d1","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2026-08-28 11:58:54.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"bab206e798903abbaf347e31338184fe93d3d8d7","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2026-08-31 17:41:03.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"fcb9336ca7aae8f630be14ee9fdc32df28a571b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 17:41:32.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"f31bae53a32e31af34c437cfecc40bda57ba3dfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:58:57.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb1a96fe4a81473cbcf5d026b1f30076\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/675c0c9f1000498fb16f59f7581431fe : SUCCESS in 3m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b24e0dea009541debf1ef1f69c922659 : SUCCESS in 1h 11m 17s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/06b7679ac992453da5146305c890bf62 : SUCCESS in 40m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7d34907aa4ee4efdb4e6c7caf7ce5419 : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":17},{"id":"d72f2ef571a2f452f6dc22f8b24c82a99e1d727b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:58:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"d4a7ddac3b44c123fa3e7af5deda58361456968e","revisions":{"ea5c06161e85399c784ae394e5142d903a664137":{"kind":"REWORK","_number":1,"created":"2026-06-25 11:06:12.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/1"}}},"commit":{"parents":[{"commit":"afb832b53c04aee8c121fd9048f1480fe6459918","subject":"Merge \"Add replication_policy metadata tests for SnapMirror policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/afb832b53c04aee8c121fd9048f1480fe6459918"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:19:00.000000000","tz":330},"subject":"[WIP] - Add tests for new snapshot_inherit_share_access share type","message":"[WIP] - Add tests for new snapshot_inherit_share_access share type\n\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ea5c06161e85399c784ae394e5142d903a664137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ea5c06161e85399c784ae394e5142d903a664137"}]},"branch":"refs/heads/master"},"3cc1144db092a73be054f92046a880b249dc131e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-25 12:42:04.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/2"}}},"commit":{"parents":[{"commit":"fc809ff749625f28a276d2f5ab7c1cb12f992569","subject":"Merge \"Replication: verify access rules in detail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc809ff749625f28a276d2f5ab7c1cb12f992569"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 12:42:04.000000000","tz":0},"subject":"[WIP] - Add tests for new snapshot_inherit_share_access share type","message":"[WIP] - Add tests for new snapshot_inherit_share_access share type\n\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3cc1144db092a73be054f92046a880b249dc131e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3cc1144db092a73be054f92046a880b249dc131e"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"afb832b53c04aee8c121fd9048f1480fe6459918","ours":"ea5c06161e85399c784ae394e5142d903a664137","theirs":"fc809ff749625f28a276d2f5ab7c1cb12f992569","merge_strategy":"recursive","contains_conflicts":false}},"81885de295b8cb6fbf4fde80c000531f87a7f6ad":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-06-25 15:09:29.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/3"}}},"commit":{"parents":[{"commit":"fc809ff749625f28a276d2f5ab7c1cb12f992569","subject":"Merge \"Replication: verify access rules in detail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc809ff749625f28a276d2f5ab7c1cb12f992569"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 15:22:04.000000000","tz":330},"subject":"[WIP] - Add tests for new snapshot_inherit_share_access share type","message":"[WIP] - Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/992850\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/81885de295b8cb6fbf4fde80c000531f87a7f6ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/81885de295b8cb6fbf4fde80c000531f87a7f6ad"}]},"branch":"refs/heads/master"},"1569fbafe2994d894a381eace39b945e98f6f598":{"kind":"NO_CHANGE","_number":4,"created":"2026-06-26 07:16:48.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/4"}}},"commit":{"parents":[{"commit":"fc809ff749625f28a276d2f5ab7c1cb12f992569","subject":"Merge \"Replication: verify access rules in detail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc809ff749625f28a276d2f5ab7c1cb12f992569"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-26 07:29:05.000000000","tz":330},"subject":"[WIP] - Add tests for new snapshot_inherit_share_access share type","message":"[WIP] - Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/992850\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1569fbafe2994d894a381eace39b945e98f6f598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1569fbafe2994d894a381eace39b945e98f6f598"}]},"branch":"refs/heads/master"},"3ef55d014a0df2f87d24864656c3dfcf92fb868a":{"kind":"REWORK","_number":5,"created":"2026-06-29 09:32:38.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/5"}}},"commit":{"parents":[{"commit":"fc809ff749625f28a276d2f5ab7c1cb12f992569","subject":"Merge \"Replication: verify access rules in detail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc809ff749625f28a276d2f5ab7c1cb12f992569"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-29 09:45:39.000000000","tz":330},"subject":"[WIP] - Add tests for new snapshot_inherit_share_access share type","message":"[WIP] - Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/992850\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3ef55d014a0df2f87d24864656c3dfcf92fb868a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3ef55d014a0df2f87d24864656c3dfcf92fb868a"}]},"branch":"refs/heads/master"},"f577db293014d883c1f9e4ad394658d04be4dc90":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-07-06 05:29:56.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/6"}}},"commit":{"parents":[{"commit":"fc809ff749625f28a276d2f5ab7c1cb12f992569","subject":"Merge \"Replication: verify access rules in detail\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc809ff749625f28a276d2f5ab7c1cb12f992569"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-07-06 05:42:53.000000000","tz":330},"subject":"Add tests for new snapshot_inherit_share_access share type","message":"Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/992850\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f577db293014d883c1f9e4ad394658d04be4dc90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f577db293014d883c1f9e4ad394658d04be4dc90"}]},"branch":"refs/heads/master"},"faa4f311662b7b36cf89a5b6031406bf26a1e4bf":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-07-06 06:43:14.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/7"}}},"commit":{"parents":[{"commit":"563704d03e7f92565f922fdd644dbf376e9a4213","subject":"Merge \"Adapt to new testtools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/563704d03e7f92565f922fdd644dbf376e9a4213"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-07-06 06:43:14.000000000","tz":0},"subject":"Add tests for new snapshot_inherit_share_access share type","message":"Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/992850\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/faa4f311662b7b36cf89a5b6031406bf26a1e4bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/faa4f311662b7b36cf89a5b6031406bf26a1e4bf"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"fc809ff749625f28a276d2f5ab7c1cb12f992569","ours":"f577db293014d883c1f9e4ad394658d04be4dc90","theirs":"563704d03e7f92565f922fdd644dbf376e9a4213","merge_strategy":"recursive","contains_conflicts":false}},"46dab9a20c753fb8bd5e181681a1971dde76ac08":{"kind":"REWORK","_number":8,"created":"2026-07-06 10:43:54.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/8"}}},"commit":{"parents":[{"commit":"563704d03e7f92565f922fdd644dbf376e9a4213","subject":"Merge \"Adapt to new testtools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/563704d03e7f92565f922fdd644dbf376e9a4213"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-07-06 10:57:28.000000000","tz":330},"subject":"Add tests for new snapshot_inherit_share_access share type","message":"Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/992850\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/46dab9a20c753fb8bd5e181681a1971dde76ac08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/46dab9a20c753fb8bd5e181681a1971dde76ac08"}]},"branch":"refs/heads/master"},"035346ce599f1bd13f4d1d044c3276f7d291f87d":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":9,"created":"2026-08-07 06:58:16.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/9"}}},"commit":{"parents":[{"commit":"0ea09d9358cb616500207e25fe0a03ded70ed2f2","subject":"Merge \"[CI] Increase cephadm loopback disk size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ea09d9358cb616500207e25fe0a03ded70ed2f2"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-08-07 07:13:58.000000000","tz":330},"subject":"Add tests for new snapshot_inherit_share_access share type","message":"Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/1000068\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/035346ce599f1bd13f4d1d044c3276f7d291f87d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/035346ce599f1bd13f4d1d044c3276f7d291f87d"}]},"branch":"refs/heads/master"},"e5a13d1b648a51c8326c2fcb99209c212d9fc9e7":{"kind":"REWORK","_number":10,"created":"2026-08-07 11:34:32.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/10"}}},"commit":{"parents":[{"commit":"0ea09d9358cb616500207e25fe0a03ded70ed2f2","subject":"Merge \"[CI] Increase cephadm loopback disk size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ea09d9358cb616500207e25fe0a03ded70ed2f2"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-08-07 11:50:15.000000000","tz":330},"subject":"Add tests for new snapshot_inherit_share_access share type","message":"Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/1000068\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e5a13d1b648a51c8326c2fcb99209c212d9fc9e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e5a13d1b648a51c8326c2fcb99209c212d9fc9e7"}]},"branch":"refs/heads/master"},"fdf3b4322ea51e37db32345b3babcbb08dd1b01d":{"kind":"REWORK","_number":11,"created":"2026-08-09 06:50:19.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/11"}}},"commit":{"parents":[{"commit":"0ea09d9358cb616500207e25fe0a03ded70ed2f2","subject":"Merge \"[CI] Increase cephadm loopback disk size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ea09d9358cb616500207e25fe0a03ded70ed2f2"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-08-09 07:05:27.000000000","tz":330},"subject":"Add tests for new snapshot_inherit_share_access share type","message":"Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/1000068\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fdf3b4322ea51e37db32345b3babcbb08dd1b01d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fdf3b4322ea51e37db32345b3babcbb08dd1b01d"}]},"branch":"refs/heads/master"},"39e03792aace4df07a848507f3dbdcee004a74dc":{"kind":"REWORK","_number":12,"created":"2026-08-09 09:55:46.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/12"}}},"commit":{"parents":[{"commit":"0ea09d9358cb616500207e25fe0a03ded70ed2f2","subject":"Merge \"[CI] Increase cephadm loopback disk size\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ea09d9358cb616500207e25fe0a03ded70ed2f2"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-08-09 10:10:13.000000000","tz":330},"subject":"Add tests for new snapshot_inherit_share_access share type","message":"Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/1000068\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/39e03792aace4df07a848507f3dbdcee004a74dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/39e03792aace4df07a848507f3dbdcee004a74dc"}]},"branch":"refs/heads/master"},"64629a5f75e4ebe80733fedfa777bd8ac9d06bd8":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-08-09 16:57:19.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/13"}}},"commit":{"parents":[{"commit":"69e8ab6d76489149aa59b4f86330991a2d6943de","subject":"Merge \"Use POSIX-compatible ls in migration test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69e8ab6d76489149aa59b4f86330991a2d6943de"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-08-09 16:57:19.000000000","tz":0},"subject":"Add tests for new snapshot_inherit_share_access share type","message":"Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/1000068\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/64629a5f75e4ebe80733fedfa777bd8ac9d06bd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/64629a5f75e4ebe80733fedfa777bd8ac9d06bd8"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"0ea09d9358cb616500207e25fe0a03ded70ed2f2","ours":"39e03792aace4df07a848507f3dbdcee004a74dc","theirs":"69e8ab6d76489149aa59b4f86330991a2d6943de","merge_strategy":"recursive","contains_conflicts":false}},"0c2d6fb6d9c848cdbf6da574941a80ae2d5cd001":{"kind":"REWORK","_number":14,"created":"2026-08-17 18:15:15.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/14"}}},"commit":{"parents":[{"commit":"d753c5552b822754c1d0256fd6713ef3bb58bc27","subject":"Merge \"Use manila-test-image for non-generic driver jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d753c5552b822754c1d0256fd6713ef3bb58bc27"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-08-17 18:31:15.000000000","tz":330},"subject":"Add tests for new snapshot_inherit_share_access share type","message":"Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/1000068\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0c2d6fb6d9c848cdbf6da574941a80ae2d5cd001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0c2d6fb6d9c848cdbf6da574941a80ae2d5cd001"}]},"branch":"refs/heads/master"},"8d4c31a7d6f32188c42a32626b1ca0971737692e":{"kind":"REWORK","_number":15,"created":"2026-08-24 14:11:04.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/15"}}},"commit":{"parents":[{"commit":"f04b07d022435027ecae2c022017ec113517e254","subject":"Merge \"Add tests for Qos Type API.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f04b07d022435027ecae2c022017ec113517e254"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-08-24 14:27:10.000000000","tz":330},"subject":"Add tests for new snapshot_inherit_share_access share type","message":"Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/1000068\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d4c31a7d6f32188c42a32626b1ca0971737692e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d4c31a7d6f32188c42a32626b1ca0971737692e"}]},"branch":"refs/heads/master"},"d2e722c40a8c0bdcb5adeaf7ac089d45395ad194":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-08-26 14:30:28.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"real_uploader":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"ref":"refs/changes/78/994878/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/16"}}},"commit":{"parents":[{"commit":"82c6bcfcfe72fc43021f4bb125b11c811d918c9b","subject":"Merge \"Add LUSTRE protocol support and CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82c6bcfcfe72fc43021f4bb125b11c811d918c9b"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-08-26 14:30:28.000000000","tz":0},"subject":"Add tests for new snapshot_inherit_share_access share type","message":"Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/1000068\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2e722c40a8c0bdcb5adeaf7ac089d45395ad194"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d2e722c40a8c0bdcb5adeaf7ac089d45395ad194"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"f04b07d022435027ecae2c022017ec113517e254","ours":"8d4c31a7d6f32188c42a32626b1ca0971737692e","theirs":"82c6bcfcfe72fc43021f4bb125b11c811d918c9b","merge_strategy":"recursive","contains_conflicts":false}},"d4a7ddac3b44c123fa3e7af5deda58361456968e":{"kind":"NO_CHANGE","_number":17,"created":"2026-08-26 17:39:53.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/78/994878/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/78/994878/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/78/994878/17"}}},"commit":{"parents":[{"commit":"82c6bcfcfe72fc43021f4bb125b11c811d918c9b","subject":"Merge \"Add LUSTRE protocol support and CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82c6bcfcfe72fc43021f4bb125b11c811d918c9b"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-08-26 17:56:05.000000000","tz":330},"subject":"Add tests for new snapshot_inherit_share_access share type","message":"Add tests for new snapshot_inherit_share_access share type\n\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/manila/+/1000068\nChange-Id: I8840e31b108f61b58361d04c434c090c81e36931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d4a7ddac3b44c123fa3e7af5deda58361456968e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d4a7ddac3b44c123fa3e7af5deda58361456968e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
