)]}'
{"id":"openstack%2Fmanila-tempest-plugin~994924","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie54155569311144add00a44ec391d9885a20b96c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"snapshotinheritshareruletests","hashtags":[],"change_id":"Ie54155569311144add00a44ec391d9885a20b96c","subject":"[WIP] - Add tests for new snapshot_inherit_share_access share type","status":"NEW","created":"2026-06-25 15:01:55.000000000","updated":"2026-06-25 17:13:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"ef5338a44b1a6676b890328e8d79979763871c44","_number":994924,"virtual_id_number":994924,"owner":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-25 17:13:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-06-25 17:13:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14b72e282a9af82e96f4ddb2b39e7bc3c23b3659","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-06-25 15:01:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8a4a92575936ff718f94a5f2fc8e8c93cd78419","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"date":"2026-06-25 15:03:37.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ef5338a44b1a6676b890328e8d79979763871c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 17:13:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a5cc464a37d42159e4e56b938174030\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d318b84f9f640f68f1852e75d571085 : SUCCESS in 2m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b780babd2aaa4385be144dcb5cfaff3a : FAILURE in 1h 30m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cf0eb778a9804332880b2907b8280bac : SUCCESS in 2h 05m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/98592e5450c243c1aa4cdca9f9588396 : SUCCESS in 1h 28m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/94efb7d23b5d4a44a00f3121ea7da691 : SUCCESS in 31m 25s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4b9fc3836efb473fb350eb27e3600493 : SUCCESS in 1h 36m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a89de10bd4ea4d5c9725cba74d2dd740 : FAILURE in 1h 15m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/35c0b09f86194f909cab96411ebe064c : SUCCESS in 35m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/749874e5ba9f4419bf3294c600ab0cfe : SUCCESS in 30m 46s\n- manila-tempest-plugin-lvm-2026.1 https://zuul.opendev.org/t/openstack/build/cf234ed7705c4ad6ad22b0069aa1bfee : SUCCESS in 1h 06m 37s\n- manila-tempest-plugin-lvm-2025.2 https://zuul.opendev.org/t/openstack/build/16503c3f42be4a72b6ca34fc77095686 : SUCCESS in 1h 23m 18s\n- manila-tempest-plugin-lvm-2025.1 https://zuul.opendev.org/t/openstack/build/337b30cd3a7c419a8aaa8773e956d3c0 : SUCCESS in 33m 06s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/27b7b6e237ea44bda9e3d21d3570af58 : SUCCESS in 25m 35s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/829ea11c27ed49dd871c5053f9a25fd7 : FAILURE in 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe385246550b35650efda74ca63f8a23fc00a46d","revisions":{"fe385246550b35650efda74ca63f8a23fc00a46d":{"kind":"REWORK","_number":1,"created":"2026-06-25 15:01:55.000000000","uploader":{"_account_id":38607,"name":"Srinesh A R","display_name":"Srinesh A R","email":"srinesh.a-r@hpe.com","username":"srineshar"},"ref":"refs/changes/24/994924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/24/994924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/994924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/994924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/994924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/24/994924/1"}}},"commit":{"parents":[{"commit":"afb832b53c04aee8c121fd9048f1480fe6459918","subject":"Merge \"Add replication_policy metadata tests for SnapMirror policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/afb832b53c04aee8c121fd9048f1480fe6459918"}]}],"author":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 11:14:35.000000000","tz":330},"committer":{"name":"Srinesh A R","email":"srinesh.a-r@hpe.com","date":"2026-06-25 15:14:43.000000000","tz":330},"subject":"[WIP] - Add tests for new snapshot_inherit_share_access share type","message":"[WIP] - Add tests for new snapshot_inherit_share_access share type\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/992850\nChange-Id: Ie54155569311144add00a44ec391d9885a20b96c\nSigned-off-by: Srinesh A R \u003csrinesh.a-r@hpe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fe385246550b35650efda74ca63f8a23fc00a46d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fe385246550b35650efda74ca63f8a23fc00a46d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
