)]}'
{"/COMMIT_MSG":[{"author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"change_message_id":"c8baa5ca383c171201fbbb08c05fb685d0197af3","unresolved":true,"context_lines":[{"line_number":11,"context_line":"Depends-on: https://review.opendev.org/c/openstack/manila/+/993501"},{"line_number":12,"context_line":"Depends-on: https://review.opendev.org/c/openstack/manila/+/991844"},{"line_number":13,"context_line":"Depends-on: https://review.opendev.org/c/openstack/manila/+/996629"},{"line_number":14,"context_line":""},{"line_number":15,"context_line":"Change-Id: Iee56c3c00b0797c6f3643ce5e90cb738cd925c48"},{"line_number":16,"context_line":"Signed-off-by: Manideep \u003cmanideep.openstack@gmail.com\u003e"}],"source_content_type":"text/x-gerrit-commit-message","patch_set":29,"id":"a5af758c_093561cb","line":14,"updated":"2026-09-01 18:27:50.000000000","message":"this should only depend on https://review.opendev.org/c/openstack/manila/+/993501 - these tests don\u0027t need the NetApp driver to be merged, the core should suffice. The NetApp driver should be tested as part of a different change.","commit_id":"b6b442986e560f31451e2ad5eb9f3da24a3e58b1"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"569360536d96d45b66d9514e8a931c587b5f4c01","unresolved":false,"context_lines":[{"line_number":11,"context_line":"Depends-on: https://review.opendev.org/c/openstack/manila/+/993501"},{"line_number":12,"context_line":"Depends-on: https://review.opendev.org/c/openstack/manila/+/991844"},{"line_number":13,"context_line":"Depends-on: https://review.opendev.org/c/openstack/manila/+/996629"},{"line_number":14,"context_line":""},{"line_number":15,"context_line":"Change-Id: Iee56c3c00b0797c6f3643ce5e90cb738cd925c48"},{"line_number":16,"context_line":"Signed-off-by: Manideep \u003cmanideep.openstack@gmail.com\u003e"}],"source_content_type":"text/x-gerrit-commit-message","patch_set":29,"id":"d9b0cb33_047fc38d","line":14,"in_reply_to":"a5af758c_093561cb","updated":"2026-09-02 07:32:59.000000000","message":"Acknowledged","commit_id":"b6b442986e560f31451e2ad5eb9f3da24a3e58b1"}],"/PATCHSET_LEVEL":[{"author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"change_message_id":"f89cb8ed7ebfb54e769c259d2c0358f67455cb42","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":9,"id":"0f79b2a9_d8925342","updated":"2026-08-17 08:04:22.000000000","message":"Please address the merge conflicts.","commit_id":"6b572c158248f9d84dbdb55d3ca6f5d89bb9ac9b"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"158e2d26397d298c33e4e97a0b6e2f28f9b584b1","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":12,"id":"8a87c5cb_010f3b7f","updated":"2026-08-23 06:22:59.000000000","message":"recheck","commit_id":"2c50ed098c1f1c44013a66079493e783c9d86324"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"812e6302858d10c3bd4adc86197136ba8fa23f52","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":13,"id":"16843211_6909422a","updated":"2026-08-23 19:08:09.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"137ff3d94f858d1c51427e73d5b83cf834517d67"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"4937659b4aef719eb81caae280557bf21b4882ef","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":13,"id":"19a6916a_e99a9d68","updated":"2026-08-23 19:30:56.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"137ff3d94f858d1c51427e73d5b83cf834517d67"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"dec62f366503b297a5a460b6d3f91064af82f49e","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":13,"id":"6154f367_2c15931d","updated":"2026-08-23 18:47:14.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"137ff3d94f858d1c51427e73d5b83cf834517d67"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"be2a79b598628deebb4c8e1875cdf5c54fe4d191","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":13,"id":"fb3e496f_f1fb6735","updated":"2026-08-24 02:37:07.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"137ff3d94f858d1c51427e73d5b83cf834517d67"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"46afbd3371bfe2267e0cf1a21f30c1d16b5d5f2e","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":14,"id":"88dba31b_719cf5de","updated":"2026-08-24 06:58:18.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"80459eef048f7f2a5eacdb8de4a863bdd962e95c"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"b3db36c50c549a172170ee989bf7c1dbff34dbfe","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":14,"id":"94818d07_5d856e25","updated":"2026-08-24 17:30:36.000000000","message":"recheck","commit_id":"80459eef048f7f2a5eacdb8de4a863bdd962e95c"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"452bf6664141e816c77af4606771b31f65137147","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":15,"id":"69a104c5_0be71d36","updated":"2026-08-24 20:11:55.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"e86655e340eb0adac8c3974c4d0b9b409f69c07d"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"8a9f93116ea03c15a6380fb6db2b7f1ff0d297e7","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":16,"id":"b3ab7218_b9939f49","updated":"2026-08-25 06:04:10.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"28f16cc84917efd840f16db81c8442f5092fa723"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"1af3c48e010e79d72da8c62094459629ddb104aa","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":17,"id":"0930279d_c58a9699","updated":"2026-08-25 07:22:06.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"b8bb16f933ba34b87d3a3e9a9e88440933dbd092"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"81715500fbe9412512839486e2b2f1ce5ced32fa","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":17,"id":"c6aa7f22_ba39eca4","updated":"2026-08-25 06:16:58.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"b8bb16f933ba34b87d3a3e9a9e88440933dbd092"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"febacc897bde4687d5bc468901d9d1c86d5cd45e","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":19,"id":"550e3be8_2eb79564","updated":"2026-08-25 12:02:43.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"a978f0344b46ba9e95b54030cb5d05a73f16248b"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"184133e42d5cf7aba17142cda69b69a01f652b56","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":20,"id":"ec55b969_fd082828","updated":"2026-08-25 14:50:40.000000000","message":"recheck","commit_id":"caa91298675e2b7d591527b7c77ab58a24842a14"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"0eab52d6d7b78ce0af74c1cea28470aa477e8c40","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":23,"id":"317fd489_cd393489","updated":"2026-08-26 14:30:59.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"91e64adf478f91f7c5868de55c30a84fa40045fe"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"8fccef3a087fb1374b5f2258728e93760b8e489b","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":23,"id":"cffa8018_9ecf2a0c","updated":"2026-08-26 13:16:12.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"91e64adf478f91f7c5868de55c30a84fa40045fe"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"5f83d4df8e7a966f6a12ecec2b013112f8a59f07","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":23,"id":"a79d55c0_ee87951a","updated":"2026-08-26 14:30:49.000000000","message":"recheck","commit_id":"91e64adf478f91f7c5868de55c30a84fa40045fe"},{"author":{"_account_id":38371,"name":"Kumar Tadepalli","display_name":"Kumar Tadepalli","email":"kumart.lead@gmail.com","username":"kumart","status":"NetApp"},"change_message_id":"4161db764771302a6c9a2fc54297e37930c91d68","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":24,"id":"5542dea0_01d53688","updated":"2026-08-26 15:21:08.000000000","message":"LGTM.","commit_id":"e95a3ae591bb386323107a8bc68e93876447968f"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"036702818cac3173c5b7c4f7015385d9f242e928","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":24,"id":"15befd1d_52eaaeb7","updated":"2026-08-26 14:43:18.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"e95a3ae591bb386323107a8bc68e93876447968f"},{"author":{"_account_id":38059,"name":"Anoop Kumar Shukla","display_name":"Anoop Shukla","email":"anoop.shukla@netapp.com","username":"anoop2","status":"NetApp"},"change_message_id":"7baeb149b6500b0fed622acb9b0222c8334a2dea","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":24,"id":"30de5e71_fedc1311","updated":"2026-08-26 17:15:55.000000000","message":"ZUUL is failing. Please check.","commit_id":"e95a3ae591bb386323107a8bc68e93876447968f"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"2314fde72204a2a3f7cd8a03136d6a30a4494987","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":24,"id":"9cbed843_f3b808e3","updated":"2026-08-26 14:43:09.000000000","message":"recheck","commit_id":"e95a3ae591bb386323107a8bc68e93876447968f"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"3e64878ff63ac14226d2b9e0a1f8044d8a25bb00","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":24,"id":"206e029a_bb4f0bb5","in_reply_to":"30de5e71_fedc1311","updated":"2026-08-26 17:22:49.000000000","message":"@anoop.shukla@netapp.com - The Zuul -1 is because of a change went into master (Implement new snapshot_inherit_share_access_support share capability), the tempest change to address it is already raised (Add tests for new snapshot_inherit_share_access share type), it should go in before this patch.","commit_id":"e95a3ae591bb386323107a8bc68e93876447968f"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"5b41d6fe6e456b7c1712ab7d38aa6194dc8675e5","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":25,"id":"fa6432af_9aa04eb4","updated":"2026-08-26 19:52:35.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"34e0a087ae828e99a54ea2efead3cb653a98ecd4"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"aa5d11225762aa7435486b548eaf2d60c4ab9e9e","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":26,"id":"45998699_455c1e49","updated":"2026-08-27 19:21:10.000000000","message":"@maurice.escher@sap.com - Thank you for the review. I have incorporated your review comments.","commit_id":"3a76f50f8ec38793d7abf1a28f307c1323418ed4"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"26c3dddee5bc2c1668cdd96c036f5a57422ba40e","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":26,"id":"1cce3107_bb27179c","updated":"2026-08-28 09:04:25.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"3a76f50f8ec38793d7abf1a28f307c1323418ed4"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"66f037c68a6fdb289c912efa4140d2518fed78a8","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":26,"id":"6c4985ba_f731e637","updated":"2026-08-28 09:10:41.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"3a76f50f8ec38793d7abf1a28f307c1323418ed4"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"17e6024724357df43bb82a4ad652acd0f2287b94","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":26,"id":"7fe2099e_226314ee","updated":"2026-08-28 14:47:49.000000000","message":"Run-NetApp-SMAS-NAS CI","commit_id":"3a76f50f8ec38793d7abf1a28f307c1323418ed4"},{"author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"change_message_id":"65f6be5ac6d2648a1cb481cbd75d220189b75117","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":26,"id":"9104438a_88bdfb05","updated":"2026-08-31 09:50:16.000000000","message":"recheck","commit_id":"3a76f50f8ec38793d7abf1a28f307c1323418ed4"},{"author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"change_message_id":"3c38916ad5bc64b8fa6cf2572126737bacb831b2","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":28,"id":"0576bf4c_bb2443e1","updated":"2026-09-01 14:54:12.000000000","message":"TThis change bumps the manila-tempest-plugin version, so it needs to depend on the share server replication change","commit_id":"268be9c5c53dae3b30b9e73f97c8a38dbc3aec20"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"f2a3b406da94658587525cc5053b57a427809c02","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":28,"id":"ba456b6f_e31c7bfe","updated":"2026-09-01 07:48:27.000000000","message":"recheck","commit_id":"268be9c5c53dae3b30b9e73f97c8a38dbc3aec20"},{"author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"change_message_id":"ccb036c4fe77366ba1dff9365c0617d7733c72d2","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":31,"id":"1487c75f_f579aede","updated":"2026-09-02 11:57:36.000000000","message":"Which config is having run_share_server_replication_tests as True ?","commit_id":"638f31caa5a7a9c8a02e91a65600bb794e5079a5"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"2126a123057970fddaaa87aa570b9cd1c80689ab","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":31,"id":"1fb9b72f_d964130e","updated":"2026-09-02 07:32:42.000000000","message":"recheck","commit_id":"638f31caa5a7a9c8a02e91a65600bb794e5079a5"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"cdbd4a0ceba3529efe85369b545361e16bbb0400","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":31,"id":"4d822c95_15d280a4","updated":"2026-09-02 03:35:28.000000000","message":"recheck","commit_id":"638f31caa5a7a9c8a02e91a65600bb794e5079a5"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"7529cc240fbeb60a1b3e45c2b56e6d1d23766e42","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":31,"id":"838f3407_86874587","updated":"2026-09-02 09:26:34.000000000","message":"recheck","commit_id":"638f31caa5a7a9c8a02e91a65600bb794e5079a5"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"50b760ab1d9cef6a235c190ace3a59c4fc412999","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":31,"id":"de35a5ed_8237e5f0","updated":"2026-09-02 09:27:34.000000000","message":"recheck","commit_id":"638f31caa5a7a9c8a02e91a65600bb794e5079a5"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"422fd63c68701ca037944dca23b7ade739dc787f","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":31,"id":"85098167_a86e0c17","in_reply_to":"1487c75f_f579aede","updated":"2026-09-02 13:19:42.000000000","message":"Build succeeded.\n\nmanila-tempest-plugin-ontap-smas-nas-dhss-rest \nhttps://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/60/998360/20/upstream-check-manila-smas-nas/manila-tempest-plugin-ontap-smas-nas-dhss-rest/05d7d40/ : SUCCESS in 51m 12s\nComment \"Run-NetApp-SMAS-NAS CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","commit_id":"638f31caa5a7a9c8a02e91a65600bb794e5079a5"},{"author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"change_message_id":"08d446de74306495a3599041ce5b555d7b502c1f","unresolved":true,"context_lines":[],"source_content_type":"","patch_set":31,"id":"c5d08721_fa939890","in_reply_to":"85098167_a86e0c17","updated":"2026-09-02 13:39:20.000000000","message":"https://netappopenstacklogserver.s3-us-west-1.amazonaws.com/logs/60/998360/31/upstream-check-manila/manila-tempest-plugin-ontap-dhss/52589eb/testr_results.html\nboth ShareServerReplicasTest and ShareServerReplicasNegativeTest skipped.\n\n\nmarking unresolved","commit_id":"638f31caa5a7a9c8a02e91a65600bb794e5079a5"},{"author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"change_message_id":"4b98ee1269063e8e182c451aedf4eda7fd4ddaf0","unresolved":true,"context_lines":[],"source_content_type":"","patch_set":31,"id":"eef6a39f_1d2369da","in_reply_to":"c5d08721_fa939890","updated":"2026-09-02 14:20:39.000000000","message":"something lik https://review.opendev.org/c/openstack/manila-tempest-plugin/+/1003099","commit_id":"638f31caa5a7a9c8a02e91a65600bb794e5079a5"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"42264c67e68a3b9304c2c73e2cb12da5b527bdce","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":31,"id":"e90af8c9_52a5fc93","in_reply_to":"eef6a39f_1d2369da","updated":"2026-09-02 15:32:37.000000000","message":"With respect to manila, NetApp CI had \"upstream-check-manila\" pipeline for quite sometime now.\n\nFor share server replication, we have introduced a new pipeline \"upstream-check-manila-smas-nas\" (Due to hardware limitations) and this pipeline runs only ShareServerReplicasTest and ShareServerReplicasNegativeTest. (run_share_server_replication_tests\u003dTrue)\n\nThe above url which you had shared corresponds to upstream-check-manila pipeline, so that is the reason the replications tests are skipped.\n\n\n\nThe below is the link for share server replication test results.\nhttps://10.193.154.68/zuul/t/netapp/build/4f1a545ba2ee49148a21b5601a0c1def\nTemest conf - https://netappopenstacklogserver.s3-us-west-1.amazonaws.com/logs/60/998360/26/upstream-check-manila-smas-nas/manila-tempest-plugin-ontap-smas-nas-dhss-rest/4f1a545/controller/logs/tempest_conf.txt","commit_id":"638f31caa5a7a9c8a02e91a65600bb794e5079a5"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"0e8b957819596fdbc1175672394abe619fbe7d3b","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":32,"id":"252262b8_ad4d1c99","updated":"2026-09-03 08:57:56.000000000","message":"The depends-ons got added by mistake.\nI will remove them after zuul run.","commit_id":"8774e81400944af60ed6fabceeae84d9104bc808"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"3b44b64454177f5544a9a634b72d0831c959b1f0","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":32,"id":"f92a4808_f8d72433","updated":"2026-09-03 08:51:49.000000000","message":"recheck","commit_id":"8774e81400944af60ed6fabceeae84d9104bc808"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"c535fc86eaf9844282b75399401ec349c04f39aa","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":33,"id":"8b8fba97_26423003","updated":"2026-09-03 16:56:14.000000000","message":"recheck","commit_id":"2cb370331d61b2f7e4e75707b4fca23323ff255b"}],"manila_tempest_tests/common/constants.py":[{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"3ae1e39e7f24f95612e1614b9fc94ff0e49ff67a","unresolved":false,"context_lines":[{"line_number":43,"context_line":"MIN_SHARE_REPLICATION_VERSION \u003d \u00272.11\u0027"},{"line_number":44,"context_line":"SHARE_REPLICA_GRADUATION_VERSION \u003d \u00272.56\u0027"},{"line_number":45,"context_line":"SHARE_REPLICA_SHARE_NET_PARAM_VERSION \u003d \u00272.72\u0027"},{"line_number":46,"context_line":"#TODO(manideepn): Update this accordingly based on manila api version"},{"line_number":47,"context_line":"MIN_SHARE_SERVER_REPLICATION_VERSION \u003d \u00272.97\u0027"},{"line_number":48,"context_line":""},{"line_number":49,"context_line":"# Access Rules"}],"source_content_type":"text/x-python","patch_set":2,"id":"fbcf92ed_5402dc1f","line":46,"updated":"2026-07-22 21:02:49.000000000","message":"pep8: E265 block comment should start with \u0027# \u0027","commit_id":"151cb229ddf40dd86159a8b24f146b8d6134c2e8"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"cd1517d80bd9e48ab2b9ff8d03c0fb57b2bbf35a","unresolved":false,"context_lines":[{"line_number":43,"context_line":"MIN_SHARE_REPLICATION_VERSION \u003d \u00272.11\u0027"},{"line_number":44,"context_line":"SHARE_REPLICA_GRADUATION_VERSION \u003d \u00272.56\u0027"},{"line_number":45,"context_line":"SHARE_REPLICA_SHARE_NET_PARAM_VERSION \u003d \u00272.72\u0027"},{"line_number":46,"context_line":"#TODO(manideepn): Update this accordingly based on manila api version"},{"line_number":47,"context_line":"MIN_SHARE_SERVER_REPLICATION_VERSION \u003d \u00272.97\u0027"},{"line_number":48,"context_line":""},{"line_number":49,"context_line":"# Access Rules"}],"source_content_type":"text/x-python","patch_set":3,"id":"119bc9e0_0b5e1aa0","line":46,"updated":"2026-08-05 10:45:38.000000000","message":"pep8: E265 block comment should start with \u0027# \u0027","commit_id":"cf4ff1f27fcf40c7116162f7ba449c6fc33f4da9"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"30ac1578b680b7ca4b87af57a51a81b6dd3571be","unresolved":false,"context_lines":[{"line_number":43,"context_line":"MIN_SHARE_REPLICATION_VERSION \u003d \u00272.11\u0027"},{"line_number":44,"context_line":"SHARE_REPLICA_GRADUATION_VERSION \u003d \u00272.56\u0027"},{"line_number":45,"context_line":"SHARE_REPLICA_SHARE_NET_PARAM_VERSION \u003d \u00272.72\u0027"},{"line_number":46,"context_line":"#TODO(manideepn): Update this accordingly based on manila api version"},{"line_number":47,"context_line":"MIN_SHARE_SERVER_REPLICATION_VERSION \u003d \u00272.98\u0027"},{"line_number":48,"context_line":""},{"line_number":49,"context_line":"# Access Rules"}],"source_content_type":"text/x-python","patch_set":6,"id":"24f8873a_c3173d16","line":46,"updated":"2026-08-05 19:58:47.000000000","message":"pep8: E265 block comment should start with \u0027# \u0027","commit_id":"ca1a81436458160c999284f4a5af939ce62d4285"}],"manila_tempest_tests/services/share/v2/json/shares_client.py":[{"author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"change_message_id":"5a0604a52eb9131da8936dfafd1694bcfcb65555","unresolved":true,"context_lines":[{"line_number":2080,"context_line":"                                    version\u003dLATEST_MICROVERSION):"},{"line_number":2081,"context_line":"        \"\"\"Add a share server replica of an existing share server.\"\"\""},{"line_number":2082,"context_line":"        uri \u003d \"share-server-replicas\""},{"line_number":2083,"context_line":"        post_body \u003d {"},{"line_number":2084,"context_line":"            \u0027share_server\u0027: share_server_id,"},{"line_number":2085,"context_line":"        }"},{"line_number":2086,"context_line":"        if availability_zone:"}],"source_content_type":"text/x-python","patch_set":31,"id":"082c3346_07756681","line":2083,"updated":"2026-09-02 16:44:53.000000000","message":"https://review.opendev.org/c/openstack/manila/+/993501/79/manila/api/v2/share_server_replicas.py#168 is expecting share_server_id and you are passing share_server. How tests are passing specifically which goes through create workflow ?","commit_id":"638f31caa5a7a9c8a02e91a65600bb794e5079a5"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"2b2fa067b9a0886fe83a0c929dc4230d5ad832b5","unresolved":false,"context_lines":[{"line_number":2080,"context_line":"                                    version\u003dLATEST_MICROVERSION):"},{"line_number":2081,"context_line":"        \"\"\"Add a share server replica of an existing share server.\"\"\""},{"line_number":2082,"context_line":"        uri \u003d \"share-server-replicas\""},{"line_number":2083,"context_line":"        post_body \u003d {"},{"line_number":2084,"context_line":"            \u0027share_server\u0027: share_server_id,"},{"line_number":2085,"context_line":"        }"},{"line_number":2086,"context_line":"        if availability_zone:"}],"source_content_type":"text/x-python","patch_set":31,"id":"8bfa8d98_f76e984b","line":2083,"in_reply_to":"082c3346_07756681","updated":"2026-09-03 08:48:47.000000000","message":"These tempest tesed needs some changes after addressing review comments.\npushed the changes.","commit_id":"638f31caa5a7a9c8a02e91a65600bb794e5079a5"}],"manila_tempest_tests/tests/api/admin/test_share_server_replicas.py":[{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"3ae1e39e7f24f95612e1614b9fc94ff0e49ff67a","unresolved":false,"context_lines":[{"line_number":28,"context_line":"LATEST_MICROVERSION \u003d CONF.share.max_api_microversion"},{"line_number":29,"context_line":"_MIN_SUPPORTED_MICROVERSION \u003d constants.MIN_SHARE_SERVER_REPLICATION_VERSION"},{"line_number":30,"context_line":"SUMMARY_KEYS \u003d [\u0027id\u0027, \u0027source_share_server_id\u0027, \u0027host\u0027, \u0027status\u0027,"},{"line_number":31,"context_line":"               \u0027replica_state\u0027]"},{"line_number":32,"context_line":"DETAIL_KEYS \u003d SUMMARY_KEYS + [\u0027created_at\u0027, \u0027updated_at\u0027]"},{"line_number":33,"context_line":""},{"line_number":34,"context_line":""}],"source_content_type":"text/x-python","patch_set":2,"id":"bad86f13_a5eef884","line":31,"updated":"2026-07-22 21:02:49.000000000","message":"pep8: E128 continuation line under-indented for visual indent","commit_id":"151cb229ddf40dd86159a8b24f146b8d6134c2e8"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"3ae1e39e7f24f95612e1614b9fc94ff0e49ff67a","unresolved":false,"context_lines":[{"line_number":280,"context_line":"        self.assertNotEqual(page1[0][\u0027id\u0027], page2[0][\u0027id\u0027])"},{"line_number":281,"context_line":"        paginated_ids \u003d sorted([page1[0][\u0027id\u0027], page2[0][\u0027id\u0027]])"},{"line_number":282,"context_line":"        self.assertEqual(full_ids, paginated_ids)"},{"line_number":283,"context_line":""}],"source_content_type":"text/x-python","patch_set":2,"id":"bbcecde5_538dc566","line":283,"updated":"2026-07-22 21:02:49.000000000","message":"pep8: W391 blank line at end of file","commit_id":"151cb229ddf40dd86159a8b24f146b8d6134c2e8"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"cd1517d80bd9e48ab2b9ff8d03c0fb57b2bbf35a","unresolved":false,"context_lines":[{"line_number":28,"context_line":"LATEST_MICROVERSION \u003d CONF.share.max_api_microversion"},{"line_number":29,"context_line":"_MIN_SUPPORTED_MICROVERSION \u003d constants.MIN_SHARE_SERVER_REPLICATION_VERSION"},{"line_number":30,"context_line":"SUMMARY_KEYS \u003d [\u0027id\u0027, \u0027source_share_server_id\u0027, \u0027host\u0027, \u0027status\u0027,"},{"line_number":31,"context_line":"               \u0027replica_state\u0027]"},{"line_number":32,"context_line":"DETAIL_KEYS \u003d SUMMARY_KEYS + [\u0027created_at\u0027, \u0027updated_at\u0027]"},{"line_number":33,"context_line":""},{"line_number":34,"context_line":""}],"source_content_type":"text/x-python","patch_set":3,"id":"3f821664_4ffd7190","line":31,"updated":"2026-08-05 10:45:38.000000000","message":"pep8: E128 continuation line under-indented for visual indent","commit_id":"cf4ff1f27fcf40c7116162f7ba449c6fc33f4da9"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"cd1517d80bd9e48ab2b9ff8d03c0fb57b2bbf35a","unresolved":false,"context_lines":[{"line_number":280,"context_line":"        self.assertNotEqual(page1[0][\u0027id\u0027], page2[0][\u0027id\u0027])"},{"line_number":281,"context_line":"        paginated_ids \u003d sorted([page1[0][\u0027id\u0027], page2[0][\u0027id\u0027]])"},{"line_number":282,"context_line":"        self.assertEqual(full_ids, paginated_ids)"},{"line_number":283,"context_line":""}],"source_content_type":"text/x-python","patch_set":3,"id":"f1bf8f37_f3aeda91","line":283,"updated":"2026-08-05 10:45:38.000000000","message":"pep8: W391 blank line at end of file","commit_id":"cf4ff1f27fcf40c7116162f7ba449c6fc33f4da9"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"30ac1578b680b7ca4b87af57a51a81b6dd3571be","unresolved":false,"context_lines":[{"line_number":28,"context_line":"LATEST_MICROVERSION \u003d CONF.share.max_api_microversion"},{"line_number":29,"context_line":"_MIN_SUPPORTED_MICROVERSION \u003d constants.MIN_SHARE_SERVER_REPLICATION_VERSION"},{"line_number":30,"context_line":"SUMMARY_KEYS \u003d [\u0027id\u0027, \u0027source_share_server_id\u0027, \u0027host\u0027, \u0027status\u0027,"},{"line_number":31,"context_line":"               \u0027replica_state\u0027]"},{"line_number":32,"context_line":"DETAIL_KEYS \u003d SUMMARY_KEYS + [\u0027created_at\u0027, \u0027updated_at\u0027]"},{"line_number":33,"context_line":""},{"line_number":34,"context_line":""}],"source_content_type":"text/x-python","patch_set":6,"id":"ab73f633_2d67f66e","line":31,"updated":"2026-08-05 19:58:47.000000000","message":"pep8: E128 continuation line under-indented for visual indent","commit_id":"ca1a81436458160c999284f4a5af939ce62d4285"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"30ac1578b680b7ca4b87af57a51a81b6dd3571be","unresolved":false,"context_lines":[{"line_number":280,"context_line":"        self.assertNotEqual(page1[0][\u0027id\u0027], page2[0][\u0027id\u0027])"},{"line_number":281,"context_line":"        paginated_ids \u003d sorted([page1[0][\u0027id\u0027], page2[0][\u0027id\u0027]])"},{"line_number":282,"context_line":"        self.assertEqual(full_ids, paginated_ids)"},{"line_number":283,"context_line":""}],"source_content_type":"text/x-python","patch_set":6,"id":"3bd8e394_ad7108e3","line":283,"updated":"2026-08-05 19:58:47.000000000","message":"pep8: W391 blank line at end of file","commit_id":"ca1a81436458160c999284f4a5af939ce62d4285"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"e0f5f2f39f488ed5359d84e87906325ea7847b59","unresolved":false,"context_lines":[{"line_number":28,"context_line":"LATEST_MICROVERSION \u003d CONF.share.max_api_microversion"},{"line_number":29,"context_line":"_MIN_SUPPORTED_MICROVERSION \u003d constants.MIN_SHARE_SERVER_REPLICATION_VERSION"},{"line_number":30,"context_line":"SUMMARY_KEYS \u003d [\u0027id\u0027, \u0027source_share_server_id\u0027, \u0027host\u0027, \u0027status\u0027,"},{"line_number":31,"context_line":"               \u0027replica_state\u0027]"},{"line_number":32,"context_line":"DETAIL_KEYS \u003d SUMMARY_KEYS + [\u0027created_at\u0027, \u0027updated_at\u0027]"},{"line_number":33,"context_line":""},{"line_number":34,"context_line":""}],"source_content_type":"text/x-python","patch_set":7,"id":"d5da7b7d_739a1c40","line":31,"updated":"2026-08-07 08:54:14.000000000","message":"pep8: E128 continuation line under-indented for visual indent","commit_id":"0f68fa739255998fb3edafc3e31a562d720e5424"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"e0f5f2f39f488ed5359d84e87906325ea7847b59","unresolved":false,"context_lines":[{"line_number":355,"context_line":"        self.assertNotEqual(page1[0][\u0027id\u0027], page2[0][\u0027id\u0027])"},{"line_number":356,"context_line":"        paginated_ids \u003d sorted([page1[0][\u0027id\u0027], page2[0][\u0027id\u0027]])"},{"line_number":357,"context_line":"        self.assertEqual(full_ids, paginated_ids)"},{"line_number":358,"context_line":""}],"source_content_type":"text/x-python","patch_set":7,"id":"62657956_f8b790c4","line":358,"updated":"2026-08-07 08:54:14.000000000","message":"pep8: W391 blank line at end of file","commit_id":"0f68fa739255998fb3edafc3e31a562d720e5424"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"bcb322254f2e524d603a16f2eabf2673bb7ead51","unresolved":false,"context_lines":[{"line_number":14,"context_line":"#    under the License."},{"line_number":15,"context_line":""},{"line_number":16,"context_line":"import ddt"},{"line_number":17,"context_line":"import time"},{"line_number":18,"context_line":""},{"line_number":19,"context_line":"from tempest import config"},{"line_number":20,"context_line":"from tempest.lib import decorators"}],"source_content_type":"text/x-python","patch_set":9,"id":"109efb61_26941a57","line":17,"updated":"2026-08-13 20:46:11.000000000","message":"pep8: I100 Import statements are in the wrong order. \u0027import time\u0027 should be before \u0027import ddt\u0027 and in a different group.","commit_id":"6b572c158248f9d84dbdb55d3ca6f5d89bb9ac9b"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"bcb322254f2e524d603a16f2eabf2673bb7ead51","unresolved":false,"context_lines":[{"line_number":14,"context_line":"#    under the License."},{"line_number":15,"context_line":""},{"line_number":16,"context_line":"import ddt"},{"line_number":17,"context_line":"import time"},{"line_number":18,"context_line":""},{"line_number":19,"context_line":"from tempest import config"},{"line_number":20,"context_line":"from tempest.lib import decorators"}],"source_content_type":"text/x-python","patch_set":9,"id":"b04d1cf7_93b47732","line":17,"updated":"2026-08-13 20:46:11.000000000","message":"pep8: I201 Missing newline between import groups. \u0027import time\u0027 is identified as Stdlib and \u0027import ddt\u0027 is identified as Third Party.","commit_id":"6b572c158248f9d84dbdb55d3ca6f5d89bb9ac9b"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"bcb322254f2e524d603a16f2eabf2673bb7ead51","unresolved":false,"context_lines":[{"line_number":30,"context_line":"LATEST_MICROVERSION \u003d CONF.share.max_api_microversion"},{"line_number":31,"context_line":"_MIN_SUPPORTED_MICROVERSION \u003d constants.MIN_SHARE_SERVER_REPLICATION_VERSION"},{"line_number":32,"context_line":"SUMMARY_KEYS \u003d [\u0027id\u0027, \u0027source_share_server_id\u0027, \u0027host\u0027, \u0027status\u0027,"},{"line_number":33,"context_line":"               \u0027replica_state\u0027]"},{"line_number":34,"context_line":"DETAIL_KEYS \u003d SUMMARY_KEYS + [\u0027created_at\u0027, \u0027updated_at\u0027]"},{"line_number":35,"context_line":""},{"line_number":36,"context_line":""}],"source_content_type":"text/x-python","patch_set":9,"id":"1edbfea5_dbde2547","line":33,"updated":"2026-08-13 20:46:11.000000000","message":"pep8: E128 continuation line under-indented for visual indent","commit_id":"6b572c158248f9d84dbdb55d3ca6f5d89bb9ac9b"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"bcb322254f2e524d603a16f2eabf2673bb7ead51","unresolved":false,"context_lines":[{"line_number":148,"context_line":"                    \"Share server replica %s did not reach replica_state \""},{"line_number":149,"context_line":"                    \"\u0027%s\u0027 within %s seconds; last seen replica_state: \""},{"line_number":150,"context_line":"                    \"\u0027%s\u0027.\" % (replica_id, expected_state, timeout,"},{"line_number":151,"context_line":"                              replica[\u0027replica_state\u0027]))"},{"line_number":152,"context_line":"            time.sleep(interval)"},{"line_number":153,"context_line":"            replica \u003d self.client.get_share_server_replica("},{"line_number":154,"context_line":"                replica_id)[\u0027share_server_replica\u0027]"}],"source_content_type":"text/x-python","patch_set":9,"id":"0a06b11b_f31a75a3","line":151,"updated":"2026-08-13 20:46:11.000000000","message":"pep8: E128 continuation line under-indented for visual indent","commit_id":"6b572c158248f9d84dbdb55d3ca6f5d89bb9ac9b"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"bcb322254f2e524d603a16f2eabf2673bb7ead51","unresolved":false,"context_lines":[{"line_number":422,"context_line":"                self.client, share[\u0027id\u0027], constants.STATUS_AVAILABLE)"},{"line_number":423,"context_line":"            share \u003d self.client.get_share(share[\u0027id\u0027])[\u0027share\u0027]"},{"line_number":424,"context_line":"            self.assertEqual(new_size, int(share[\u0027size\u0027]))"},{"line_number":425,"context_line":""}],"source_content_type":"text/x-python","patch_set":9,"id":"30f9bd31_f0e99dc1","line":425,"updated":"2026-08-13 20:46:11.000000000","message":"pep8: W391 blank line at end of file","commit_id":"6b572c158248f9d84dbdb55d3ca6f5d89bb9ac9b"},{"author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"change_message_id":"0fa2b3deca0662c55cc3e48c15fa44a8361c4649","unresolved":true,"context_lines":[{"line_number":126,"context_line":"        start \u003d int(time.time())"},{"line_number":127,"context_line":"        replica \u003d self.client.get_share_server_replica("},{"line_number":128,"context_line":"            replica_id)[\u0027share_server_replica\u0027]"},{"line_number":129,"context_line":"        while replica[\u0027replica_state\u0027] !\u003d expected_state:"},{"line_number":130,"context_line":"            if int(time.time()) - start \u003e\u003d timeout:"},{"line_number":131,"context_line":"                self.fail("},{"line_number":132,"context_line":"                    \"Share server replica %s did not reach replica_state \""}],"source_content_type":"text/x-python","patch_set":25,"id":"e244cedc_69c551a2","line":129,"updated":"2026-08-27 12:51:18.000000000","message":"what if the state transitions to `error`? I think it should fail fast and not wait for the timeout then.","commit_id":"34e0a087ae828e99a54ea2efead3cb653a98ecd4"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"aa5d11225762aa7435486b548eaf2d60c4ab9e9e","unresolved":false,"context_lines":[{"line_number":126,"context_line":"        start \u003d int(time.time())"},{"line_number":127,"context_line":"        replica \u003d self.client.get_share_server_replica("},{"line_number":128,"context_line":"            replica_id)[\u0027share_server_replica\u0027]"},{"line_number":129,"context_line":"        while replica[\u0027replica_state\u0027] !\u003d expected_state:"},{"line_number":130,"context_line":"            if int(time.time()) - start \u003e\u003d timeout:"},{"line_number":131,"context_line":"                self.fail("},{"line_number":132,"context_line":"                    \"Share server replica %s did not reach replica_state \""}],"source_content_type":"text/x-python","patch_set":25,"id":"f084d751_e9c68503","line":129,"in_reply_to":"e244cedc_69c551a2","updated":"2026-08-27 19:21:10.000000000","message":"Added logic to fail fast.","commit_id":"34e0a087ae828e99a54ea2efead3cb653a98ecd4"},{"author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"change_message_id":"b5657365064ba67e9ba97124924bbf5e4bad08fe","unresolved":true,"context_lines":[{"line_number":197,"context_line":"            self.client, replica[\u0027id\u0027], constants.SERVER_STATE_INACTIVE,"},{"line_number":198,"context_line":"            resource_name\u003d\u0027share_server_replica\u0027)"},{"line_number":199,"context_line":"        self._wait_for_replica_in_sync("},{"line_number":200,"context_line":"            replica[\u0027id\u0027], timeout\u003d15, interval\u003d5)"},{"line_number":201,"context_line":""},{"line_number":202,"context_line":"        self.delete_share_server_replica(replica[\u0027id\u0027], client\u003dself.client)"},{"line_number":203,"context_line":"        self.assertRaises(exceptions.NotFound,"}],"source_content_type":"text/x-python","patch_set":32,"id":"0d8a92c6_c85f2bf6","line":200,"updated":"2026-09-03 11:20:27.000000000","message":"Why do we set a shorter timeout than default here? I think it would be better keep the conservative 300 seconds","commit_id":"8774e81400944af60ed6fabceeae84d9104bc808"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"847203bdc903cb823b98c2bd4beef24a9924d006","unresolved":false,"context_lines":[{"line_number":197,"context_line":"            self.client, replica[\u0027id\u0027], constants.SERVER_STATE_INACTIVE,"},{"line_number":198,"context_line":"            resource_name\u003d\u0027share_server_replica\u0027)"},{"line_number":199,"context_line":"        self._wait_for_replica_in_sync("},{"line_number":200,"context_line":"            replica[\u0027id\u0027], timeout\u003d15, interval\u003d5)"},{"line_number":201,"context_line":""},{"line_number":202,"context_line":"        self.delete_share_server_replica(replica[\u0027id\u0027], client\u003dself.client)"},{"line_number":203,"context_line":"        self.assertRaises(exceptions.NotFound,"}],"source_content_type":"text/x-python","patch_set":32,"id":"6c1b7629_fba8703b","line":200,"in_reply_to":"0d8a92c6_c85f2bf6","updated":"2026-09-03 11:27:14.000000000","message":"I wanted to keep short time out so that the update made by periodic resync job can be separated.\nBut, if you feel we should use the default, I can make the change.","commit_id":"8774e81400944af60ed6fabceeae84d9104bc808"}],"manila_tempest_tests/tests/api/admin/test_share_server_replicas_negative.py":[{"author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"change_message_id":"1ecb87575d65449f4f88c0f56571493b651221eb","unresolved":true,"context_lines":[{"line_number":151,"context_line":""},{"line_number":152,"context_line":"    @decorators.idempotent_id(\u0027cabae8ae-dd2a-452d-9c20-b4be72e9973e\u0027)"},{"line_number":153,"context_line":"    @tc.attr(base.TAG_NEGATIVE, base.TAG_API)"},{"line_number":154,"context_line":"    def test_create_share_server_replica_with_invalid_az(self):"},{"line_number":155,"context_line":"        self.assertRaises("},{"line_number":156,"context_line":"            lib_exc.NotFound,"},{"line_number":157,"context_line":"            self.create_share_server_replica,"}],"source_content_type":"text/x-python","patch_set":31,"id":"cd91a2fb_4ad585dd","line":154,"updated":"2026-09-02 14:00:28.000000000","message":"https://zuul.opendev.org/t/openstack/build/ca6f67bc7a04405a9ef75cf88a0a6b81\n\ntest is failing here.","commit_id":"638f31caa5a7a9c8a02e91a65600bb794e5079a5"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"42264c67e68a3b9304c2c73e2cb12da5b527bdce","unresolved":false,"context_lines":[{"line_number":151,"context_line":""},{"line_number":152,"context_line":"    @decorators.idempotent_id(\u0027cabae8ae-dd2a-452d-9c20-b4be72e9973e\u0027)"},{"line_number":153,"context_line":"    @tc.attr(base.TAG_NEGATIVE, base.TAG_API)"},{"line_number":154,"context_line":"    def test_create_share_server_replica_with_invalid_az(self):"},{"line_number":155,"context_line":"        self.assertRaises("},{"line_number":156,"context_line":"            lib_exc.NotFound,"},{"line_number":157,"context_line":"            self.create_share_server_replica,"}],"source_content_type":"text/x-python","patch_set":31,"id":"f45fb177_3e3e39be","line":154,"in_reply_to":"cd91a2fb_4ad585dd","updated":"2026-09-02 15:32:37.000000000","message":"Acknowledged","commit_id":"638f31caa5a7a9c8a02e91a65600bb794e5079a5"}],"manila_tempest_tests/tests/api/base.py":[{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"bcb322254f2e524d603a16f2eabf2673bb7ead51","unresolved":false,"context_lines":[{"line_number":802,"context_line":"        except exceptions.NotFound:"},{"line_number":803,"context_line":"            pass"},{"line_number":804,"context_line":"        except (exceptions.BadRequest,"},{"line_number":805,"context_line":"               share_exceptions.ResourceReleaseFailed):"},{"line_number":806,"context_line":"            # The replica may already be in \u0027error_deleting\u0027 from a prior"},{"line_number":807,"context_line":"            # failed backend deletion attempt, which the API rejects with"},{"line_number":808,"context_line":"            # \"already in error_deleting status.\" Reset it back to \u0027error\u0027"}],"source_content_type":"text/x-python","patch_set":9,"id":"82ce3c7a_6faa6a80","line":805,"updated":"2026-08-13 20:46:11.000000000","message":"pep8: E128 continuation line under-indented for visual indent","commit_id":"6b572c158248f9d84dbdb55d3ca6f5d89bb9ac9b"},{"author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"tag":"autogenerated:zuul:check","change_message_id":"bcb322254f2e524d603a16f2eabf2673bb7ead51","unresolved":false,"context_lines":[{"line_number":1053,"context_line":"                                client.wait_for_resource_deletion("},{"line_number":1054,"context_line":"                                    share_server_replica_id\u003dres_id)"},{"line_number":1055,"context_line":"                            except (exceptions.BadRequest,"},{"line_number":1056,"context_line":"                                   share_exceptions.ResourceReleaseFailed):"},{"line_number":1057,"context_line":"                                # The replica may already be in"},{"line_number":1058,"context_line":"                                # \u0027error_deleting\u0027 from a prior failed"},{"line_number":1059,"context_line":"                                # backend deletion attempt, which the API"}],"source_content_type":"text/x-python","patch_set":9,"id":"9ba42565_12794dbc","line":1056,"updated":"2026-08-13 20:46:11.000000000","message":"pep8: E128 continuation line under-indented for visual indent","commit_id":"6b572c158248f9d84dbdb55d3ca6f5d89bb9ac9b"},{"author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"change_message_id":"0fa2b3deca0662c55cc3e48c15fa44a8361c4649","unresolved":true,"context_lines":[{"line_number":812,"context_line":"                    \"Share server replica %s failed to delete (attempt \""},{"line_number":813,"context_line":"                    \"%s/%s); resetting status to \u0027error\u0027 and retrying.\","},{"line_number":814,"context_line":"                    replica_id, attempt, max_attempts)"},{"line_number":815,"context_line":"                client.share_server_reset_state("},{"line_number":816,"context_line":"                    replica_id, status\u003dconstants.SERVER_STATE_ERROR)"},{"line_number":817,"context_line":""},{"line_number":818,"context_line":"    @classmethod"}],"source_content_type":"text/x-python","patch_set":25,"id":"00b616b5_766426e7","line":815,"updated":"2026-08-27 12:51:18.000000000","message":"This needs to target the replica.\nA `reset_share_server_replica_status` is not available, though, it needs to be implemented first","commit_id":"34e0a087ae828e99a54ea2efead3cb653a98ecd4"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"aa5d11225762aa7435486b548eaf2d60c4ab9e9e","unresolved":false,"context_lines":[{"line_number":812,"context_line":"                    \"Share server replica %s failed to delete (attempt \""},{"line_number":813,"context_line":"                    \"%s/%s); resetting status to \u0027error\u0027 and retrying.\","},{"line_number":814,"context_line":"                    replica_id, attempt, max_attempts)"},{"line_number":815,"context_line":"                client.share_server_reset_state("},{"line_number":816,"context_line":"                    replica_id, status\u003dconstants.SERVER_STATE_ERROR)"},{"line_number":817,"context_line":""},{"line_number":818,"context_line":"    @classmethod"}],"source_content_type":"text/x-python","patch_set":25,"id":"92ed08bc_8a725816","line":815,"in_reply_to":"00b616b5_766426e7","updated":"2026-08-27 19:21:10.000000000","message":"Addressed.","commit_id":"34e0a087ae828e99a54ea2efead3cb653a98ecd4"},{"author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"change_message_id":"0fa2b3deca0662c55cc3e48c15fa44a8361c4649","unresolved":true,"context_lines":[{"line_number":1055,"context_line":"                                    \"Share server replica %s failed to \""},{"line_number":1056,"context_line":"                                    \"delete; resetting status to \u0027error\u0027 \""},{"line_number":1057,"context_line":"                                    \"and force deleting it.\", res_id)"},{"line_number":1058,"context_line":"                                client.share_server_reset_state("},{"line_number":1059,"context_line":"                                    res_id,"},{"line_number":1060,"context_line":"                                    status\u003dconstants.SERVER_STATE_ERROR)"},{"line_number":1061,"context_line":"                                client.force_delete_share_server_replica("}],"source_content_type":"text/x-python","patch_set":25,"id":"3a3871b8_74a22ff6","line":1058,"updated":"2026-08-27 12:51:18.000000000","message":"same like above, needs to target the replica","commit_id":"34e0a087ae828e99a54ea2efead3cb653a98ecd4"},{"author":{"_account_id":38373,"name":"Manideep","display_name":"Manideep","email":"manideep.openstack@gmail.com","username":"manideepn","status":"NetApp"},"change_message_id":"aa5d11225762aa7435486b548eaf2d60c4ab9e9e","unresolved":false,"context_lines":[{"line_number":1055,"context_line":"                                    \"Share server replica %s failed to \""},{"line_number":1056,"context_line":"                                    \"delete; resetting status to \u0027error\u0027 \""},{"line_number":1057,"context_line":"                                    \"and force deleting it.\", res_id)"},{"line_number":1058,"context_line":"                                client.share_server_reset_state("},{"line_number":1059,"context_line":"                                    res_id,"},{"line_number":1060,"context_line":"                                    status\u003dconstants.SERVER_STATE_ERROR)"},{"line_number":1061,"context_line":"                                client.force_delete_share_server_replica("}],"source_content_type":"text/x-python","patch_set":25,"id":"8e107a34_7b110217","line":1058,"in_reply_to":"3a3871b8_74a22ff6","updated":"2026-08-27 19:21:10.000000000","message":"Addressed.","commit_id":"34e0a087ae828e99a54ea2efead3cb653a98ecd4"}]}
