)]}'
{"id":"openstack%2Fmanila-test-image~746565","triplet_id":"openstack%2Fmanila-test-image~master~Ifc6d592b8b7cf0e802ee11be837ebd4a3b55ea9b","project":"openstack/manila-test-image","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"Ifc6d592b8b7cf0e802ee11be837ebd4a3b55ea9b","subject":"Drop legacy-base from manila-test-image-build","status":"MERGED","created":"2020-08-17 16:56:39.000000000","updated":"2020-09-24 22:03:32.000000000","submitted":"2020-09-24 22:03:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746565-1600985012705-b51fe808","meta_rev_id":"36f6f96803a65ae577ba486a8d8f82b5b5363456","_number":746565,"virtual_id_number":746565,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 22:03:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2020-09-24 20:33:37.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2020-09-24 21:54:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 20:33:37.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-09-24 22:03:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2fc7c596b9b79c9c78cd865cd4e5f71d2ceca7e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-17 16:56:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bd99d88cfbd7d91340e8b84b5661e8e2286fbee","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-17 17:05:15.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"9e21e68834b0ef8b1748313403973ecb3b574054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 17:09:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e83d12afa4864ed388ac380dd802e982 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d6072fe2e4c2d5d9a345f64bc6b9fd261beb53f8","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 17:21:50.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- manila-test-image-build https://zuul.opendev.org/t/openstack/build/8b6398ec5f224ffd94ccc4d968d6c68f : FAILURE in 10m 50s","accounts_in_message":[],"_revision_number":1},{"id":"bc039830b830e7de2aefa8bd5e4e670bbbe5d026","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-24 17:06:45.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"59aac840b49d4f67d91c48652711a1b79cd1976d","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 18:09:10.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- manila-test-image-build https://zuul.opendev.org/t/openstack/build/2330eea8fda74f0ead1df66268e2553d : FAILURE in 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a561eb10d93a4ee7cfa82b16acc2aa94a9158dac","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-09-24 20:33:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8bafe34b692d0f1612ced2734244b261804a7267","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-24 21:54:25.000000000","message":"Patch Set 1: Workflow+1\n\nThis job cleanup does not contribute to the build failure, so merging this..","accounts_in_message":[],"_revision_number":1},{"id":"5fd6fabce15d4629ca2335305b8b3422cc1c5944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 21:56:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"993bf0486cec02d7dd486c0e518cf6645af1788f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 22:03:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/fa0b8879ae404308a41f1f6170cf08d0 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"668a147b8f00218eb3659c7a2abee9d381047310","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 22:03:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c16a41181aa7c77cbe6c59700d10a2e0270444d","revisions":{"1c16a41181aa7c77cbe6c59700d10a2e0270444d":{"kind":"REWORK","_number":1,"created":"2020-08-17 16:56:39.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/65/746565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-test-image","ref":"refs/changes/65/746565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-test-image refs/changes/65/746565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-test-image refs/changes/65/746565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-test-image refs/changes/65/746565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-test-image refs/changes/65/746565/1"}}},"commit":{"parents":[{"commit":"037158d7608d063e5567512343c4280229de4bd0","subject":"Remove tox env for python2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-test-image/commit/037158d7608d063e5567512343c4280229de4bd0"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-17 16:56:22.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-17 16:56:22.000000000","tz":-420},"subject":"Drop legacy-base from manila-test-image-build","message":"Drop legacy-base from manila-test-image-build\n\nWe don\u0027t need the legacy and deprecated zuul-cloner\napproach if we rewrite the job in native zuulv3 style.\n\nChange-Id: Ifc6d592b8b7cf0e802ee11be837ebd4a3b55ea9b\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-test-image/commit/1c16a41181aa7c77cbe6c59700d10a2e0270444d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-test-image/commit/1c16a41181aa7c77cbe6c59700d10a2e0270444d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
