)]}'
{"id":"openstack%2Fmanila-ui~688813","triplet_id":"openstack%2Fmanila-ui~stable%2Frocky~I5fb9cef577b109530fde9a4bafe930ea05f3fed8","project":"openstack/manila-ui","branch":"stable/rocky","topic":"bug/1842119-stable/rocky","hashtags":[],"change_id":"I5fb9cef577b109530fde9a4bafe930ea05f3fed8","subject":"Updated to get quotas data for Modify Quotas dialog Share tab","status":"MERGED","created":"2019-10-15 20:23:26.000000000","updated":"2019-10-16 17:57:52.000000000","submitted":"2019-10-16 17:56:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688813-1571248586703-47151255","meta_rev_id":"835e4eef8b228c09bdbb9226909ab5e135473201","_number":688813,"virtual_id_number":688813,"owner":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2019-10-16 02:11:38.000000000","_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 17:56:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2019-10-16 17:13:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},{"value":2,"date":"2019-10-16 10:03:06.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":1,"date":"2019-10-16 16:23:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2019-10-16 17:13:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},{"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 20:23:26.000000000","updated_by":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"reviewer":{"_account_id":11880,"name":"Gloria Gu","email":"gloriafanggu@gmail.com","username":"gugl"},"state":"REVIEWER"},{"updated":"2019-10-15 21:54:33.000000000","updated_by":{"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},"reviewer":{"_account_id":12156,"name":"Jeremy Moffitt","email":"jmoffitt@suse.com","username":"jmoffitt"},"state":"REVIEWER"},{"updated":"2019-10-16 00:54:26.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2019-10-16 10:03:06.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2019-10-16 16:23:24.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2019-10-16 17:13:24.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-10-16 17:56:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1fbb41f7e639be289d1643c587a0d97638f30b24","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2019-10-15 20:23:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"404c488fdf14d60f0f67b1f4316463d40b8b9501","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2019-10-15 20:23:45.000000000","message":"Patch Set 1: Code-Review-1\n\nWIP, waiting for Train and Stein to merge in that order","accounts_in_message":[],"_revision_number":1},{"id":"af400b5f686ab0d4ef3ef5329f2f21ede36f15c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 21:13:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ff9e66a79c74086b155b0ba33fce369 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bfb2b5b834944ea988b46510cce09a5f : SUCCESS in 5m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dc3f6fb83a1a4629b94a9e45add8838a : SUCCESS in 7m 15s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ce8f221faab34324b6538c53fadde601 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e3d1f44bc184d079abf77754cb92718 : SUCCESS in 4m 49s\n- manila-ui-dsvm https://zuul.opendev.org/t/openstack/build/babf49a0d4b84228987ede3f09370975 : SUCCESS in 42m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b88bc58a4334185bc3425b99ebfd301 : SUCCESS in 5m 25s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/5be8f0cc7e2049bda1507d1c007f36c8 : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/160ea673c22a45cba82c410abd5901c8 : SUCCESS in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71e38038d5b5e7a50cf6e8824608ee0685196ece","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2019-10-15 21:51:53.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3f6b96c8643bcbdea9f844a23bd24a00373deb5e","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2019-10-15 21:58:20.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ab886d1cc2da6cc2bccea6c9cf2bd7ff32cae591","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2019-10-15 21:59:04.000000000","message":"Patch Set 3: Code-Review-1\n\nuntil stein merges","accounts_in_message":[],"_revision_number":3},{"id":"b9afbd394d0de79de278b73145b187b5c669beac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:34:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de749cb00d54220be4e07b6eb1c6b73 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84a6539ee9ce47a39a7318ef0a986b34 : SUCCESS in 4m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d90f445cd8694388be6dae295ad26734 : SUCCESS in 5m 07s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/19ba2571f48e453eb7a04a4f85ac69fd : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27cce8cb4f374b1ba41e1edb4a3ef351 : SUCCESS in 6m 15s\n- manila-ui-dsvm https://zuul.opendev.org/t/openstack/build/8bf9d2b2b183483ba7d9b712ce5fab03 : SUCCESS in 34m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1dc75af290b40f68fc5e7b9f6146382 : SUCCESS in 4m 59s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/b01c6048f4904714b9603ebee7430924 : SUCCESS in 5m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b684ead39ce48578f11666bc9b4bf73 : SUCCESS in 7m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ca11d6644693829a804ef87983df4c028394a77d","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2019-10-16 00:53:39.000000000","message":"Patch Set 3: -Code-Review\n\nremove -1 as stein merged","accounts_in_message":[],"_revision_number":3},{"id":"7942ba97e6018f99ad7c4c0037a35cb894409878","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-10-16 01:22:19.000000000","message":"Patch Set 3: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":3},{"id":"ede9e6e973e13e7aa4bd29a2546307a2568fb0e4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-10-16 01:47:39.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nClean cherry-pick, commit message has an incorrect hash","accounts_in_message":[],"_revision_number":3},{"id":"3d8fbd3a75f063986388d0ba1afd076fc5312083","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2019-10-16 02:10:44.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"3f68c55922ac385872e167541d290912bdda7517","author":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"date":"2019-10-16 02:11:38.000000000","message":"Patch Set 4:\n\nSorry not sure how that copy paste did not come through correctly.  Looking at too many patches today","accounts_in_message":[],"_revision_number":4},{"id":"df9ffbdbef545819601b1e6644232d4ed27c2646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 02:54:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b39c5d8c4db5442c9882be3342f134db : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b544e601d6a4c61af792967b734f7a6 : SUCCESS in 5m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a650896b1f7c438db6678c9acdda2852 : SUCCESS in 4m 53s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6fd031e60ac4426ebe193d7a8ad6500d : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0b21cb96d30450f94838d38beba515c : SUCCESS in 5m 16s\n- manila-ui-dsvm https://zuul.opendev.org/t/openstack/build/a042f601cc9c4d929f5580ac411c14c1 : SUCCESS in 31m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76e7e59c7c9a47859bbe39d6cd08827a : SUCCESS in 5m 49s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/8560cddc93e84560ab2a8f6bb65c3bd9 : SUCCESS in 4m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5a42adef1764e1abd182f07277c4eab : SUCCESS in 6m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ca7e4539f763aefd18584d4c641418542207cd4c","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-10-16 10:03:06.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2411d0fe7362d44ba4d3cc65b38e81b337a310a5","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2019-10-16 16:23:24.000000000","message":"Patch Set 4: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":4},{"id":"488d6e8d7eaed5e68f4e3998838030271ece06dd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-10-16 17:13:24.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks, LGTM","accounts_in_message":[],"_revision_number":4},{"id":"085c920860b61b6bc4a0536c1e37430be75e0467","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 17:13:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3207b9a707080fe448e94646665715f605c26e67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 17:56:26.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59773b2d05074c6da63315c5beb37db1 : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c22ba885cb14fc9a347d8fde8b9c43c : SUCCESS in 7m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f4f183a1f3c6415f91bd1b5fc90b10bb : SUCCESS in 6m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7a98128c1d024abebca4b44b20f189e6 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e60f700cb0d84a398838213ea2645a99 : SUCCESS in 5m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c6da7602512441786470f0a483acc38 : SUCCESS in 5m 07s\n- horizon-openstack-tox-python3-django111 https://zuul.opendev.org/t/openstack/build/e622fbdad64144fd800765c91af2649d : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":4},{"id":"0e5eb42212c08676eb975b3e16b7f7161aa969e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 17:56:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"6433f492f207c52c1b474dd2196a682c0d794930","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 17:57:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c526dd56f1b4f12a7272944f32d26b6 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bfc45c381aebe6aaf83f0098a21a473a60134bf0","revisions":{"5c00e9ca2fa536d0d6333e7a480f0158d3e62fc2":{"kind":"REWORK","_number":1,"created":"2019-10-15 20:23:26.000000000","uploader":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"ref":"refs/changes/13/688813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/13/688813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/1"}}},"commit":{"parents":[{"commit":"2365e124915f82e21b1c1d11ea37316d8cefdf17","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/2365e124915f82e21b1c1d11ea37316d8cefdf17"}]}],"author":{"name":"Gloria Gu","email":"gfgu@suse.com","date":"2019-09-04 04:10:56.000000000","tz":-420},"committer":{"name":"Keith Berger","email":"keith.berger@suse.com","date":"2019-10-15 20:23:26.000000000","tz":0},"subject":"Updated to get quotas data for Modify Quotas dialog Share tab","message":"Updated to get quotas data for Modify Quotas dialog Share tab\n\nCurrent Modfiy Quotas action of a project in Identity-\u003eProjects\nfor manila ui does not get quotas data for the project. If user\nupdated quotas for manila ui, the saved data will not be shown\nin the Modify Quotas dialog Share tab.\n\nThis commit includes:\n1) Added a call to get quotas data and also maps the data\nfields retuned to the UI data fields.\n2) Added unit tests.\n3) Added a release note.\n\nChange-Id: I5fb9cef577b109530fde9a4bafe930ea05f3fed8\nCloses-bug: #1842119\nDepends-on: https://review.opendev.org/#/c/679513/\n(cherry picked from commit c763ebe6cd89657aa820ea14774fa3c95e4d0a6b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/5c00e9ca2fa536d0d6333e7a480f0158d3e62fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/5c00e9ca2fa536d0d6333e7a480f0158d3e62fc2"}]},"branch":"refs/heads/stable/rocky"},"3dba20d5490b4f68b307adf235ee48e765623954":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-15 21:51:53.000000000","uploader":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"ref":"refs/changes/13/688813/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/13/688813/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/2"}}},"commit":{"parents":[{"commit":"2365e124915f82e21b1c1d11ea37316d8cefdf17","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/2365e124915f82e21b1c1d11ea37316d8cefdf17"}]}],"author":{"name":"Gloria Gu","email":"gfgu@suse.com","date":"2019-09-04 04:10:56.000000000","tz":-420},"committer":{"name":"Keith Berger","email":"keith.berger@suse.com","date":"2019-10-15 21:51:29.000000000","tz":0},"subject":"Updated to get quotas data for Modify Quotas dialog Share tab","message":"Updated to get quotas data for Modify Quotas dialog Share tab\n\nCurrent Modfiy Quotas action of a project in Identity-\u003eProjects\nfor manila ui does not get quotas data for the project. If user\nupdated quotas for manila ui, the saved data will not be shown\nin the Modify Quotas dialog Share tab.\n\nThis commit includes:\n1) Added a call to get quotas data and also maps the data\nfields retuned to the UI data fields.\n2) Added unit tests.\n3) Added a release note.\n\nChange-Id: I5fb9cef577b109530fde9a4bafe930ea05f3fed8\nCloses-bug: #1842119\nDepends-on: https://review.opendev.org/#/c/679513/\n(cherry picked from commit fb6daa507841b266df89b0261a73728c4fcd0917)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/3dba20d5490b4f68b307adf235ee48e765623954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/3dba20d5490b4f68b307adf235ee48e765623954"}]},"branch":"refs/heads/stable/rocky"},"fd60b968f2f8644b518f0e9271b36e2e7c65fc10":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-15 21:58:20.000000000","uploader":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"ref":"refs/changes/13/688813/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/13/688813/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/3"}}},"commit":{"parents":[{"commit":"2365e124915f82e21b1c1d11ea37316d8cefdf17","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/2365e124915f82e21b1c1d11ea37316d8cefdf17"}]}],"author":{"name":"Gloria Gu","email":"gfgu@suse.com","date":"2019-09-04 04:10:56.000000000","tz":-420},"committer":{"name":"Keith Berger","email":"keith.berger@suse.com","date":"2019-10-15 21:58:14.000000000","tz":0},"subject":"Updated to get quotas data for Modify Quotas dialog Share tab","message":"Updated to get quotas data for Modify Quotas dialog Share tab\n\nCurrent Modfiy Quotas action of a project in Identity-\u003eProjects\nfor manila ui does not get quotas data for the project. If user\nupdated quotas for manila ui, the saved data will not be shown\nin the Modify Quotas dialog Share tab.\n\nThis commit includes:\n1) Added a call to get quotas data and also maps the data\nfields retuned to the UI data fields.\n2) Added unit tests.\n3) Added a release note.\n\nChange-Id: I5fb9cef577b109530fde9a4bafe930ea05f3fed8\nCloses-bug: #1842119\nDepends-on: https://review.opendev.org/#/c/679513/\n(cherry picked from commit c763ebe6cd89657aa820ea14774fa3c95e4d0a6b)\n(cherry picked from commit 7f0c6ad276d42289c61761e9cdcaa3720f7be764)\n(cherry picked from commit fb6daa507841b266df89b0261a73728c4fcd0917)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/fd60b968f2f8644b518f0e9271b36e2e7c65fc10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/fd60b968f2f8644b518f0e9271b36e2e7c65fc10"}]},"branch":"refs/heads/stable/rocky"},"bfc45c381aebe6aaf83f0098a21a473a60134bf0":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-10-16 02:10:44.000000000","uploader":{"_account_id":14892,"name":"Keith Berger","email":"keith.berger@suse.com","username":"kberger"},"ref":"refs/changes/13/688813/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/13/688813/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/13/688813/4"}}},"commit":{"parents":[{"commit":"2365e124915f82e21b1c1d11ea37316d8cefdf17","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/2365e124915f82e21b1c1d11ea37316d8cefdf17"}]}],"author":{"name":"Gloria Gu","email":"gfgu@suse.com","date":"2019-09-04 04:10:56.000000000","tz":-420},"committer":{"name":"Keith Berger","email":"keith.berger@suse.com","date":"2019-10-16 02:10:31.000000000","tz":0},"subject":"Updated to get quotas data for Modify Quotas dialog Share tab","message":"Updated to get quotas data for Modify Quotas dialog Share tab\n\nCurrent Modfiy Quotas action of a project in Identity-\u003eProjects\nfor manila ui does not get quotas data for the project. If user\nupdated quotas for manila ui, the saved data will not be shown\nin the Modify Quotas dialog Share tab.\n\nThis commit includes:\n1) Added a call to get quotas data and also maps the data\nfields retuned to the UI data fields.\n2) Added unit tests.\n3) Added a release note.\n\nChange-Id: I5fb9cef577b109530fde9a4bafe930ea05f3fed8\nCloses-bug: #1842119\nDepends-on: https://review.opendev.org/#/c/679513/\n(cherry picked from commit c763ebe6cd89657aa820ea14774fa3c95e4d0a6b)\n(cherry picked from commit 7f0c6ad276d42289c61761e9cdcaa3720f7be764)\n(cherry picked from commit cc52163a36cd4863ebe96bfef0a163b4190d01b2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/bfc45c381aebe6aaf83f0098a21a473a60134bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/bfc45c381aebe6aaf83f0098a21a473a60134bf0"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
