)]}'
{"id":"openstack%2Fmanila-ui~793713","triplet_id":"openstack%2Fmanila-ui~master~I686257fe9ffaefde4d2cdca8ff60a83b908f3b6b","project":"openstack/manila-ui","branch":"master","topic":"update-min-tox","hashtags":[],"change_id":"I686257fe9ffaefde4d2cdca8ff60a83b908f3b6b","subject":"Changed minversion in tox to 3.18.0","status":"MERGED","created":"2021-05-31 02:02:08.000000000","updated":"2021-07-06 16:29:05.000000000","submitted":"2021-07-06 16:27:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793713-update-min-tox","meta_rev_id":"083162cb9e56cba1b1ad8532feff253531505ed1","_number":793713,"virtual_id_number":793713,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-06 16:27:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2021-07-06 15:34:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-06 15:15:07.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2021-07-06 15:34:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-31 02:51:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-06 15:15:07.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-07-06 15:34:31.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"c6205b02931b90ae98273ca3c6af602e74209356","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-05-31 02:02:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"faa83c81472752c94d499be0647864c79160c172","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-05-31 02:03:47.000000000","message":"Topic set to update-min-tox","accounts_in_message":[],"_revision_number":1},{"id":"489ca93a3e25fbfeba58f388671eb835be1ea0de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 02:51:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/786c3f55e0eb49a8b8a9cad81f73e0ba : SUCCESS in 7m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cce275653928468d9ca9225aa9170697 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24a62b316b194602a0fff29437380dbf : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b35828897d9342ee98b293089fcea2a1 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b59580363c2c4824becbb599e61d9c7f : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be17325a47d640759bf96af3b3aada61 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a8b0f1966e4fa78bb0e7122b6cf570 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3448c649efb74472b8a201c8fb82c0f2 : SUCCESS in 6m 36s\n- manila-ui-integration-tests https://zuul.opendev.org/t/openstack/build/9cea5c6d6a4a4006a850a59d844971fe : FAILURE in 43m 27s","accounts_in_message":[],"_revision_number":1},{"id":"191dec077fe18b7a86ad5eb23e2aa8a5e052d3b4","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-05-31 03:48:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"78170afe1dbaab923df78a9a10a28773ea9eef50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 04:33:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c3c4957c9fd43f98a49e0d73235d2bc : SUCCESS in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6c0614432614c9a9ebe23e20a6002e3 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f50a3662b6704e1aa1cafc6d2363fc2a : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c7b663898144c59b73adc3c42009f3e : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4785828254a144f0aec469e5097e6c08 : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/614d4acb4e734a79aafede070cbc3387 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb838ba27dc34fd4b7743c000e981a9a : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1ad05ef46754978be2e6148fd37d1e4 : SUCCESS in 5m 40s\n- manila-ui-integration-tests https://zuul.opendev.org/t/openstack/build/6d57891d1cb1495da25e012cf57f889c : FAILURE in 41m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e254357043bef0565498a9392715155d4592cc42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-07-06 07:56:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d2c972fef65b2405e7a1914d6eb4495496993c61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 08:58:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17910ab3d05c48728a81a5e54c925053 : SUCCESS in 7m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f463da4004f48d4a1f07f703018caa2 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8888593c71c484a93a0255094584469 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/673e02f728ec483eb0eafe5a234d0f17 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f935ed60816b4d10a542c980d9930166 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5221e9188ef14d73ad621efcc72a2ede : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a7f0cd1b99467fa9058f03a15e0616 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a330ff8e72214b69bbbc74dee44e7c29 : SUCCESS in 7m 19s\n- manila-ui-integration-tests https://zuul.opendev.org/t/openstack/build/ab4d62901c834d5f981440e801084bed : FAILURE in 50m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c108127bf0165247c5031c50cad79a3285679c72","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-07-06 15:15:07.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks wushiming, looks good.","accounts_in_message":[],"_revision_number":2},{"id":"0234c66df3b1c3663e312186ef9ebdfc0a77efc5","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-07-06 15:34:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM. Thank you!","accounts_in_message":[],"_revision_number":2},{"id":"4a64543a67b6efc09395f76b3495a1542819e5ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 15:37:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc921d145a60305432a871643af6be504d89e087","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 16:27:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b1da299a3c64ad68b9b9243db407333 : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/236c4ad783a34ceeb7672f2ed4722aec : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d68e62e27a4412a0e980fb4e3729da : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfec3eb1433445c1b9ca355a86efc450 : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/298edc0444314d679d66d399083e243c : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/697d80201d4d4304a3896e4efa15f152 : SUCCESS in 6m 48s\n- manila-ui-integration-tests https://zuul.opendev.org/t/openstack/build/9c85efab5e0347ddb9a8df0770b254b3 : FAILURE in 46m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c728f20081a0e7a428d705e6acc4a3bf5d2e9477","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 16:27:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"083162cb9e56cba1b1ad8532feff253531505ed1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 16:29:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7de5e4e10f46402ba8d65a91eb8eba7f : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/887b199cb5cb44c8901e5ad336135d03 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eb52c16c65593c5a277935a9575bc4cfeabdfaea","revisions":{"7e621c3a2f6c347b09117b27f7243f444b9cce24":{"kind":"REWORK","_number":1,"created":"2021-05-31 02:02:08.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/13/793713/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/13/793713/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/793713/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/793713/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/793713/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/13/793713/1"}}},"commit":{"parents":[{"commit":"cc8cf7378a1433ef41fce8e0bebf6d29211afcf6","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/cc8cf7378a1433ef41fce8e0bebf6d29211afcf6"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-05-31 02:01:51.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-05-31 02:01:51.000000000","tz":480},"subject":"Changed minversion in tox to 3.18.0","message":"Changed minversion in tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace tox\u0027s whitelist_externals by allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nChange-Id: I686257fe9ffaefde4d2cdca8ff60a83b908f3b6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/7e621c3a2f6c347b09117b27f7243f444b9cce24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/7e621c3a2f6c347b09117b27f7243f444b9cce24"}]},"branch":"refs/heads/master"},"eb52c16c65593c5a277935a9575bc4cfeabdfaea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-06 07:56:10.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/13/793713/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/13/793713/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/793713/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/793713/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/13/793713/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/13/793713/2"}}},"commit":{"parents":[{"commit":"4827a4c3366c94b307518d1b6ad44ce440210e3e","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/4827a4c3366c94b307518d1b6ad44ce440210e3e"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-05-31 02:01:51.000000000","tz":480},"committer":{"name":"wushiming","email":"wushiming@yovole.com","date":"2021-07-06 07:56:10.000000000","tz":0},"subject":"Changed minversion in tox to 3.18.0","message":"Changed minversion in tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace tox\u0027s whitelist_externals by allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nChange-Id: I686257fe9ffaefde4d2cdca8ff60a83b908f3b6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/eb52c16c65593c5a277935a9575bc4cfeabdfaea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/eb52c16c65593c5a277935a9575bc4cfeabdfaea"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
