)]}'
{"id":"openstack%2Fmanila-ui~796553","triplet_id":"openstack%2Fmanila-ui~stable%2Fwallaby~I92e87282020047c7fad055cae92f2382c4193597","project":"openstack/manila-ui","branch":"stable/wallaby","topic":"bp/remove-nova-net-plugin","hashtags":[],"change_id":"I92e87282020047c7fad055cae92f2382c4193597","subject":"Remove nova networking","status":"MERGED","created":"2021-06-15 21:33:06.000000000","updated":"2021-06-16 22:57:12.000000000","submitted":"2021-06-16 22:55:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796553-bp/remove-nova-net-plugin","cherry_pick_of_change":786748,"cherry_pick_of_patch_set":1,"meta_rev_id":"c533f92ededc9386f7567d6e8ad98438d24fd380","_number":796553,"virtual_id_number":796553,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 22:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2021-06-16 22:04:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2021-06-16 10:26:23.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2021-06-16 22:04:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 22:15:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 10:26:23.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-06-16 19:49:51.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-06-16 19:50:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"8a2ddf6f79caa2473126c3da6412687f00302344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-15 21:33:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b3ee72be43229cbcff0a25dadc381705bab4e2bc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-15 21:37:11.000000000","message":"Topic bp/remove-nova-net-plugin-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"4d83c8f2be6a11000e6ba40a2ab7b23244d77f6a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-15 21:37:45.000000000","message":"Topic set to bp/remove-nova-net-plugin","accounts_in_message":[],"_revision_number":1},{"id":"8acf56b21e1af0dc8e42d4988d39ecd1e73f250b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 22:15:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e115bb1e1764fba83390855313a42f8 : SUCCESS in 9m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8e89534baaf46ba8aab2cbff44909fa : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a493088886414c55b65d8e98cee8cc0f : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a660795aa296452dbb838008cc0e4004 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c038019b35bf4d9a9198162cfa3b55b5 : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c7ab038eafc430dac51ad5e0b9490e2 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e052040cf67a4ebbb4eb4470a612fc13 : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/664a0af826d048d7ab7a14bb095d6037 : SUCCESS in 5m 29s\n- manila-ui-integration-tests https://zuul.opendev.org/t/openstack/build/345d6dd8ffbc4b5b8a79cfded3a87df8 : SUCCESS in 40m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0cb6442de3583f2f679de8953a5b9f17dd45b7dd","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-16 10:26:23.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks, Goutham!  Wow, this is overdue.","accounts_in_message":[],"_revision_number":1},{"id":"97abf46a11b6ad273748860e76f0e5ebd085b2e9","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-06-16 22:04:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean cherry pick, LGTM, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"a43049996df6a3bc9c1a8ff1aaa85bd233701d9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 22:04:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"879405265a6e17c91be61201df37bf87314ec69d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 22:55:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/857e7fad08914aa8a2eb2438e4a861e9 : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26fe4828a85d4215b61a3b164db36443 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1f902f013a6441c9ec19c85af5cf57b : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ae27e0b8bc04ebab36d8278c7d89dea : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/050a0ed91f244643b70b3e7dd0af2baf : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35e836e97bf340ce889958891dcfa635 : SUCCESS in 5m 07s\n- manila-ui-integration-tests https://zuul.opendev.org/t/openstack/build/6c71149b8edc4dd9a178557db58cca9a : SUCCESS in 46m 47s","accounts_in_message":[],"_revision_number":1},{"id":"483b8a4811b011d64fc03d5d43fc0f756fcdee73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 22:55:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c533f92ededc9386f7567d6e8ad98438d24fd380","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 22:57:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce6a3c3aed54727960b4ca1ee4f3536 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3318281952b4a419ad4860d9ee56beb : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"85137cf757d848a51d82f78f770fb507532279ee","revisions":{"85137cf757d848a51d82f78f770fb507532279ee":{"kind":"REWORK","_number":1,"created":"2021-06-15 21:33:06.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/53/796553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/53/796553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/53/796553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/53/796553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/53/796553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/53/796553/1"}}},"commit":{"parents":[{"commit":"1f48e83eae979cb6bca78a024d7023ae70102c1b","subject":"Support policy-in-code and deprecated policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/1f48e83eae979cb6bca78a024d7023ae70102c1b"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-04-17 01:31:22.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-06-15 21:33:06.000000000","tz":0},"subject":"Remove nova networking","message":"Remove nova networking\n\nThis is essentially dead code elimination\nfrom manila-ui at this point since manila\nand python-manilaclient dropped support\nin the Ocata release.\n\nChange-Id: I92e87282020047c7fad055cae92f2382c4193597\nPartially-implements: bp remove-nova-net-plugin\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 4edb837a1b5492b14b05d00236337525bdcde686)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/85137cf757d848a51d82f78f770fb507532279ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/85137cf757d848a51d82f78f770fb507532279ee"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
