)]}'
{"id":"openstack%2Fmanila-ui~830786","triplet_id":"openstack%2Fmanila-ui~stable%2Fwallaby~Ib57e0dc00c9f47d589d8723d7feac01f27b0f3a0","project":"openstack/manila-ui","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-02-24 11:42:18.000000000","reason":"Carlos Eduardo replied on the change"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-02-24 20:31:32.000000000","reason":"Goutham Pacha Ravi replied on the change"},"34319":{"account":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"last_update":"2022-02-24 21:19:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib57e0dc00c9f47d589d8723d7feac01f27b0f3a0","subject":"Add platform annotation in bindep.txt","status":"MERGED","created":"2022-02-24 08:43:34.000000000","updated":"2022-02-24 21:22:45.000000000","submitted":"2022-02-24 21:19:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830786","cherry_pick_of_change":830809,"cherry_pick_of_patch_set":1,"meta_rev_id":"f116f608b977d24621ca209af2a38945bcc7287f","_number":830786,"virtual_id_number":830786,"owner":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-24 21:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-02-24 20:04:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-02-24 20:31:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2022-02-24 20:31:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-24 08:44:41.000000000","updated_by":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-02-24 08:44:41.000000000","updated_by":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-02-24 09:40:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c229a3cf653a68ad611db1f2b7020efb316536a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"date":"2022-02-24 08:43:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1275486c4e5eaef6720c6ade8fa1236b17d8f4d","author":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"date":"2022-02-24 08:44:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa690714f00e3737bdf5344e5521d94c6d16f272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 09:40:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f8b1468d1624e2ca1688e55750d4f02 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/260de3dc947d418a80f68834cf25f2dc : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a7a07d9a5754d3bae66d16bb769c7e9 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e668afe1f5984cbda8b311a88e43d540 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a28d996f4b1486e921330c3c265a9a0 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d35f3c7938414aa58fec11eac475cfe6 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69eb5ec8c23e435abcdeb93b1f79ac2f : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecb4ab25ca3f4cc5893dde184ddae2b5 : SUCCESS in 5m 07s\n- manila-ui-integration-tests https://zuul.opendev.org/t/openstack/build/52cdf510d1da49ef8d4d0b3e764fa326 : SUCCESS in 49m 48s","accounts_in_message":[],"_revision_number":1},{"id":"927bd061d13a97b13bf67fafe5bd736dc6577d46","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-24 11:42:18.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b62b1f535c9d02b261fed29dc26959d394054bd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-24 17:25:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"79efd8738acc77e408c4b7b4b77ba0561cd46508","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-24 17:26:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd521fd0afb8a2394f1810e2fbb67185f865488d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 18:13:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cd08548edab4efd846461b51ae56302 : SUCCESS in 7m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e77e0a8e444d470eb5e7825eaee79751 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a0dcee376074e1fb8df8f6130266497 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d510180ce664b0cb4166c3f461af765 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c43ef2b11c440ea8b86e1f75e0accfd : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43b1aa53eb664222a72e19ee036b6934 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e70141a7e5aa4e74bf7765764d28da41 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/055f59f499514f8e991e784edaf47aa3 : SUCCESS in 6m 43s\n- manila-ui-integration-tests https://zuul.opendev.org/t/openstack/build/73cd6be5690d4725a1fdfab907a5eb70 : SUCCESS in 45m 00s","accounts_in_message":[],"_revision_number":2},{"id":"7c3f39171aff504df3ad2e7a3dc8ae07eadcd6a7","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-24 20:04:06.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"50751c5f293b38ae86bfb2c8f090d6115b24388f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-24 20:31:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39d8ea9ebc92f7def10e6047a8f0d7068fb60963","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 20:33:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"07b07c4465d525aab7ae01d3085b05b636bc6bea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 21:19:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77682b1d3c884ea78a568d8de2fc6f53 : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebcad2d46f2a4186bd53eb049bce34d5 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c05da8b7dea346cc9d3f059aff745d0f : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b504e5715899449b801a2dd3ce575a3c : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87213f2bc6cb47f69a77a6b8eabeb256 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1909868926a94affa1e67e9a430849b1 : SUCCESS in 6m 03s\n- manila-ui-integration-tests https://zuul.opendev.org/t/openstack/build/9b09844f06164e6ea7f83a22d7d441dd : SUCCESS in 45m 38s","accounts_in_message":[],"_revision_number":2},{"id":"1a744a5f000a4f829d616695770492784593833e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 21:19:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f116f608b977d24621ca209af2a38945bcc7287f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 21:22:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0372615eeae94adeaf9f59ae7829b647 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6242d85b858749208d77c42ea904100f : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f7a8a8a6a736bba8abf76f50ebd26ed2e391f7c7","revisions":{"7e12c0653b80931e2602e4b48d6c550a59a5d4b1":{"kind":"REWORK","_number":1,"created":"2022-02-24 08:43:34.000000000","uploader":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"ref":"refs/changes/86/830786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/86/830786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/86/830786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/86/830786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/86/830786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/86/830786/1"}}},"commit":{"parents":[{"commit":"f421aeb1e18664411cbb20420fe2c4847767573b","subject":"Fix share group creation with share types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/f421aeb1e18664411cbb20420fe2c4847767573b"}]}],"author":{"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","date":"2022-02-22 10:43:22.000000000","tz":60},"committer":{"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","date":"2022-02-24 08:43:15.000000000","tz":60},"subject":"Add platform annotation in bindep.txt","message":"Add platform annotation in bindep.txt\n\nAdd dpkg platform for the packages and their\nrespective package name for rpm too. There are some\npackages not available in rpm so I added just those that have it.\n\nRelated-Bug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d2054776\nChange-Id: Ib57e0dc00c9f47d589d8723d7feac01f27b0f3a0\n(cherry picked from commit 014ed7d41e9e819ecad2f2a0286b0c1d7e7e002e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/7e12c0653b80931e2602e4b48d6c550a59a5d4b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/7e12c0653b80931e2602e4b48d6c550a59a5d4b1"}]},"branch":"refs/heads/stable/wallaby"},"f7a8a8a6a736bba8abf76f50ebd26ed2e391f7c7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-24 17:25:23.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/86/830786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/86/830786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/86/830786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/86/830786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/86/830786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/86/830786/2"}}},"commit":{"parents":[{"commit":"f421aeb1e18664411cbb20420fe2c4847767573b","subject":"Fix share group creation with share types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/f421aeb1e18664411cbb20420fe2c4847767573b"}]}],"author":{"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","date":"2022-02-22 10:43:22.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-02-24 17:25:23.000000000","tz":0},"subject":"Add platform annotation in bindep.txt","message":"Add platform annotation in bindep.txt\n\nAdd dpkg platform for the packages and their\nrespective package name for rpm too. There are some\npackages not available in rpm so I added just those that have it.\n\nRelated-Bug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d2054776\nChange-Id: Ib57e0dc00c9f47d589d8723d7feac01f27b0f3a0\n(cherry picked from commit 014ed7d41e9e819ecad2f2a0286b0c1d7e7e002e)\n(cherry picked from commit d3c364fc6594b11a8a1d3fc5ba539bb7540731cd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/f7a8a8a6a736bba8abf76f50ebd26ed2e391f7c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/f7a8a8a6a736bba8abf76f50ebd26ed2e391f7c7"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
