)]}'
{"id":"openstack%2Fmanila-ui~830809","triplet_id":"openstack%2Fmanila-ui~stable%2Fxena~Ib57e0dc00c9f47d589d8723d7feac01f27b0f3a0","project":"openstack/manila-ui","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-02-24 12:40:34.000000000","reason":"Carlos Eduardo replied on the change"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-02-24 17:24:26.000000000","reason":"Goutham Pacha Ravi replied on the change"},"34319":{"account":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"last_update":"2022-02-24 18:14:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib57e0dc00c9f47d589d8723d7feac01f27b0f3a0","subject":"Add platform annotation in bindep.txt","status":"MERGED","created":"2022-02-24 12:32:39.000000000","updated":"2022-02-24 18:17:24.000000000","submitted":"2022-02-24 18:14:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830809","meta_rev_id":"fa5dc0141b24090a0cbc629d576a8818e824a698","_number":830809,"virtual_id_number":830809,"owner":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-24 18:14:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-02-24 14:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-02-24 17:24:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2022-02-24 17:24:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-24 12:35:45.000000000","updated_by":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-02-24 12:35:45.000000000","updated_by":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-02-24 13:36:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"774da253b0febfe35580f781954d944bb57b8caf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"date":"2022-02-24 12:32:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e0c8877d39c8576cc271298050c62eef8189168","author":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"date":"2022-02-24 12:35:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbd1f7ef79c1f08930b5d3698be03d533afea302","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-24 12:40:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c603dc852bcc784480ad3f55cb7bccb569aea943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 13:36:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29f5ddd6da8f477e90598843ea852a5d : SUCCESS in 6m 55s\n- horizon-tox-python3-django32 https://zuul.opendev.org/t/openstack/build/117637f093fe4915aa2179bba792ad1c : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/642c85f9634044878e501deb072b4383 : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9dbb5fc088a547e18bb987ad8711fdca : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60535f9c92ae414ab840c384b7124f88 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e139d6519ee646e88ce6dc636b9accf8 : SUCCESS in 8m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be320db611b14af0851291cecaf2afee : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75255c42d6ac42b6b82de55f2459115f : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9152bbbd13d541da824f61db8bb221b8 : SUCCESS in 5m 26s\n- manila-ui-integration-tests https://zuul.opendev.org/t/openstack/build/257004a1c46e48369273afb66a916286 : FAILURE in 57m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f66a6f046f6e2bc6689b2b2656ca1b2222e6a37","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-24 14:32:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a5145b4ba6369978e3d77c130f723e558bdb3aa","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-24 17:24:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b17e0ed140cf54822c394b76ce3f4bdf5ddae81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 17:25:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c11c28ff0749dbe223b9b2ff4337c607b77101ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 18:14:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2b3223fc8a046528ba982f8932bdf3b : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd0dfe1e1e034dfca370b2e0a554ce90 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab23b18fe66245e9b518aa71db107c6d : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6693b22a6ea449a96df970ff33c2a1f : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e61c88cdc8478ba1be9f2c0a6770e4 : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a29e847c8bf64f71912a336a35e38290 : SUCCESS in 6m 21s\n- manila-ui-integration-tests https://zuul.opendev.org/t/openstack/build/a33cffa8c9b4482489b8bf6421f6742b : FAILURE in 47m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"449de06f19dac83e59af76b36ea5a413d08a3361","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 18:14:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fa5dc0141b24090a0cbc629d576a8818e824a698","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 18:17:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df2cc6caaa44f03b5ee0332effdf588 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d87b92da90ad40898acc640a974c75a8 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3c364fc6594b11a8a1d3fc5ba539bb7540731cd","revisions":{"d3c364fc6594b11a8a1d3fc5ba539bb7540731cd":{"kind":"REWORK","_number":1,"created":"2022-02-24 12:32:39.000000000","uploader":{"_account_id":34319,"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","username":"adrianfuscoar"},"ref":"refs/changes/09/830809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/09/830809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/09/830809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/09/830809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/09/830809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/09/830809/1"}}},"commit":{"parents":[{"commit":"544305443a2002c0658a93fdb9a129f57d2977bb","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/544305443a2002c0658a93fdb9a129f57d2977bb"}]}],"author":{"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","date":"2022-02-22 10:43:22.000000000","tz":60},"committer":{"name":"Adrian Fusco Arnejo","email":"afuscoar@redhat.com","date":"2022-02-24 12:32:28.000000000","tz":60},"subject":"Add platform annotation in bindep.txt","message":"Add platform annotation in bindep.txt\n\nAdd dpkg platform for the packages and their\nrespective package name for rpm too. There are some\npackages not available in rpm so I added just those that have it.\n\nRelated-Bug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d2054776\nChange-Id: Ib57e0dc00c9f47d589d8723d7feac01f27b0f3a0\n(cherry picked from commit 014ed7d41e9e819ecad2f2a0286b0c1d7e7e002e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/d3c364fc6594b11a8a1d3fc5ba539bb7540731cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/d3c364fc6594b11a8a1d3fc5ba539bb7540731cd"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
