)]}'
{"id":"openstack%2Fmanila-ui~987136","triplet_id":"openstack%2Fmanila-ui~master~Id09c9c6bff4dc77729dfaab2eabd735363e5d8a1","project":"openstack/manila-ui","branch":"master","attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-05-06 22:19:17.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-20 11:00:15.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Id09c9c6bff4dc77729dfaab2eabd735363e5d8a1","subject":"Update packaging configuration","status":"NEW","created":"2026-05-04 10:28:57.000000000","updated":"2026-08-10 14:02:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"47abc6588c4668ab3bc3371d0619892c88ca2fc6","_number":987136,"virtual_id_number":987136,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-10 14:02:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-04 10:54:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-06 22:19:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"e917a97d219947522cfa12400a0893be10540d01","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 10:28:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"243f2077a2c1a729132759b3ea0526a2b7c8b63e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 10:29:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7318b6d169a617d5437ca9cd711427c22e1bd52a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 10:30:06.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"5bf1b4adbeb8b3d0d1a43c08cb5e04a2ec02b282","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 10:30:19.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"2508f035eb0b21e194becba607e1c8554ddc5d92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 10:54:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f370f039f9344eecb356501cbd0d237a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d080742248094f54944c878f5ac07525 : SUCCESS in 3m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ae8fe27f190d4ae9aa4dee3683f3c6a0 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61290b8edf18471bb7696e60cc93d6db : SUCCESS in 9m 04s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/d8fe287d0bc844268423f4812f6cdf6b : SUCCESS in 5m 19s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/adcc79be5a2441d38e8973b3d4622afe : SUCCESS in 11m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0aef25a568e44ec9d771a1bbbb274ec : SUCCESS in 18m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3c4c8f96ec48d0aafd5eb7811b4939 : SUCCESS in 7m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a2a04e17c3b34029a66e974ff8b23aff : SUCCESS in 8m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8bfbf513748f43e48743c460b13b68ca : FAILURE in 10m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/688aa2031fd84fb9836e000fd24d657c : SUCCESS in 9m 00s\n- manila-ui-tox-bandit https://zuul.opendev.org/t/openstack/build/f854977dde2b4124bbab0941115688b9 : SUCCESS in 3m 46s\n- manila-ui-integration-pytest https://zuul.opendev.org/t/openstack/build/7ee409af924b48bd97095375e9199a82 : SUCCESS in 20m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7bb07f14e7cd5c52c1beb07ab045099a33adc6aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-04 16:47:32.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"15633212ba6eec056a33c9f5f0c79595a987ffb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-04 17:27:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8606f610e8474013a485247a52b7ba0f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5bc02d27e3594222ad4642873b0b655f : SUCCESS in 2m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4aaf2daab87c4b4eb53c0aaf653a918d : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1f784d0fb5e4084896f32efaef4093b : SUCCESS in 8m 49s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/0b2d5463498d4edeb2634a4e447a1771 : SUCCESS in 8m 46s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/d50aa444e9bb49f28429af4a7369a320 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66307c71f3bf4ec0add91867ff79c1a0 : SUCCESS in 17m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06438d6bae4c4c9ea3a7ebd64eb84521 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6fa75c49696041acb7f397a1cc53b69c : SUCCESS in 8m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fb5add3e9fd940ffacb058207574c685 : FAILURE in 11m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7cc90da8c574bc793ee333660333aa5 : SUCCESS in 8m 12s\n- manila-ui-tox-bandit https://zuul.opendev.org/t/openstack/build/3db6d42ba09f476d81033e74e5ae7170 : SUCCESS in 6m 33s\n- manila-ui-integration-pytest https://zuul.opendev.org/t/openstack/build/f21edaa72d1449c9a4fd6df9805719a4 : SUCCESS in 37m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"52bee07b85377de58b6dde117d4b4dba21200f91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-14 19:10:54.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"b2ea669685c73688c5264b30699ab4a9e199b742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 19:50:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/703afe4fcc5e48acab18362554697ddf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/90df1ccd8ad44a1eae67c4e72aace44f : SUCCESS in 2m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd7772c9d3d1414aacb6d27b2feecf88 : SUCCESS in 6m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc05354853f14c1d8f4c52c905db7408 : SUCCESS in 10m 30s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/c157035005ac4881804b5f0da0c93f3e : SUCCESS in 10m 06s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/98e589407fc84622a516635c8823e7ac : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e204202d712431f9050430a0860b1a9 : SUCCESS in 17m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03954d3a8dc345a2acf7f97b95f4c35c : SUCCESS in 8m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b67e77ceb47b47b68e8bd34cdc90bc2c : SUCCESS in 7m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f838298282064f87b0870a5af2397bf7 : FAILURE in 12m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73fbd89dc2354c91a04e6166b425166a : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2b58ddd993c497c8d0a9a242bf15ea3 : FAILURE in 7m 00s\n- manila-ui-tox-bandit https://zuul.opendev.org/t/openstack/build/068930981a10464db47f5c932f8d3eff : SUCCESS in 6m 33s\n- manila-ui-integration-pytest https://zuul.opendev.org/t/openstack/build/58bca829305c4524b24e5ed51b45734e : SUCCESS in 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e27801eb43cf5c4c919823489037ceae8e93dbf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-20 10:30:21.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"051df6c7705c55a90ce87499e2463b64267fde3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 10:58:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd2a26265f7149058e232c2790dcb31f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ee610267d88c4dd9af8f253bb060c12b : SUCCESS in 1m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4c06f0036e4544c0b845df1c346c73d7 : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0157d6f895f94dc69e08810bd5b8aa0f : SUCCESS in 4m 33s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/f86b44e495d94b6191d28f4f94131047 : SUCCESS in 6m 47s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/198b04a4684e48659d2397a8adb8b19f : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8d125f25dca4196b74eb3321b2994f4 : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77c6d6e6c5274b259aa6096b26fcb5bd : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bec76566bb464b3aa24f8203f177a9b6 : SUCCESS in 6m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0752a04d1dd14a74970afb1ecb2cba55 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac2d77b90fc9466a81cb8cb832c7bdcd : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/984714af3ba94281902cb5112e5d27d9 : FAILURE in 4m 45s\n- manila-ui-tox-bandit https://zuul.opendev.org/t/openstack/build/9e320a07b0a5476c841c58b5b45cf36d : SUCCESS in 4m 53s\n- manila-ui-integration-pytest https://zuul.opendev.org/t/openstack/build/0a5bb233dade4148b545468a9ee28e03 : SUCCESS in 26m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c5c0860d375e319a993736b2463d4b46a9b45754","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-20 11:00:10.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"e9ff9b6414c3819a287d982bd90cc54d968a058b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 11:24:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11eea120b33b484cb790c8008c917e8e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a42114621080443da45b8154a62525c7 : SUCCESS in 3m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/975cd40db13e426b8cc916af7131d806 : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/588adef31fb5427f9b745ff46445c1e5 : SUCCESS in 8m 08s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/af444c453bc649b59b2ad37c2645a0ea : SUCCESS in 7m 49s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/ea16a245143347b6b54826c8d123b930 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ccb580fac944a9a9cf52d60393d5453 : SUCCESS in 17m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d99410e19e1448aaba965a01dac47f4 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0fb8d260442d4e6f8f2dc3f3d64a4135 : SUCCESS in 4m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/54615c54d2944c9b821067835322a8a5 : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f62c0b1acc74f99a8589bf39a873c4c : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75b8541a8de343a28461b092f80821b4 : SUCCESS in 6m 13s\n- manila-ui-tox-bandit https://zuul.opendev.org/t/openstack/build/85d0e7c38ecf49dd9d780dfc3ea337a4 : SUCCESS in 6m 08s\n- manila-ui-integration-pytest https://zuul.opendev.org/t/openstack/build/6afa4dbf800f4515880f5d8b760ffe40 : SUCCESS in 19m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f69cdb32636fa3b0bec7e411ef16a5b942d1bc1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-10 13:31:55.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"79a0981abcc79d8e3ab449f4d69dc823a95aaeda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-10 13:32:17.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"5556c0d31d9aa9c14fc7118ea3b7bec9a69df595","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-10 13:32:27.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"47abc6588c4668ab3bc3371d0619892c88ca2fc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 14:02:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eefbfe2b4fc46fb89488cef3a74ba3a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e0188f2c3978435398d6774e7d0c7db4 : SUCCESS in 2m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/35ac4d5005f54e2a972ac18ba1b672f0 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee2912519d1f4dc4b3b9aa9e21a00f09 : SUCCESS in 5m 23s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/58841f3ba85d46a28e1f1607dceafb65 : SUCCESS in 12m 04s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/2c1f2980c19b4b219572435b94f35865 : SUCCESS in 10m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c6dcc1ef5db4575b9be9cf9728490fd : SUCCESS in 18m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5efe153c784c60a7a800a57a219d40 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3b3509bd96c400aadbb5ecd3b24cd98 : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6264b13397954b69b43bdb5f0d1047bc : SUCCESS in 12m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f6855d3d3340eda80163bd16f074b9 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdaf8290a6fb49cf86ed4f5da56935c5 : SUCCESS in 6m 39s\n- manila-ui-tox-bandit https://zuul.opendev.org/t/openstack/build/532d3267043f4012aaf59be57154c139 : SUCCESS in 3m 22s\n- manila-ui-integration-pytest https://zuul.opendev.org/t/openstack/build/d9b2d20032064e4194b15c4f5440e4eb : SUCCESS in 27m 17s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"c15d1b83b78f26d2da1ec3b3f4b680727526bbe9","revisions":{"8d0b1f84006c6d6896c5df3a44a09f0ba36e216d":{"kind":"REWORK","_number":1,"created":"2026-05-04 10:28:57.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/987136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/36/987136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/1"}}},"commit":{"parents":[{"commit":"a77ec044764a7d362c9fc189fcc3b57ae7fe1e6d","subject":"Fix neutron code with openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/a77ec044764a7d362c9fc189fcc3b57ae7fe1e6d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 10:28:57.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 10:28:57.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Id09c9c6bff4dc77729dfaab2eabd735363e5d8a1\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/8d0b1f84006c6d6896c5df3a44a09f0ba36e216d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/8d0b1f84006c6d6896c5df3a44a09f0ba36e216d"}]},"branch":"refs/heads/master","conflicts":{"contains_conflicts":false}},"18258de3efb3aeed254537eaa5691114c86a3867":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-05-04 10:29:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/987136/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/36/987136/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/2"}}},"commit":{"parents":[{"commit":"a77ec044764a7d362c9fc189fcc3b57ae7fe1e6d","subject":"Fix neutron code with openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/a77ec044764a7d362c9fc189fcc3b57ae7fe1e6d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 10:28:57.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 10:29:24.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nChange-Id: Id09c9c6bff4dc77729dfaab2eabd735363e5d8a1\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/18258de3efb3aeed254537eaa5691114c86a3867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/18258de3efb3aeed254537eaa5691114c86a3867"}]},"branch":"refs/heads/master","description":"Edit commit message"},"14b8836221e3cb56eeb213540254bbf93eb32f97":{"kind":"REWORK","_number":3,"created":"2026-05-04 10:30:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/987136/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/36/987136/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/3"}}},"commit":{"parents":[{"commit":"a77ec044764a7d362c9fc189fcc3b57ae7fe1e6d","subject":"Fix neutron code with openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/a77ec044764a7d362c9fc189fcc3b57ae7fe1e6d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 10:28:57.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 10:30:06.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nChange-Id: Id09c9c6bff4dc77729dfaab2eabd735363e5d8a1\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/14b8836221e3cb56eeb213540254bbf93eb32f97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/14b8836221e3cb56eeb213540254bbf93eb32f97"}]},"branch":"refs/heads/master"},"803c192377c6e2eee35f225478c74290b684b1cc":{"kind":"REWORK","_number":4,"created":"2026-05-04 16:47:32.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/987136/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/36/987136/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/4"}}},"commit":{"parents":[{"commit":"a77ec044764a7d362c9fc189fcc3b57ae7fe1e6d","subject":"Fix neutron code with openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/a77ec044764a7d362c9fc189fcc3b57ae7fe1e6d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 10:28:57.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 16:47:31.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nChange-Id: Id09c9c6bff4dc77729dfaab2eabd735363e5d8a1\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/803c192377c6e2eee35f225478c74290b684b1cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/803c192377c6e2eee35f225478c74290b684b1cc"}]},"branch":"refs/heads/master"},"eb10b5c5fa8c30ea8d123cb1ad1575b1696addc0":{"kind":"REWORK","_number":5,"created":"2026-05-14 19:10:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/987136/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/36/987136/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/5"}}},"commit":{"parents":[{"commit":"a77ec044764a7d362c9fc189fcc3b57ae7fe1e6d","subject":"Fix neutron code with openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/a77ec044764a7d362c9fc189fcc3b57ae7fe1e6d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 10:28:57.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-14 19:10:54.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nChange-Id: Id09c9c6bff4dc77729dfaab2eabd735363e5d8a1\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/eb10b5c5fa8c30ea8d123cb1ad1575b1696addc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/eb10b5c5fa8c30ea8d123cb1ad1575b1696addc0"}]},"branch":"refs/heads/master"},"a1852b64e4c628eaa7ecbff1ee36c6bad8dded24":{"kind":"REWORK","_number":6,"created":"2026-05-20 10:30:21.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/987136/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/36/987136/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/6"}}},"commit":{"parents":[{"commit":"a77ec044764a7d362c9fc189fcc3b57ae7fe1e6d","subject":"Fix neutron code with openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/a77ec044764a7d362c9fc189fcc3b57ae7fe1e6d"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 10:28:57.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-20 10:30:21.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nChange-Id: Id09c9c6bff4dc77729dfaab2eabd735363e5d8a1\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/a1852b64e4c628eaa7ecbff1ee36c6bad8dded24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/a1852b64e4c628eaa7ecbff1ee36c6bad8dded24"}]},"branch":"refs/heads/master"},"dc00932fb3689534d86014cdae0c653a31d81c33":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-05-20 11:00:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/987136/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/36/987136/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/7"}}},"commit":{"parents":[{"commit":"4e7ab543416bd215ea9ddd1dcb5055a8fd298249","subject":"Sync releasenotes tox env with previous zuul job command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/4e7ab543416bd215ea9ddd1dcb5055a8fd298249"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 10:28:57.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-20 11:00:10.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nChange-Id: Id09c9c6bff4dc77729dfaab2eabd735363e5d8a1\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/dc00932fb3689534d86014cdae0c653a31d81c33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/dc00932fb3689534d86014cdae0c653a31d81c33"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"a1852b64e4c628eaa7ecbff1ee36c6bad8dded24","theirs":"4e7ab543416bd215ea9ddd1dcb5055a8fd298249","contains_conflicts":false}},"391f5a142bf0d9e64197b643e7739b4ec11602a4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-08-10 13:31:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/987136/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/36/987136/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/8"}}},"commit":{"parents":[{"commit":"c1f5b0496b9f32e6431fe4ea96af70b8f7b52ab4","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/c1f5b0496b9f32e6431fe4ea96af70b8f7b52ab4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 10:28:57.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-10 13:31:55.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nChange-Id: Id09c9c6bff4dc77729dfaab2eabd735363e5d8a1\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/391f5a142bf0d9e64197b643e7739b4ec11602a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/391f5a142bf0d9e64197b643e7739b4ec11602a4"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"4e7ab543416bd215ea9ddd1dcb5055a8fd298249","ours":"dc00932fb3689534d86014cdae0c653a31d81c33","theirs":"c1f5b0496b9f32e6431fe4ea96af70b8f7b52ab4","merge_strategy":"recursive","contains_conflicts":false}},"231c93e37652534df81851e5f91a746a435d63e6":{"kind":"REWORK","_number":9,"created":"2026-08-10 13:32:17.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/987136/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/36/987136/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/9"}}},"commit":{"parents":[{"commit":"c1f5b0496b9f32e6431fe4ea96af70b8f7b52ab4","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/c1f5b0496b9f32e6431fe4ea96af70b8f7b52ab4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 10:28:57.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-10 13:32:17.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nChange-Id: Id09c9c6bff4dc77729dfaab2eabd735363e5d8a1\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/231c93e37652534df81851e5f91a746a435d63e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/231c93e37652534df81851e5f91a746a435d63e6"}]},"branch":"refs/heads/master"},"c15d1b83b78f26d2da1ec3b3f4b680727526bbe9":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-08-10 13:32:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/987136/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-ui","ref":"refs/changes/36/987136/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-ui refs/changes/36/987136/10"}}},"commit":{"parents":[{"commit":"c1f5b0496b9f32e6431fe4ea96af70b8f7b52ab4","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/c1f5b0496b9f32e6431fe4ea96af70b8f7b52ab4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-04 10:28:57.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-10 13:32:27.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nResolve setuptools deprecation warnings\naround license configuration.\n\nChange-Id: Id09c9c6bff4dc77729dfaab2eabd735363e5d8a1\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/c15d1b83b78f26d2da1ec3b3f4b680727526bbe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-ui/commit/c15d1b83b78f26d2da1ec3b3f4b680727526bbe9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
