)]}'
{"id":"openstack%2Fmanila~1007365","triplet_id":"openstack%2Fmanila~master~I346ede573ae8440cef1b1e58049229eefd0e8ccd","project":"openstack/manila","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I346ede573ae8440cef1b1e58049229eefd0e8ccd","subject":"Imported Translations from Zanata","status":"MERGED","created":"2026-09-25 03:55:44.000000000","updated":"2026-09-28 23:08:08.000000000","submitted":"2026-09-28 20:32:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007365-zanata/translations","meta_rev_id":"259536d895977b5c2a0450b88839cc09de4e8863","_number":1007365,"virtual_id_number":1007365,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-28 20:32:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2026-09-28 19:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2026-09-28 19:30:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":39277,"name":"ExaScaler CI","email":"alexeykhod0@gmail.com","username":"ExaScaler-CI"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-25 04:13:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-25 04:54:52.000000000","updated_by":{"_account_id":39277,"name":"ExaScaler CI","email":"alexeykhod0@gmail.com","username":"ExaScaler-CI"},"reviewer":{"_account_id":39277,"name":"ExaScaler CI","email":"alexeykhod0@gmail.com","username":"ExaScaler-CI"},"state":"CC"},{"updated":"2026-09-25 05:38:52.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2026-09-28 19:30:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"0ec8cb699c7531816635257cdabb11155f5bf127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-09-25 03:55:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5951e23fa2472cb5fd4df9f36aa8773527c91f96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 04:13:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53314b7b49284b7a8b3b4018a741b5b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee02fd3f696436487be869a09c2a682 : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dd1b2d99ef443189db0dbe6e6f90e09 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee93003ef3ab49dc94f7eb67c8914faa : SUCCESS in 10m 40s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2086dc340e904b238f3b873130435ebc : SUCCESS in 6m 14s\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/98b8a7ad68a84979a94c55a7ce836a9b : SUCCESS in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"510e32e0200f6e3abe3bf3c77ab26e92393332de","author":{"_account_id":39277,"name":"ExaScaler CI","email":"alexeykhod0@gmail.com","username":"ExaScaler-CI"},"date":"2026-09-25 04:54:52.000000000","message":"Patch Set 1:\n\nExaScaler Manila Driver https://d3krvozkty9bs0.cloudfront.net/manila_build_568/index.html : SUCCESS in 58m","accounts_in_message":[],"_revision_number":1},{"id":"d05a88354d734a5bffb1918da5a545b47f2bb8c4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2026-09-25 05:38:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/1007365/1/upstream-check-manila/manila-tempest-plugin-ontap-dhss/56d0250/ : SUCCESS in 1h 17m 58s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/1007365/1/upstream-check-manila/manila-tempest-plugin-ontap-no-dhss/812b3e5/ : FAILURE in 1h 24m 25s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"3ac1a2ce59baf63c1ab4c8c1761db4e4b6173d26","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-28 19:30:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d5823678425b52974f69e27518230f4d1d17521e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 19:31:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca983550aca4fee97fd751b19598fee9e51862e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 20:32:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7e47193ae7642239071755b56b626af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e28a943eb0e44c9f818d9a3357ffb415 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ff5af59308246aa855df28aa641d43b : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab50157856864379b8335f587ed2be2b : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0bc81ebd55588c094e90c3a20600f34a06389090","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 20:32:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d48573dc60b8f1fc4e74fb3cba9c54d54fa2dacd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 20:33:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22e7d60f35fa4f59ba135392a9d54eb5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/274b53fda11f445c84281dab7ff8e2b4 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/923c6926b1f34a91be0b9d9019f4ff31 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1},{"id":"259536d895977b5c2a0450b88839cc09de4e8863","author":{"_account_id":39277,"name":"ExaScaler CI","email":"alexeykhod0@gmail.com","username":"ExaScaler-CI"},"date":"2026-09-28 23:08:08.000000000","message":"Patch Set 1:\n\nExaScaler Manila Driver https://d3krvozkty9bs0.cloudfront.net/manila_build_578/index.html : FAILURE in 87m","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e19cdcb5c7b618c1ce7ab5cc25c06d2983858e48","revisions":{"e19cdcb5c7b618c1ce7ab5cc25c06d2983858e48":{"kind":"REWORK","_number":1,"created":"2026-09-25 03:55:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/1007365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/65/1007365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/65/1007365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/65/1007365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/65/1007365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/65/1007365/1"}}},"commit":{"parents":[{"commit":"32e719b8d1c433ef8671e8edc1272b96676010c1","subject":"Fix typos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/32e719b8d1c433ef8671e8edc1272b96676010c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-25 03:55:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-09-25 03:55:40.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I346ede573ae8440cef1b1e58049229eefd0e8ccd\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/openstack-zuul-jobs:roles/prepare-zanata-client/files/common_translation_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e19cdcb5c7b618c1ce7ab5cc25c06d2983858e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e19cdcb5c7b618c1ce7ab5cc25c06d2983858e48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
