)]}'
{"id":"openstack%2Fmanila~636770","triplet_id":"openstack%2Fmanila~master~Ie4785873e7ea7e3c025ad216483daed0cb35a443","project":"openstack/manila","branch":"master","topic":"vision-reflection","hashtags":[],"change_id":"Ie4785873e7ea7e3c025ad216483daed0cb35a443","subject":"Add a high level vision reflection document","status":"NEW","created":"2019-02-13 22:32:22.000000000","updated":"2020-11-20 13:19:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":30,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dc92d4416a9f40696b2c58629efb453da114caae","_number":636770,"virtual_id_number":636770,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"date":"2020-11-05 18:49:36.000000000","_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-22 21:44:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"date":"2020-11-17 21:02:57.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"date":"2020-09-22 22:10:10.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"date":"2020-09-22 23:11:21.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"date":"2020-09-22 22:32:13.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"date":"2020-09-23 01:12:51.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2020-11-17 12:28:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-05 15:11:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-13 22:46:27.000000000","updated_by":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"reviewer":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"state":"REVIEWER"},{"updated":"2019-02-13 23:43:49.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-02-14 00:35:24.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-14 01:18:47.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-14 02:41:43.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-02-14 22:29:32.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-09-22 21:44:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-22 22:10:10.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-22 22:32:13.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-22 23:11:21.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-23 01:12:51.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-11-05 15:11:18.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-11-05 18:49:36.000000000","updated_by":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"reviewer":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"state":"REVIEWER"},{"updated":"2020-11-17 12:28:43.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-11-17 21:02:57.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"}],"messages":[{"id":"07784a4babe7f6f0d6556503a31b0519ffc13b38","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-02-13 22:32:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3a0697fefd1227966fed18153afea3c1164b6b7","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-02-13 22:32:48.000000000","message":"Topic set to vision-reflection","accounts_in_message":[],"_revision_number":1},{"id":"024ecfcbcc1ab668e564eec8b1c22896698311a8","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-02-13 22:33:05.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-manila-driver http://52.42.67.99/70/636770/1/check/dsvm-veritas-access-manila-driver/f9b947e : FAILURE in 12s","accounts_in_message":[],"_revision_number":1},{"id":"60c5b9fde27ece13ea2f35f1b88c98cb6b479cdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-13 22:42:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/636770/1/check/openstack-tox-docs/45abf9f/html/ : SUCCESS in 5m 18s\n- openstack-tox-pep8 http://logs.openstack.org/70/636770/1/check/openstack-tox-pep8/2e29466/ : SUCCESS in 6m 32s\n- manila-tox-genconfig http://logs.openstack.org/70/636770/1/check/manila-tox-genconfig/7c8ed04/ : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c19f6934a6e39abcc3efa5cc479687a245e8e5d8","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-02-13 22:45:47.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f484959ae362f84517885cd3c7a579ca6ad11d9f","author":{"_account_id":24863,"name":"Veritas Access CI","email":"DL-VTAS-ENG-SDIO-Access-Opensource@veritas.com","username":"vtas-access-ci"},"date":"2019-02-13 22:46:27.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-veritas-access-manila-driver http://52.42.67.99/70/636770/2/check/dsvm-veritas-access-manila-driver/738d73b : FAILURE in 11s","accounts_in_message":[],"_revision_number":2},{"id":"fb430608e19f10cf44b61e9fd8b2cc316204e0b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-13 22:54:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/636770/2/check/openstack-tox-docs/bd6ab03/html/ : SUCCESS in 7m 33s\n- openstack-tox-pep8 http://logs.openstack.org/70/636770/2/check/openstack-tox-pep8/2e2584b/ : SUCCESS in 6m 12s\n- manila-tox-genconfig http://logs.openstack.org/70/636770/2/check/manila-tox-genconfig/78a2ea3/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"39dbfe0e1f31c605048f27d779ca6258825be4e6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-02-13 23:43:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/70/636770/2/check/dsvm-tempest-infinibox-nfs/1215924 : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c98bbd864a4206435d9df3a689e822e464f02895","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-02-14 00:34:41.000000000","message":"Patch Set 1:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-70-636770-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"adf72bff1b00edc88cae0bbbe45f31e7b7586e04","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-02-14 00:35:24.000000000","message":"Patch Set 2:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/70/636770/2/check/manila-master-client-pip-ssa03-dsvm-neutron/4386331 : FAILURE in 1h 37m 50s","accounts_in_message":[],"_revision_number":2},{"id":"2f1536028356d6e1a8b07381762ec1b7bc7d9160","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-02-14 01:18:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-as13000 http://218.29.111.78:9000/70/636770/2/check/manila-as13000/a729358 : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":2},{"id":"e78498fd532bad8beec1b6bdf68289ce956318fd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-02-14 02:22:48.000000000","message":"Patch Set 2:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-70-636770-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"0511a095937a0a0003fc25a5ad582c2543cb7725","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-02-14 02:41:43.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_Manila http://publiclogs.emc.com/636770/2/EMC_VMAX_Manila/430 : FAILURE in 2h 41m 27s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":2},{"id":"6cb141dc00ab836b37a459d305bd8c76a6d5b5d6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-02-14 05:52:19.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_Unity_Manila http://publiclogs.emc.com/70/636770/2/check/EMC_Unity_Manila/b5b3500/EMC_Unity_Manila/400 : FAILURE in 2h 27m 36s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"64aa89e6548ccdb10d1074c134a1db9357d4a0be","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-02-14 22:29:32.000000000","message":"Patch Set 2:\n\n(8 comments)\n\nThanks Tom, good start, I\u0027ll read some more.. but thoughts inline","accounts_in_message":[],"_revision_number":2},{"id":"4f3f70fbc54aa016911730a4b3246c4f871aa56b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-22 21:14:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0bff5cd6007ca1cb91d070a497f7d84122a54235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 21:44:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0316cccda1de439ab472f6c03288a714 : SUCCESS in 20m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7a61372cac344c383905cc17da941c5 : SUCCESS in 5m 14s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/36fe2c515756439ea9f94dd45aaf5ae9 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":3},{"id":"7b90366f175afa456f2045003352bd8c51880bd8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-22 22:10:10.000000000","message":"Patch Set 3:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-70-636770-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"7adea7d2e69ce1690cd5182dfd8988ff60139b07","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-22 22:32:13.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-70-636770-3/ns5_manila : SUCCESS in 1h 17m 9s","accounts_in_message":[],"_revision_number":3},{"id":"b32b162966c57637d58a199ce7986b032405f7d4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-22 23:11:21.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/70/636770/3/check/DellEMC_VNX_Manila/4ec5db7/DellEMC_VNX_Manila/265 : SUCCESS in 1h 13m 08s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"3cb5a5b626c1c7e00098ce9b9e94fbe6223a04f0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-23 01:12:51.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/70/636770/3/check/DellEMC_Unity_Manila_dhss_true/8688e70/DellEMC_Unity_Manila_dhss_true/482 : FAILURE in 54m 17s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/70/636770/3/check/DellEMC_Unity_Manila_dhss_false/9f2b606/DellEMC_Unity_Manila_dhss_false/465 : FAILURE in 47m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"15614a032106d5f963bf3b3df0cac9651b992506","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2020-11-05 18:49:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2dc66c63b8855ba8083399e9ddb6000001d15769","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-11-17 12:28:43.000000000","message":"Patch Set 3:\n\nThanks for working on this, Tom :)\nI have no comments to add... I just agree and would like to enforce the points added by Goutham and Rodrigo.","accounts_in_message":[],"_revision_number":3},{"id":"fc84a2881afdf46c86f726b92b3f2441a1af9818","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-11-17 21:02:57.000000000","message":"Patch Set 3:\n\n(14 comments)\n\nAre we rebasing and merging this work? :)\n\nThanks Tom","accounts_in_message":[],"_revision_number":3},{"id":"bbe6b92e8c370c1727c9360217f88880ee0266c6","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-11-20 13:19:16.000000000","message":"Patch Set 2:\n\n(7 comments)\n\nGoutham, thanks for this review and though I\u0027m sorry for not following up sooner, with the passage of time it is clear that this draft was and perhaps remains murky on lots of important points, including those that you emphasized here.\n\nLooks like we have more interest as a community in working on this reflection document and finally may be able to complete it.  Thank you for moving the agenda :D","accounts_in_message":[],"_revision_number":2}],"current_revision_number":3,"current_revision":"51a0500961e72c136c388ba0024d9bf6082a28d5","revisions":{"01935b13604e8002884b898649218d8c0ed2b220":{"kind":"REWORK","_number":1,"created":"2019-02-13 22:32:22.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/70/636770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/70/636770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/636770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/636770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/636770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/70/636770/1"}}},"commit":{"parents":[{"commit":"1471376131859fcafeb3e88c2058549c5fc24fad","subject":"Bump timeout on dsvm jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1471376131859fcafeb3e88c2058549c5fc24fad"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-02-08 16:58:01.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-02-13 22:32:15.000000000","tz":-300},"subject":"Add a high level vision relection document","message":"Add a high level vision relection document\n\nThe TC has asked that each project add a document reflecting\non gaps relative to the recently approved ``Vision for\nOpenStack Clouds`` document [1].\n\nThis review starts the work in manila for such a reflection\ndocument.\n\n[1] https://governance.openstack.org/tc/reference/technical-vision.html\n\nChange-Id: Ie4785873e7ea7e3c025ad216483daed0cb35a443\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/01935b13604e8002884b898649218d8c0ed2b220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/01935b13604e8002884b898649218d8c0ed2b220"}]},"branch":"refs/heads/master"},"9321befc439bad498c8a2e8e70bbe7e3e468542a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-02-13 22:45:47.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/70/636770/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/70/636770/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/636770/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/636770/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/636770/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/70/636770/2"}}},"commit":{"parents":[{"commit":"1471376131859fcafeb3e88c2058549c5fc24fad","subject":"Bump timeout on dsvm jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1471376131859fcafeb3e88c2058549c5fc24fad"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-02-08 16:58:01.000000000","tz":-300},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-02-13 22:45:37.000000000","tz":0},"subject":"Add a high level vision reflection document","message":"Add a high level vision reflection document\n\nThe TC has asked that each project add a document reflecting\non gaps relative to the recently approved ``Vision for\nOpenStack Clouds`` document [1].\n\nThis review starts the work in manila for such a reflection\ndocument.\n\n[1] https://governance.openstack.org/tc/reference/technical-vision.html\n\nChange-Id: Ie4785873e7ea7e3c025ad216483daed0cb35a443\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9321befc439bad498c8a2e8e70bbe7e3e468542a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9321befc439bad498c8a2e8e70bbe7e3e468542a"}]},"branch":"refs/heads/master"},"51a0500961e72c136c388ba0024d9bf6082a28d5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-22 21:14:41.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/70/636770/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/70/636770/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/636770/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/636770/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/636770/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/70/636770/3"}}},"commit":{"parents":[{"commit":"6a05a2d92213bb17d830c4b8595a4b00ce558a95","subject":"Merge \"Update api-ref for share groups graduation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6a05a2d92213bb17d830c4b8595a4b00ce558a95"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-02-08 16:58:01.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-09-22 21:14:41.000000000","tz":0},"subject":"Add a high level vision reflection document","message":"Add a high level vision reflection document\n\nThe TC has asked that each project add a document reflecting\non gaps relative to the recently approved ``Vision for\nOpenStack Clouds`` document [1].\n\nThis review starts the work in manila for such a reflection\ndocument.\n\n[1] https://governance.openstack.org/tc/reference/technical-vision.html\n\nChange-Id: Ie4785873e7ea7e3c025ad216483daed0cb35a443\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/51a0500961e72c136c388ba0024d9bf6082a28d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/51a0500961e72c136c388ba0024d9bf6082a28d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
