)]}'
{"id":"openstack%2Fmanila~645010","triplet_id":"openstack%2Fmanila~master~Ia85736693d7f536766e0e0980f88805c61aa8325","project":"openstack/manila","branch":"master","topic":"bug/1796986","hashtags":[],"change_id":"Ia85736693d7f536766e0e0980f88805c61aa8325","subject":"Adding documentation for User Messages in Manila Documentation","status":"MERGED","created":"2019-03-20 21:14:05.000000000","updated":"2019-06-27 01:11:34.000000000","submitted":"2019-06-26 12:33:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":40,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"645010-1561552396592-0124e28e","meta_rev_id":"bf9fcec674ebf895b3a826e21580acf7a9e41c51","_number":645010,"virtual_id_number":645010,"owner":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},{"value":0,"date":"2019-06-24 21:10:18.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"date":"2019-06-26 00:16:25.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":2,"date":"2019-06-26 12:33:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-26 13:03:52.000000000","post_submit":true,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},{"value":0,"date":"2019-06-26 02:34:19.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-26 01:13:26.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-26 00:17:09.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2417,"name":"Ben Swartzlander","email":"ben@swartzlander.org","username":"bswartz"},{"value":0,"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},{"value":0,"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},{"value":0,"date":"2019-06-25 22:43:19.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":15100,"name":"zhongjun","email":"jun.zhongjun2@gmail.com","username":"zhongjun2"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2019-06-25 22:10:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":2,"date":"2019-06-26 11:02:55.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":2417,"name":"Ben Swartzlander","email":"ben@swartzlander.org","username":"bswartz"},{"value":0,"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},{"value":0,"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":15100,"name":"zhongjun","email":"jun.zhongjun2@gmail.com","username":"zhongjun2"},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2019-06-26 00:23:06.000000000","_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"value":1,"date":"2019-06-26 11:02:55.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-26 00:23:06.000000000","_account_id":2417,"name":"Ben Swartzlander","email":"ben@swartzlander.org","username":"bswartz"},{"value":0,"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},{"value":0,"date":"2019-06-26 00:23:06.000000000","_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"date":"2019-06-26 00:23:06.000000000","_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"date":"2019-06-26 00:23:06.000000000","_account_id":15100,"name":"zhongjun","email":"jun.zhongjun2@gmail.com","username":"zhongjun2"},{"value":0,"date":"2019-06-26 00:23:06.000000000","_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2417,"name":"Ben Swartzlander","email":"ben@swartzlander.org","username":"bswartz"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":15100,"name":"zhongjun","email":"jun.zhongjun2@gmail.com","username":"zhongjun2"},{"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 20:39:27.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2019-06-12 19:39:21.000000000","updated_by":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"reviewer":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"state":"REVIEWER"},{"updated":"2019-06-22 13:01:31.000000000","updated_by":{"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},"reviewer":{"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},"state":"REVIEWER"},{"updated":"2019-06-22 13:05:23.000000000","updated_by":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"reviewer":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"state":"REVIEWER"},{"updated":"2019-06-24 15:41:36.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2019-06-24 21:10:18.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-06-25 22:10:41.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-06-25 22:43:19.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2019-06-26 00:16:25.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-26 00:17:09.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-26 00:23:06.000000000","updated_by":{"_account_id":2417,"name":"Ben Swartzlander","email":"ben@swartzlander.org","username":"bswartz"},"reviewer":{"_account_id":2417,"name":"Ben Swartzlander","email":"ben@swartzlander.org","username":"bswartz"},"state":"REVIEWER"},{"updated":"2019-06-26 00:23:06.000000000","updated_by":{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},"reviewer":{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},"state":"REVIEWER"},{"updated":"2019-06-26 00:23:06.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2019-06-26 00:23:06.000000000","updated_by":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"reviewer":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"state":"REVIEWER"},{"updated":"2019-06-26 00:23:06.000000000","updated_by":{"_account_id":15100,"name":"zhongjun","email":"jun.zhongjun2@gmail.com","username":"zhongjun2"},"reviewer":{"_account_id":15100,"name":"zhongjun","email":"jun.zhongjun2@gmail.com","username":"zhongjun2"},"state":"REVIEWER"},{"updated":"2019-06-26 00:23:06.000000000","updated_by":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"reviewer":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"state":"REVIEWER"},{"updated":"2019-06-26 01:13:26.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-26 02:34:19.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-26 11:02:55.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2019-06-26 12:33:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-26 13:03:52.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ada56aed078beff022f2c1d8678ce7e537cf44ef","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-20 21:14:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85a7b0014edf8f5d7e00791a69b43a03d148b446","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-20 21:19:35.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"af9076870bb8a6713119dc9e504690f1e4a73faf","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-20 21:19:50.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"43d9cd1401b6a9d2da6e10ec863cb0fe18d090b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-20 21:32:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/2/check/openstack-tox-docs/569bef6/ : FAILURE in 7m 20s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/2/check/openstack-tox-pep8/d434df7/ : SUCCESS in 7m 26s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/2/check/manila-tox-genconfig/e6c3af1/ : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"5f1cb15fe1b9fd4f46eb7158ac2696356900c479","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-03-20 22:20:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/2/check/dsvm-tempest-infinibox-nfs/7e97d6d : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":2},{"id":"927dddf1fa4949e415a198fe9937e3b046efd4a1","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-21 03:15:52.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"fde5723d99f02b90bb52da97ac03ec2cfe24ec9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 03:25:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/3/check/openstack-tox-docs/2fe405d/ : FAILURE in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/3/check/openstack-tox-pep8/4b55d4f/ : SUCCESS in 7m 30s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/3/check/manila-tox-genconfig/4fe0ae0/ : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":3},{"id":"02abbe97d00e6b2ec0fe98a446eddf0452105a01","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-03-21 04:13:16.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/3/check/dsvm-tempest-infinibox-nfs/0376214 : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":3},{"id":"d8e82a2ab1d4a834299d219396ee5744b81a52a4","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-03-21 14:21:18.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- manila-as13000 http://218.29.111.78:9000/10/645010/3/check/manila-as13000/c56d9a4 : FAILURE in 1h 42m 45s\n- manila-instorage http://218.29.111.78:9000/10/645010/3/check/manila-instorage/e0edbea : SUCCESS in 3h 09m 37s","accounts_in_message":[],"_revision_number":3},{"id":"c48139c58d685852b27258b8f7151fe804e9c5ef","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-21 15:16:55.000000000","message":"Patch Set 3: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"d4e3369446c57fe33b9ab54f813584996bffc35c","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-21 16:01:53.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"4fda7bb34ddb7bdb75c45a9de2157f6716ea9f99","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-21 16:02:20.000000000","message":"Patch Set 4: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"c678234b656447008056d7663a4c4b1c05144a57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 16:25:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/4/check/openstack-tox-docs/9114667/ : FAILURE in 5m 58s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/4/check/openstack-tox-pep8/4553ce9/ : SUCCESS in 7m 20s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/4/check/manila-tox-genconfig/049d17e/ : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":4},{"id":"c55e1921521fe93ae0b34891d0d903ff9646c62b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-03-21 17:52:42.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/4/check/dsvm-tempest-infinibox-nfs/7d3e95f : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":4},{"id":"544f0d874c0a2fdf261d01fb89283b244d4b64f6","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-21 20:52:54.000000000","message":"Abandoned\n\nI\u0027m going to upload it again","accounts_in_message":[],"_revision_number":4},{"id":"e17b9daca954dbe011113a31019f819e22874073","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-22 03:24:22.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"3e1817963c822233f800408628dfcbdae6574b78","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-22 03:32:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5d36466ccfea06102bf0a0c7f46fa10adc45a897","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-03-22 04:29:38.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/5/check/dsvm-tempest-infinibox-nfs/dee633e : SUCCESS in 57m 03s","accounts_in_message":[],"_revision_number":5},{"id":"2fb430f8da27247de6ea15444560679aaec4c058","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 05:16:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/5/check/openstack-tox-docs/d050859/ : FAILURE in 6m 23s\n- openstack-tox-cover http://logs.openstack.org/10/645010/5/check/openstack-tox-cover/12f0e69/cover/ : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/10/645010/5/check/openstack-tox-lower-constraints/3f064d9/ : SUCCESS in 7m 30s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/5/check/openstack-tox-pep8/62967f6/ : SUCCESS in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/10/645010/5/check/openstack-tox-py27/a76e647/ : SUCCESS in 7m 11s\n- openstack-tox-py36 http://logs.openstack.org/10/645010/5/check/openstack-tox-py36/6c0744b/ : SUCCESS in 7m 08s\n- openstack-tox-py37 http://logs.openstack.org/10/645010/5/check/openstack-tox-py37/0af738b/ : SUCCESS in 10m 45s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/5/check/manila-tox-genconfig/d7a26fa/ : SUCCESS in 5m 35s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/10/645010/5/check/manila-tempest-dsvm-mysql-generic/483a378/ : SUCCESS in 1h 41m 38s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/10/645010/5/check/manila-tempest-dsvm-postgres-container/2825a3b/ : SUCCESS in 45m 29s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/10/645010/5/check/manila-tempest-dsvm-postgres-zfsonlinux/270e1de/ : FAILURE in 1h 03m 58s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/10/645010/5/check/manila-tempest-dsvm-postgres-generic-singlebackend/fc2ad86/ : SUCCESS in 1h 01m 11s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/10/645010/5/check/manila-tempest-dsvm-generic-no-share-servers/34e1387/ : SUCCESS in 1h 01m 53s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/10/645010/5/check/manila-tempest-dsvm-scenario/609693c/ : SUCCESS in 1h 21m 19s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/10/645010/5/check/manila-tempest-minimal-dsvm-cephfs-native/b9f6ed9/ : SUCCESS in 23m 37s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/10/645010/5/check/manila-tempest-minimal-dsvm-cephfs-nfs/7c4b114/ : SUCCESS in 30m 16s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/10/645010/5/check/manila-tempest-minimal-dsvm-dummy/0480077/ : SUCCESS in 38m 09s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/10/645010/5/check/manila-tempest-minimal-dsvm-lvm/51e1dd3/ : SUCCESS in 1h 07m 39s\n- manila-grenade http://logs.openstack.org/10/645010/5/check/manila-grenade/f1c209f/ : SUCCESS in 22m 25s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/10/645010/5/check/manila-rally-multibackend/8523d4f/ : SUCCESS in 19m 22s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/10/645010/5/check/manila-rally-multibackend-no-ss/2778452/ : SUCCESS in 18m 17s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/10/645010/5/check/openstack-tox-pylint/e67af05/ : SUCCESS in 5m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3beb912833f2e9085a217a2177bfe5112eaf7577","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-22 15:55:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0dcfccfb830f915233354696edf08523e679fd12","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-03-22 16:48:12.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/6/check/dsvm-tempest-infinibox-nfs/f146fee : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":6},{"id":"31735dca594ca0b502b6f47143002ad1c03b45d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 17:40:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/6/check/openstack-tox-docs/9b72054/ : FAILURE in 5m 25s\n- openstack-tox-cover http://logs.openstack.org/10/645010/6/check/openstack-tox-cover/0622ed3/cover/ : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/10/645010/6/check/openstack-tox-lower-constraints/35705e3/ : SUCCESS in 8m 28s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/6/check/openstack-tox-pep8/cf00925/ : SUCCESS in 6m 40s\n- openstack-tox-py27 http://logs.openstack.org/10/645010/6/check/openstack-tox-py27/f9f5c96/ : SUCCESS in 7m 05s\n- openstack-tox-py36 http://logs.openstack.org/10/645010/6/check/openstack-tox-py36/e53f46f/ : SUCCESS in 8m 27s\n- openstack-tox-py37 http://logs.openstack.org/10/645010/6/check/openstack-tox-py37/95961bd/ : SUCCESS in 13m 05s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/6/check/manila-tox-genconfig/7b663fb/ : SUCCESS in 7m 02s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/10/645010/6/check/manila-tempest-dsvm-mysql-generic/00e07a8/ : FAILURE in 1h 25m 28s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/10/645010/6/check/manila-tempest-dsvm-postgres-container/5407776/ : SUCCESS in 1h 02m 08s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/10/645010/6/check/manila-tempest-dsvm-postgres-zfsonlinux/6ae9a88/ : SUCCESS in 1h 13m 07s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/10/645010/6/check/manila-tempest-dsvm-postgres-generic-singlebackend/258b445/ : SUCCESS in 1h 05m 57s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/10/645010/6/check/manila-tempest-dsvm-generic-no-share-servers/4b8ac0a/ : SUCCESS in 55m 53s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/10/645010/6/check/manila-tempest-dsvm-scenario/7d34bd1/ : SUCCESS in 1h 37m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/10/645010/6/check/manila-tempest-minimal-dsvm-cephfs-native/4b52486/ : SUCCESS in 32m 03s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/10/645010/6/check/manila-tempest-minimal-dsvm-cephfs-nfs/ed46187/ : SUCCESS in 23m 20s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/10/645010/6/check/manila-tempest-minimal-dsvm-dummy/f91c08c/ : SUCCESS in 45m 36s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/10/645010/6/check/manila-tempest-minimal-dsvm-lvm/a4dd140/ : SUCCESS in 1h 24m 04s\n- manila-grenade http://logs.openstack.org/10/645010/6/check/manila-grenade/5c46f4f/ : POST_FAILURE in 23m 11s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/10/645010/6/check/manila-rally-multibackend/73d6f13/ : SUCCESS in 27m 30s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/10/645010/6/check/manila-rally-multibackend-no-ss/b1d90a0/ : SUCCESS in 25m 25s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/10/645010/6/check/openstack-tox-pylint/6ab8133/ : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5d62a4f0ba86f8b7b8ca7397738b0e67e9e73a00","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-22 18:20:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"13d81f0cac47888f863eb3f4e6f89454696796ea","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-03-22 19:26:48.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/7/check/dsvm-tempest-infinibox-nfs/79b6bdc : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":7},{"id":"9703ef96e7db4ca66fb24d6a77ec04f7213b6774","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-22 19:59:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/7/check/openstack-tox-docs/34caaa9/html/ : SUCCESS in 6m 09s\n- openstack-tox-cover http://logs.openstack.org/10/645010/7/check/openstack-tox-cover/89c7d76/cover/ : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/10/645010/7/check/openstack-tox-lower-constraints/7911ba1/ : SUCCESS in 6m 48s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/7/check/openstack-tox-pep8/84af8c8/ : SUCCESS in 6m 47s\n- openstack-tox-py27 http://logs.openstack.org/10/645010/7/check/openstack-tox-py27/32f2301/ : SUCCESS in 8m 49s\n- openstack-tox-py36 http://logs.openstack.org/10/645010/7/check/openstack-tox-py36/1d91534/ : SUCCESS in 8m 00s\n- openstack-tox-py37 http://logs.openstack.org/10/645010/7/check/openstack-tox-py37/2c9221e/ : SUCCESS in 8m 30s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/7/check/manila-tox-genconfig/724cfbf/ : SUCCESS in 6m 20s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/10/645010/7/check/manila-tempest-dsvm-mysql-generic/66807aa/ : SUCCESS in 1h 31m 38s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/10/645010/7/check/manila-tempest-dsvm-postgres-container/257614e/ : SUCCESS in 1h 05m 58s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/10/645010/7/check/manila-tempest-dsvm-postgres-zfsonlinux/a36b4fe/ : FAILURE in 1h 06m 46s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/10/645010/7/check/manila-tempest-dsvm-postgres-generic-singlebackend/2de3b6d/ : SUCCESS in 56m 34s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/10/645010/7/check/manila-tempest-dsvm-generic-no-share-servers/a05a09e/ : SUCCESS in 1h 12m 55s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/10/645010/7/check/manila-tempest-dsvm-scenario/1f7840f/ : FAILURE in 1h 24m 29s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/10/645010/7/check/manila-tempest-minimal-dsvm-cephfs-native/4fc766c/ : SUCCESS in 26m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/10/645010/7/check/manila-tempest-minimal-dsvm-cephfs-nfs/9e8aef6/ : SUCCESS in 29m 34s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/10/645010/7/check/manila-tempest-minimal-dsvm-dummy/c6f3a18/ : SUCCESS in 44m 46s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/10/645010/7/check/manila-tempest-minimal-dsvm-lvm/d6027cb/ : SUCCESS in 1h 24m 14s\n- manila-grenade http://logs.openstack.org/10/645010/7/check/manila-grenade/c82c0a0/ : SUCCESS in 27m 59s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/10/645010/7/check/manila-rally-multibackend/f7e9f92/ : SUCCESS in 25m 17s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/10/645010/7/check/manila-rally-multibackend-no-ss/d39f894/ : SUCCESS in 32m 46s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/10/645010/7/check/openstack-tox-pylint/4bd2a2f/ : SUCCESS in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6323d6b8ba7fff27333268a1d181813857378825","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-03-22 20:14:37.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VNX_Manila http://publiclogs.emc.com/10/645010/7/check/EMC_VNX_Manila/a45151f/EMC_VNX_Manila/238 : FAILURE in 14m 57s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-vnx-ci","accounts_in_message":[],"_revision_number":7},{"id":"e44303073d756cbed9f62ffe04b6c732063464ee","author":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"date":"2019-03-22 21:39:17.000000000","message":"Patch Set 7:\n\n* nexenta-manila_ns5_nfs http://140.174.232.106/refs-changes-10-645010-7/manila_ns5_nfs : SUCCESS in 70m 20s","accounts_in_message":[],"_revision_number":7},{"id":"265890f113d103aefba9525e8102167dd97178c3","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-03-22 23:00:52.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_Manila http://publiclogs.emc.com/645010/7/EMC_VMAX_Manila/587 : FAILURE in 1h 29m 16s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":7},{"id":"a31670fe7b733bde4669c10dea19a80e4a75c443","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-22 23:10:10.000000000","message":"Patch Set 7:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-10-645010-7 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":7},{"id":"24a8be8cd0e05ce0b70e6dabe35b0214f685947a","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-25 01:16:44.000000000","message":"Patch Set 7: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"64339e737b2def2e66da0b6616e4b7d1d1e3bb99","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-25 16:01:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d3647ddad7e273f78a7d5330ed60b380b306ee8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 16:16:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/8/check/openstack-tox-docs/54f8f86/ : FAILURE in 5m 43s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/8/check/openstack-tox-pep8/402d15f/ : SUCCESS in 6m 42s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/8/check/manila-tox-genconfig/a2a7d1a/ : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":8},{"id":"8619729930e329e1b1454a6eea17ee1727ccf402","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-25 16:27:51.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ca872147fbe0af86d79776b9502a20d0f805e74c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 16:58:55.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/9/check/openstack-tox-docs/a3a7ef6/ : FAILURE in 6m 38s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/9/check/openstack-tox-pep8/985bcf8/ : SUCCESS in 7m 38s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/9/check/manila-tox-genconfig/0c8e344/ : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":9},{"id":"43a103223c99e9debee6f8cd6f31a380cbb811c8","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-25 17:26:26.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"610c210c14f19a71ad4ea57e32d035efb3842732","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-25 17:27:11.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f16b2914d5c248f724f5d6bba61855231a107e3e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-03-25 17:35:31.000000000","message":"Patch Set 10: Code-Review-1\n\n(4 comments)\n\nThanks for adding this Soledad. Please see comments inline.","accounts_in_message":[],"_revision_number":10},{"id":"a6aab73a039fbd2388c93505927be518a86a0d9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 17:40:23.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/10/check/openstack-tox-docs/b65e218/ : FAILURE in 5m 29s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/10/check/openstack-tox-pep8/f8c68c6/ : SUCCESS in 7m 33s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/10/check/manila-tox-genconfig/5d43f7b/ : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":10},{"id":"2cc913e6c92de9a16c9ab5dc20aca9129bb34a71","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-25 18:11:51.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"4e21e2cd9c26e12778a8353ce4709d6653d325c1","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-25 18:14:12.000000000","message":"Patch Set 11: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"3a7bc687ba05db4f75d8f4d41568526bb157775d","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-25 18:20:36.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"732612854c8d18fbc84347943d71e0b9fc307565","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 18:36:54.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/12/check/openstack-tox-docs/1ce0b43/ : FAILURE in 5m 44s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/12/check/openstack-tox-pep8/69d9789/ : SUCCESS in 6m 37s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/12/check/manila-tox-genconfig/5badfa7/ : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":12},{"id":"974f70a4d65fa3809d2046aa50518b209152f61f","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-25 18:40:16.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d83448e2e8c85a248a33366a6166c805203d7b14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-25 18:53:36.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/13/check/openstack-tox-docs/90806a5/html/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/13/check/openstack-tox-pep8/c879390/ : SUCCESS in 6m 46s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/13/check/manila-tox-genconfig/54c0265/ : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":13},{"id":"39d6d68b46430768c352af00e62be596c1414557","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-03-25 19:03:15.000000000","message":"Patch Set 13:\n\nIs it ok with the removal of the ID example that isn\u0027t used? Or is it better add the new share created too?","accounts_in_message":[],"_revision_number":13},{"id":"7ba751325466bdfaad9c8efef7b21d646da3e3ef","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-03-25 19:41:16.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/13/check/dsvm-tempest-infinibox-nfs/e8dbde2 : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":13},{"id":"cf8691e98861204af7f1b0141818cc30cba7ef85","author":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"date":"2019-03-25 20:05:54.000000000","message":"Patch Set 13:\n\n* nexenta-manila_ns5_nfs http://140.174.232.106/refs-changes-10-645010-13/manila_ns5_nfs : SUCCESS in 71m 36s","accounts_in_message":[],"_revision_number":13},{"id":"fdef110740d62c89081a1ea4af024b0c8338f3a8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-25 20:26:11.000000000","message":"Patch Set 13:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-10-645010-13 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":13},{"id":"f19796111fce55041d7dba936de0a1f0fac96b0e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-03-25 20:29:12.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_Manila http://publiclogs.emc.com/645010/13/EMC_VMAX_Manila/596 : FAILURE in 1h 35m 26s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":13},{"id":"93d0e3a678826eee22ce6749e10b5172c71152c0","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-03-25 20:42:34.000000000","message":"Patch Set 13:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/10/645010/13/check/manila-master-client-pip-ssa03-dsvm-neutron/bc2c16d : FAILURE in 1h 46m 27s","accounts_in_message":[],"_revision_number":13},{"id":"3f3998b266ce301e47543f2d3ac36be4c3f8a804","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-03-25 20:53:49.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- EMC_VNX_Manila http://publiclogs.emc.com/10/645010/13/check/EMC_VNX_Manila/7e5d206/EMC_VNX_Manila/247 : SUCCESS in 2h 00m 02s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-vnx-ci","accounts_in_message":[],"_revision_number":13},{"id":"abee0abadd124070a861785e104c6eac0165e3a4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-03-25 21:00:09.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- EMC_Unity_Manila http://publiclogs.emc.com/10/645010/13/check/EMC_Unity_Manila/e74adf5/EMC_Unity_Manila/506 : SUCCESS in 2h 19m 43s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"41d0963186cf21a23df7da6eb4b4bf7dd6294c47","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-03-25 21:58:39.000000000","message":"Patch Set 13:\n\n\u003e Is it ok with the removal of the ID example that isn\u0027t used? Or is it better add the new share created too?\n\n\nI would suggest the whole example, Soledad... It makes sense to walk through the whole scenario.","accounts_in_message":[],"_revision_number":13},{"id":"df0dabb88dfd604c075593b483a092b9e7c384f6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-03-25 22:00:39.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"01da44777999a4792263e7148afdb4c17a3aabee","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-04-09 21:26:42.000000000","message":"Patch Set 13:\n\nSoledad: Do you need any help updating this patch?","accounts_in_message":[],"_revision_number":13},{"id":"6cac07e2ed473de076549103f60899863b2229bb","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-04-11 22:53:56.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"11520f3974fb18b9f9a8e4a7cdbd9fedaf19dfdf","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-04-11 22:54:33.000000000","message":"Patch Set 14: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":14},{"id":"11871d24a1289ac4e97c2f79618ff092b2343684","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-04-11 22:55:04.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_Manila http://publiclogs.emc.com/645010/14/EMC_VMAX_Manila/641 : FAILURE in 52s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":14},{"id":"a0d207abddea609656e175ec60a487378ac106a0","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-04-11 23:03:34.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"273e209a8c8ffee80611cc5182df2c301aaa655b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-04-11 23:04:10.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_Manila http://publiclogs.emc.com/645010/15/EMC_VMAX_Manila/642 : FAILURE in 27s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":15},{"id":"f470fa19f0ac43edbb09327651e2249934db545d","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-04-11 23:10:13.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"20d79ccccd7dd55b2618864555c1cb666799ac21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 23:22:53.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/15/check/openstack-tox-docs/0c4c110/ : FAILURE in 6m 44s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/15/check/openstack-tox-pep8/e68aa6f/ : SUCCESS in 5m 54s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/15/check/manila-tox-genconfig/70a552d/ : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":15},{"id":"d2081875de5416e018f121cf6831606804be0b15","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-04-12 00:03:07.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/15/check/dsvm-tempest-infinibox-nfs/f255732 : SUCCESS in 59m 08s","accounts_in_message":[],"_revision_number":15},{"id":"5311889ff121dd1ad1af442a11eb2af48644831d","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-04-12 17:23:26.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"6207b80211c45c1ed95f38261db3850446fbc173","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 17:24:02.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_Manila http://publiclogs.emc.com/645010/16/EMC_VMAX_Manila/643 : FAILURE in 26s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":16},{"id":"6d62e50feca71f441cf8de0f5f3cc7762c50db48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 17:40:55.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/16/check/openstack-tox-docs/8a0a3b4/html/ : SUCCESS in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/16/check/openstack-tox-pep8/60f94b3/ : SUCCESS in 6m 36s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/16/check/manila-tox-genconfig/5d4925c/ : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":16},{"id":"ef553d7c8d1eb3030dc1f51b137d97e388f78f86","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 17:41:34.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VMAX_Manila http://publiclogs.emc.com/645010/16/EMC_VMAX_Manila/644 : FAILURE in 30s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027","accounts_in_message":[],"_revision_number":16},{"id":"073184cd808d07a90c31410403af9ed9d14fbf3d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 17:42:15.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VNX_Manila http://publiclogs.emc.com/10/645010/16/check/EMC_VNX_Manila/b05ecfb/EMC_VNX_Manila/286 : FAILURE in 1m 09s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/emc-vnx-ci","accounts_in_message":[],"_revision_number":16},{"id":"8d8b78a7a8cd23d81c56191dd4abb605ab54264c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-04-12 18:18:26.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/16/check/dsvm-tempest-infinibox-nfs/098a1df : SUCCESS in 54m 33s","accounts_in_message":[],"_revision_number":16},{"id":"3181369c11c9fc31e0545773d541ee4a9868bfb9","author":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"date":"2019-04-12 18:53:58.000000000","message":"Patch Set 16:\n\n* nexenta-manila_ns5_nfs http://140.174.232.106/refs-changes-10-645010-16/manila_ns5_nfs : SUCCESS in 72m 16s","accounts_in_message":[],"_revision_number":16},{"id":"40e9bb9ec772fe9eb148639d3d8e0927318cbbaf","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-04-12 19:17:09.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/10/645010/16/check/manila-master-client-pip-ssa03-dsvm-neutron/8c0b1cb : SUCCESS in 1h 33m 35s","accounts_in_message":[],"_revision_number":16},{"id":"0579398f4ac018faac7860f4e567d1b131ef5ee2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-12 19:37:35.000000000","message":"Patch Set 16:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-10-645010-16 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":16},{"id":"4b0e24a28af433c588e7198750900fe671d547e0","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-04-15 19:33:18.000000000","message":"Patch Set 16:\n\nthe two failures of EMC, I don\u0027t know if they have something to do with me...","accounts_in_message":[],"_revision_number":16},{"id":"3fcbe7fab68e5350888f34fdbe84a6af118b19e3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-04-15 20:21:03.000000000","message":"Patch Set 16:\n\n\u003e the two failures of EMC, I don\u0027t know if they have something to do with me...\n\n\nYeah, you can ignore third party CI failures on docs-only patches. Those CIs seem to be having some infrastructure/scripting issues.","accounts_in_message":[],"_revision_number":16},{"id":"a1c8f0810f64475c90c349ed455acdb6f5442177","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-04-15 20:30:53.000000000","message":"Patch Set 16: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":16},{"id":"0f9bdb6d2ce2ad6fdaa708d8ef22677dc80e919f","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-04-17 00:57:48.000000000","message":"Patch Set 16: Code-Review-1 Workflow-1\n\nI was uploading my changes to use the automated tests from Gerrit because I couldn\u0027t do them locally but now I\u0027m trying to see if I can use the local tests. Because since I\u0027m not really understanding what has to be written I think I need at least to know that it is well formatted. I will complete it better considering what you told me; I think in the summary I tried to do I left out too much.","accounts_in_message":[],"_revision_number":16},{"id":"1539ea86bbe247b2a5385ae23c1557ec34fadb0b","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-04-20 20:32:21.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"261a91c8a8bddb224e4fac1ef46c446f2bf01a0f","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-04-20 20:34:29.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VMAX_Manila http://publiclogs.emc.com/10/645010/17/check/DellEMC_VMAX_Manila/2e3ff10/DellEMC_VMAX_Manila/2 : FAILURE in 1m 54s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VMAX_CI","accounts_in_message":[],"_revision_number":17},{"id":"f136804e138cbea4eca5b3b51b8e92a57975261e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-04-20 20:41:49.000000000","message":"Patch Set 17:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/17/check/dsvm-tempest-infinibox-nfs/b32949b : FAILURE in 8m 10s","accounts_in_message":[],"_revision_number":17},{"id":"7f1ede63cf0fc88519f5c6ca8e7ca0620c7ecc06","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-04-20 22:09:19.000000000","message":"Patch Set 17:\n\nI uploaded the entire scenario","accounts_in_message":[],"_revision_number":17},{"id":"466ac746d093dcb50a2cf55e72a10de13e0570e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 22:11:09.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/17/check/openstack-tox-docs/9726a67/html/ : SUCCESS in 5m 17s\n- openstack-tox-cover http://logs.openstack.org/10/645010/17/check/openstack-tox-cover/dad6a0c/cover/ : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/10/645010/17/check/openstack-tox-lower-constraints/dd9eb39/ : SUCCESS in 6m 39s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/17/check/openstack-tox-pep8/6d2eae0/ : FAILURE in 6m 10s\n- openstack-tox-py27 http://logs.openstack.org/10/645010/17/check/openstack-tox-py27/ce81905/ : SUCCESS in 6m 44s\n- openstack-tox-py36 http://logs.openstack.org/10/645010/17/check/openstack-tox-py36/8d5d47f/ : SUCCESS in 6m 48s\n- openstack-tox-py37 http://logs.openstack.org/10/645010/17/check/openstack-tox-py37/0e4af07/ : SUCCESS in 7m 03s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/17/check/manila-tox-genconfig/3911027/ : SUCCESS in 4m 40s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/10/645010/17/check/manila-tempest-dsvm-mysql-generic/cd163e7/ : FAILURE in 1h 21m 48s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/10/645010/17/check/manila-tempest-dsvm-postgres-container/359cd34/ : SUCCESS in 51m 18s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/10/645010/17/check/manila-tempest-dsvm-postgres-zfsonlinux/5338cac/ : SUCCESS in 1h 09m 44s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/10/645010/17/check/manila-tempest-dsvm-postgres-generic-singlebackend/c6a1f26/ : SUCCESS in 58m 35s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/10/645010/17/check/manila-tempest-dsvm-generic-no-share-servers/b5d4b61/ : SUCCESS in 1h 09m 38s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/10/645010/17/check/manila-tempest-dsvm-scenario/d3e6cb0/ : SUCCESS in 1h 33m 10s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/10/645010/17/check/manila-tempest-minimal-dsvm-cephfs-native/dbb5a23/ : SUCCESS in 27m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/10/645010/17/check/manila-tempest-minimal-dsvm-cephfs-nfs/d54b845/ : SUCCESS in 24m 56s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/10/645010/17/check/manila-tempest-minimal-dsvm-dummy/46b6fec/ : SUCCESS in 35m 31s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/10/645010/17/check/manila-tempest-minimal-dsvm-lvm/0c8c153/ : SUCCESS in 1h 17m 22s\n- manila-grenade http://logs.openstack.org/10/645010/17/check/manila-grenade/5553a74/ : SUCCESS in 26m 31s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/10/645010/17/check/manila-rally-multibackend/55cfa4f/ : SUCCESS in 26m 54s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/10/645010/17/check/manila-rally-multibackend-no-ss/3f87a97/ : SUCCESS in 25m 15s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/10/645010/17/check/openstack-tox-pylint/a22ada4/ : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"b3a3ed2c15251e4743ac7d9377360e9288f4cbd9","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-04-21 06:42:14.000000000","message":"Patch Set 17:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- manila-as13000 http://218.29.111.78:9000/10/645010/17/check/manila-as13000/2f85372 : FAILURE in 15m 19s\n- manila-instorage http://218.29.111.78:9000/10/645010/17/check/manila-instorage/361c9dd : FAILURE in 8m 38s","accounts_in_message":[],"_revision_number":17},{"id":"104edd633994eeb217eb67f044c526e7cbb09ffe","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2019-05-01 23:11:47.000000000","message":"Patch Set 17:\n\nrecheck pep8","accounts_in_message":[],"_revision_number":17},{"id":"4169701587b05c0ff62fc364f3ecf4b5a00cb2e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 01:13:08.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/17/check/openstack-tox-docs/c948de8/html/ : SUCCESS in 5m 08s\n- openstack-tox-cover http://logs.openstack.org/10/645010/17/check/openstack-tox-cover/7889055/cover/ : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/10/645010/17/check/openstack-tox-lower-constraints/a712983/ : SUCCESS in 7m 25s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/17/check/openstack-tox-pep8/2d87bf3/ : FAILURE in 5m 40s\n- openstack-tox-py27 http://logs.openstack.org/10/645010/17/check/openstack-tox-py27/67e90f6/ : SUCCESS in 7m 46s\n- openstack-tox-py36 http://logs.openstack.org/10/645010/17/check/openstack-tox-py36/7780371/ : SUCCESS in 7m 34s\n- openstack-tox-py37 http://logs.openstack.org/10/645010/17/check/openstack-tox-py37/beea5d2/ : SUCCESS in 7m 18s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/17/check/manila-tox-genconfig/ef64cd1/ : SUCCESS in 4m 45s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/10/645010/17/check/manila-tempest-dsvm-mysql-generic/b3d5dc6/ : FAILURE in 31m 43s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/10/645010/17/check/manila-tempest-dsvm-postgres-container/2ff0aed/ : SUCCESS in 45m 08s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/10/645010/17/check/manila-tempest-dsvm-postgres-zfsonlinux/8a262d1/ : FAILURE in 33m 06s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/10/645010/17/check/manila-tempest-dsvm-postgres-generic-singlebackend/7a7b58b/ : SUCCESS in 1h 01m 40s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/10/645010/17/check/manila-tempest-dsvm-generic-no-share-servers/4646c66/ : SUCCESS in 1h 10m 16s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/10/645010/17/check/manila-tempest-dsvm-scenario/5ed7b98/ : FAILURE in 33m 46s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/10/645010/17/check/manila-tempest-minimal-dsvm-cephfs-native/8882943/ : SUCCESS in 27m 36s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/10/645010/17/check/manila-tempest-minimal-dsvm-cephfs-nfs/64c8cea/ : SUCCESS in 21m 18s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/10/645010/17/check/manila-tempest-minimal-dsvm-dummy/eeeca66/ : SUCCESS in 30m 56s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/10/645010/17/check/manila-tempest-minimal-dsvm-lvm/e78dddd/ : SUCCESS in 1h 24m 39s\n- manila-grenade http://logs.openstack.org/10/645010/17/check/manila-grenade/46c437c/ : SUCCESS in 28m 29s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/10/645010/17/check/manila-rally-multibackend/ea97960/ : SUCCESS in 19m 15s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/10/645010/17/check/manila-rally-multibackend-no-ss/41071d2/ : SUCCESS in 26m 01s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/10/645010/17/check/openstack-tox-pylint/3574685/ : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"1dd7fbc3ddac96367da78d560d90f8f61da612a4","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-05-04 12:14:11.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"ed83ac19bc295894965af3cee6a8f23bf31e4053","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-05-04 12:15:56.000000000","message":"Patch Set 18:\n\nRemoved pip","accounts_in_message":[],"_revision_number":18},{"id":"928a2ef0e9c937b300ad1e3c1f68edcb5723b377","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 12:24:18.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/18/check/openstack-tox-docs/e3a117a/html/ : SUCCESS in 5m 47s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/18/check/openstack-tox-pep8/a819e9c/ : SUCCESS in 6m 58s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/18/check/manila-tox-genconfig/fca7b3e/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":18},{"id":"965ac4d3b93590141cb76c3ac1f9bb813ad2cf54","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-05-04 12:53:05.000000000","message":"Patch Set 18:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-10-645010-18 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":18},{"id":"626dddbf39b54433f06c8abf1973cdcf6244451d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-05-04 13:12:18.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/18/check/dsvm-tempest-infinibox-nfs/c64f745 : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":18},{"id":"7acc8be2991ff9f9790338fdd00fca2227aa6c69","author":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"date":"2019-05-04 13:33:49.000000000","message":"Patch Set 18:\n\n* nexenta-manila_ns5_nfs http://140.174.232.106/refs-changes-10-645010-18/manila_ns5_nfs : SUCCESS in 69m 20s","accounts_in_message":[],"_revision_number":18},{"id":"f4f4c8f4a3f4c82136cc6f65b051cdbe056a2d1d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-05-04 14:13:10.000000000","message":"Patch Set 18:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/10/645010/18/check/manila-master-client-pip-ssa03-dsvm-neutron/53913c3 : FAILURE in 1h 46m 20s","accounts_in_message":[],"_revision_number":18},{"id":"266701273d086f9111a41896d44d67485f6b0007","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-05-04 14:19:06.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/10/645010/18/check/DellEMC_VNX_Manila/c701ee9/DellEMC_VNX_Manila/10 : UNSTABLE in 1h 54m 32s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":18},{"id":"9d48996c7ddd3b71fa5b8cb681e45313c3e70186","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-05-04 14:33:59.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_VMAX_Manila http://publiclogs.emc.com/10/645010/18/check/DellEMC_VMAX_Manila/428c467/DellEMC_VMAX_Manila/9 : SUCCESS in 2h 19m 37s\n\nLeave a comment with \u0027run-dell-emc-vmax-ci\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VMAX_CI","accounts_in_message":[],"_revision_number":18},{"id":"1d6ab9bd27bcd05a256ec9347188fdbd83ed4b50","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-05-04 14:52:43.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/10/645010/18/check/DellEMC_Unity_Manila/f62e94c/DellEMC_Unity_Manila/5 : SUCCESS in 2h 38m 21s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"9698f77b8f6b153ce7e4abd32464882482147556","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2019-05-04 16:06:37.000000000","message":"Patch Set 18: Code-Review-1\n\n(7 comments)\n\nFew comments inline Sol, great work!","accounts_in_message":[],"_revision_number":18},{"id":"df059292ba10f68076458a466060985db8a2cdda","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-05-10 16:23:56.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"e85108e5cdd5df465f0300d23909c518a55609eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 16:42:09.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/19/check/openstack-tox-docs/42662b5/html/ : SUCCESS in 6m 39s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/19/check/openstack-tox-pep8/12727bb/ : SUCCESS in 7m 22s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/19/check/manila-tox-genconfig/0343091/ : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":19},{"id":"d29830289214cb327a2f4673908306308965b133","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-05-10 17:19:45.000000000","message":"Patch Set 19:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-10-645010-19 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":19},{"id":"59a6ae8b29f177f81e6b8cbf8e378c6e65d2cfdd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-05-10 17:21:09.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/19/check/dsvm-tempest-infinibox-nfs/7b1b7d0 : SUCCESS in 56m 45s","accounts_in_message":[],"_revision_number":19},{"id":"fc1cf37384e7cb8ccfff14885f19477149d80bb6","author":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"date":"2019-05-10 17:53:32.000000000","message":"Patch Set 19:\n\n* nexenta-manila_ns5_nfs http://140.174.232.106/refs-changes-10-645010-19/manila_ns5_nfs : SUCCESS in 70m 48s","accounts_in_message":[],"_revision_number":19},{"id":"fa3c48404b00005681cf8655b03d537f333a94ea","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-05-10 18:21:45.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/10/645010/19/check/manila-master-client-pip-ssa03-dsvm-neutron/63ed942 : SUCCESS in 1h 37m 01s","accounts_in_message":[],"_revision_number":19},{"id":"fd05a2a370879dbd4ff498a2b32d623db209116c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-05-10 18:44:53.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/10/645010/19/check/DellEMC_Unity_Manila/5a1895a/DellEMC_Unity_Manila/14 : SUCCESS in 2h 20m 46s\n\nLeave a comment with \u0027run-dell-emc-unity\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"0ba321db3b8af92708c793a64db1bccf16738686","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-05-10 18:47:37.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/10/645010/19/check/DellEMC_VNX_Manila/63008bf/DellEMC_VNX_Manila/16 : SUCCESS in 2h 05m 12s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":19},{"id":"36eb85edacb5258a2edb4459e0d17e372c5eba5e","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2019-05-29 17:11:52.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)\n\nLooks good Sol! Just some more comments. The cli docs look great, but the user guide docs needs one extra change. Could you create a new document, say, doc/source/user/troubleshooting-asynchronous-failures.rst and put all the content you generated there. You will also need to add it to the index (doc/source/user/index.rst). That way the rendering will look better. Thanks!","accounts_in_message":[],"_revision_number":19},{"id":"97957554ca22f86a5b851deea2a2ec491fb93ad6","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-05-29 17:22:51.000000000","message":"Patch Set 19: Code-Review-1 Workflow-1\n\n(6 comments)\n\nGreat, got it!","accounts_in_message":[],"_revision_number":19},{"id":"1b09b0ed9219de1d981f5dcb9abf372ddde92201","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-11 20:23:05.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"3e30129db6bee71a0f410d6caf7d8d9d11111c74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 20:36:06.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/20/check/openstack-tox-docs/8ebc238/ : FAILURE in 7m 27s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/20/check/openstack-tox-pep8/10eeedb/ : SUCCESS in 7m 24s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/20/check/manila-tox-genconfig/2704d07/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":20},{"id":"d0dffd1520007de52160ccfc57e7f069d3a60095","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2019-06-11 20:39:27.000000000","message":"Patch Set 20: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"2f36906fac851b112c40c366c6d33555505020ab","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-11 20:43:52.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"bd37e6ec43c98af165fd605700cd7a2dc5476081","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:24:01.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/21/check/openstack-tox-docs/8cc4d8a/html/ : SUCCESS in 5m 27s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/21/check/openstack-tox-pep8/4c6f742/ : SUCCESS in 6m 21s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/21/check/manila-tox-genconfig/130d5e0/ : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":21},{"id":"457e207f26e0726bc2b2eb82ad3e6e75e3fcb7e9","author":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"date":"2019-06-11 21:25:53.000000000","message":"Patch Set 21:\n\n* nexenta-manila_ns5_nfs http://140.174.232.106/refs-changes-10-645010-21/manila_ns5_nfs : FAILURE in 0m 12s","accounts_in_message":[],"_revision_number":21},{"id":"21c5d3c07949260af5140b7cbc82682fc3094bf8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-11 21:49:44.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/21/check/dsvm-tempest-infinibox-nfs/a283045 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":21},{"id":"9a7f2a5aaf7e1818ef5cdbe151340f1efbe55423","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-06-11 22:22:29.000000000","message":"Patch Set 21:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/10/645010/21/check/manila-master-client-pip-ssa03-dsvm-neutron/de04f68 : FAILURE in 55m 51s","accounts_in_message":[],"_revision_number":21},{"id":"560dfeb5ff8fe493557c52258d24cc807e332037","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-06-11 23:44:14.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VMAX_Manila http://publiclogs.emc.com/10/645010/21/check/DellEMC_VMAX_Manila/3047116/DellEMC_VMAX_Manila/78 : ABORTED in 3h 00m 05s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":21},{"id":"7d3be0f9441be5ccfe8c5f5aa79d12fa7acae54c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-12 00:13:07.000000000","message":"Patch Set 21:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-10-645010-21 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":21},{"id":"d956e42e77dcab7b3d6e458cf7573a7a1488623b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-06-12 01:32:36.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/10/645010/21/check/DellEMC_VNX_Manila/076608d/DellEMC_VNX_Manila/57 : SUCCESS in 2h 08m 17s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":21},{"id":"1834b22c8f1ce6d12d2a4e86cc875d2e9c339c56","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2019-06-12 14:14:16.000000000","message":"Patch Set 21: Code-Review-1\n\n(1 comment)\n\nI think you copy\u0026pasted a truncated line","accounts_in_message":[],"_revision_number":21},{"id":"973a618021a5e8f817e556d0c3a18b817519793c","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-12 16:56:00.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"bf5c79dee780488425f6ca20e31caf6e3315dde5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-12 18:08:01.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/22/check/dsvm-tempest-infinibox-nfs/a2cca09 : SUCCESS in 1h 11m 32s","accounts_in_message":[],"_revision_number":22},{"id":"bc40932a1c3aeb72e7d8b4aa7c352a5c7e532196","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 18:15:38.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/22/check/openstack-tox-docs/1dee1a5/html/ : SUCCESS in 6m 37s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/22/check/openstack-tox-pep8/ad72e41/ : SUCCESS in 6m 42s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/22/check/manila-tox-genconfig/494b374/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":22},{"id":"1ddf8e45c77bf565dd3a913780231e76c92d8147","author":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"date":"2019-06-12 18:16:57.000000000","message":"Patch Set 22:\n\n* nexenta-manila_ns5_nfs http://140.174.232.106/refs-changes-10-645010-22/manila_ns5_nfs : FAILURE in 0m 10s","accounts_in_message":[],"_revision_number":22},{"id":"e3ec4e5f1806513feb2bd960bb937c424cfa36f3","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-06-12 18:30:27.000000000","message":"Patch Set 22:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- manila-as13000 http://218.29.111.78:9000/10/645010/22/check/manila-as13000/13c1e16 : FAILURE in 1h 18m 06s\n- manila-instorage http://218.29.111.78:9000/10/645010/22/check/manila-instorage/575e1e3 : SUCCESS in 1h 27m 16s","accounts_in_message":[],"_revision_number":22},{"id":"bcee0dc21a6e1f90c45e539096ae177e708f9045","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-06-12 19:07:36.000000000","message":"Patch Set 22:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/10/645010/22/check/manila-master-client-pip-ssa03-dsvm-neutron/1229c5f : FAILURE in 49m 24s","accounts_in_message":[],"_revision_number":22},{"id":"a979d0f5a9ae1cc59453ff06d45e89ce7e6b0ee1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-12 19:32:07.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/10/645010/22/check/DellEMC_Unity_Manila/18ec641/DellEMC_Unity_Manila/60 : SUCCESS in 2h 35m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":22},{"id":"7b1b47edebfa0080f82d6f24dbfdb58c8f5df7fe","author":{"_account_id":18058,"name":"Lucio Seki","email":"lseki@redhat.com","username":"lseki"},"date":"2019-06-12 19:39:21.000000000","message":"Patch Set 22: Code-Review-1\n\nFor some reason, the code snippets for the steps 7 (the 2nd one), 10, 11 and 12 are being rendered as block quotes.\nThey have gray vertical bars at the left side, indicating that they\u0027re block quotes [0].\n\nMaybe the renderer is getting confused with the indented `.. code-block::` directives? I\u0027m not sure how to solve this...\n\n[0] http://logs.openstack.org/10/645010/22/check/openstack-tox-docs/1dee1a5/html/user/troubleshooting-asynchronous-failures.html","accounts_in_message":[],"_revision_number":22},{"id":"f5cb453852c5aed9cf970c4d5e550679ace80b48","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-12 19:43:38.000000000","message":"Patch Set 22: Code-Review-1\n\n(3 comments)\n\nGood stuff so far Sol, please take a look at the comments inline.","accounts_in_message":[],"_revision_number":22},{"id":"25c4d1091b61697ab464f4c9026c6b1b590d7d2e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-06-12 19:56:16.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VMAX_Manila http://publiclogs.emc.com/10/645010/22/check/DellEMC_VMAX_Manila/1e886d3/DellEMC_VMAX_Manila/81 : ABORTED in 3h 00m 05s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":22},{"id":"ad1e16e2b171a2244902dfc59372b54692ff1079","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-12 20:15:40.000000000","message":"Patch Set 22:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-10-645010-22 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":22},{"id":"d15ecf50634521964f3f1460c851f6b8f5cd966b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-06-12 20:29:42.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/10/645010/22/check/DellEMC_VNX_Manila/15c50a3/DellEMC_VNX_Manila/58 : SUCCESS in 2h 13m 47s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":22},{"id":"feb71c21f10086cb963f8a5c94bb0fb9b3349cbe","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-14 17:15:46.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"834e63f22bb02ccff9d37510be5c9e54af8cf72e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-14 17:16:15.000000000","message":"Patch Set 23:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/23/check/dsvm-tempest-infinibox-nfs/38060b9 : FAILURE in 1s","accounts_in_message":[],"_revision_number":23},{"id":"4b02b279c6afae1991f30de73315d932aa78eecf","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-14 17:21:44.000000000","message":"Patch Set 23:\n\n(2 comments)\n\nI think finally I was able to use tox build well... All this time doing it blindly! So sorry...","accounts_in_message":[],"_revision_number":23},{"id":"7aefd6ba1a258ac67a931d44c0c8104260a3a911","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 17:26:58.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/23/check/openstack-tox-docs/b04a444/html/ : SUCCESS in 5m 39s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/23/check/openstack-tox-pep8/36dff7f/ : SUCCESS in 5m 59s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/23/check/manila-tox-genconfig/e2c8ddb/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":23},{"id":"33470abd8af453271be6e83cef35662719680225","author":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"date":"2019-06-14 17:41:08.000000000","message":"Patch Set 23:\n\n* nexenta-manila_ns5_nfs http://140.174.232.106/refs-changes-10-645010-23/manila_ns5_nfs : FAILURE in 13m 4s","accounts_in_message":[],"_revision_number":23},{"id":"567c1d169765ec2fb03c3bbfd42ef99b1e180dbe","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-06-14 19:11:49.000000000","message":"Patch Set 23:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/10/645010/23/check/manila-master-client-pip-ssa03-dsvm-neutron/7de9e36 : FAILURE in 1h 42m 19s","accounts_in_message":[],"_revision_number":23},{"id":"3d54a7769cc4ec16e123031b28168d33b4c31b12","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-14 19:15:13.000000000","message":"Patch Set 23:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-10-645010-23 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":23},{"id":"e9ed17f59d654dcf7eec5fa221b0f2993f36a6cc","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-06-14 19:29:30.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_VMAX_Manila http://publiclogs.emc.com/10/645010/23/check/DellEMC_VMAX_Manila/ab88854/DellEMC_VMAX_Manila/88 : SUCCESS in 2h 13m 33s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":23},{"id":"5e62100df8cfd438aead7856adc262fb112c721d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-06-14 19:41:55.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/10/645010/23/check/DellEMC_VNX_Manila/aa76ff2/DellEMC_VNX_Manila/62 : FAILURE in 2h 14m 41s\n\nLeave a comment with \u0027run-dell-emc-vnx\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VNX_CI","accounts_in_message":[],"_revision_number":23},{"id":"0a43d5c1d2405c10b29822502212c85ad97208b6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-14 19:52:57.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/10/645010/23/check/DellEMC_Unity_Manila/abc1eb4/DellEMC_Unity_Manila/64 : SUCCESS in 2h 37m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"b0b8a8b5c96a90dd6f1f15bed3f2d0a88220202c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-17 20:27:25.000000000","message":"Patch Set 23: Code-Review-1\n\n(3 comments)\n\nHi, requesting changes that I asked for on a previous patchset.","accounts_in_message":[],"_revision_number":23},{"id":"33e507760b2aca3620d7c7911d40b52237b74213","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-19 18:44:57.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"de3eb258b1a89261a99de99768c5a0d8b1b5a95b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-19 18:50:56.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"2f978dc567eb47f968d4244a40f8ea14eef5b732","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-21 20:25:45.000000000","message":"Patch Set 23: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"b8a475433c8754a067b640f6a5f0148d50a8b4eb","author":{"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},"date":"2019-06-22 13:01:18.000000000","message":"Patch Set 23:\n\nrun-nexenta ci","accounts_in_message":[],"_revision_number":23},{"id":"e61a7bc11c46c8df9c71847c080a336b3e6c3bc7","author":{"_account_id":15439,"name":"Alexey Khodos","email":"akhodos@ddn.com","username":"Alexey"},"date":"2019-06-22 13:01:31.000000000","message":"Patch Set 23:\n\nrun-nexenta manila ci","accounts_in_message":[],"_revision_number":23},{"id":"3949b12200aac1154deb455ffe12cff75e93b3ea","author":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"date":"2019-06-22 13:05:23.000000000","message":"Patch Set 23:\n\n* nexenta-manila_ns5_nfs http://140.174.232.106/refs-changes-10-645010-23/manila_ns5_nfs : FAILURE in 2m 51s","accounts_in_message":[],"_revision_number":23},{"id":"a203d24ced0eb88e778de5a4de82da72a94cc6d2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-22 18:05:55.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"0b318d632c2849462f1a6c7e65001eb767572a1d","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2019-06-24 15:41:36.000000000","message":"Patch Set 23:\n\nAgree with Goutham\u0027s comment. Seems logic to have delete share first, then clean up the user messages queue. Other than that, the docs look good. Thanks Sol!","accounts_in_message":[],"_revision_number":23},{"id":"90840241b004e0d28ddb571d9179f3ffc6ae8cf2","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-24 19:08:01.000000000","message":"Patch Set 23:\n\nAlright. I analyzed it better and asking myself why we show only the message-list output, I saw that the docs already show the command for delete shares and for message-list don\u0027t. I guess is to avoid repeating. Thanks","accounts_in_message":[],"_revision_number":23},{"id":"6f064388d51ddf13c871d0db99c2128e7403c7bc","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-24 19:25:30.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"9a94832a8b6c7eb89cdb3dcb043b69d5a52203a3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-24 19:27:29.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"d79a49d32ce13d04b977c6d2aba720470924dcab","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-24 19:46:03.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/10/645010/24/check/DellEMC_Unity_Manila/c8b02ce/DellEMC_Unity_Manila/80 : FAILURE in 20m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":24},{"id":"8f27b6d7277858456639de5033bd850322474c95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 19:58:18.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/24/check/openstack-tox-docs/b7cebc3/ : FAILURE in 3m 41s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/24/check/openstack-tox-pep8/5ef36d2/ : RETRY_LIMIT in 2m 07s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/24/check/manila-tox-genconfig/1f5d621/ : RETRY_LIMIT in 2m 32s","accounts_in_message":[],"_revision_number":24},{"id":"39077bc0efb7f07087192d90e1e541f4f1258961","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-24 20:10:46.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"d96d2748de0433b444b1eabc2b81f39fb2562027","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-24 20:31:12.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/10/645010/25/check/DellEMC_Unity_Manila/dbb90a0/DellEMC_Unity_Manila/81 : FAILURE in 20m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":25},{"id":"8c533ada8c944d9c0297faca2ef0b36fb6635656","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 20:41:39.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/25/check/openstack-tox-docs/9889f9b/ : FAILURE in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/25/check/openstack-tox-pep8/f211d9d/ : RETRY_LIMIT in 1m 50s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/25/check/manila-tox-genconfig/ce16854/ : RETRY_LIMIT in 2m 17s","accounts_in_message":[],"_revision_number":25},{"id":"307d832e758f5b7af75f0ba4d406f111c6038c9b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-24 21:10:18.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/10/645010/25/check/dsvm-tempest-infinibox-nfs/5c8de4e : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":25},{"id":"c0653cff26d18fc1005a0d87911ba88cb2235184","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-06-24 22:27:21.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/10/645010/25/check/DellEMC_PowerMAX_Manila/b8e07ca/DellEMC_PowerMAX_Manila/110 : SUCCESS in 2h 16m 18s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":25},{"id":"77e16b1030404109fa134fb3747759e52374f07f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-25 22:10:41.000000000","message":"Patch Set 25: Code-Review+2\n\nThanks for your patience and perseverance with this patch. This is indeed a great addition to our documentation.","accounts_in_message":[],"_revision_number":25},{"id":"a1e9e759e4ee7d513dd8da83bccb83e7338bcc70","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-25 22:10:57.000000000","message":"Patch Set 25:\n\nrecheck\n\nWe had a CI issue, it\u0027s now been fixed,","accounts_in_message":[],"_revision_number":25},{"id":"70f0dab270784f761112fbc5cd1ce84a1aa59804","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 22:23:28.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/25/check/openstack-tox-docs/8b520fb/html/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/25/check/openstack-tox-pep8/e1a307d/ : SUCCESS in 5m 48s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/25/check/manila-tox-genconfig/af70e5a/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":25},{"id":"aaf762e275d7eff00dc7ec3268f74aa121193440","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-25 22:43:19.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/10/645010/25/check/DellEMC_Unity_Manila/24b2fe6/DellEMC_Unity_Manila/91 : FAILURE in 19m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":25},{"id":"0b72a899d02f29349be70d78c0deaf382249e992","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-06-26 00:16:25.000000000","message":"Patch Set 25:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/10/645010/25/check/manila-master-client-pip-ssa03-dsvm-neutron/44e3c32 : FAILURE in 1h 50m 24s","accounts_in_message":[],"_revision_number":25},{"id":"05ccc81c310611d05109d2caade0d0fe3e60cad3","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-06-26 00:17:09.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/10/645010/25/check/DellEMC_VNX_Manila/e5c5be8/DellEMC_VNX_Manila/75 : SUCCESS in 1h 46m 10s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":25},{"id":"c11b5c698c4f5b81e7c2c7a0539267e026585b25","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-26 01:13:26.000000000","message":"Patch Set 25:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-10-645010-25 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":25},{"id":"11f1e474a4aacc2c34fb9c7eb051a8f5cb7a56e4","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-06-26 02:34:19.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/10/645010/25/check/DellEMC_PowerMAX_Manila/898a7ad/DellEMC_PowerMAX_Manila/116 : SUCCESS in 2h 15m 59s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":25},{"id":"ad21a8d699d588bccb6ba9eb95c4e2dbd59a3857","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-06-26 11:02:33.000000000","message":"Patch Set 25:\n\nThanks, Soledad, for filling this important documentation gap!","accounts_in_message":[],"_revision_number":25},{"id":"1128066b0b2b0cd0c7692b10dc0d36116133430d","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-06-26 11:02:55.000000000","message":"Patch Set 25: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":25},{"id":"6b9499fd2d82277326aa8dd9452c535bab21256a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 11:03:09.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"e1bef05f99005219037ddfac7d2bc40ee5121286","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 11:26:05.000000000","message":"Patch Set 25: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/25/gate/openstack-tox-docs/f568fcb/html/ : SUCCESS in 6m 25s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/25/gate/openstack-tox-pep8/6d8fba0/ : RETRY_LIMIT in 1m 54s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/25/gate/manila-tox-genconfig/6a46fcd/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":25},{"id":"c1a4f9348edd624cc7532fb91f16399b2fa46cc9","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-06-26 11:47:50.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"7d73c799c9df821bd03d8e49036e62501428ac72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 11:58:34.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/25/check/openstack-tox-docs/70b59c3/html/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/25/check/openstack-tox-pep8/4274c24/ : SUCCESS in 5m 20s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/25/check/manila-tox-genconfig/d66b25a/ : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":25},{"id":"2519e455098c07e61e4b1e8779f7e938904a063e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 11:58:48.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"d48f4f4a3088242d1181b2059b84bdcce7935550","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 12:33:16.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/10/645010/25/gate/openstack-tox-docs/b9fc0d4/html/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/10/645010/25/gate/openstack-tox-pep8/1d20927/ : SUCCESS in 5m 06s\n- manila-tox-genconfig http://logs.openstack.org/10/645010/25/gate/manila-tox-genconfig/96bfb38/ : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":25},{"id":"5f5f56eccfb78af94cfba8012f2e46b7f0fa54e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 12:33:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":25},{"id":"d1809700c230c6c680504483727e5890adddfef0","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-06-26 13:03:52.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- manila-as13000 http://218.29.111.78:9000/10/645010/25/check/manila-as13000/2283495 : SUCCESS in 55m 27s\n- manila-instorage http://218.29.111.78:9000/10/645010/25/check/manila-instorage/81b562e : SUCCESS in 1h 14m 11s","accounts_in_message":[],"_revision_number":25},{"id":"b703f0f2b600cd70aea1d316d06c612e8b0abff7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-26 13:23:59.000000000","message":"Patch Set 25:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-10-645010-25 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":25},{"id":"fea77dd3823cb904c8e84e4abf76f36a237872fe","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-26 17:46:53.000000000","message":"Patch Set 25:\n\nthanks to you, Goutham!, you had the patience part, I learned a lot.","accounts_in_message":[],"_revision_number":25},{"id":"b4ebce3d06a0512b0b31fd2a80397d05dfa9c2c3","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-26 17:47:44.000000000","message":"Patch Set 25:\n\nGlad to hear Tom! finally is there :)","accounts_in_message":[],"_revision_number":25},{"id":"baff052b0fb1df10f8eeec5c45b718226ebb38e0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-26 20:37:34.000000000","message":"Patch Set 25:\n\nHi Soledad, would you be so kind to back port this documentation to the stable/stein branch?\n\nUser messages were added in the Pike release, so this documentation can be back ported to all the releases we support. So, once it merges in stable/stein, we\u0027ll get it backported to stable/rocky, and then to stable/queens and finally to stable/pike.\n\n\nTo backport this patch, look for the \"Cherry pick\" button on this review, and type \"stable/stein\" in the branch. Or, you can also:\n\n  $ cd manila\n  $ git fetch --all\n  $ git checkout origin/stable/stein\n  $ git cherry-pick -x  84e7f4e5426fcee5941b8e4cb6d53040be8cd0ee\n  $ # Resolve any merge conflicts\n  $ git review","accounts_in_message":[],"_revision_number":25},{"id":"585fe631e47807ea2d68e7b0c73306115ad2e40e","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-27 01:09:51.000000000","message":"Patch Set 25:\n\nSure Goutham","accounts_in_message":[],"_revision_number":25},{"id":"2a7db4a9d20a110c33e55c29023055e8a54bd902","author":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"date":"2019-06-27 01:11:34.000000000","message":"Patch Set 25: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit bf829be9999c97acd260bbc16aa4e6a921637886","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"84e7f4e5426fcee5941b8e4cb6d53040be8cd0ee","revisions":{"8bcb13f4ec7dc395fdd6affac14dc52460b09728":{"kind":"REWORK","_number":1,"created":"2019-03-20 21:14:05.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/1"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 21:13:33.000000000","tz":0},"subject":"Adding documentation for user messages in Manila","message":"Adding documentation for user messages in Manila\n\n-Reference to the CLI usage\n\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\nPartial-Bug: #1796986\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8bcb13f4ec7dc395fdd6affac14dc52460b09728"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8bcb13f4ec7dc395fdd6affac14dc52460b09728"}]},"branch":"refs/heads/master"},"4162f92fdc3e50da359475b2b42f8a395ab36907":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-03-20 21:19:35.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/2"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 21:19:02.000000000","tz":0},"subject":"Adding documentation for user messages in Manila","message":"Adding documentation for user messages in Manila\n\n-Reference to the CLI usage\n\nPartial-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4162f92fdc3e50da359475b2b42f8a395ab36907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4162f92fdc3e50da359475b2b42f8a395ab36907"}]},"branch":"refs/heads/master"},"2574345930b7ddd2bb1ce473ff24f27c8da5cafc":{"kind":"REWORK","_number":3,"created":"2019-03-21 03:15:52.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/3"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","date":"2019-03-21 03:15:48.000000000","tz":0},"subject":"Adding documentation for user messages in Manila","message":"Adding documentation for user messages in Manila\n\n-Reference to the CLI usage\n\nPartial-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2574345930b7ddd2bb1ce473ff24f27c8da5cafc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2574345930b7ddd2bb1ce473ff24f27c8da5cafc"}]},"branch":"refs/heads/master"},"0e64ba61353605efa16bfbb21a9731858ec1a623":{"kind":"REWORK","_number":4,"created":"2019-03-21 16:01:53.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/4"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","date":"2019-03-21 15:57:07.000000000","tz":0},"subject":"Adding documentation for user messages in Manila","message":"Adding documentation for user messages in Manila\n\n-Reference to the CLI usage\n\nPartial-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0e64ba61353605efa16bfbb21a9731858ec1a623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0e64ba61353605efa16bfbb21a9731858ec1a623"}]},"branch":"refs/heads/master"},"bde4bccb98e9966826bc5787e6b63cef7365ca46":{"kind":"REWORK","_number":5,"created":"2019-03-22 03:32:10.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/5"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-21 21:53:32.000000000","tz":0},"subject":"Adding documentation for user messages in Manila Documentation","message":"Adding documentation for user messages in Manila Documentation\n\n-Reference to the user messages CLI usage at\nhttps://docs.openstack.org/manila/latest/cli/manila.html\n\nPartial-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bde4bccb98e9966826bc5787e6b63cef7365ca46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bde4bccb98e9966826bc5787e6b63cef7365ca46"}]},"branch":"refs/heads/master"},"6b59074f415793c13b25b928841d15cefeef4506":{"kind":"REWORK","_number":6,"created":"2019-03-22 15:55:58.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/6"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-22 15:54:22.000000000","tz":0},"subject":"Adding documentation for user messages in Manila Documentation","message":"Adding documentation for user messages in Manila Documentation\n\n-Reference to the user messages CLI usage at\nhttps://docs.openstack.org/manila/latest/cli/manila.html\n\nPartial-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6b59074f415793c13b25b928841d15cefeef4506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6b59074f415793c13b25b928841d15cefeef4506"}]},"branch":"refs/heads/master"},"6f02643669b3a8886323414928969b00c158e7fd":{"kind":"REWORK","_number":7,"created":"2019-03-22 18:20:24.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/7"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-22 18:02:08.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at\nhttps://docs.openstack.org/manila/latest/cli/manila.html\n\nPartial-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6f02643669b3a8886323414928969b00c158e7fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6f02643669b3a8886323414928969b00c158e7fd"}]},"branch":"refs/heads/master"},"d78f03e88f6a8910cc46151dcff7a31418267609":{"kind":"REWORK","_number":8,"created":"2019-03-25 16:01:05.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/8"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-25 15:59:08.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d78f03e88f6a8910cc46151dcff7a31418267609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d78f03e88f6a8910cc46151dcff7a31418267609"}]},"branch":"refs/heads/master"},"4f7106ce0b88354003a5baadb881f0c8fe38cbce":{"kind":"REWORK","_number":9,"created":"2019-03-25 16:27:51.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/9"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-25 16:27:27.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4f7106ce0b88354003a5baadb881f0c8fe38cbce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4f7106ce0b88354003a5baadb881f0c8fe38cbce"}]},"branch":"refs/heads/master"},"4a229953ce190d61001a64fddf8ef30e1a681dfe":{"kind":"REWORK","_number":10,"created":"2019-03-25 17:27:11.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/10"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-25 17:25:55.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4a229953ce190d61001a64fddf8ef30e1a681dfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4a229953ce190d61001a64fddf8ef30e1a681dfe"}]},"branch":"refs/heads/master"},"896301d5a6bdf909c04fd574316ad5d6b8e43478":{"kind":"REWORK","_number":11,"created":"2019-03-25 18:11:51.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/11"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-25 18:11:23.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/896301d5a6bdf909c04fd574316ad5d6b8e43478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/896301d5a6bdf909c04fd574316ad5d6b8e43478"}]},"branch":"refs/heads/master"},"ac4c1d643b75be5dbf55a9dcbe00f18f466e1e06":{"kind":"REWORK","_number":12,"created":"2019-03-25 18:20:36.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/12"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-25 18:20:17.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ac4c1d643b75be5dbf55a9dcbe00f18f466e1e06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ac4c1d643b75be5dbf55a9dcbe00f18f466e1e06"}]},"branch":"refs/heads/master"},"0f3cf2341a002847ceda8e695482ad0a5162e58f":{"kind":"REWORK","_number":13,"created":"2019-03-25 18:40:16.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/13"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-25 18:39:56.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0f3cf2341a002847ceda8e695482ad0a5162e58f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0f3cf2341a002847ceda8e695482ad0a5162e58f"}]},"branch":"refs/heads/master"},"590084359b3432fb19fb01b41e1b130bb3554f3f":{"kind":"REWORK","_number":14,"created":"2019-04-11 22:53:56.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/14"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-04-11 22:49:35.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/590084359b3432fb19fb01b41e1b130bb3554f3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/590084359b3432fb19fb01b41e1b130bb3554f3f"}]},"branch":"refs/heads/master"},"0ef941ac69270b439a03bda8cf6c8447ab775598":{"kind":"REWORK","_number":15,"created":"2019-04-11 23:03:34.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/15"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-04-11 22:58:34.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0ef941ac69270b439a03bda8cf6c8447ab775598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0ef941ac69270b439a03bda8cf6c8447ab775598"}]},"branch":"refs/heads/master"},"ff59af501c58425e51077513cda6b19f158d2302":{"kind":"REWORK","_number":16,"created":"2019-04-12 17:23:26.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/16"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-04-12 17:22:54.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ff59af501c58425e51077513cda6b19f158d2302"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ff59af501c58425e51077513cda6b19f158d2302"}]},"branch":"refs/heads/master"},"0122bfaf701d0f7f4a57f5d3d5e820be19bdd453":{"kind":"REWORK","_number":17,"created":"2019-04-20 20:32:21.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/17"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-04-19 17:45:20.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0122bfaf701d0f7f4a57f5d3d5e820be19bdd453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0122bfaf701d0f7f4a57f5d3d5e820be19bdd453"}]},"branch":"refs/heads/master"},"250accc77529bcabf89c15eac0c4d828f28968cb":{"kind":"REWORK","_number":18,"created":"2019-05-04 12:14:11.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/18"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-05-04 12:13:26.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/250accc77529bcabf89c15eac0c4d828f28968cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/250accc77529bcabf89c15eac0c4d828f28968cb"}]},"branch":"refs/heads/master"},"ee272bbfbc49b8085dfcc038b0f020f955abf273":{"kind":"REWORK","_number":19,"created":"2019-05-10 16:23:56.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/19"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-05-10 16:23:15.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ee272bbfbc49b8085dfcc038b0f020f955abf273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ee272bbfbc49b8085dfcc038b0f020f955abf273"}]},"branch":"refs/heads/master"},"7b697e632eb2ac01974acb067f9ee2a198d92a21":{"kind":"REWORK","_number":20,"created":"2019-06-11 20:23:05.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/20"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-06-11 20:22:35.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7b697e632eb2ac01974acb067f9ee2a198d92a21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7b697e632eb2ac01974acb067f9ee2a198d92a21"}]},"branch":"refs/heads/master"},"3e2b90db0dcf2074d90319b4b8f6d1bbac9786c0":{"kind":"REWORK","_number":21,"created":"2019-06-11 20:43:52.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/21"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-06-11 20:43:26.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3e2b90db0dcf2074d90319b4b8f6d1bbac9786c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3e2b90db0dcf2074d90319b4b8f6d1bbac9786c0"}]},"branch":"refs/heads/master"},"2556bf7b321781027c419c82fb54d98a0670fcf0":{"kind":"REWORK","_number":22,"created":"2019-06-12 16:56:00.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/22"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-06-12 16:53:40.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2556bf7b321781027c419c82fb54d98a0670fcf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2556bf7b321781027c419c82fb54d98a0670fcf0"}]},"branch":"refs/heads/master"},"52b5c0d67bb5a34940746d86a012d232b2d4ce1b":{"kind":"REWORK","_number":23,"created":"2019-06-14 17:15:46.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/23"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-06-14 17:12:37.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/52b5c0d67bb5a34940746d86a012d232b2d4ce1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/52b5c0d67bb5a34940746d86a012d232b2d4ce1b"}]},"branch":"refs/heads/master"},"874c58f4d25626146dbf64f865c3f80375e52c98":{"kind":"REWORK","_number":24,"created":"2019-06-24 19:25:30.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/24"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-06-24 19:25:01.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/874c58f4d25626146dbf64f865c3f80375e52c98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/874c58f4d25626146dbf64f865c3f80375e52c98"}]},"branch":"refs/heads/master"},"84e7f4e5426fcee5941b8e4cb6d53040be8cd0ee":{"kind":"REWORK","_number":25,"created":"2019-06-24 20:10:46.000000000","uploader":{"_account_id":29958,"name":"Soledad Kuczala","email":"sol.kuczala@gmail.com","username":"SolKuczala"},"ref":"refs/changes/10/645010/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/10/645010/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/10/645010/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/10/645010/25"}}},"commit":{"parents":[{"commit":"15878d96f3145fc8baf4df3aa02c1f0b963a4336","subject":"Merge \"Check all_tenants value in share_networks api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/15878d96f3145fc8baf4df3aa02c1f0b963a4336"}]}],"author":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-03-20 20:56:09.000000000","tz":0},"committer":{"name":"SolKuczala","email":"sol.kuczala@gmail.com","date":"2019-06-24 20:10:22.000000000","tz":0},"subject":"Adding documentation for User Messages in Manila Documentation","message":"Adding documentation for User Messages in Manila Documentation\n\n-Reference to the User Messages CLI usage at:\nhttps://docs.openstack.org/manila/latest/cli/manila.html\nhttps://docs.openstack.org/manila/latest/user/create-and-manage-shares.\n\nCloses-Bug: #1796986\nChange-Id: Ia85736693d7f536766e0e0980f88805c61aa8325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84e7f4e5426fcee5941b8e4cb6d53040be8cd0ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84e7f4e5426fcee5941b8e4cb6d53040be8cd0ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
