)]}'
{"id":"openstack%2Fmanila~666213","triplet_id":"openstack%2Fmanila~master~Ibcac3ebc1d665fd8357e4716e31ac7464161e497","project":"openstack/manila","branch":"master","topic":"constraints","hashtags":[],"change_id":"Ibcac3ebc1d665fd8357e4716e31ac7464161e497","subject":"Modify the url of upper_constraints_file","status":"ABANDONED","created":"2019-06-19 04:29:10.000000000","updated":"2019-12-24 06:28:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"519f70468d6f9d9b74318bca2bbb5d3a38a2ea84","_number":666213,"virtual_id_number":666213,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"date":"2019-07-05 04:26:07.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":-1,"date":"2019-07-05 04:53:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"date":"2019-07-05 03:49:02.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":2417,"name":"Ben Swartzlander","email":"ben@swartzlander.org","username":"bswartz"},{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},{"date":"2019-07-05 05:23:05.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":15100,"name":"zhongjun","email":"jun.zhongjun2@gmail.com","username":"zhongjun2"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2417,"name":"Ben Swartzlander","email":"ben@swartzlander.org","username":"bswartz"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15100,"name":"zhongjun","email":"jun.zhongjun2@gmail.com","username":"zhongjun2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":2417,"name":"Ben Swartzlander","email":"ben@swartzlander.org","username":"bswartz"},{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":15100,"name":"zhongjun","email":"jun.zhongjun2@gmail.com","username":"zhongjun2"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2417,"name":"Ben Swartzlander","email":"ben@swartzlander.org","username":"bswartz"},{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":15100,"name":"zhongjun","email":"jun.zhongjun2@gmail.com","username":"zhongjun2"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 05:01:01.000000000","updated_by":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"reviewer":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"state":"REVIEWER"},{"updated":"2019-06-19 09:48:50.000000000","updated_by":{"_account_id":2417,"name":"Ben Swartzlander","email":"ben@swartzlander.org","username":"bswartz"},"reviewer":{"_account_id":2417,"name":"Ben Swartzlander","email":"ben@swartzlander.org","username":"bswartz"},"state":"REVIEWER"},{"updated":"2019-06-19 09:48:50.000000000","updated_by":{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},"reviewer":{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},"state":"REVIEWER"},{"updated":"2019-06-19 09:48:50.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2019-06-19 09:48:50.000000000","updated_by":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"reviewer":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"state":"REVIEWER"},{"updated":"2019-06-19 09:48:50.000000000","updated_by":{"_account_id":15100,"name":"zhongjun","email":"jun.zhongjun2@gmail.com","username":"zhongjun2"},"reviewer":{"_account_id":15100,"name":"zhongjun","email":"jun.zhongjun2@gmail.com","username":"zhongjun2"},"state":"REVIEWER"},{"updated":"2019-06-19 09:48:50.000000000","updated_by":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"reviewer":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"state":"REVIEWER"},{"updated":"2019-06-19 20:06:10.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-07-01 06:07:38.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-01 06:11:21.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-01 06:29:46.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-01 07:23:37.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-02 05:42:34.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-04 10:13:06.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2019-07-05 03:49:02.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-05 04:26:07.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-07-05 04:53:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-05 05:23:05.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"}],"messages":[{"id":"706ec5fd59baa262571bce915ffdbe45b4150c72","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-19 04:29:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b90f316bb36bc18a383cf8429fb6ea0a261920d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-19 04:52:07.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/13/666213/1/check/DellEMC_Unity_Manila/8c8ca64/DellEMC_Unity_Manila/70 : FAILURE in 22m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"767535d0c66899d5171f030991c0f42ed11bd388","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 04:56:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/13/666213/1/check/build-openstack-api-ref/1d45926/html/ : SUCCESS in 6m 29s\n- openstack-tox-docs http://logs.openstack.org/13/666213/1/check/openstack-tox-docs/b3af0ae/html/ : SUCCESS in 5m 54s\n- openstack-tox-cover http://logs.openstack.org/13/666213/1/check/openstack-tox-cover/16f332d/cover/ : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/13/666213/1/check/openstack-tox-lower-constraints/6b28628/ : SUCCESS in 7m 06s\n- openstack-tox-pep8 http://logs.openstack.org/13/666213/1/check/openstack-tox-pep8/d01d893/ : SUCCESS in 6m 01s\n- openstack-tox-py27 http://logs.openstack.org/13/666213/1/check/openstack-tox-py27/0b4be57/ : SUCCESS in 6m 29s\n- openstack-tox-py36 http://logs.openstack.org/13/666213/1/check/openstack-tox-py36/f2cec51/ : SUCCESS in 13m 55s\n- openstack-tox-py37 http://logs.openstack.org/13/666213/1/check/openstack-tox-py37/e0885ae/ : SUCCESS in 9m 33s\n- build-openstack-releasenotes http://logs.openstack.org/13/666213/1/check/build-openstack-releasenotes/5369153/html/ : SUCCESS in 7m 53s\n- manila-tox-genconfig http://logs.openstack.org/13/666213/1/check/manila-tox-genconfig/b03ab4f/ : SUCCESS in 5m 28s\n- openstack-tox-pylint http://logs.openstack.org/13/666213/1/check/openstack-tox-pylint/1fe9052/ : SUCCESS in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"832ed40ce1cb58fbf141ee8c3e43d65de81559e3","author":{"_account_id":20695,"name":"Nexenta Manila CI","email":"nexenta.manila@gmail.com","username":"nexentamanilaci"},"date":"2019-06-19 05:01:01.000000000","message":"Patch Set 1:\n\n* nexenta-manila_ns5_nfs http://140.174.232.106/refs-changes-13-666213-1/manila_ns5_nfs : FAILURE in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b456055488b98f3559968e2810ae2415dc89fdf5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-19 05:19:28.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/13/666213/1/check/DellEMC_Unity_Manila/e27fe15/DellEMC_Unity_Manila/71 : FAILURE in 22m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"5fd67457b8aa424ae302f4a8bc3249bb4c5dfdfc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-19 05:26:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/13/666213/1/check/dsvm-tempest-infinibox-nfs/ed075c7 : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0d03dbcd635a58a4539689ef09654848c3c788c9","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-06-19 05:33:41.000000000","message":"Patch Set 1:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/13/666213/1/check/manila-master-client-pip-ssa03-dsvm-neutron/018a230 : FAILURE in 34m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4ad98159c161361471d51157eca31bd63899648a","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-06-19 05:43:19.000000000","message":"Patch Set 1:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- manila-as13000 http://218.29.111.78:9000/13/666213/1/check/manila-as13000/ba27c1a : FAILURE in 4m 05s\n- manila-instorage http://218.29.111.78:9000/13/666213/1/check/manila-instorage/1675e2e : SUCCESS in 1h 13m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f5a2f944ced109203ad947d6af51be6ea3a90332","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-06-19 05:52:16.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/13/666213/1/check/DellEMC_PowerMAX_Manila/57c958c/DellEMC_PowerMAX_Manila/100 : FAILURE in 1h 22m 55s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"9127d3a60a0e60cbc6c0ae66610973d2cfc893d8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-19 06:17:55.000000000","message":"Patch Set 1:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-13-666213-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"37de6fb7bafde7de2bb87a2d4fc62d21ca717302","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-06-19 20:06:10.000000000","message":"Patch Set 1: Code-Review-1\n\nThanks for the change, reading through the email http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html, it appears that we need more changes. \n\nFor example \"Make sure projects do not specify a constraint file in install_command\" isn\u0027t covered in your change.","accounts_in_message":[],"_revision_number":1},{"id":"41eae830d352a9ff67aca2433eaf9a536943d1c0","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-06-21 01:27:47.000000000","message":"Patch Set 1:\n\n\u003e Thanks for the change, reading through the email http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html,\n \u003e it appears that we need more changes.\n \u003e \n \u003e For example \"Make sure projects do not specify a constraint file in\n \u003e install_command\" isn\u0027t covered in your change.\n\nThis feature already exists in L102-L107 and does not need to be added","accounts_in_message":[],"_revision_number":1},{"id":"b0b1874e25ea907c22fc3870c342333accb689e1","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-01 04:16:20.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"97589b5df3137d23da70e29962755fcc379c64fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 04:39:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/13/666213/2/check/build-openstack-api-ref/acc5b72/html/ : SUCCESS in 5m 07s\n- openstack-tox-docs http://logs.openstack.org/13/666213/2/check/openstack-tox-docs/740c23d/html/ : SUCCESS in 5m 41s\n- openstack-tox-cover http://logs.openstack.org/13/666213/2/check/openstack-tox-cover/5ac1f3b/cover/ : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/13/666213/2/check/openstack-tox-lower-constraints/ac90a9c/ : SUCCESS in 9m 36s\n- openstack-tox-pep8 http://logs.openstack.org/13/666213/2/check/openstack-tox-pep8/a7be8d1/ : SUCCESS in 5m 40s\n- openstack-tox-py27 http://logs.openstack.org/13/666213/2/check/openstack-tox-py27/cf2888c/ : SUCCESS in 9m 59s\n- openstack-tox-py36 http://logs.openstack.org/13/666213/2/check/openstack-tox-py36/2c3274b/ : SUCCESS in 6m 19s\n- openstack-tox-py37 http://logs.openstack.org/13/666213/2/check/openstack-tox-py37/a94675e/ : SUCCESS in 9m 30s\n- build-openstack-releasenotes http://logs.openstack.org/13/666213/2/check/build-openstack-releasenotes/019f02c/html/ : SUCCESS in 6m 08s\n- manila-tox-genconfig http://logs.openstack.org/13/666213/2/check/manila-tox-genconfig/fe00ff8/ : SUCCESS in 6m 14s\n- openstack-tox-pylint http://logs.openstack.org/13/666213/2/check/openstack-tox-pylint/c8f1abb/ : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d347593d0acbb59daabac2774558cede1b0a6695","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-01 05:17:23.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/13/666213/2/check/dsvm-tempest-infinibox-nfs/edf729b : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":2},{"id":"219f527b9fa9e0078334e620a5c3e78d37db236e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-01 05:27:50.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/13/666213/2/check/DellEMC_Unity_Manila/47c73ae/DellEMC_Unity_Manila/98 : FAILURE in 1h 11m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"bdd21561f82db5d1854e37179c56fd79e6585314","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-07-01 06:07:38.000000000","message":"Patch Set 2:\n\n* nexenta-manila_ns5_nfs http://104.254.242.106/refs-changes-13-666213-2/manila_ns5_nfs : SUCCESS in 76m 44s","accounts_in_message":[],"_revision_number":2},{"id":"c3c4c44a0f52fba6cc01834ef052215e69a03557","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-07-01 06:11:21.000000000","message":"Patch Set 2:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-13-666213-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"b86334d0968b44368c65b6dc4b667b1a1aa182db","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-07-01 06:29:46.000000000","message":"Patch Set 2:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/13/666213/2/check/manila-master-client-pip-ssa03-dsvm-neutron/19a6af9 : FAILURE in 1h 47m 40s","accounts_in_message":[],"_revision_number":2},{"id":"011d2823423e8a290471f6e7b0a48ffccfa14999","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-07-01 07:01:01.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- manila-as13000 http://218.29.111.78:9000/13/666213/2/check/manila-as13000/bac7925 : SUCCESS in 1h 12m 47s\n- manila-instorage http://218.29.111.78:9000/13/666213/2/check/manila-instorage/8c10a5e : FAILURE in 1h 32m 02s","accounts_in_message":[],"_revision_number":2},{"id":"df12e4d591b1103e2b706646b8e41841c6530093","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-07-01 07:05:29.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/13/666213/2/check/DellEMC_PowerMAX_Manila/b36186b/DellEMC_PowerMAX_Manila/127 : FAILURE in 1h 52m 33s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"dc72e768edd1a90ca6563c69ecef98746bba4dfc","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-07-01 07:23:37.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/13/666213/2/check/DellEMC_VNX_Manila/586453b/DellEMC_VNX_Manila/83 : SUCCESS in 1h 57m 29s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"c744357d0c2142ec1aeec0640c66b9dd313030ef","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-01 10:28:33.000000000","message":"Patch Set 2: Code-Review-1\n\nPlease see comments in https://review.opendev.org/#/c/666212","accounts_in_message":[],"_revision_number":2},{"id":"c9c60353304bff07cf3b54e5608c19cb667818d6","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-02 02:06:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a649d1db21c49e0c2c4b0e826b43eb5d172b2e9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 02:29:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/13/666213/3/check/build-openstack-api-ref/2999bc0/html/ : SUCCESS in 3m 54s\n- openstack-tox-docs http://logs.openstack.org/13/666213/3/check/openstack-tox-docs/021b943/html/ : SUCCESS in 5m 06s\n- openstack-tox-cover http://logs.openstack.org/13/666213/3/check/openstack-tox-cover/bffddaf/cover/ : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/13/666213/3/check/openstack-tox-lower-constraints/981f4b2/ : FAILURE in 3m 03s\n- openstack-tox-pep8 http://logs.openstack.org/13/666213/3/check/openstack-tox-pep8/326c8be/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/13/666213/3/check/openstack-tox-py27/cc533d7/ : SUCCESS in 7m 00s\n- openstack-tox-py36 http://logs.openstack.org/13/666213/3/check/openstack-tox-py36/2c31ac8/ : SUCCESS in 6m 03s\n- openstack-tox-py37 http://logs.openstack.org/13/666213/3/check/openstack-tox-py37/02cce12/ : SUCCESS in 6m 50s\n- build-openstack-releasenotes http://logs.openstack.org/13/666213/3/check/build-openstack-releasenotes/cca5de4/html/ : SUCCESS in 6m 13s\n- manila-tox-genconfig http://logs.openstack.org/13/666213/3/check/manila-tox-genconfig/9fc6667/ : SUCCESS in 4m 09s\n- openstack-tox-pylint http://logs.openstack.org/13/666213/3/check/openstack-tox-pylint/3e4ea60/ : SUCCESS in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eaed5830cc2fdb8926fdd5059fa0dd7fc8f93361","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-02 02:37:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2be73b8382ba98b19277d3eeda053ba40bfef566","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 02:52:22.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/13/666213/3/check/build-openstack-api-ref/50ba37f/html/ : SUCCESS in 3m 50s\n- openstack-tox-docs http://logs.openstack.org/13/666213/3/check/openstack-tox-docs/d2185ad/html/ : SUCCESS in 4m 59s\n- openstack-tox-cover http://logs.openstack.org/13/666213/3/check/openstack-tox-cover/5e928e7/cover/ : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/13/666213/3/check/openstack-tox-lower-constraints/dd4578a/ : FAILURE in 3m 04s\n- openstack-tox-pep8 http://logs.openstack.org/13/666213/3/check/openstack-tox-pep8/bd17269/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/13/666213/3/check/openstack-tox-py27/cf5623c/ : SUCCESS in 5m 44s\n- openstack-tox-py36 http://logs.openstack.org/13/666213/3/check/openstack-tox-py36/469a577/ : SUCCESS in 5m 37s\n- openstack-tox-py37 http://logs.openstack.org/13/666213/3/check/openstack-tox-py37/9ad2714/ : SUCCESS in 5m 52s\n- build-openstack-releasenotes http://logs.openstack.org/13/666213/3/check/build-openstack-releasenotes/b109562/html/ : SUCCESS in 6m 19s\n- manila-tox-genconfig http://logs.openstack.org/13/666213/3/check/manila-tox-genconfig/492ea5c/ : SUCCESS in 4m 03s\n- openstack-tox-pylint http://logs.openstack.org/13/666213/3/check/openstack-tox-pylint/835649e/ : SUCCESS in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e2ffcd8b0ad3030895bbc1ca0b54a75d5336854d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-02 03:02:36.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/13/666213/3/check/dsvm-tempest-infinibox-nfs/a9152a0 : SUCCESS in 55m 24s","accounts_in_message":[],"_revision_number":3},{"id":"f24acf4557299ceb3cf0d2e511b97052cb8e03a6","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-02 03:09:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2b757cf5e4e784633e23618ada0bef87aa6a71ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 03:37:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/13/666213/4/check/build-openstack-api-ref/a97596b/html/ : SUCCESS in 5m 16s\n- openstack-tox-docs http://logs.openstack.org/13/666213/4/check/openstack-tox-docs/110bcf3/html/ : SUCCESS in 6m 15s\n- openstack-tox-cover http://logs.openstack.org/13/666213/4/check/openstack-tox-cover/ab9b7f6/cover/ : SUCCESS in 15m 33s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/13/666213/4/check/openstack-tox-lower-constraints/94203bd/ : FAILURE in 3m 22s\n- openstack-tox-pep8 http://logs.openstack.org/13/666213/4/check/openstack-tox-pep8/bfb33a4/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/13/666213/4/check/openstack-tox-py27/a2ad20d/ : SUCCESS in 10m 55s\n- openstack-tox-py36 http://logs.openstack.org/13/666213/4/check/openstack-tox-py36/ab7cd20/ : SUCCESS in 6m 32s\n- openstack-tox-py37 http://logs.openstack.org/13/666213/4/check/openstack-tox-py37/fcde94c/ : SUCCESS in 11m 42s\n- build-openstack-releasenotes http://logs.openstack.org/13/666213/4/check/build-openstack-releasenotes/34de05e/html/ : SUCCESS in 6m 30s\n- manila-tox-genconfig http://logs.openstack.org/13/666213/4/check/manila-tox-genconfig/088fe7b/ : SUCCESS in 3m 41s\n- openstack-tox-pylint http://logs.openstack.org/13/666213/4/check/openstack-tox-pylint/94b80bb/ : SUCCESS in 4m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d8a63caa8f31f53d9a8254dad6cafcbfdf979992","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-02 04:08:49.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/13/666213/4/check/dsvm-tempest-infinibox-nfs/c070412 : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":4},{"id":"68ec60b26aeadd1ac1fa6ebfc9707ea1998ed744","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-07-02 05:04:55.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/13/666213/4/check/DellEMC_PowerMAX_Manila/b2547cc/DellEMC_PowerMAX_Manila/135 : FAILURE in 1h 54m 46s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"34fb7b867eb61580b1c3d915e2789421e4c5e3b7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-02 05:10:46.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/13/666213/4/check/DellEMC_Unity_Manila/1e52421/DellEMC_Unity_Manila/107 : FAILURE in 2h 00m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"f3c0e6feb897458408f8342c64c4f6f22a3fa9a4","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-07-02 05:42:34.000000000","message":"Patch Set 4:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- manila-as13000 http://218.29.111.78:9000/13/666213/4/check/manila-as13000/d232613 : SUCCESS in 1h 07m 47s\n- manila-instorage http://218.29.111.78:9000/13/666213/4/check/manila-instorage/b7b72eb : FAILURE in 1h 48m 02s","accounts_in_message":[],"_revision_number":4},{"id":"a9d6fdf4a7f8ccdc7462be29408d0730930d0866","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-02 11:08:44.000000000","message":"Topic changed from Bug#3 to constraints","accounts_in_message":[],"_revision_number":4},{"id":"533727f6b0edfbdfbd667942b1e54081725ba376","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-02 11:10:50.000000000","message":"Patch Set 4:\n\nI set the Topic on this and the related review to \"constraints\" so that they are grouped together and not with arbitrary other reviews.","accounts_in_message":[],"_revision_number":4},{"id":"21ef143c49414ff1cc9d26a3686f66f53cf6a948","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-02 11:40:19.000000000","message":"Patch Set 4:\n\nThis looks like the right change, but if so we have unmasked an issue with the lower constraints[1]. It looks like this problem [2] and that nova was forced to up the lower constraint for psycopg2 because of it [3].\n\n[1] http://logs.openstack.org/13/666213/4/check/openstack-tox-lower-constraints/94203bd/job-output.txt.gz#_2019-07-02_03_17_23_329906\n\n[2] https://github.com/psycopg/psycopg2/issues/594\n\n[3] https://review.opendev.org/#/c/622972/","accounts_in_message":[],"_revision_number":4},{"id":"16597cab8a6aff062e0940a0760c32cd2056c33f","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-03 01:44:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"818a2e464d3065a1c5e3ce861180971f8afd04a4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-03 02:42:13.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/13/666213/5/check/dsvm-tempest-infinibox-nfs/676b25c : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":5},{"id":"cc697636ce84ae5ae9a22affc3ea0a35a08c1f6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 03:20:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/13/666213/5/check/build-openstack-api-ref/a513632/html/ : SUCCESS in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/13/666213/5/check/openstack-tox-docs/6820ff9/html/ : SUCCESS in 5m 43s\n- openstack-tox-cover http://logs.openstack.org/13/666213/5/check/openstack-tox-cover/b3ff19a/cover/ : SUCCESS in 11m 26s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/13/666213/5/check/openstack-tox-lower-constraints/8a39c4d/ : TIMED_OUT in 40m 52s\n- openstack-tox-pep8 http://logs.openstack.org/13/666213/5/check/openstack-tox-pep8/e5e78dc/ : SUCCESS in 5m 51s\n- openstack-tox-py27 http://logs.openstack.org/13/666213/5/check/openstack-tox-py27/53e5e18/ : SUCCESS in 6m 38s\n- openstack-tox-py36 http://logs.openstack.org/13/666213/5/check/openstack-tox-py36/61d110a/ : SUCCESS in 6m 42s\n- openstack-tox-py37 http://logs.openstack.org/13/666213/5/check/openstack-tox-py37/ae13116/ : SUCCESS in 7m 14s\n- requirements-check http://logs.openstack.org/13/666213/5/check/requirements-check/7674eab/ : FAILURE in 3m 01s\n- build-openstack-releasenotes http://logs.openstack.org/13/666213/5/check/build-openstack-releasenotes/14429a3/html/ : SUCCESS in 6m 28s\n- manila-tox-genconfig http://logs.openstack.org/13/666213/5/check/manila-tox-genconfig/72d0572/ : SUCCESS in 4m 14s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/13/666213/5/check/manila-tempest-dsvm-mysql-generic/b9ae352/ : FAILURE in 1h 24m 26s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/13/666213/5/check/manila-tempest-dsvm-postgres-container/29dfa5e/ : SUCCESS in 46m 55s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/13/666213/5/check/manila-tempest-dsvm-postgres-zfsonlinux/dc5d783/ : SUCCESS in 1h 06m 02s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/13/666213/5/check/manila-tempest-dsvm-postgres-generic-singlebackend/3fc2811/ : SUCCESS in 49m 48s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/13/666213/5/check/manila-tempest-dsvm-generic-no-share-servers/11f07e1/ : SUCCESS in 59m 13s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/13/666213/5/check/manila-tempest-dsvm-scenario/8f45093/ : SUCCESS in 1h 24m 35s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/13/666213/5/check/manila-tempest-minimal-dsvm-cephfs-native/a20631c/ : SUCCESS in 30m 17s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/13/666213/5/check/manila-tempest-minimal-dsvm-cephfs-nfs/f943aa5/ : SUCCESS in 1h 17m 53s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/13/666213/5/check/manila-tempest-minimal-dsvm-dummy/9b0da3f/ : SUCCESS in 36m 33s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/13/666213/5/check/manila-tempest-minimal-dsvm-lvm/b4392c8/ : FAILURE in 54m 17s\n- manila-grenade http://logs.openstack.org/13/666213/5/check/manila-grenade/b4c5363/ : SUCCESS in 30m 28s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/13/666213/5/check/manila-rally-multibackend/ee01d52/ : SUCCESS in 24m 59s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/13/666213/5/check/manila-rally-multibackend-no-ss/563f85d/ : SUCCESS in 17m 47s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/13/666213/5/check/openstack-tox-pylint/80b65f7/ : SUCCESS in 3m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f9efc0fd18d06a99221ad91d81c8df913d00b611","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-03 03:25:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e14a988a6b282793d76df182555e2418fbc93876","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-03 04:22:43.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/13/666213/6/check/dsvm-tempest-infinibox-nfs/1a88336 : SUCCESS in 56m 50s","accounts_in_message":[],"_revision_number":6},{"id":"731f1ad7feb68072074021c50951930cb52307b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 05:12:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/13/666213/6/check/build-openstack-api-ref/2bd8335/html/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/13/666213/6/check/openstack-tox-docs/c87ec08/html/ : SUCCESS in 4m 16s\n- openstack-tox-cover http://logs.openstack.org/13/666213/6/check/openstack-tox-cover/17d5b98/cover/ : SUCCESS in 9m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/13/666213/6/check/openstack-tox-lower-constraints/ec87c93/ : TIMED_OUT in 41m 14s\n- openstack-tox-pep8 http://logs.openstack.org/13/666213/6/check/openstack-tox-pep8/f98d5df/ : SUCCESS in 6m 48s\n- openstack-tox-py27 http://logs.openstack.org/13/666213/6/check/openstack-tox-py27/9ccb527/ : SUCCESS in 6m 16s\n- openstack-tox-py36 http://logs.openstack.org/13/666213/6/check/openstack-tox-py36/3b58fcc/ : SUCCESS in 8m 38s\n- openstack-tox-py37 http://logs.openstack.org/13/666213/6/check/openstack-tox-py37/4fd60b5/ : SUCCESS in 11m 13s\n- requirements-check http://logs.openstack.org/13/666213/6/check/requirements-check/c92ecc2/ : SUCCESS in 2m 39s\n- build-openstack-releasenotes http://logs.openstack.org/13/666213/6/check/build-openstack-releasenotes/a12216b/html/ : SUCCESS in 7m 14s\n- manila-tox-genconfig http://logs.openstack.org/13/666213/6/check/manila-tox-genconfig/be8a0dc/ : SUCCESS in 3m 49s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-mysql-generic/90fc14e/ : SUCCESS in 1h 34m 22s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-postgres-container/65df7f8/ : SUCCESS in 48m 51s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-postgres-zfsonlinux/b4a218e/ : SUCCESS in 1h 12m 35s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-postgres-generic-singlebackend/c0d0934/ : SUCCESS in 1h 05m 19s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-generic-no-share-servers/41b0010/ : SUCCESS in 59m 53s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-scenario/1622496/ : FAILURE in 1h 40m 41s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-cephfs-native/de433ee/ : SUCCESS in 24m 26s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-cephfs-nfs/b175cbe/ : SUCCESS in 1h 22m 29s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-dummy/17649e1/ : SUCCESS in 40m 30s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-lvm/17460ea/ : SUCCESS in 59m 59s\n- manila-grenade http://logs.openstack.org/13/666213/6/check/manila-grenade/9a0da57/ : SUCCESS in 24m 51s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/13/666213/6/check/manila-rally-multibackend/694254f/ : SUCCESS in 24m 16s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/13/666213/6/check/manila-rally-multibackend-no-ss/f5920c8/ : SUCCESS in 20m 20s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/13/666213/6/check/openstack-tox-pylint/0827119/ : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5f8dd2bb5c9ea3038cedfe05ce933f1591df94d4","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-07-03 08:19:33.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/13/666213/6/check/DellEMC_PowerMAX_Manila/d355062/DellEMC_PowerMAX_Manila/140 : SUCCESS in 2h 24m 26s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"8af43393e25884fb0ac21685de9d5273ed9d0ed5","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-03 13:43:18.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"4575603ff916be997983bc7e3396878b4184f9bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 15:21:36.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/13/666213/6/check/build-openstack-api-ref/672a374/html/ : SUCCESS in 6m 07s\n- openstack-tox-docs http://logs.openstack.org/13/666213/6/check/openstack-tox-docs/742a1f1/html/ : SUCCESS in 6m 02s\n- openstack-tox-cover http://logs.openstack.org/13/666213/6/check/openstack-tox-cover/5b60919/cover/ : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/13/666213/6/check/openstack-tox-lower-constraints/d2afc66/ : TIMED_OUT in 40m 53s\n- openstack-tox-pep8 http://logs.openstack.org/13/666213/6/check/openstack-tox-pep8/fbee9f8/ : SUCCESS in 6m 01s\n- openstack-tox-py27 http://logs.openstack.org/13/666213/6/check/openstack-tox-py27/6be9621/ : SUCCESS in 6m 17s\n- openstack-tox-py36 http://logs.openstack.org/13/666213/6/check/openstack-tox-py36/c2e19e0/ : SUCCESS in 7m 23s\n- openstack-tox-py37 http://logs.openstack.org/13/666213/6/check/openstack-tox-py37/0b445ae/ : SUCCESS in 7m 01s\n- requirements-check http://logs.openstack.org/13/666213/6/check/requirements-check/335f69f/ : SUCCESS in 3m 01s\n- build-openstack-releasenotes http://logs.openstack.org/13/666213/6/check/build-openstack-releasenotes/a55865c/html/ : SUCCESS in 7m 08s\n- manila-tox-genconfig http://logs.openstack.org/13/666213/6/check/manila-tox-genconfig/883dc5a/ : SUCCESS in 5m 52s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-mysql-generic/00b4dce/ : SUCCESS in 1h 28m 56s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-postgres-container/15ded03/ : SUCCESS in 48m 35s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-postgres-zfsonlinux/8b58a2f/ : FAILURE in 1h 01m 23s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-postgres-generic-singlebackend/37a18d9/ : SUCCESS in 56m 06s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-generic-no-share-servers/a2533ed/ : FAILURE in 1h 14m 50s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-scenario/a4b37d4/ : SUCCESS in 1h 27m 17s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-cephfs-native/6eafb3b/ : SUCCESS in 27m 05s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-cephfs-nfs/ecfd6e9/ : SUCCESS in 1h 31m 11s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-dummy/303b028/ : SUCCESS in 40m 51s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-lvm/051c864/ : SUCCESS in 58m 15s\n- manila-grenade http://logs.openstack.org/13/666213/6/check/manila-grenade/7bb3b92/ : SUCCESS in 23m 01s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/13/666213/6/check/manila-rally-multibackend/9b0b9bf/ : SUCCESS in 22m 44s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/13/666213/6/check/manila-rally-multibackend-no-ss/c97a792/ : SUCCESS in 24m 47s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/13/666213/6/check/openstack-tox-pylint/c15929e/ : SUCCESS in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a1704a906b0fcc9d4390de780784e764c4b7c937","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-04 02:05:24.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a9ba94af7c9efe988b86e797a8dc838ba78abbfb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 03:48:21.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/13/666213/6/check/build-openstack-api-ref/0906436/html/ : SUCCESS in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/13/666213/6/check/openstack-tox-docs/ffcce1e/html/ : SUCCESS in 6m 30s\n- openstack-tox-cover http://logs.openstack.org/13/666213/6/check/openstack-tox-cover/ecd0cd7/cover/ : SUCCESS in 13m 05s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/13/666213/6/check/openstack-tox-lower-constraints/dfbe3b6/ : TIMED_OUT in 41m 02s\n- openstack-tox-pep8 http://logs.openstack.org/13/666213/6/check/openstack-tox-pep8/3d8c7ce/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/13/666213/6/check/openstack-tox-py27/29f93a0/ : SUCCESS in 6m 26s\n- openstack-tox-py36 http://logs.openstack.org/13/666213/6/check/openstack-tox-py36/101db5c/ : SUCCESS in 7m 16s\n- openstack-tox-py37 http://logs.openstack.org/13/666213/6/check/openstack-tox-py37/5dc8b86/ : SUCCESS in 7m 25s\n- requirements-check http://logs.openstack.org/13/666213/6/check/requirements-check/4a93e4a/ : SUCCESS in 3m 32s\n- build-openstack-releasenotes http://logs.openstack.org/13/666213/6/check/build-openstack-releasenotes/425c4f9/html/ : SUCCESS in 6m 26s\n- manila-tox-genconfig http://logs.openstack.org/13/666213/6/check/manila-tox-genconfig/0619e95/ : SUCCESS in 4m 10s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-mysql-generic/8e42d17/ : SUCCESS in 1h 33m 47s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-postgres-container/1dbe3b6/ : SUCCESS in 58m 50s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-postgres-zfsonlinux/ff94156/ : SUCCESS in 1h 11m 20s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-postgres-generic-singlebackend/cc1c70e/ : FAILURE in 56m 04s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-generic-no-share-servers/19558a4/ : SUCCESS in 1h 00m 16s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-scenario/f0a270f/ : SUCCESS in 1h 27m 33s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-cephfs-native/d0afb33/ : SUCCESS in 24m 56s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-cephfs-nfs/dd27b56/ : SUCCESS in 1h 07m 56s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-dummy/5b463da/ : SUCCESS in 40m 43s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-lvm/d5c0587/ : SUCCESS in 1h 03m 34s\n- manila-grenade http://logs.openstack.org/13/666213/6/check/manila-grenade/882bf6a/ : SUCCESS in 19m 57s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/13/666213/6/check/manila-rally-multibackend/6035f71/ : SUCCESS in 21m 20s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/13/666213/6/check/manila-rally-multibackend-no-ss/05a14ac/ : SUCCESS in 21m 04s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/13/666213/6/check/openstack-tox-pylint/8721e06/ : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d55d42ba5b60dcf6aede206cd07909255dfc333e","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-04 06:00:07.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"d48a9ff334c6da05ce3071aa38b63c732fa69963","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 07:43:08.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/13/666213/6/check/build-openstack-api-ref/1f5676a/html/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/13/666213/6/check/openstack-tox-docs/c41af15/html/ : SUCCESS in 5m 10s\n- openstack-tox-cover http://logs.openstack.org/13/666213/6/check/openstack-tox-cover/b572499/cover/ : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/13/666213/6/check/openstack-tox-lower-constraints/53d0d4c/ : TIMED_OUT in 40m 55s\n- openstack-tox-pep8 http://logs.openstack.org/13/666213/6/check/openstack-tox-pep8/f71f455/ : SUCCESS in 6m 57s\n- openstack-tox-py27 http://logs.openstack.org/13/666213/6/check/openstack-tox-py27/8b700e0/ : SUCCESS in 6m 37s\n- openstack-tox-py36 http://logs.openstack.org/13/666213/6/check/openstack-tox-py36/d2eabab/ : SUCCESS in 5m 47s\n- openstack-tox-py37 http://logs.openstack.org/13/666213/6/check/openstack-tox-py37/fa8bb9a/ : SUCCESS in 8m 33s\n- requirements-check http://logs.openstack.org/13/666213/6/check/requirements-check/67bf447/ : SUCCESS in 3m 34s\n- build-openstack-releasenotes http://logs.openstack.org/13/666213/6/check/build-openstack-releasenotes/ced547d/html/ : SUCCESS in 6m 44s\n- manila-tox-genconfig http://logs.openstack.org/13/666213/6/check/manila-tox-genconfig/24ab3e1/ : SUCCESS in 5m 40s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-mysql-generic/318be01/ : FAILURE in 1h 20m 12s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-postgres-container/8039882/ : SUCCESS in 56m 34s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-postgres-zfsonlinux/f39f050/ : SUCCESS in 1h 09m 27s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-postgres-generic-singlebackend/c5d783e/ : SUCCESS in 1h 04m 18s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-generic-no-share-servers/8b455ed/ : FAILURE in 1h 00m 09s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/13/666213/6/check/manila-tempest-dsvm-scenario/eec113d/ : SUCCESS in 1h 35m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-cephfs-native/5e13faa/ : SUCCESS in 23m 52s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-cephfs-nfs/5f4bc1c/ : SUCCESS in 1h 28m 46s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-dummy/72a1798/ : SUCCESS in 41m 40s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/13/666213/6/check/manila-tempest-minimal-dsvm-lvm/4c006a7/ : SUCCESS in 1h 00m 10s\n- manila-grenade http://logs.openstack.org/13/666213/6/check/manila-grenade/0395388/ : SUCCESS in 25m 18s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/13/666213/6/check/manila-rally-multibackend/fb21a6a/ : SUCCESS in 29m 59s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/13/666213/6/check/manila-rally-multibackend-no-ss/077f77e/ : SUCCESS in 26m 56s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/13/666213/6/check/openstack-tox-pylint/da61901/ : SUCCESS in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"439cb24e48db2c01a5597bfd730bf26f768e4346","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-04 10:13:06.000000000","message":"Patch Set 6: Code-Review-1\n\nlower constraints job has a problem with eventlet at 0.18.2 [1]\n\nCheck the nova review mentioned earlier [2] and see that they bumped its version as well.\n\n[1] http://logs.openstack.org/13/666213/6/check/openstack-tox-lower-constraints/53d0d4c/job-output.txt.gz#_2019-07-04_06_11_30_575787\n\n[2] https://review.opendev.org/#/c/622972/","accounts_in_message":[],"_revision_number":6},{"id":"f374d903da46d14d4bf7e946aa0a07fd2a5da5ac","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-05 01:21:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"99cb69f2bec18bc64a8720ab2f0200cd8ce24003","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-05 02:21:45.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/13/666213/7/check/dsvm-tempest-infinibox-nfs/08334d8 : SUCCESS in 59m 41s","accounts_in_message":[],"_revision_number":7},{"id":"1e4e23e47bfd150bccc268e22db6cc6f4cf1414e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 03:12:09.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/13/666213/7/check/build-openstack-api-ref/6bda045/html/ : SUCCESS in 3m 54s\n- openstack-tox-docs http://logs.openstack.org/13/666213/7/check/openstack-tox-docs/8ebfa37/html/ : SUCCESS in 5m 01s\n- openstack-tox-cover http://logs.openstack.org/13/666213/7/check/openstack-tox-cover/3933fd1/cover/ : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/13/666213/7/check/openstack-tox-lower-constraints/094450b/ : TIMED_OUT in 40m 54s\n- openstack-tox-pep8 http://logs.openstack.org/13/666213/7/check/openstack-tox-pep8/e26e5e9/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/13/666213/7/check/openstack-tox-py27/5cdc570/ : SUCCESS in 7m 27s\n- openstack-tox-py36 http://logs.openstack.org/13/666213/7/check/openstack-tox-py36/91e96f0/ : SUCCESS in 6m 50s\n- openstack-tox-py37 http://logs.openstack.org/13/666213/7/check/openstack-tox-py37/ab2e942/ : SUCCESS in 7m 04s\n- requirements-check http://logs.openstack.org/13/666213/7/check/requirements-check/146274e/ : FAILURE in 3m 02s\n- build-openstack-releasenotes http://logs.openstack.org/13/666213/7/check/build-openstack-releasenotes/6dfb3a8/html/ : SUCCESS in 6m 12s\n- manila-tox-genconfig http://logs.openstack.org/13/666213/7/check/manila-tox-genconfig/e8c96f0/ : SUCCESS in 4m 23s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/13/666213/7/check/manila-tempest-dsvm-mysql-generic/c576bd4/ : FAILURE in 1h 13m 18s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/13/666213/7/check/manila-tempest-dsvm-postgres-container/4f6b3a2/ : SUCCESS in 40m 18s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/13/666213/7/check/manila-tempest-dsvm-postgres-zfsonlinux/b497fb2/ : SUCCESS in 54m 06s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/13/666213/7/check/manila-tempest-dsvm-postgres-generic-singlebackend/974c020/ : SUCCESS in 1h 37m 16s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/13/666213/7/check/manila-tempest-dsvm-generic-no-share-servers/b970ae0/ : SUCCESS in 1h 03m 52s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/13/666213/7/check/manila-tempest-dsvm-scenario/538b63f/ : SUCCESS in 1h 31m 43s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/13/666213/7/check/manila-tempest-minimal-dsvm-cephfs-native/7ac2faf/ : SUCCESS in 21m 24s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/13/666213/7/check/manila-tempest-minimal-dsvm-cephfs-nfs/4ff041d/ : SUCCESS in 1h 13m 44s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/13/666213/7/check/manila-tempest-minimal-dsvm-dummy/729467f/ : SUCCESS in 36m 33s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/13/666213/7/check/manila-tempest-minimal-dsvm-lvm/51fa623/ : SUCCESS in 1h 06m 13s\n- manila-grenade http://logs.openstack.org/13/666213/7/check/manila-grenade/65d2f90/ : SUCCESS in 21m 37s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/13/666213/7/check/manila-rally-multibackend/4ef3f8f/ : SUCCESS in 22m 40s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/13/666213/7/check/manila-rally-multibackend-no-ss/66739a6/ : SUCCESS in 21m 15s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/13/666213/7/check/openstack-tox-pylint/293d144/ : SUCCESS in 5m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"52eacc90a2a876c6cfb5352e3e83b2b7ba4547da","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-05 03:17:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d8f473ad25fefb0618490efb8f7ef669801169b0","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-07-05 03:49:02.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/13/666213/8/check/DellEMC_PowerMAX_Manila/b8dd9dd/DellEMC_PowerMAX_Manila/149 : FAILURE in 31m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"62087ab6b12c2779c28f375883c59c8b5fa52b32","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-05 04:26:07.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/13/666213/8/check/dsvm-tempest-infinibox-nfs/56708d7 : SUCCESS in 1h 08m 00s","accounts_in_message":[],"_revision_number":8},{"id":"551c6815ea9690fe965c65c3670090b0bba059c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 04:53:23.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/13/666213/8/check/build-openstack-api-ref/ab43c92/html/ : SUCCESS in 4m 20s\n- openstack-tox-docs http://logs.openstack.org/13/666213/8/check/openstack-tox-docs/6cbb3de/html/ : SUCCESS in 4m 30s\n- openstack-tox-cover http://logs.openstack.org/13/666213/8/check/openstack-tox-cover/f5c7916/cover/ : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/13/666213/8/check/openstack-tox-lower-constraints/6a62266/ : TIMED_OUT in 41m 00s\n- openstack-tox-pep8 http://logs.openstack.org/13/666213/8/check/openstack-tox-pep8/7721505/ : SUCCESS in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/13/666213/8/check/openstack-tox-py27/3bc5f1e/ : SUCCESS in 6m 11s\n- openstack-tox-py36 http://logs.openstack.org/13/666213/8/check/openstack-tox-py36/21937eb/ : SUCCESS in 5m 43s\n- openstack-tox-py37 http://logs.openstack.org/13/666213/8/check/openstack-tox-py37/d6565f6/ : SUCCESS in 6m 14s\n- requirements-check http://logs.openstack.org/13/666213/8/check/requirements-check/1047ce8/ : SUCCESS in 2m 54s\n- build-openstack-releasenotes http://logs.openstack.org/13/666213/8/check/build-openstack-releasenotes/835aac6/html/ : SUCCESS in 6m 37s\n- manila-tox-genconfig http://logs.openstack.org/13/666213/8/check/manila-tox-genconfig/58b5dc2/ : SUCCESS in 4m 04s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/13/666213/8/check/manila-tempest-dsvm-mysql-generic/67839b6/ : SUCCESS in 1h 29m 32s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/13/666213/8/check/manila-tempest-dsvm-postgres-container/80fb2ad/ : SUCCESS in 53m 13s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/13/666213/8/check/manila-tempest-dsvm-postgres-zfsonlinux/3a9a41b/ : FAILURE in 1h 08m 01s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/13/666213/8/check/manila-tempest-dsvm-postgres-generic-singlebackend/df5abd4/ : FAILURE in 56m 17s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/13/666213/8/check/manila-tempest-dsvm-generic-no-share-servers/f7f26da/ : SUCCESS in 1h 04m 43s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/13/666213/8/check/manila-tempest-dsvm-scenario/0ad2c58/ : SUCCESS in 1h 28m 07s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/13/666213/8/check/manila-tempest-minimal-dsvm-cephfs-native/d07b7a3/ : SUCCESS in 25m 52s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/13/666213/8/check/manila-tempest-minimal-dsvm-cephfs-nfs/9ac86ea/ : FAILURE in 1h 28m 47s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/13/666213/8/check/manila-tempest-minimal-dsvm-dummy/0762c91/ : SUCCESS in 43m 02s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/13/666213/8/check/manila-tempest-minimal-dsvm-lvm/b74b0ef/ : SUCCESS in 1h 00m 54s\n- manila-grenade http://logs.openstack.org/13/666213/8/check/manila-grenade/05f65b4/ : SUCCESS in 28m 30s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/13/666213/8/check/manila-rally-multibackend/3240877/ : SUCCESS in 27m 24s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/13/666213/8/check/manila-rally-multibackend-no-ss/6472482/ : SUCCESS in 23m 54s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/13/666213/8/check/openstack-tox-pylint/dc375ab/ : SUCCESS in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5849d886290eb11ccc103d8a46df6c4a415843b6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-05 05:23:05.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/13/666213/8/check/DellEMC_Unity_Manila/33a4504/DellEMC_Unity_Manila/119 : SUCCESS in 2h 05m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"519f70468d6f9d9b74318bca2bbb5d3a38a2ea84","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-12-24 06:28:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"962e880291c65601b8b0288f4b689b660327790f","revisions":{"dc5f5e3c8d4bea5702e5be6e6801d65d63adbc44":{"kind":"REWORK","_number":1,"created":"2019-06-19 04:29:10.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/13/666213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/666213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/666213/1"}}},"commit":{"parents":[{"commit":"302a142b63095b5c4ea94cfea9f59269a7047180","subject":"Merge \"[CI] Run scenario tests in the cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/302a142b63095b5c4ea94cfea9f59269a7047180"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-19 04:09:02.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-19 04:09:08.000000000","tz":480},"subject":"Modify the url of upper_constraints_file","message":"Modify the url of upper_constraints_file\n\nDepends-On: http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Ibcac3ebc1d665fd8357e4716e31ac7464161e497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dc5f5e3c8d4bea5702e5be6e6801d65d63adbc44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dc5f5e3c8d4bea5702e5be6e6801d65d63adbc44"}]},"branch":"refs/heads/master"},"a4b9aad67d31e5b248f9249256cc8e3d951a15df":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-01 04:16:20.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/13/666213/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/666213/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/666213/2"}}},"commit":{"parents":[{"commit":"302a142b63095b5c4ea94cfea9f59269a7047180","subject":"Merge \"[CI] Run scenario tests in the cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/302a142b63095b5c4ea94cfea9f59269a7047180"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-19 04:09:02.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-01 04:15:26.000000000","tz":480},"subject":"Modify the url of upper_constraints_file","message":"Modify the url of upper_constraints_file\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Ibcac3ebc1d665fd8357e4716e31ac7464161e497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a4b9aad67d31e5b248f9249256cc8e3d951a15df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a4b9aad67d31e5b248f9249256cc8e3d951a15df"}]},"branch":"refs/heads/master"},"fb342fbc0247a552535350c2691ebafb66eb44c3":{"kind":"REWORK","_number":3,"created":"2019-07-02 02:06:46.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/13/666213/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/666213/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/666213/3"}}},"commit":{"parents":[{"commit":"302a142b63095b5c4ea94cfea9f59269a7047180","subject":"Merge \"[CI] Run scenario tests in the cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/302a142b63095b5c4ea94cfea9f59269a7047180"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-19 04:09:02.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-02 02:06:31.000000000","tz":480},"subject":"Modify the url of upper_constraints_file","message":"Modify the url of upper_constraints_file\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Ibcac3ebc1d665fd8357e4716e31ac7464161e497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fb342fbc0247a552535350c2691ebafb66eb44c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fb342fbc0247a552535350c2691ebafb66eb44c3"}]},"branch":"refs/heads/master"},"356a96b09e06efe1372f950cdec8625b9b5ec6ff":{"kind":"REWORK","_number":4,"created":"2019-07-02 03:09:51.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/13/666213/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/666213/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/666213/4"}}},"commit":{"parents":[{"commit":"302a142b63095b5c4ea94cfea9f59269a7047180","subject":"Merge \"[CI] Run scenario tests in the cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/302a142b63095b5c4ea94cfea9f59269a7047180"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-19 04:09:02.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-02 03:09:16.000000000","tz":480},"subject":"Modify the url of upper_constraints_file","message":"Modify the url of upper_constraints_file\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Ibcac3ebc1d665fd8357e4716e31ac7464161e497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/356a96b09e06efe1372f950cdec8625b9b5ec6ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/356a96b09e06efe1372f950cdec8625b9b5ec6ff"}]},"branch":"refs/heads/master"},"872f541d6fd7a2f8249a20f3099688dfbb212dfb":{"kind":"REWORK","_number":5,"created":"2019-07-03 01:44:11.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/13/666213/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/666213/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/666213/5"}}},"commit":{"parents":[{"commit":"302a142b63095b5c4ea94cfea9f59269a7047180","subject":"Merge \"[CI] Run scenario tests in the cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/302a142b63095b5c4ea94cfea9f59269a7047180"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-19 04:09:02.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-03 01:43:53.000000000","tz":480},"subject":"Modify the url of upper_constraints_file","message":"Modify the url of upper_constraints_file\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Ibcac3ebc1d665fd8357e4716e31ac7464161e497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/872f541d6fd7a2f8249a20f3099688dfbb212dfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/872f541d6fd7a2f8249a20f3099688dfbb212dfb"}]},"branch":"refs/heads/master"},"9d6d1e55c8e3a1b35212b32ac5826b712905e1c0":{"kind":"REWORK","_number":6,"created":"2019-07-03 03:25:26.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/13/666213/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/666213/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/666213/6"}}},"commit":{"parents":[{"commit":"302a142b63095b5c4ea94cfea9f59269a7047180","subject":"Merge \"[CI] Run scenario tests in the cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/302a142b63095b5c4ea94cfea9f59269a7047180"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-19 04:09:02.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-03 03:24:17.000000000","tz":480},"subject":"Modify the url of upper_constraints_file","message":"Modify the url of upper_constraints_file\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Ibcac3ebc1d665fd8357e4716e31ac7464161e497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9d6d1e55c8e3a1b35212b32ac5826b712905e1c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9d6d1e55c8e3a1b35212b32ac5826b712905e1c0"}]},"branch":"refs/heads/master"},"04af2ce41bebc346a699f271ef8f5b77ad128634":{"kind":"REWORK","_number":7,"created":"2019-07-05 01:21:28.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/13/666213/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/666213/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/666213/7"}}},"commit":{"parents":[{"commit":"302a142b63095b5c4ea94cfea9f59269a7047180","subject":"Merge \"[CI] Run scenario tests in the cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/302a142b63095b5c4ea94cfea9f59269a7047180"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-19 04:09:02.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-05 01:20:34.000000000","tz":480},"subject":"Modify the url of upper_constraints_file","message":"Modify the url of upper_constraints_file\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Ibcac3ebc1d665fd8357e4716e31ac7464161e497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/04af2ce41bebc346a699f271ef8f5b77ad128634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/04af2ce41bebc346a699f271ef8f5b77ad128634"}]},"branch":"refs/heads/master"},"962e880291c65601b8b0288f4b689b660327790f":{"kind":"REWORK","_number":8,"created":"2019-07-05 03:17:40.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/13/666213/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/666213/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/666213/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/666213/8"}}},"commit":{"parents":[{"commit":"302a142b63095b5c4ea94cfea9f59269a7047180","subject":"Merge \"[CI] Run scenario tests in the cephfs-nfs job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/302a142b63095b5c4ea94cfea9f59269a7047180"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-06-19 04:09:02.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-07-05 03:17:22.000000000","tz":480},"subject":"Modify the url of upper_constraints_file","message":"Modify the url of upper_constraints_file\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Ibcac3ebc1d665fd8357e4716e31ac7464161e497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/962e880291c65601b8b0288f4b689b660327790f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/962e880291c65601b8b0288f4b689b660327790f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
