)]}'
{"id":"openstack%2Fmanila~667166","triplet_id":"openstack%2Fmanila~master~Iff070aad667d3d1b4aa83642099f8f410606bb9e","project":"openstack/manila","branch":"master","hashtags":[],"change_id":"Iff070aad667d3d1b4aa83642099f8f410606bb9e","subject":"DNM","status":"ABANDONED","created":"2019-06-24 17:31:25.000000000","updated":"2019-09-22 20:53:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"28d68f0eb393a34c86b7270b3b81e78f792d0ddf","_number":667166,"virtual_id_number":667166,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-27 16:14:05.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"date":"2019-07-27 16:08:00.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"date":"2019-07-27 15:35:21.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"date":"2019-07-27 12:35:29.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"date":"2019-07-27 15:10:46.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"date":"2019-07-27 14:33:26.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"date":"2019-07-29 00:50:13.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"date":"2019-07-27 14:34:29.000000000","_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"date":"2019-07-27 14:13:32.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":1,"date":"2019-07-27 13:14:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-27 12:35:29.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-07-27 13:14:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-27 14:13:32.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2019-07-27 14:33:26.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-27 14:34:29.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-27 15:10:46.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-27 15:35:21.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-27 16:08:00.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-27 16:14:05.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-29 00:50:13.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3178ce1b6f61f6043c4b5bbf458b12a59e2536c2","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-06-24 17:31:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8440bbf4d0a7c011f438a83cf3735ad5520df74","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-24 17:52:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/66/667166/1/check/DellEMC_Unity_Manila/69952ba/DellEMC_Unity_Manila/78 : FAILURE in 20m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d0ae3ce2ca921d78da502279c3d35eecb59a7ec5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-24 18:29:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/66/667166/1/check/dsvm-tempest-infinibox-nfs/cdfa001 : SUCCESS in 57m 58s","accounts_in_message":[],"_revision_number":1},{"id":"bb77c8c8d88a01ca4247fbaad53e792b538a7fd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 19:03:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/66/667166/1/check/openstack-tox-docs/5d99223/html/ : SUCCESS in 7m 11s\n- openstack-tox-cover http://logs.openstack.org/66/667166/1/check/openstack-tox-cover/9ecdb4c/cover/ : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/66/667166/1/check/openstack-tox-lower-constraints/6bf85c7/ : SUCCESS in 6m 31s\n- openstack-tox-pep8 http://logs.openstack.org/66/667166/1/check/openstack-tox-pep8/1373f05/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/66/667166/1/check/openstack-tox-py27/a24c088/ : SUCCESS in 9m 45s\n- openstack-tox-py36 http://logs.openstack.org/66/667166/1/check/openstack-tox-py36/f4f385e/ : SUCCESS in 6m 30s\n- openstack-tox-py37 http://logs.openstack.org/66/667166/1/check/openstack-tox-py37/a48e593/ : SUCCESS in 6m 53s\n- manila-tox-genconfig http://logs.openstack.org/66/667166/1/check/manila-tox-genconfig/f1ea0e8/ : SUCCESS in 5m 02s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/66/667166/1/check/manila-tempest-dsvm-mysql-generic/34538bc/ : FAILURE in 1h 15m 57s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/66/667166/1/check/manila-tempest-dsvm-postgres-container/e338450/ : SUCCESS in 56m 34s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/66/667166/1/check/manila-tempest-dsvm-postgres-zfsonlinux/f0ad043/ : SUCCESS in 1h 06m 50s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/66/667166/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/1e59da0/ : FAILURE in 54m 04s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/66/667166/1/check/manila-tempest-dsvm-generic-no-share-servers/cf8a941/ : SUCCESS in 1h 04m 17s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/66/667166/1/check/manila-tempest-dsvm-scenario/fa0c5d0/ : SUCCESS in 1h 25m 18s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/66/667166/1/check/manila-tempest-minimal-dsvm-cephfs-native/dd8e8c3/ : SUCCESS in 22m 25s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/66/667166/1/check/manila-tempest-minimal-dsvm-cephfs-nfs/3af8c8d/ : FAILURE in 1h 11m 23s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/66/667166/1/check/manila-tempest-minimal-dsvm-dummy/eb3855a/ : SUCCESS in 38m 20s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/66/667166/1/check/manila-tempest-minimal-dsvm-lvm/436cd92/ : SUCCESS in 58m 03s\n- manila-grenade http://logs.openstack.org/66/667166/1/check/manila-grenade/8696159/ : SUCCESS in 24m 31s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/66/667166/1/check/manila-rally-multibackend/d2237ae/ : SUCCESS in 24m 50s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/66/667166/1/check/manila-rally-multibackend-no-ss/1cb0da7/ : SUCCESS in 22m 30s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/66/667166/1/check/openstack-tox-pylint/c77bcf3/ : SUCCESS in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13a633d7b36b10a52d164a8da896318c22910109","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-06-24 19:23:59.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/66/667166/1/check/DellEMC_Unity_Manila/92e3794/DellEMC_Unity_Manila/79 : FAILURE in 20m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"75ba9c7643f06c49408e0d515f36c1fea4268cd4","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-06-24 19:46:11.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/66/667166/1/check/DellEMC_PowerMAX_Manila/0766d12/DellEMC_PowerMAX_Manila/108 : SUCCESS in 2h 14m 35s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"2a74ff37deff68e89b05034e946a603c78449e6d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-06-24 20:20:45.000000000","message":"Patch Set 1:\n\n* nexenta-manila_ns5_nfs http://104.254.242.106/refs-changes-66-667166-1/manila_ns5_nfs : SUCCESS in 76m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b7cb491f768010716d973c78c516a7fa0fd5a4d2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-24 20:27:42.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-ss http://13.56.179.158/logs/66/667166/1/upstream-check/manila-cDOT-ss/6316589/ : SUCCESS in 1h 09m 09s\n- manila-cDOT-no-ss http://13.56.179.158/logs/66/667166/1/upstream-check/manila-cDOT-no-ss/3737a2c/ : SUCCESS in 1h 23m 55s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f289bb02c64eac0d4d273b6cedbdf331c0ffa8ab","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-06-24 20:51:09.000000000","message":"Patch Set 1:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/66/667166/1/check/manila-master-client-pip-ssa03-dsvm-neutron/ee76727 : FAILURE in 1h 45m 13s","accounts_in_message":[],"_revision_number":1},{"id":"521599e6d39c459f54fb4e2449748c9427ab175f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-24 20:58:55.000000000","message":"Patch Set 1:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-66-667166-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"665b9657ed72f3a2628f469fbfa5513068205a0a","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-06-27 12:25:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"39bc3837944be93067fbd26e9e8013d7063fdaa8","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-06-27 13:47:55.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- manila-as13000 http://218.29.111.78:9000/66/667166/2/check/manila-as13000/3eb3bbd : SUCCESS in 1h 04m 24s\n- manila-instorage http://218.29.111.78:9000/66/667166/2/check/manila-instorage/8217a58 : FAILURE in 1h 21m 39s","accounts_in_message":[],"_revision_number":2},{"id":"1dda46fb07b8fc23c6082a7b6634642644e5e57c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-06-27 13:48:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/66/667166/2/check/dsvm-tempest-infinibox-nfs/d594a84 : SUCCESS in 1h 22m 04s","accounts_in_message":[],"_revision_number":2},{"id":"cdd1a98c367d5eed5a2c8574f0529b321587670a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:00:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/66/667166/2/check/openstack-tox-docs/2477386/html/ : SUCCESS in 5m 19s\n- openstack-tox-cover http://logs.openstack.org/66/667166/2/check/openstack-tox-cover/e35d3c5/cover/ : SUCCESS in 10m 44s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/66/667166/2/check/openstack-tox-lower-constraints/942e5de/ : SUCCESS in 6m 44s\n- openstack-tox-pep8 http://logs.openstack.org/66/667166/2/check/openstack-tox-pep8/2e7e43f/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/66/667166/2/check/openstack-tox-py27/0776b20/ : SUCCESS in 6m 23s\n- openstack-tox-py36 http://logs.openstack.org/66/667166/2/check/openstack-tox-py36/7b76f51/ : SUCCESS in 7m 20s\n- openstack-tox-py37 http://logs.openstack.org/66/667166/2/check/openstack-tox-py37/ff82716/ : SUCCESS in 6m 35s\n- manila-tox-genconfig http://logs.openstack.org/66/667166/2/check/manila-tox-genconfig/12ece54/ : SUCCESS in 4m 24s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/66/667166/2/check/manila-tempest-dsvm-mysql-generic/c7e9466/ : FAILURE in 1h 28m 35s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/66/667166/2/check/manila-tempest-dsvm-postgres-container/e18eeca/ : SUCCESS in 58m 29s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/66/667166/2/check/manila-tempest-dsvm-postgres-zfsonlinux/cb9eb63/ : SUCCESS in 1h 02m 19s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/66/667166/2/check/manila-tempest-dsvm-postgres-generic-singlebackend/590d961/ : SUCCESS in 56m 10s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/66/667166/2/check/manila-tempest-dsvm-generic-no-share-servers/f572d30/ : FAILURE in 1h 20m 42s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/66/667166/2/check/manila-tempest-dsvm-scenario/9848eef/ : SUCCESS in 1h 27m 36s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/66/667166/2/check/manila-tempest-minimal-dsvm-cephfs-native/04107b1/ : SUCCESS in 25m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/66/667166/2/check/manila-tempest-minimal-dsvm-cephfs-nfs/5ad2d4c/ : FAILURE in 1h 23m 39s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/66/667166/2/check/manila-tempest-minimal-dsvm-dummy/43016dd/ : SUCCESS in 39m 08s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/66/667166/2/check/manila-tempest-minimal-dsvm-lvm/ed04830/ : SUCCESS in 1h 14m 48s\n- manila-grenade http://logs.openstack.org/66/667166/2/check/manila-grenade/4da28c0/ : SUCCESS in 25m 50s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/66/667166/2/check/manila-rally-multibackend/54392f4/ : SUCCESS in 24m 51s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/66/667166/2/check/manila-rally-multibackend-no-ss/109a339/ : SUCCESS in 21m 25s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/66/667166/2/check/openstack-tox-pylint/d6d1ad2/ : SUCCESS in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e99cb6889cae12caee96bbabd35fa315ed362355","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-06-27 14:46:09.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/66/667166/2/check/DellEMC_PowerMAX_Manila/b53c0a8/DellEMC_PowerMAX_Manila/120 : SUCCESS in 2h 20m 14s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"682d947264931889b9022762e010227c5b59c961","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-06-27 15:20:30.000000000","message":"Patch Set 2:\n\n* nexenta-manila_ns5_nfs http://104.254.242.106/refs-changes-66-667166-2/manila_ns5_nfs : SUCCESS in 78m 59s","accounts_in_message":[],"_revision_number":2},{"id":"77f8f26d77d9f048b42dd929c62c36195cb5e946","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-06-27 15:26:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss http://13.56.179.158/logs/66/667166/2/upstream-check/manila-cDOT-ss/0520b3b/ : FAILURE in 3m 11s\n- manila-cDOT-no-ss http://13.56.179.158/logs/66/667166/2/upstream-check/manila-cDOT-no-ss/504136b/ : SUCCESS in 1h 24m 48s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"4521923850d261dc265ae0e311965972e63355b0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-27 15:39:19.000000000","message":"Patch Set 2:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-66-667166-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"c97fd06fcb91e03292ae9c2ba66bfdd164b5c3f6","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-06-27 15:54:39.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/66/667166/2/check/DellEMC_VNX_Manila/a7443d5/DellEMC_VNX_Manila/79 : SUCCESS in 1h 53m 35s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"b27df8a8add957549d6bf9a3fba9ec77dbbba76b","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-06-27 15:54:53.000000000","message":"Patch Set 2:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/66/667166/2/check/manila-master-client-pip-ssa03-dsvm-neutron/4014373 : FAILURE in 1h 49m 56s","accounts_in_message":[],"_revision_number":2},{"id":"50d2aa30d6a34fb115b7dcd13eaeec0bdefbff1f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-16 13:11:28.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"adc01775b984f124f8536fd6d49a6066d258f23c","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-16 13:13:58.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"07f4406b764ef8b08fab9c88c7d980567087043e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-16 13:17:39.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/66/667166/3/check/dsvm-tempest-infinibox-nfs/f42b286 : FAILURE in 5m 42s","accounts_in_message":[],"_revision_number":3},{"id":"1c7038bd86eb6a536841f6ddafc7399451329802","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:08:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/66/667166/3/check/openstack-tox-docs/829e14c/html/ : SUCCESS in 5m 20s\n- openstack-tox-cover http://logs.openstack.org/66/667166/3/check/openstack-tox-cover/4f8990e/cover/ : SUCCESS in 15m 01s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/66/667166/3/check/openstack-tox-lower-constraints/97918bf/ : SUCCESS in 11m 34s\n- openstack-tox-pep8 http://logs.openstack.org/66/667166/3/check/openstack-tox-pep8/0a0c534/ : SUCCESS in 10m 34s\n- openstack-tox-py27 http://logs.openstack.org/66/667166/3/check/openstack-tox-py27/c86c8bf/ : SUCCESS in 6m 32s\n- openstack-tox-py36 http://logs.openstack.org/66/667166/3/check/openstack-tox-py36/5ce6aa8/ : SUCCESS in 6m 41s\n- openstack-tox-py37 http://logs.openstack.org/66/667166/3/check/openstack-tox-py37/cbbca6e/ : SUCCESS in 6m 59s\n- manila-tox-genconfig http://logs.openstack.org/66/667166/3/check/manila-tox-genconfig/7784afe/ : SUCCESS in 9m 31s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/66/667166/3/check/manila-tempest-dsvm-mysql-generic/10680af/ : FAILURE in 1h 33m 35s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/66/667166/3/check/manila-tempest-dsvm-postgres-container/2237371/ : SUCCESS in 1h 04m 24s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/66/667166/3/check/manila-tempest-dsvm-postgres-zfsonlinux/d98b153/ : FAILURE in 1h 23m 55s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/66/667166/3/check/manila-tempest-dsvm-postgres-generic-singlebackend/25ed2bf/ : SUCCESS in 1h 09m 51s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/66/667166/3/check/manila-tempest-dsvm-generic-no-share-servers/97fd596/ : FAILURE in 1h 04m 19s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/66/667166/3/check/manila-tempest-dsvm-scenario/b464456/ : FAILURE in 1h 47m 37s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/66/667166/3/check/manila-tempest-minimal-dsvm-cephfs-native/9c07e8a/ : SUCCESS in 22m 58s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/66/667166/3/check/manila-tempest-minimal-dsvm-cephfs-nfs/fef7290/ : FAILURE in 1h 20m 13s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/66/667166/3/check/manila-tempest-minimal-dsvm-dummy/c68afce/ : SUCCESS in 42m 34s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/66/667166/3/check/manila-tempest-minimal-dsvm-lvm/f7eecab/ : SUCCESS in 1h 09m 49s\n- manila-grenade http://logs.openstack.org/66/667166/3/check/manila-grenade/8a8fa34/ : SUCCESS in 30m 02s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/66/667166/3/check/manila-rally-multibackend/6b96f4d/ : SUCCESS in 26m 29s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/66/667166/3/check/manila-rally-multibackend-no-ss/dfa6440/ : SUCCESS in 23m 29s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/66/667166/3/check/openstack-tox-pylint/7a2169b/ : SUCCESS in 6m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4e63f454e3655be52aacf2f7c3d86fb58bf3107e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-07-16 16:18:23.000000000","message":"Patch Set 3:\n\n* nexenta-manila_ns5_nfs http://104.254.242.106/refs-changes-66-667166-3/manila_ns5_nfs : FAILURE in 67m 55s","accounts_in_message":[],"_revision_number":3},{"id":"1521b032550104da3021dd6b3dd4c6cf29aeef50","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-07-16 16:56:36.000000000","message":"Patch Set 3:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/66/667166/3/check/manila-master-client-pip-ssa03-dsvm-neutron/41f4030 : FAILURE in 1h 45m 16s","accounts_in_message":[],"_revision_number":3},{"id":"b0825a6637af7b02b273fe98289a6c91891cdbd8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-16 17:07:25.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-ss http://13.56.179.158/logs/66/667166/3/upstream-check/manila-cDOT-ss/f48acde/ : SUCCESS in 1h 12m 59s\n- manila-cDOT-no-ss http://13.56.179.158/logs/66/667166/3/upstream-check/manila-cDOT-no-ss/76b10d8/ : SUCCESS in 1h 26m 36s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"048e1e3c816f32f16173887c7b979694a1e68e65","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-16 17:21:42.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/66/667166/3/check/DellEMC_Unity_Manila/b5b9c0a/DellEMC_Unity_Manila/167 : FAILURE in 28m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"0c133391460e3f6f98a38a6197d059e1fb1d3fb5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-07-16 18:04:23.000000000","message":"Patch Set 3:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-66-667166-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"ce50910348b0246b7e6f6cef067361c0a03e6f75","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-07-16 19:55:57.000000000","message":"Patch Set 3:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- manila-as13000 http://218.29.111.78:9000/66/667166/3/check/manila-as13000/e89bbf6 : FAILURE in 28m 58s\n- manila-instorage http://218.29.111.78:9000/66/667166/3/check/manila-instorage/05786c1 : FAILURE in 1h 01m 18s","accounts_in_message":[],"_revision_number":3},{"id":"43faa87b95144468978f438c2118c02c0cfb7407","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-07-16 20:10:52.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/66/667166/3/check/DellEMC_VNX_Manila/a32349c/DellEMC_VNX_Manila/115 : SUCCESS in 1h 56m 37s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"638d417974c4c56b639da451270c258a775ea837","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-27 11:37:49.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"7efbc94518ed9a2810b95680d81a7d55cac8b060","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-27 12:35:29.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/66/667166/4/check/dsvm-tempest-infinibox-nfs/f8de746 : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":4},{"id":"6b0a856e3323d0970d07711ae96dadf2493decc0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 13:14:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/66/667166/4/check/openstack-tox-docs/ac5d429/html/ : SUCCESS in 5m 58s\n- openstack-tox-cover http://logs.openstack.org/66/667166/4/check/openstack-tox-cover/c21e56a/cover/ : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/66/667166/4/check/openstack-tox-lower-constraints/41062cb/ : SUCCESS in 8m 41s\n- openstack-tox-pep8 http://logs.openstack.org/66/667166/4/check/openstack-tox-pep8/078f151/ : SUCCESS in 5m 35s\n- openstack-tox-py27 http://logs.openstack.org/66/667166/4/check/openstack-tox-py27/053fa48/ : SUCCESS in 9m 36s\n- openstack-tox-py36 http://logs.openstack.org/66/667166/4/check/openstack-tox-py36/357d1a3/ : SUCCESS in 6m 26s\n- openstack-tox-py37 http://logs.openstack.org/66/667166/4/check/openstack-tox-py37/3f1f080/ : SUCCESS in 9m 20s\n- manila-tox-genconfig http://logs.openstack.org/66/667166/4/check/manila-tox-genconfig/b8173d3/ : SUCCESS in 4m 18s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/66/667166/4/check/manila-tempest-dsvm-mysql-generic/668fb31/ : SUCCESS in 1h 31m 06s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/66/667166/4/check/manila-tempest-dsvm-postgres-container/27de5bd/ : SUCCESS in 54m 13s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/66/667166/4/check/manila-tempest-dsvm-postgres-zfsonlinux/35738dc/ : FAILURE in 59m 18s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/66/667166/4/check/manila-tempest-dsvm-postgres-generic-singlebackend/7fbfb49/ : SUCCESS in 58m 51s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/66/667166/4/check/manila-tempest-dsvm-generic-no-share-servers/16d554f/ : FAILURE in 1h 10m 20s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/66/667166/4/check/manila-tempest-dsvm-scenario/bdac159/ : FAILURE in 1h 11m 44s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native http://logs.openstack.org/66/667166/4/check/manila-tempest-minimal-dsvm-cephfs-native/ffb6578/ : SUCCESS in 33m 42s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs http://logs.openstack.org/66/667166/4/check/manila-tempest-minimal-dsvm-cephfs-nfs/68d0d6a/ : SUCCESS in 1h 23m 45s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/66/667166/4/check/manila-tempest-minimal-dsvm-dummy/f2df968/ : SUCCESS in 44m 18s\n- manila-tempest-minimal-dsvm-dummy-py2 http://logs.openstack.org/66/667166/4/check/manila-tempest-minimal-dsvm-dummy-py2/628ce05/ : SUCCESS in 40m 26s\n- manila-tempest-minimal-dsvm-lvm http://logs.openstack.org/66/667166/4/check/manila-tempest-minimal-dsvm-lvm/c48c152/ : SUCCESS in 58m 50s\n- manila-grenade http://logs.openstack.org/66/667166/4/check/manila-grenade/dc9664a/ : SUCCESS in 29m 25s (non-voting)\n- manila-rally-multibackend http://logs.openstack.org/66/667166/4/check/manila-rally-multibackend/f703d09/ : SUCCESS in 21m 56s (non-voting)\n- manila-rally-multibackend-no-ss http://logs.openstack.org/66/667166/4/check/manila-rally-multibackend-no-ss/0ade2ae/ : SUCCESS in 25m 06s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/66/667166/4/check/openstack-tox-pylint/b7c37e1/ : SUCCESS in 3m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"77cf815e331d2264e9a4b4ed818a79f4478abeae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-07-27 14:13:32.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila http://publiclogs.emc.com/66/667166/4/check/DellEMC_Unity_Manila/2673d4e/DellEMC_Unity_Manila/252 : FAILURE in 1h 18m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"d80fc58f3f631d2ed208e45153f7c1470a537294","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-07-27 14:33:26.000000000","message":"Patch Set 4:\n\n* nexenta-manila_ns5_nfs http://104.254.242.106/refs-changes-66-667166-4/manila_ns5_nfs : SUCCESS in 78m 57s","accounts_in_message":[],"_revision_number":4},{"id":"a313794327d555dcc36331d32095cb78f3e117b4","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-07-27 14:34:29.000000000","message":"Patch Set 4:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- manila-as13000 http://218.29.111.78:9000/66/667166/4/check/manila-as13000/6aee76e : FAILURE in 39m 46s\n- manila-instorage http://218.29.111.78:9000/66/667166/4/check/manila-instorage/91c7952 : SUCCESS in 1h 16m 43s","accounts_in_message":[],"_revision_number":4},{"id":"9ec442911bff90ad751969ba2c89c255e2647d0d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-07-27 15:10:46.000000000","message":"Patch Set 4:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/66/667166/4/check/manila-master-client-pip-ssa03-dsvm-neutron/0a7b2a2 : FAILURE in 1h 53m 53s","accounts_in_message":[],"_revision_number":4},{"id":"93facfb718d9ef7887084fb110dd8935e01955f5","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-07-27 15:35:21.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/66/667166/4/check/DellEMC_VNX_Manila/07f5763/DellEMC_VNX_Manila/164 : SUCCESS in 1h 58m 05s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"65cce6bd586e0d88a76384911f6542f6e90c90d8","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-07-27 16:08:00.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/66/667166/4/check/DellEMC_PowerMAX_Manila/6fb4a83/DellEMC_PowerMAX_Manila/256 : SUCCESS in 2h 30m 49s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"10b3015ff8d044ac3422b755e7282fe9389903f9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-07-27 16:14:05.000000000","message":"Patch Set 4:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-66-667166-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"c00b7908a259c09cc1dfd92b8f9e39a8a4f31ea0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-07-29 00:50:13.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-ss http://13.56.161.150/logs/66/667166/4/upstream-check/manila-cDOT-ss/1c7aa29/ : FAILURE in 4m 27s\n- manila-cDOT-no-ss http://13.56.161.150/logs/66/667166/4/upstream-check/manila-cDOT-no-ss/2271a80/ : FAILURE in 3m 22s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"28d68f0eb393a34c86b7270b3b81e78f792d0ddf","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-09-22 20:53:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"37f16a40dfa79377a1c22d6d7fd7e023b578da71","revisions":{"912456da875a9c47e4b34ba0a7899609f3b8db8c":{"kind":"REWORK","_number":1,"created":"2019-06-24 17:31:25.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/66/667166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/66/667166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/667166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/667166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/667166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/66/667166/1"}}},"commit":{"parents":[{"commit":"5d1ef1570b3ae84d3cdeed03feda132297774824","subject":"Fix typo in Manila docs in manila.rst file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5d1ef1570b3ae84d3cdeed03feda132297774824"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-06-24 17:31:17.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-06-24 17:31:17.000000000","tz":-240},"subject":"DNM","message":"DNM\n\nChange-Id: Iff070aad667d3d1b4aa83642099f8f410606bb9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/912456da875a9c47e4b34ba0a7899609f3b8db8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/912456da875a9c47e4b34ba0a7899609f3b8db8c"}]},"branch":"refs/heads/master"},"2c6c681214dbc694a92abc566fa0c0e3028d982e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-27 12:25:43.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/66/667166/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/66/667166/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/667166/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/667166/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/667166/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/66/667166/2"}}},"commit":{"parents":[{"commit":"8a1b465ea83789665b12f10a6bee382568b86370","subject":"Merge \"Adding documentation for User Messages in Manila Documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8a1b465ea83789665b12f10a6bee382568b86370"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-06-24 17:31:17.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-06-27 12:25:43.000000000","tz":0},"subject":"DNM","message":"DNM\n\nChange-Id: Iff070aad667d3d1b4aa83642099f8f410606bb9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2c6c681214dbc694a92abc566fa0c0e3028d982e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2c6c681214dbc694a92abc566fa0c0e3028d982e"}]},"branch":"refs/heads/master"},"38921d548343a5d9c6ca856988f32bf8c2a6acbe":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-16 13:11:28.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/66/667166/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/66/667166/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/667166/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/667166/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/667166/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/66/667166/3"}}},"commit":{"parents":[{"commit":"19d169e083ed055505943dda74e0401916d2cc66","subject":"Merge \"NeutronBindNetworkPlugin: fix multi segment mtu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/19d169e083ed055505943dda74e0401916d2cc66"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-06-24 17:31:17.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-16 13:11:28.000000000","tz":0},"subject":"DNM","message":"DNM\n\nChange-Id: Iff070aad667d3d1b4aa83642099f8f410606bb9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/38921d548343a5d9c6ca856988f32bf8c2a6acbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/38921d548343a5d9c6ca856988f32bf8c2a6acbe"}]},"branch":"refs/heads/master"},"37f16a40dfa79377a1c22d6d7fd7e023b578da71":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-27 11:37:49.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/66/667166/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/66/667166/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/667166/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/667166/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/667166/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/66/667166/4"}}},"commit":{"parents":[{"commit":"cc830350f9089999c052fa1b30501f2380e133bc","subject":"Merge \"[api-ref] Correct share metadata API ref\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cc830350f9089999c052fa1b30501f2380e133bc"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-06-24 17:31:17.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-27 11:37:49.000000000","tz":0},"subject":"DNM","message":"DNM\n\nChange-Id: Iff070aad667d3d1b4aa83642099f8f410606bb9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/37f16a40dfa79377a1c22d6d7fd7e023b578da71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/37f16a40dfa79377a1c22d6d7fd7e023b578da71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
