)]}'
{"id":"openstack%2Fmanila~668908","triplet_id":"openstack%2Fmanila~stable%2Fqueens~I90cda99692359dfb7d228a031688b2ce245365ff","project":"openstack/manila","branch":"stable/queens","hashtags":[],"change_id":"I90cda99692359dfb7d228a031688b2ce245365ff","subject":"DNM - test only","status":"ABANDONED","created":"2019-07-03 13:45:19.000000000","updated":"2019-07-03 23:09:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9cbe819c07f1c39b9e3ba9366cf2eed9a3ac5d60","_number":668908,"virtual_id_number":668908,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-03 16:09:04.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"date":"2019-07-03 14:13:10.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":-1,"date":"2019-07-03 15:06:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 14:13:10.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-07-03 15:06:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-03 16:09:04.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"88bb4c9ca4aa221df05556ab68bc3a944865970d","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-03 13:45:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa9cd991c27529278e4d38dce99fc6c3053c5d1e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-07-03 14:13:10.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/08/668908/1/check/dsvm-tempest-infinibox-nfs/012d856 : FAILURE in 25m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d38463c1e597b1e9b6c28c947955d7877caa00a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 15:06:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/08/668908/1/check/build-openstack-sphinx-docs/9dd260f/ : FAILURE in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/08/668908/1/check/openstack-tox-pep8/035e60a/ : RETRY_LIMIT in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/08/668908/1/check/openstack-tox-py27/2e13bbb/ : RETRY_LIMIT in 2m 35s\n- openstack-tox-py35 http://logs.openstack.org/08/668908/1/check/openstack-tox-py35/c5fb9a9/ : RETRY_LIMIT in 2m 09s\n- manila-tox-genconfig http://logs.openstack.org/08/668908/1/check/manila-tox-genconfig/afacf30/ : RETRY_LIMIT in 2m 05s\n- manila-tempest-dsvm-mysql-generic http://logs.openstack.org/08/668908/1/check/manila-tempest-dsvm-mysql-generic/c3abe73/ : SUCCESS in 1h 02m 34s (non-voting)\n- manila-tempest-dsvm-postgres-container http://logs.openstack.org/08/668908/1/check/manila-tempest-dsvm-postgres-container/24f6a59/ : SUCCESS in 41m 02s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux http://logs.openstack.org/08/668908/1/check/manila-tempest-dsvm-postgres-zfsonlinux/6ea0908/ : SUCCESS in 1h 00m 34s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend http://logs.openstack.org/08/668908/1/check/manila-tempest-dsvm-postgres-generic-singlebackend/35fa159/ : SUCCESS in 43m 54s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers http://logs.openstack.org/08/668908/1/check/manila-tempest-dsvm-generic-no-share-servers/2a45a8d/ : SUCCESS in 1h 03m 53s (non-voting)\n- manila-tempest-dsvm-scenario http://logs.openstack.org/08/668908/1/check/manila-tempest-dsvm-scenario/a37d1bd/ : SUCCESS in 1h 11m 00s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 http://logs.openstack.org/08/668908/1/check/manila-tempest-minimal-dsvm-cephfs-native-centos-7/35cfddf/ : FAILURE in 20m 53s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 http://logs.openstack.org/08/668908/1/check/manila-tempest-minimal-dsvm-cephfs-nfs-centos-7/8c19b28/ : FAILURE in 23m 59s (non-voting)\n- manila-tempest-minimal-dsvm-dummy http://logs.openstack.org/08/668908/1/check/manila-tempest-minimal-dsvm-dummy/864e2c1/ : SUCCESS in 44m 13s\n- manila-tempest-minimal-dsvm-lvm-centos-7 http://logs.openstack.org/08/668908/1/check/manila-tempest-minimal-dsvm-lvm-centos-7/71247a3/ : FAILURE in 24m 58s\n- grenade-dsvm-manila http://logs.openstack.org/08/668908/1/check/grenade-dsvm-manila/6defa1a/ : SUCCESS in 22m 13s (non-voting)\n- rally-dsvm-manila-multibackend http://logs.openstack.org/08/668908/1/check/rally-dsvm-manila-multibackend/9133432/ : FAILURE in 19m 47s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss http://logs.openstack.org/08/668908/1/check/rally-dsvm-manila-multibackend-no-ss/d2e493f/ : FAILURE in 18m 34s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/08/668908/1/check/openstack-tox-cover/34bd49a/ : RETRY_LIMIT in 2m 08s (non-voting)\n- openstack-tox-pylint http://logs.openstack.org/08/668908/1/check/openstack-tox-pylint/23b3eed/ : RETRY_LIMIT in 2m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b5731a9d8eef8557db911d63767ed1ad474f5c4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-07-03 16:09:04.000000000","message":"Patch Set 1:\n\nWe need the bindep.txt changes on this branch: https://review.opendev.org/#/c/667721/\nThat\u0027s held up because the LVM job is failing. I\u0027m still stuck trying to backport the devstack fix for that. Currently at stable/rocky: https://review.opendev.org/#/c/668787/","accounts_in_message":[],"_revision_number":1},{"id":"9cbe819c07f1c39b9e3ba9366cf2eed9a3ac5d60","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-07-03 23:09:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6ea5844920ffcbb2b98a8cdc7a40a97595d369b","revisions":{"a6ea5844920ffcbb2b98a8cdc7a40a97595d369b":{"kind":"REWORK","_number":1,"created":"2019-07-03 13:45:19.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/08/668908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/08/668908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/668908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/668908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/668908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/08/668908/1"}}},"commit":{"parents":[{"commit":"390aa443fe40edb3b02fbc962747ad60870b8536","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/390aa443fe40edb3b02fbc962747ad60870b8536"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-03 13:45:04.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2019-07-03 13:45:04.000000000","tz":-240},"subject":"DNM - test only","message":"DNM - test only\n\nChange-Id: I90cda99692359dfb7d228a031688b2ce245365ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a6ea5844920ffcbb2b98a8cdc7a40a97595d369b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a6ea5844920ffcbb2b98a8cdc7a40a97595d369b"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
