)]}'
{"id":"openstack%2Fmanila~680328","triplet_id":"openstack%2Fmanila~master~Iacbde5c4337f892fe1b9ec903d32ca3d138774a3","project":"openstack/manila","branch":"master","topic":"eventlet","hashtags":[],"change_id":"Iacbde5c4337f892fe1b9ec903d32ca3d138774a3","subject":"Blacklist eventlet 0.21.0,0.23.0,0.25.0","status":"ABANDONED","created":"2019-09-05 08:33:12.000000000","updated":"2020-10-04 19:43:58.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"40c88d8a773682fa7e884a711a40c6c86e0e8713","_number":680328,"virtual_id_number":680328,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-06 03:23:17.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"date":"2019-09-06 05:46:34.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:jenkins-gerrit-trigger","date":"2019-09-06 07:14:15.000000000","_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"date":"2019-09-06 04:15:32.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"date":"2019-09-06 03:40:02.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"date":"2019-09-06 04:46:12.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"date":"2019-09-06 07:39:29.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":1,"date":"2019-09-06 03:23:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-06 05:17:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 03:23:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-06 03:23:17.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-06 03:40:02.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-06 04:15:32.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-09-06 04:46:12.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-06 05:17:46.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-09-06 05:46:34.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-06 07:14:15.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2019-09-06 07:39:29.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38a9af22d234e009e0684453c0221b447ede442b","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-05 08:33:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1353f60a6b3809c823648103a6c85e32e9735fd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 08:53:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acac39f04f9b4eeb930d9c66b67bd9e7 : SUCCESS in 10m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78ca5274012a4adf87199f36cbe15295 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3d340398c34429684e55e1414fbb96a : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a8e078475d04cb6af47717b8e8f04bd : SUCCESS in 8m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c36c4debf9284774b2cf9a2f9f00b12e : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9181af2675b946cd97a6b27c092e70fe : SUCCESS in 7m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a4d25dfb53a407db20f9643704521b0 : SUCCESS in 7m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/39440814791c4b919a145cecbf07e35a : SUCCESS in 3m 59s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ebb2599eb2f94d8dbd142023742c8d02 : SUCCESS in 4m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e4706461fc314ef5a12f3ab0b9059e1d : SUCCESS in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52b04d663047ea7306ca1ae0c26e5a491e451d5d","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-09-05 09:11:37.000000000","message":"Patch Set 1:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/28/680328/1/check/manila-master-client-pip-ssa03-dsvm-neutron/cd8bdc8 : FAILURE in 15m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6c30be8ece2e34596464f8cda5b5a0ccbcf1cd4c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-09-05 09:34:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/28/680328/1/check/dsvm-tempest-infinibox-nfs/b106b35 : SUCCESS in 1h 01m 06s","accounts_in_message":[],"_revision_number":1},{"id":"90c66d08de03a2c4918066dd9e6850b07078fed1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-09-05 10:15:49.000000000","message":"Patch Set 1:\n\n* nexenta-manila_ns5_nfs http://104.254.242.106/refs-changes-28-680328-1/manila_ns5_nfs : SUCCESS in 81m 58s","accounts_in_message":[],"_revision_number":1},{"id":"13925147d3b521ba167e8a8819181e61530c3a93","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-09-05 10:35:40.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/597/680328/1/check/infortrend-manila-ci/d3fd00192b0211a4858eaa7b3a7541f2972f8485 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"39695c97f1f5f785dee137a3b99e36151ca4b6a1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-09-05 11:08:11.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/28/680328/1/check/DellEMC_VNX_Manila/8c6fe74/DellEMC_VNX_Manila/251 : SUCCESS in 2h 14m 18s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"2b0a8addd5b84224ac5ed85fc1e8de0a95676587","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-05 11:24:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd6a26823f69a453225a2ffd9e289eddc7827d80","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-06 03:03:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"267a99f25e198de0a5274da461e03a1884b9495e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-09-06 03:03:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/28/680328/2/check/DellEMC_PowerMAX_Manila/f24616d/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"6c287355360cbf0f04726bd53b6bc97111c79439","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 03:23:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d243b3679c684967913a7936377f9019 : SUCCESS in 9m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0909557c9874c8fb8681d17fc2a8160 : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3659efa2bf11499382f85490977595c5 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d8ba82aa9d42df9cd83f952e4a3f5f : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf62213a312947e982412a770f0e7e10 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a7a8ad02b084fefaf1dda2a46a30488 : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37a77830b5724ec39723574220948f58 : SUCCESS in 7m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a1e759ff3d434f1b90cd85b651e6fc32 : SUCCESS in 2m 49s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9f8a0babb5ce4b759ac6d48d9efaf36b : SUCCESS in 5m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/95bf775b221d4cdebae96e3f58cdcdb1 : SUCCESS in 4m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a67a7a7dedc0bee790a4b6c1d7a17d8469a8dcdc","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-09-06 03:23:17.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/28/680328/2/check/DellEMC_PowerMAX_Manila/0d04ec5/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"0852ea6ac289d045aa33be1225c0851f3e83d5ef","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-09-06 03:40:02.000000000","message":"Patch Set 2:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/28/680328/2/check/manila-master-client-pip-ssa03-dsvm-neutron/aafa027 : FAILURE in 14m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b8d7ca43e8e3822ff426d6fb0c83e69b22bfc642","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-09-06 04:15:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/28/680328/2/check/dsvm-tempest-infinibox-nfs/1f37b48 : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":2},{"id":"2bcdcdd160fc71a5d5bd2b07dab94b2e9cfd2436","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-09-06 04:46:12.000000000","message":"Patch Set 2:\n\n* nexenta-manila_ns5_nfs http://104.254.242.106/refs-changes-28-680328-2/manila_ns5_nfs : SUCCESS in 82m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f81228716586300ffb210019757b5aaf7805b21c","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-09-06 05:17:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"57897c67286fa7e26f8c188e37eb68c410011407","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-09-06 05:46:34.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/28/680328/2/check/DellEMC_VNX_Manila/a819401/DellEMC_VNX_Manila/255 : SUCCESS in 2h 23m 12s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"1f037319c0d9c37fe216315c5956830fe22cb3e9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-09-06 07:14:15.000000000","message":"Patch Set 2:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/605/680328/2/check/infortrend-manila-ci/351effc12143eae71bf5e8b2226d4003c98fb09b : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"75b985e452011f2606788b47a334a5670059d854","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-09-06 07:39:29.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-ss http://13.56.159.105/logs/28/680328/2/upstream-check/manila-cDOT-ss/2fb83e7/ : SUCCESS in 1h 14m 57s\n- manila-cDOT-no-ss http://13.56.159.105/logs/28/680328/2/upstream-check/manila-cDOT-no-ss/b9161a9/ : SUCCESS in 1h 27m 50s\n- manila-cDOT-replication-ss http://13.56.159.105/logs/28/680328/2/upstream-check/manila-cDOT-replication-ss/9fdb965/ : FAILURE in 2m 54s (non-voting)\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"3d3bc62768b4fc6db9aa66c23aa7e43ceceb416d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-10-04 19:43:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"351effc12143eae71bf5e8b2226d4003c98fb09b","revisions":{"d3fd00192b0211a4858eaa7b3a7541f2972f8485":{"kind":"REWORK","_number":1,"created":"2019-09-05 08:33:12.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/28/680328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/28/680328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/28/680328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/28/680328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/28/680328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/28/680328/1"}}},"commit":{"parents":[{"commit":"f1e10ec2ff44445a7a6e10eba3f270e1dbd3a889","subject":"Merge \"[Nexenta] Refactored NexentaStor5 NFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f1e10ec2ff44445a7a6e10eba3f270e1dbd3a889"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-05 08:32:56.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-05 08:33:00.000000000","tz":480},"subject":"Blacklist eventlet 0.25.0","message":"Blacklist eventlet 0.25.0\n\nmore details https://review.opendev.org/#/c/678078/\n\nChange-Id: Iacbde5c4337f892fe1b9ec903d32ca3d138774a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d3fd00192b0211a4858eaa7b3a7541f2972f8485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d3fd00192b0211a4858eaa7b3a7541f2972f8485"}]},"branch":"refs/heads/master"},"351effc12143eae71bf5e8b2226d4003c98fb09b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-06 03:03:47.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/28/680328/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/28/680328/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/28/680328/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/28/680328/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/28/680328/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/28/680328/2"}}},"commit":{"parents":[{"commit":"f1e10ec2ff44445a7a6e10eba3f270e1dbd3a889","subject":"Merge \"[Nexenta] Refactored NexentaStor5 NFS driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f1e10ec2ff44445a7a6e10eba3f270e1dbd3a889"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-05 08:32:56.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-06 03:02:42.000000000","tz":480},"subject":"Blacklist eventlet 0.21.0,0.23.0,0.25.0","message":"Blacklist eventlet 0.21.0,0.23.0,0.25.0\n\nmore details https://review.opendev.org/#/c/678078/\n\nChange-Id: Iacbde5c4337f892fe1b9ec903d32ca3d138774a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/351effc12143eae71bf5e8b2226d4003c98fb09b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/351effc12143eae71bf5e8b2226d4003c98fb09b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
