)]}'
{"id":"openstack%2Fmanila~696002","triplet_id":"openstack%2Fmanila~master~I2fbbd9d3950d4b68eca9905839a1f910e5a8ce4a","project":"openstack/manila","branch":"master","topic":"bug/1853805","hashtags":[],"change_id":"I2fbbd9d3950d4b68eca9905839a1f910e5a8ce4a","subject":"VNX/Powermax: Make it work under python3","status":"MERGED","created":"2019-11-26 01:43:29.000000000","updated":"2019-12-13 02:26:35.000000000","submitted":"2019-12-11 15:50:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696002-1576079415390-1e73c108","meta_rev_id":"106fa15da8f579a5544e0d89187b3daf0eaafcf6","_number":696002,"virtual_id_number":696002,"owner":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-11 05:37:02.000000000","_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2019-12-10 08:05:32.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2019-12-11 14:06:18.000000000","_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-11 11:57:06.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-11 15:50:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-10 10:13:40.000000000","_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-11 15:46:51.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-11 14:25:00.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-11 03:29:04.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2019-12-10 07:46:13.000000000","_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2019-12-11 12:10:05.000000000","_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"date":"2019-12-11 13:53:26.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-10 11:53:18.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-11 14:44:43.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"recommended":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":1,"date":"2019-12-10 12:19:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2019-12-10 12:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-10 12:19:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2019-12-11 07:00:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"date":"2019-12-10 12:09:49.000000000","_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 08:05:32.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-12-10 10:13:40.000000000","updated_by":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-10 11:53:18.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-10 12:09:49.000000000","updated_by":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"reviewer":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"state":"REVIEWER"},{"updated":"2019-12-10 12:19:24.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2019-12-10 12:19:27.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2019-12-11 03:29:04.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-11 05:37:02.000000000","updated_by":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"reviewer":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"state":"REVIEWER"},{"updated":"2019-12-11 07:00:50.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2019-12-11 11:57:06.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2019-12-11 12:10:05.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2019-12-11 13:53:26.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-11 14:06:18.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-11 14:25:00.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-11 14:44:43.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2019-12-11 15:46:51.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-11 15:50:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38a5f1a2344bd02629e9a69977958c844904bd63","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2019-11-26 01:43:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45b2245dc58ef09b2f9d3953a19e1eb93bba5ecd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-11-26 01:50:15.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/02/696002/1/check/dsvm-tempest-infinibox-nfs/acae435 : FAILURE in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"850652da55e558e6dc508e3962426946bda36527","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2019-11-26 01:51:50.000000000","message":"Patch Set 1:\n\nrun-DellEMC VNX CI","accounts_in_message":[],"_revision_number":1},{"id":"25cb5f8d2f787d8fa3dcfe45c38b345ea051b198","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-11-26 02:31:27.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/02/696002/1/check/DellEMC_Unity_Manila_dhss_true/1f2749b/DellEMC_Unity_Manila_dhss_true/43 : SUCCESS in 47m 41s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/02/696002/1/check/DellEMC_Unity_Manila_dhss_false/4fb608a/DellEMC_Unity_Manila_dhss_false/44 : SUCCESS in 34m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"fa4aadb6af9ca491a3a228abaea830639d5d46e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 03:24:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c3999767e04f979ad387590e8bdc24 : SUCCESS in 8m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f78edeb80b44a1e92fae4114df0d796 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef550406dffc477aa5fc306d9f7d3345 : FAILURE in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffadf96d02264426a5613dba45e24da5 : SUCCESS in 7m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eac5384484c64a759ef632c5650bf6f6 : FAILURE in 16m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a510c036b6eb4abb8e4e2f09641a54e8 : FAILURE in 17m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb8b2b43ceac44ed8dca9b9d0cef1ed1 : FAILURE in 8m 41s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/918ed3ba481b43089015921fb35f80bd : SUCCESS in 4m 14s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/1e8f202276ec45a98ecd024b4e48735c : FAILURE in 1h 29m 56s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1aa8169a6e074b7b8148e4f68df23a87 : SUCCESS in 46m 40s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/9f3a0454fd4f42ad9baa1c54a7cd65a2 : SUCCESS in 59m 14s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/b2a4dbd0c408463bbd88342fbe84b4aa : FAILURE in 46m 42s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/f5eb783cdff84b7d865f9af1ca316635 : SUCCESS in 1h 09m 00s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1a1fdef895424a3d8640bf608d2a6df3 : FAILURE in 1h 38m 31s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/66ca2c03e79345158f00079abe7646c6 : SUCCESS in 22m 26s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b88a5ee933ba4f2b8f92d1375691473f : SUCCESS in 1h 34m 12s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f81a659503044ae5a009f7c0a9517812 : SUCCESS in 1h 24m 06s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/e7f27ad99784443baa36a8d8b47c45c8 : SUCCESS in 1h 14m 27s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/be5caaef947a41749681128e4ae65ece : SUCCESS in 42m 17s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/10120fd84b7741d7b6e795aae42f398d : SUCCESS in 44m 52s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/37207f345d9c46aaa329f8203d959d4f : SUCCESS in 53m 28s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/b6cd85d076f04b79902f18c8a123255a : SUCCESS in 1h 04m 23s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ccfc3bfa4b0345e09fbb5ec98f4a9864 : FAILURE in 23m 59s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4c3b32d1980747d18529440a5bd218b6 : SUCCESS in 38m 34s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/985c266e548847e594c5c61a84922667 : SUCCESS in 37m 26s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cc8df55465fd4f80997c1f2dd32d8d96 : SUCCESS in 5m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"702220fd772b9b328687ab94866edd6146062ab2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-11-26 03:25:08.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/811/696002/1/check/infortrend-manila-ci/58a6be1454d1c94c1c1d994a9a4b2b8d2b00be2d : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"ac11c8c96b60aa67c76af4b34080c4def9b25f6a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-11-26 03:48:19.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/02/696002/1/check/DellEMC_VNX_Manila/4e689e9/DellEMC_VNX_Manila/24 : FAILURE in 1h 56m 11s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"16455c625496443acb1bddace1d0f8dfbd43da81","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-11-26 04:33:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/02/696002/1/check/DellEMC_PowerMAX_Manila/acacf16/DellEMC_PowerMAX_Manila/50 : FAILURE in 1h 58m 45s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"8b6cf463b41e5f98164cc9745168b6368cb90334","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2019-11-26 05:16:23.000000000","message":"Patch Set 1:\n\nrun-DellEMC VNX CI","accounts_in_message":[],"_revision_number":1},{"id":"cdd4521997937c1c4779046408105c3a1c53f1aa","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2019-11-26 05:17:30.000000000","message":"Patch Set 1:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":1},{"id":"4200aec73a569b50acfc14aebc3484b1b68ef5b9","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-11-26 06:56:33.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/02/696002/1/check/DellEMC_PowerMAX_Manila/fe2db6d/DellEMC_PowerMAX_Manila/51 : SUCCESS in 1h 38m 53s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"e0ebe78ae695ee2efa53815ce2fdda79a8db0f4e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-11-26 07:07:51.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/02/696002/1/check/DellEMC_VNX_Manila/01f6bef/DellEMC_VNX_Manila/25 : SUCCESS in 1h 51m 12s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"4cb8f6f7d84e25841856f710dd6d375b8b8cc0fb","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2019-12-09 04:56:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c9c4142c900579b46218e1025202328670f7186","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-12-09 05:09:34.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/828/696002/1/check/infortrend-manila-ci/58a6be1454d1c94c1c1d994a9a4b2b8d2b00be2d : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a1600e9f943dd055419964ae3636c25de850cd37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 06:52:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cace0eb3c34de081e6797312e4bb4f : SUCCESS in 7m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66ad2053f41f4ac2bdff52c795367110 : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4909c0d2e8d48499aaf7b0ecb94c5bc : FAILURE in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a807cf48a3bb40ce9e1aca005c2bce43 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ca7f825988c94f15aebcbd621918259d : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/331b7e12ee044b179673595261f2444e : FAILURE in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76c8d81ac69b48e7990889e277040189 : FAILURE in 6m 23s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/724fe308407f47d1a62b9aecb4486caa : SUCCESS in 4m 29s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/7cc3c3cd69cc4d469b9aac77d9e828da : SUCCESS in 1h 35m 52s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/3b581c28c50c431b89274fbed4687a7e : SUCCESS in 1h 00m 33s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/6812a904ed1e45d3a6049e60d9f00343 : SUCCESS in 1h 10m 12s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/c00ce9122c724f1b82717099a70a5bbc : FAILURE in 53m 05s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e48f087179a34c77844f141b55434538 : SUCCESS in 1h 04m 09s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2db31cd5be3240398a592ba48a71d022 : FAILURE in 1h 51m 53s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/b33e86f4152644fe975ad8aed7195a44 : SUCCESS in 26m 25s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/560b201bc17c4166a10c87c76ba45462 : SUCCESS in 1h 21m 39s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c774e35e348d464ba1b53ae7d405b41a : SUCCESS in 1h 19m 15s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/f1797ed9b5f744088b52d16d176395f8 : FAILURE in 1h 08m 29s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/338ec7ce30224159865bd67978297818 : SUCCESS in 42m 35s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/a1c1831f11194c27a8234a85b8afc771 : SUCCESS in 42m 32s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/a453448c06f94b70956c09997c9998e6 : SUCCESS in 1h 15m 46s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/485eb041a72e4c558850f540f206c4b4 : SUCCESS in 59m 47s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0bf9f0a9de2a41ff9496c47a0506b8b7 : FAILURE in 22m 28s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/029bab436acf45c9a2ccde9adb5888db : SUCCESS in 37m 06s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0d8aae66c4d148f6a15ff99e725a9227 : SUCCESS in 39m 05s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/82391bfc7b8d491788f94f60602b4b8b : SUCCESS in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eca916b6e5f2a12af23ec3e09886433c9fe33cc8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-12-09 21:22:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26e87df91316d47d0647b9dd93d4dd72c9952eb1","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2019-12-10 03:02:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c44d3d740453645500f527aecca6586124a7c3f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-12-10 04:40:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38da9ffdd0b3b3863fe3fe6de7415a308e614587","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2019-12-10 07:44:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"998a1d872d92362b553da2babf6af61fe7b7885f","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2019-12-10 07:46:13.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nSorry, didn\u0027t notice the py36 failures. Fixed them in patch #2.\nThanks for your comments.","accounts_in_message":[],"_revision_number":2},{"id":"7c08a4d0fe366fae619147bec7d818425c880d75","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-12-10 07:57:06.000000000","message":"Patch Set 2:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/833/696002/2/check/infortrend-manila-ci/ddfb07987b15e53747d457dfc2e77fee5171fb40 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"a66f482cc7d590035c273073e31a01009bc7745a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-12-10 08:05:32.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/02/696002/2/check/dsvm-tempest-infinibox-nfs/e18993e : FAILURE in 20m 32s","accounts_in_message":[],"_revision_number":2},{"id":"0431f69dfd8231e2d89a5c96086c1876a9879291","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-12-10 08:14:44.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/02/696002/2/check/DellEMC_PowerMAX_Manila/cbc02a0/DellEMC_PowerMAX_Manila/74 : FAILURE in 30m 21s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"b4e568b11f156723e9c0e10f8a61d642e51c0d63","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-12-10 08:40:07.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/02/696002/2/check/DellEMC_Unity_Manila_dhss_true/1a469b5/DellEMC_Unity_Manila_dhss_true/66 : SUCCESS in 55m 44s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/02/696002/2/check/DellEMC_Unity_Manila_dhss_false/9a0aded/DellEMC_Unity_Manila_dhss_false/67 : SUCCESS in 38m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"3d9ff37a6875ebf95567f323b9bb668c5c9bf9cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 09:48:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39bf468951dc4afba20c719745afde1c : SUCCESS in 15m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80057261f7574dd389ad6f9f908b0d38 : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/203b978803b34caf81be9d3ec43b7e05 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e1e7e408cd34d1ea80debf5da70f692 : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0c1a3651ab64ca2ad67b413643c0684 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b2f47a754ff45c7bb782457f0e4dbd0 : SUCCESS in 12m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0e3cd173c0743a08a75e7a7cb5b8212 : SUCCESS in 13m 48s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/42635f14341245ee82f81d9450cad480 : SUCCESS in 5m 30s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/e94dabea52fb4b5b88f6364adeb395f3 : FAILURE in 1h 10m 46s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/504ed64c1caa467683f20e658d9bdff3 : SUCCESS in 56m 27s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/8d3a207cc1f448fb8d26b0bb40918dcb : FAILURE in 1h 06m 07s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/3eba7413f9c6424cad9bdf44786588d7 : SUCCESS in 53m 13s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e01f40ca5211464fbb0ca3eb6d9a9c46 : SUCCESS in 1h 02m 44s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71db3dfeebb74e078302083cfcf5265f : FAILURE in 1h 55m 07s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/d67a917aa42c4245bc202a2e72d9c573 : SUCCESS in 21m 42s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/dc4598a0dc49494481a94da140e0bc80 : SUCCESS in 1h 22m 26s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a2e13144a79f42a29d5a63eeee2c2e46 : FAILURE in 59m 58s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/30e785479efe4591ade165943ddc703b : SUCCESS in 1h 15m 22s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c3ba57b6c7064f67ae0c7a9de72614a0 : SUCCESS in 45m 40s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/c16fe10af8c54343a5820ba6796eb224 : SUCCESS in 42m 14s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/9fa595e78cdb4170b4b957e2a43dc2f2 : SUCCESS in 1h 01m 57s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/7499dc33bf09415f9cfddbbfa94eefdb : SUCCESS in 56m 37s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2e0b5391ca894667b3d471e88f51d1ea : FAILURE in 26m 31s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/36ed7c8d8595498aa5ff1e681d8544b5 : SUCCESS in 42m 49s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1d8bcf5fa5ec445e96de290768115541 : SUCCESS in 43m 30s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/19dddb6daa10473f94b4f8791618cc13 : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b79c696ffd82f1b72a087e3564f8db5079b1b59f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-12-10 09:54:09.000000000","message":"Patch Set 2:\n\n* nexenta-manila_ns5_nfs http://104.254.242.106/refs-changes-02-696002-2/manila_ns5_nfs : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"137c76c7de8f60d3c8b11030ecfd4099b0454f70","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-12-10 09:58:28.000000000","message":"Patch Set 2:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-02-696002-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"f40ac264812c72c7194cdcb611f081722bae5e12","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-12-10 10:07:25.000000000","message":"Patch Set 2:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/02/696002/2/check/manila-master-client-pip-ssa03-dsvm-neutron/d0280b2 : FAILURE in 16m 02s","accounts_in_message":[],"_revision_number":2},{"id":"6c28396d86247f20706b2e6772d742142cb7e087","author":{"_account_id":24236,"name":"INSPUR CI","email":"inspur.ci@inspur.com","username":"INSPUR-CI","tags":["SERVICE_USER"]},"date":"2019-12-10 10:13:40.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"27eb98f460b18ef976a4772a94bf3fe850d1db1e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-12-10 10:16:50.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/02/696002/2/check/DellEMC_VNX_Manila/0d64b28/DellEMC_VNX_Manila/38 : FAILURE in 27m 52s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"c8fbc19d11a2e76d3efc70d5b786dfa1081c1bd9","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-12-10 10:18:49.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/02/696002/2/check/DellEMC_PowerMAX_Manila/0ea3345/DellEMC_PowerMAX_Manila/75 : FAILURE in 29m 59s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"5c4e9b550be0a436d0279d6430c256ccfe70cfbc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-12-10 10:45:21.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/02/696002/2/check/DellEMC_Unity_Manila_dhss_true/3e865de/DellEMC_Unity_Manila_dhss_true/67 : SUCCESS in 56m 30s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/02/696002/2/check/DellEMC_Unity_Manila_dhss_false/7114b8c/DellEMC_Unity_Manila_dhss_false/68 : SUCCESS in 42m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"e2c3f10ab4afce26a8fed8c83a56e6666ec4d4b9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2019-12-10 11:53:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-ss http://logs.openstack.netapp.com/logs/02/696002/2/upstream-check/manila-cDOT-ss/e8e0e89/ : SUCCESS in 2h 04m 10s\n- manila-cDOT-no-ss http://logs.openstack.netapp.com/logs/02/696002/2/upstream-check/manila-cDOT-no-ss/c82d31e/ : SUCCESS in 1h 29m 50s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"ad21d253381ebf1190cff1540145c741d4a922bb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-12-10 12:09:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b51440fa13117a35aa8b2cd85eea9b02ed70128f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-12-10 12:19:24.000000000","message":"Patch Set 2: Code-Review+1\n\nHi, Ryan! Code looks good to me.","accounts_in_message":[],"_revision_number":2},{"id":"5906f7dde0cb080781305de80379043e9082d1ce","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2019-12-10 12:19:27.000000000","message":"Patch Set 2: Code-Review+2\n\nThe patch looks good. But VNX and Powermax CI\u0027s are failing. Please, take a look when you have a time.","accounts_in_message":[],"_revision_number":2},{"id":"ec78c8a35d65991f1f58e15239f6af38fe512ca7","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2019-12-11 01:35:17.000000000","message":"Patch Set 2:\n\nrun-DellEMC VNX CI","accounts_in_message":[],"_revision_number":2},{"id":"523c2712a1482b31374d41b745f11d4549b31d0c","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2019-12-11 01:37:39.000000000","message":"Patch Set 2:\n\nrun-DellEMC PowerMAX CI","accounts_in_message":[],"_revision_number":2},{"id":"f5de3763410b3451d0c17a13fe9c5d2fd16453b4","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-12-11 03:26:58.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/02/696002/2/check/DellEMC_PowerMAX_Manila/30bdba2/DellEMC_PowerMAX_Manila/78 : SUCCESS in 1h 49m 08s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"5aa5d2273532dd2ec852184d3907429ec1dafb22","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-12-11 03:29:04.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/02/696002/2/check/DellEMC_VNX_Manila/2f3d4a2/DellEMC_VNX_Manila/39 : SUCCESS in 1h 53m 32s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"7fcdc2a0384bd8c8ab1661675d993c6af167bb79","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2019-12-11 05:37:02.000000000","message":"Patch Set 2:\n\n\u003e The patch looks good. But VNX and Powermax CI\u0027s are failing.\n \u003e Please, take a look when you have a time.\n\nVNX/PowerMax CI\u0027s are happy now :)","accounts_in_message":[],"_revision_number":2},{"id":"b10857a0d2ffb30c28689011d15793e72bab96cd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-12-11 07:00:50.000000000","message":"Patch Set 2: Workflow+1\n\nThanks Sam!","accounts_in_message":[],"_revision_number":2},{"id":"3fedaeadc1e640f9d39daa60633f5d44f175239d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 07:01:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7105eae174327aeca20d858f89a31daf002501aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 08:26:32.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee46b67c7384450391f0ed6f926a0864 : SUCCESS in 8m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ecdb822c57547789c9a829a1f2c1777 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70b94b6805864a20b925f4996afd4cbc : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/093a2f340f014879af9f07ad3aebaf8e : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11c0688fad864c029faf24e008d4df41 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c115f68460fa4c58b0880ae284a0dc75 : SUCCESS in 7m 53s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4b1a006fb86d42858b1c731230b9b1a5 : SUCCESS in 4m 15s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a2d02e3a00f14f96a41b76bcca38a482 : FAILURE in 48m 51s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/e049117052414d5b927a8c42cbbeda07 : SUCCESS in 1h 03m 22s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/a907cdb39433458994c7a29049eebd6f : SUCCESS in 1h 17m 58s","accounts_in_message":[],"_revision_number":2},{"id":"d7e4ae4e9ff489c3bd5ddffa858013f82dc51918","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2019-12-11 11:57:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0deb63aabc57a7401387b643ff1fd88bd7678bdf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-12-11 12:10:05.000000000","message":"Patch Set 2:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/837/696002/2/check/infortrend-manila-ci/ddfb07987b15e53747d457dfc2e77fee5171fb40 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"ab9f7b51b69106e1547c70a054ebb0e3ead24b0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 13:47:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38dac4e7109442df84bab01d0b4ce3b0 : SUCCESS in 9m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d043febd46f4d73bd506481b6923202 : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e06dcf38c5204addaaff1594fbf3a220 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/247b23170a9348f4b6c0a182d81d99bf : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d55d655f13e24f4994995cb31e08248f : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f494151e912b4a32aa19a2e32e0748d3 : SUCCESS in 7m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/584471165cfc4ed0af4ccdf1636740d3 : SUCCESS in 6m 58s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/21ddd021e4d84e1a9f3adbdc6d3bb187 : SUCCESS in 4m 55s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/17cfcac66fb746c8bad2a071743ea10e : FAILURE in 1h 30m 30s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/775b61a7e8f74bf28f8810503191d382 : SUCCESS in 52m 38s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/a24a2c7df9974d3e9262903dd4448300 : FAILURE in 1h 11m 34s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/e45ec94561064f9494790421f5a1ce6c : SUCCESS in 1h 11m 57s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/5e0f268672b740938c8cd5b2a08956db : SUCCESS in 1h 04m 42s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aff0c1b540124d65865281d2924c9b64 : FAILURE in 1h 44m 15s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/a94a85007d55495c8145574c0c4e6de5 : SUCCESS in 25m 17s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8cdf5b0263c04a0d9a1c74d4ca8be22f : SUCCESS in 1h 27m 11s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c57b284e65a84099ac3197c8bd4b7ca6 : SUCCESS in 1h 09m 44s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/f5fd8ccae55f433f9c9f55f16ec5db2e : SUCCESS in 1h 21m 21s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3ed167df87e44c25aab8322123af1d2b : SUCCESS in 40m 55s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/cd971d3543094515a1dfeb5f026fed01 : SUCCESS in 39m 59s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/b289c921fad24de4b0b3a15bcf38b5e1 : SUCCESS in 1h 06m 11s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/da9512438bcf48ad8cd6e10083f54c5b : SUCCESS in 57m 12s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8279ff7ef24240998b07f8f779286a02 : FAILURE in 25m 46s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/be2ea10333a249e3af8313e6856f9a91 : SUCCESS in 1h 13m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/dde75dda66a4449081d31bf82334dfcb : SUCCESS in 44m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c846651897d64b85a2eedcebf806a383 : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"83e91464fcefe07117d0ee4ec859cc980f145406","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 13:47:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a55cf029938d00b10dccfb067fea893b7d43ef46","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-12-11 13:53:26.000000000","message":"Patch Set 2:\n\n* nexenta-manila_ns5_nfs http://104.254.242.106/refs-changes-02-696002-2/manila_ns5_nfs : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"50e85ff8e56f74962bb58287ef59f6ed37dfe150","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-12-11 14:06:18.000000000","message":"Patch Set 2:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/02/696002/2/check/manila-master-client-pip-ssa03-dsvm-neutron/3f723e5 : FAILURE in 16m 10s","accounts_in_message":[],"_revision_number":2},{"id":"3fa62f4ef02d05c6695e9b8ca065bef8169cf5ba","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-12-11 14:25:00.000000000","message":"Patch Set 2:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-02-696002-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"e8459f2f13b81d0e7a49db2bdc75ff839ac837c9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-12-11 14:44:43.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/02/696002/2/check/DellEMC_Unity_Manila_dhss_true/1cf8d1a/DellEMC_Unity_Manila_dhss_true/71 : FAILURE in 56m 58s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/02/696002/2/check/DellEMC_Unity_Manila_dhss_false/30c11d8/DellEMC_Unity_Manila_dhss_false/72 : SUCCESS in 42m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"d6950802195587d3492e1a572697ff662d214c34","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-12-11 15:46:51.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/02/696002/2/check/DellEMC_PowerMAX_Manila/076c5dd/DellEMC_PowerMAX_Manila/80 : FAILURE in 1h 59m 07s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"64d6dc597d4e6fc56a37bc9718c71b7c2568b901","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 15:50:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc2da4dcce154e3dbb76d3e9fb76bfeb : SUCCESS in 11m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9407e5e4d45948ec83502251d07cedbe : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aed222d2f3594167943afbe348c507b6 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2923933d365e47b4b8e8d03928b8de7b : SUCCESS in 8m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddc9b3b57ad34aa8bc053ddffba96503 : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b37da8c8342a4eacb5cea02435265c0b : SUCCESS in 9m 03s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/496f381048d94102b534d5de148d0f2e : SUCCESS in 4m 56s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e3e2037911f94d779d8ea137d36cc311 : SUCCESS in 36m 48s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/4bb304db32c84da88f6fd8c01e01fa0a : SUCCESS in 1h 20m 52s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/ff750c49a0c944519423042156d29ac5 : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":2},{"id":"ea24c9a6351d2bb0b9f33549f88e40e0db901468","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 15:50:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a280fcf73811715ece8a8dce5b638b3c8436d787","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 15:51:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fc444fd78e54069b1db74e61d0f1a2b : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2},{"id":"73a1bfe070859f20445195a37e90f5ddb5b5a893","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2019-12-12 14:28:50.000000000","message":"Patch Set 2:\n\nHi Ryan, \n\nPlease back port this fix to stable branches. Many distros run older branches of manila on python3 now. (e.g.: RDO/OSP have manila Stein running on python3)","accounts_in_message":[],"_revision_number":2},{"id":"b2d940e22a21e9c55462b8d3eec42970422c3e07","author":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"date":"2019-12-13 02:26:35.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit ce302d04ecb4ef69ed14f2a8e982e8170f3e60a0","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ddfb07987b15e53747d457dfc2e77fee5171fb40","revisions":{"58a6be1454d1c94c1c1d994a9a4b2b8d2b00be2d":{"kind":"REWORK","_number":1,"created":"2019-11-26 01:43:29.000000000","uploader":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"ref":"refs/changes/02/696002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/02/696002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/696002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/696002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/696002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/02/696002/1"}}},"commit":{"parents":[{"commit":"9f06403d73094aa35a98b3c6b59f0c7f210aac27","subject":"inspur: transfer \u0027rw\u0027 to \u0027rwx\u0027 when Shared File Systems protocol is cifs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f06403d73094aa35a98b3c6b59f0c7f210aac27"}]}],"author":{"name":"Ryan Liang","email":"ryan.liang@dell.com","date":"2019-07-25 01:41:18.000000000","tz":480},"committer":{"name":"Ryan Liang","email":"ryan.liang@dell.com","date":"2019-11-26 01:42:47.000000000","tz":480},"subject":"VNX/Powermax: Make it work under python3","message":"VNX/Powermax: Make it work under python3\n\nAdd and remove the unnecessary `encode`.\n\nChange-Id: I2fbbd9d3950d4b68eca9905839a1f910e5a8ce4a\nCloses-bug: #1853805\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/58a6be1454d1c94c1c1d994a9a4b2b8d2b00be2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/58a6be1454d1c94c1c1d994a9a4b2b8d2b00be2d"}]},"branch":"refs/heads/master"},"ddfb07987b15e53747d457dfc2e77fee5171fb40":{"kind":"REWORK","_number":2,"created":"2019-12-10 07:44:10.000000000","uploader":{"_account_id":18742,"name":"Ryan Liang","email":"ryan.liang@emc.com","username":"RyanLiang"},"ref":"refs/changes/02/696002/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/02/696002/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/696002/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/696002/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/696002/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/02/696002/2"}}},"commit":{"parents":[{"commit":"9f06403d73094aa35a98b3c6b59f0c7f210aac27","subject":"inspur: transfer \u0027rw\u0027 to \u0027rwx\u0027 when Shared File Systems protocol is cifs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9f06403d73094aa35a98b3c6b59f0c7f210aac27"}]}],"author":{"name":"Ryan Liang","email":"ryan.liang@dell.com","date":"2019-07-25 01:41:18.000000000","tz":480},"committer":{"name":"Ryan Liang","email":"ryan.liang@dell.com","date":"2019-12-10 07:43:32.000000000","tz":480},"subject":"VNX/Powermax: Make it work under python3","message":"VNX/Powermax: Make it work under python3\n\nAdd and remove the unnecessary `encode`.\n\nChange-Id: I2fbbd9d3950d4b68eca9905839a1f910e5a8ce4a\nCloses-bug: #1853805\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ddfb07987b15e53747d457dfc2e77fee5171fb40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ddfb07987b15e53747d457dfc2e77fee5171fb40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
