)]}'
{"id":"openstack%2Fmanila~696136","triplet_id":"openstack%2Fmanila~master~I4bd02918ff08711dfadbd129b70b79d843ef7634","project":"openstack/manila","branch":"master","topic":"limit-share-replicas-per-share","hashtags":[],"change_id":"I4bd02918ff08711dfadbd129b70b79d843ef7634","subject":"[WIP] Fix unlimited share replica per share","status":"NEW","created":"2019-11-26 17:12:29.000000000","updated":"2020-01-29 13:11:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7fa330904e4c85de7cafe46b891a3fbc0ca4f0a7","_number":696136,"virtual_id_number":696136,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-13 14:51:57.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"date":"2019-12-13 13:28:12.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"tag":"autogenerated:jenkins-gerrit-trigger","date":"2019-12-13 13:21:42.000000000","_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"date":"2019-12-24 02:01:23.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-13 15:08:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-12-13 14:02:35.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-13 13:21:42.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2019-12-13 13:28:12.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2019-12-13 14:02:35.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2019-12-13 14:51:57.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-13 15:08:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-24 02:01:23.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae7836ed9e9c848678ff95d4aad8a743d353c986","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-11-26 17:12:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c31771d27f7b3ce11dc284bd9b9e99ba203fa91d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-11-26 18:03:29.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/36/696136/1/check/DellEMC_Unity_Manila_dhss_true/3f6fd3f/DellEMC_Unity_Manila_dhss_true/46 : SUCCESS in 50m 46s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/36/696136/1/check/DellEMC_Unity_Manila_dhss_false/2209916/DellEMC_Unity_Manila_dhss_false/47 : SUCCESS in 35m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"462782e7b6814a4e017c377135c8b22821cd4f97","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-11-26 18:04:51.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/813/696136/1/check/infortrend-manila-ci/1faf9727ecf9746cdb63368167ba36d2708d887f : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"05c7169abac113b0d08faa0156d17fa32b12b02e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-11-26 18:04:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/36/696136/1/check/dsvm-tempest-infinibox-nfs/12efa6a : SUCCESS in 51m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4295dcbf91b773ca86aba816dc014ea8378bc4c0","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-11-26 18:23:16.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/36/696136/1/check/DellEMC_PowerMAX_Manila/2aec363/DellEMC_PowerMAX_Manila/54 : FAILURE in 1h 10m 36s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"8148ca2e3f0134a51339896db5d90b9d103b2b63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 19:10:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de8179d502c04d17ba97c6aa7532e227 : SUCCESS in 10m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c4d67e8e3704f4cbb6a38897059a43c : TIMED_OUT in 51m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6079d23bb5141c098baaf7375acd0b7 : TIMED_OUT in 41m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6354bff0480a40739430f77389f5d13b : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbb93fc8d8304b718fcce4be2364d768 : TIMED_OUT in 41m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f5744b28808424796ad7251960a867b : TIMED_OUT in 41m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cdc369e1828d433fa4f9783f3b075140 : TIMED_OUT in 40m 59s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9706a6db5aad45c0ac16502ab4f04b5f : SUCCESS in 5m 23s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/15c4ccbcce424af0a74b59b7b6b1dbe7 : FAILURE in 6m 49s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/f3d589769a2c4e7b98be84e232b85ca1 : FAILURE in 6m 51s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/41e64fe1527d4aa9ac3daac9e2f4cabf : FAILURE in 59m 58s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/dd01ee8c7f9c48f2a61912cd54f1260e : SUCCESS in 49m 22s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/ac64e5d9b5564406b276da7d7d82f4da : FAILURE in 1h 03m 38s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b409b39895a9403f8c71b89221fad09a : FAILURE in 1h 51m 10s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/3774a468dd0e44ec9a444d2f2cf69039 : FAILURE in 7m 14s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5b5ec474201c41baa6bf97d971bef661 : FAILURE in 8m 52s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f8d8f74e7c6041bc89468680592db392 : FAILURE in 6m 58s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/ce14e766021745d5b511b093f5779da9 : SUCCESS in 1h 15m 18s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f51d030db2cd426aa1992252538c93f2 : FAILURE in 46m 16s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/bf8bdbab28ff4a179f8fe9f7b14b39b0 : FAILURE in 46m 37s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/31ffcb9f3ba8466b8b881e4543f3c2bc : SUCCESS in 1h 11m 00s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/669ebcc65068494a9b34292668d52d80 : SUCCESS in 1h 06m 33s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e8efda4bb0e84ff29d197094a2b82a8e : FAILURE in 23m 59s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6880382a94f445ad84fe65a28bc66750 : SUCCESS in 39m 39s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3960b48d029a441a8401a1cb589b3a6e : SUCCESS in 45m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a831e356af5400c9f5a169621e52271 : FAILURE in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a25c556af75556819c447e9b4098ade603bcac8","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-11-27 18:39:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"031a93136718fce9045be4a3e2c069a99f5df3c4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-11-27 19:51:02.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/815/696136/1/check/infortrend-manila-ci/1faf9727ecf9746cdb63368167ba36d2708d887f : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"fada49bfe0e64b29371ee48cd57f05459465ba63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 20:32:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cabb0fcf343c4b6ba0dc9f2a54b09dc5 : SUCCESS in 14m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f27cc9b170ed44dfa0b940388bd843c4 : TIMED_OUT in 51m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a39c598924de476685281922ba5ed1a8 : TIMED_OUT in 41m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34dc8ad1352c45f5b949582c16ced0fe : SUCCESS in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e4f93c092e8d40569ae8cc136e4cc0c0 : TIMED_OUT in 40m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0037dd4e1ca342cb9da8fe40cfbefb77 : TIMED_OUT in 41m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35dc83a8e2f34189b725cd48b6ca238e : TIMED_OUT in 41m 33s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e053ecaa4f554e85bacbe1107b756830 : SUCCESS in 5m 51s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/cdf0f4b351ba41f9a075a55e1add136f : FAILURE in 1h 14m 08s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/8c58184649a74dc18c361f045a5c73a5 : SUCCESS in 50m 56s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/3e42056b3bfa46f58b70ddabe63b0232 : FAILURE in 1h 03m 39s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/8a026599054f43a3a8cba36e1c2efa17 : SUCCESS in 49m 47s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/aedaae8910d64b35b102a1708673b43e : SUCCESS in 56m 59s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/865d62a712054dd6b92693bb47794a29 : FAILURE in 1h 49m 58s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/572472c6e36445808e514b3b142109a9 : SUCCESS in 32m 47s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d184e9d1e5b344af9d7a428e3e4c4f72 : SUCCESS in 1h 17m 33s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d4ded3c82cd6493f9ee85cb92b810721 : FAILURE in 1h 29m 18s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/f34434c31f2e4f3b95c48e5c405746f8 : FAILURE in 1h 16m 33s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/807ed8775c5d4c859c97a523373aadbf : FAILURE in 44m 54s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/48d487ac0b92402fac0e53880469b037 : FAILURE in 44m 09s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/9e44cd0c970e4439bb2cb7927a935658 : SUCCESS in 59m 00s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/cc0538b6d21345d8bc9cd19a413aca3b : SUCCESS in 1h 09m 44s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0f5cf7fed1b6442bbd25f2d0126cb5c3 : FAILURE in 21m 54s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c22bc7d168b34810b107af671ee19ad2 : SUCCESS in 54m 09s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/752f2892abe9444487a0a14565d3fd6c : SUCCESS in 40m 37s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e07f5b029b54d468bd3f4e9892eae66 : FAILURE in 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c1eb21a90b5230c05f5e0a615286fb765455cd9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-12-04 14:33:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ff5815d5d998e741294b0443aeb245ae79e16c91","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-12-04 15:28:31.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/821/696136/1/check/infortrend-manila-ci/1faf9727ecf9746cdb63368167ba36d2708d887f : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"c95f6ae8989b750e49421b211675abd6130088e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 16:33:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf9ed4d6469c4b289c385623d330a647 : SUCCESS in 8m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e21e4ac8aa240de92500829d4f70b2e : TIMED_OUT in 52m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35e66b0d1adc4c819c12d21900bf2123 : TIMED_OUT in 40m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d2bc18bf9a415facaac20b630fa61f : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c379f5b0ebb149818c15456960ea3493 : TIMED_OUT in 41m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66af5a815f4c43fda0525eec99615436 : TIMED_OUT in 40m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9510f19fe616419785dd889aeeca6559 : TIMED_OUT in 41m 31s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/565707820d544c99b5b88c626e478f01 : SUCCESS in 5m 38s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/fcd14bc2c26845bfb7a7aa29e74991fc : FAILURE in 1h 12m 26s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/7d076a25c46f4bd582dda4e996cae92d : SUCCESS in 45m 56s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/d83d8b49d7b342978157c6b4eff5ce2c : FAILURE in 59m 18s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/da42561ec3fb47f28c9c28bf7027db05 : FAILURE in 49m 25s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/5f6aa59106ac45aa89d0989ccfd28ac1 : SUCCESS in 1h 11m 43s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d12863a6e2b248dc84920d0718b93a71 : FAILURE in 1h 57m 00s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/0adced882c9d4cb9b0fae9eec4acc5e2 : SUCCESS in 29m 31s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/097ef1f76703473b9e649d28d713834a : FAILURE in 36m 00s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e1993179683443d19b6e57beb5b25774 : SUCCESS in 1h 24m 50s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/168572b7607f4e7ea2d07a2ef393538b : SUCCESS in 1h 17m 41s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f9bff068d6b242b096f6ce5b9c710310 : FAILURE in 46m 57s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/3719f5aac4f04e728bccf489d0018f4c : FAILURE in 42m 44s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/12697c5557aa463f869fdd9cff8072ec : FAILURE in 31m 04s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/35736353b27a490cbf0ca0a717c21e84 : FAILURE in 36m 21s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/76a6a25a45464c7aa9a228d4adcfa7bd : FAILURE in 21m 22s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d60c2e38f1a74cf1808ce0bda209a3ac : SUCCESS in 37m 57s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/664e9bd4eb3e4b5aaa2d7ce527dcfce2 : SUCCESS in 40m 41s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/22d2fc4d334e4343b20f2e60d48de6dc : FAILURE in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"239272ea0831772e702d6b8d9dbb47d35380389b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-12-06 20:14:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8aedcef54caf008aefdd7f2dc0318a8df08151a1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-12-06 20:27:45.000000000","message":"Patch Set 2:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/826/696136/2/check/infortrend-manila-ci/a05dc20741b0f0c9246ab9b39c7af2eb2140aa90 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"4d0601aa1bd67e2e8f644ec4047872b46cc0bd3a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-12-06 20:33:15.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/36/696136/2/check/dsvm-tempest-infinibox-nfs/567856a : FAILURE in 18m 12s","accounts_in_message":[],"_revision_number":2},{"id":"b95605d930a6dd9a37908e0275bad4332f7c4f8f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-12-06 20:39:27.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/36/696136/2/check/DellEMC_Unity_Manila_dhss_true/c8e636d/DellEMC_Unity_Manila_dhss_true/59 : FAILURE in 24m 42s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/36/696136/2/check/DellEMC_Unity_Manila_dhss_false/98c2486/DellEMC_Unity_Manila_dhss_false/60 : FAILURE in 22m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"fa6a872d3fa969d59d5440cd506c393277a3f80c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-12-06 20:42:47.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/36/696136/2/check/DellEMC_PowerMAX_Manila/4d586e7/DellEMC_PowerMAX_Manila/67 : FAILURE in 28m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"50234312cfc4861bb103531cbeda951eb6c33f1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 21:06:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd99d2ddd24b4711accab2d1ca69dbfc : SUCCESS in 9m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09f2bae595214bd99e4e98201d8e233e : TIMED_OUT in 51m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3092697b5d4482fa47aa560922f9fa3 : TIMED_OUT in 41m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1204aa7b2822411791d83686d46fdbc6 : FAILURE in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3399afc7e223466aa4454d360faa3381 : TIMED_OUT in 41m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/baceb8bb465c488c8b9d13fd557f5a82 : TIMED_OUT in 41m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd9f9e898b624e40af221182144f6c3c : TIMED_OUT in 40m 51s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/90aa50149efe42df835b9000c6d558e7 : SUCCESS in 4m 00s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/1c07597484b84a929e4fe4b6af0a6500 : FAILURE in 30m 35s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/6451cac41d9e46ee8b3c5cf0b8c90ab2 : FAILURE in 35m 59s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/a6f920d7ce0944da9da0f6f8f6cae926 : FAILURE in 33m 35s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/949b699bf46d40f49842283a933e43e2 : FAILURE in 30m 52s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/15c2924b434a40b59ec06182e9766491 : FAILURE in 30m 12s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31a972a8527d48ec9e3312ca0656a331 : FAILURE in 30m 47s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/0499c466b1ae48febc1210c5c02f4839 : FAILURE in 18m 31s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c86ba9608b2b4973aa3cdb7d9d44081a : FAILURE in 27m 59s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/590eaf10f6be4ed28d44e146a7bd0585 : FAILURE in 30m 50s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/b49aac77d823444285fa99a6cccfbf48 : FAILURE in 31m 33s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c555c4ea84ec4dc3a8787f8e86756bb5 : FAILURE in 16m 31s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/a95af96c97284dbca4fa86ac540fd231 : FAILURE in 17m 22s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/9ce79d5135ef410991fdf8654f8cf42e : FAILURE in 27m 31s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/f2f388c07a8140048857752b6a0ded3a : FAILURE in 27m 58s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c888e251e8ac4f6f9edd1c4a70db8a95 : FAILURE in 18m 04s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f298004e714b40ce9843348f0b552cc6 : FAILURE in 32m 41s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e46c2210f598447d9c7927b0a29aa866 : FAILURE in 29m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/838a4159b3c44a0491d4f780da06a888 : FAILURE in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e796bdd2772115548f6831f09fdf9dff2bbbccd","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-12-09 11:30:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"889b6b1332b6d122721f1a7d7c2c7a5083c83b74","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-12-09 11:43:24.000000000","message":"Patch Set 2:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/832/696136/2/check/infortrend-manila-ci/a05dc20741b0f0c9246ab9b39c7af2eb2140aa90 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"843f08da7455d2e7fc27894470eb15e9335f50cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 12:57:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c75f4b8517b64292889882cb6c7f5ce5 : SUCCESS in 8m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a67f89ba3179428a832ec17e1269b6f6 : TIMED_OUT in 51m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b881d71af6b437286708eb8122d84bb : TIMED_OUT in 41m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b828b4059a1454db53a80617906b9f4 : FAILURE in 8m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d840d3a115f948ab991f662be5a51e54 : TIMED_OUT in 40m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a78058a1eca4efab8f512650eb8543c : TIMED_OUT in 41m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5951f13022d34df0b501621ed0fc5879 : TIMED_OUT in 41m 06s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/724ad7b19e1048789a2de1ed563bf0e7 : SUCCESS in 5m 09s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/d42f819d14464d8e8797d28a66a6e0f4 : FAILURE in 40m 51s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1ad13afa405d44a1aed7550dfa4bf40d : FAILURE in 32m 36s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/e3af7ca13ed645ed9bceba659d1b4513 : FAILURE in 34m 20s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/84e8da0c81264a82991eb55d3c610c16 : FAILURE in 35m 12s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/8c20ae0c03164c30a1065b0d8941ca3f : FAILURE in 31m 55s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f637bd7bb14c41cc93a111a134d701a8 : FAILURE in 37m 30s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/e4dfb788afe245b1a524a40b4102680d : FAILURE in 20m 50s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/32c5b44d17744b07a1bac6e6f030fdc3 : FAILURE in 43m 13s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1733712254b7469aa637e3884cbed986 : FAILURE in 41m 43s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/7d614b3d5f454df291b8b92638db4633 : FAILURE in 40m 43s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3179419d9426470595cdfc9c9c4ba55c : FAILURE in 28m 38s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/e43a81a44b914dcc82f6dbc94e0c8f0f : FAILURE in 23m 03s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/fb5c884b39424aa196f61680b18037fd : FAILURE in 29m 36s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/f67e5df1632a40b7b34fe4abe428b2e8 : FAILURE in 45m 06s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/33c0abdfdf9044f0ad60dfd431674c0b : FAILURE in 22m 42s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d5091af502ca4ce1b1255a241b96b754 : FAILURE in 39m 09s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/74d9c008637544a2b268c5449bfd550b : FAILURE in 33m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba063dbe2d264fae9841191e8b5a5f48 : FAILURE in 8m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05b26bb661e62794f5271362a83f42fc7113381e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-12-13 13:09:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4e7a154d83a4dea3339560067e1bfe766da26bd0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-12-13 13:21:42.000000000","message":"Patch Set 3:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/840/696136/3/check/infortrend-manila-ci/b55f214921b9db874befe1f2ded119e10dc1e0c9 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"80a92800a9c4ac039be2adeb1930524ae5b53469","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-12-13 13:28:12.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/36/696136/3/check/dsvm-tempest-infinibox-nfs/e09dc8d : FAILURE in 18m 30s","accounts_in_message":[],"_revision_number":3},{"id":"bee6c93d22ade23ec60ad3dcb0e53b6b67bf2e89","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-12-13 14:02:35.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/36/696136/3/check/DellEMC_Unity_Manila_dhss_true/fd34f14/DellEMC_Unity_Manila_dhss_true/75 : FAILURE in 53m 16s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/36/696136/3/check/DellEMC_Unity_Manila_dhss_false/48342b5/DellEMC_Unity_Manila_dhss_false/76 : FAILURE in 38m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d219e77cb4d6810ccff9308d7fc0e25ce542d634","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-12-13 14:51:57.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/36/696136/3/check/DellEMC_PowerMAX_Manila/7ed6f95/DellEMC_PowerMAX_Manila/84 : FAILURE in 1h 42m 38s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"8fd88c1d4eefcf0966d5ca42522c57bbe047c1f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 15:08:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d75baaf8d3984ee7abee995a731c236f : SUCCESS in 9m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df7848d6e4d5497fb9ec29582bdf3133 : TIMED_OUT in 52m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53f98d0e4f714f1c8d6071e91b66110c : TIMED_OUT in 40m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d8e4b0c2f6d483f9fb5a2bd0368c125 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07171ff60ced40338a67e9151e1a6fab : TIMED_OUT in 41m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b042205ac7364819afe53e806a2d0f9d : TIMED_OUT in 41m 19s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/419c1cd838a4421983a932730de863bf : SUCCESS in 5m 21s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/99948a3eebe64cbd87b7b4bdf241374c : FAILURE in 1h 00m 34s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/7b42259262f44ef18c850332bf74a5af : SUCCESS in 57m 52s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/30eeb487d2e54a559a7613d0a52e28f8 : FAILURE in 59m 22s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ee54239302894dbdb7499321c1d00602 : SUCCESS in 1h 02m 12s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/2343adb4fa2f42449e014410c0c7ebc8 : FAILURE in 55m 09s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd8c4c4de2b04fb38c6ce062dde011c6 : FAILURE in 1h 52m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/c95804cde1214e7db26c7da9a08b9720 : SUCCESS in 25m 09s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b118614994af4e1aa597f3ca0b9900fb : SUCCESS in 1h 14m 02s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/abd96b12d1c94c29818b86b8ab03c14f : SUCCESS in 1h 06m 37s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/1b6f306d2b524c3bb15a398627620e70 : SUCCESS in 1h 07m 30s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/0ff9a322ebb74ead88d5f643dd9c3cf3 : FAILURE in 36m 52s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/0f59fc00505640ffb07cc47a48adde49 : FAILURE in 54m 20s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/182f634837114d6ea0b2b2fcec4c138e : FAILURE in 54m 37s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/594a704d5dd1491a9c5545cd766174cd : FAILURE in 18m 44s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/21d7f15a9ea14bc9957565913244ec3c : SUCCESS in 36m 26s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/21688a1f599146328f1c8616f565d287 : SUCCESS in 39m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/35922e745b034035a072c5244f716e9d : FAILURE in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"84dcadd6f64b994ba3913285c732a1545bc3bf14","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-12-22 20:04:40.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-36-696136-3/ns5_manila : FAILURE in 18m 17s","accounts_in_message":[],"_revision_number":3},{"id":"f06f634f4d2a4c3d070061eb67c2c59e59a8d9a5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-12-22 20:46:00.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-36-696136-3/ns5_manila : FAILURE in 1m 6s","accounts_in_message":[],"_revision_number":3},{"id":"5201d5ac201e8de89bfdca22241a47fd08970c0d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-12-22 23:05:20.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-36-696136-3/ns5_manila : FAILURE in 2h 17m 30s","accounts_in_message":[],"_revision_number":3},{"id":"147402f3865512ca267f42863ef5b0ac32ee2309","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-12-23 21:50:08.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-36-696136-3/ns5_manila : FAILURE in 1h 58m 13s","accounts_in_message":[],"_revision_number":3},{"id":"c95826a5429e572d2a09ab208e90764d7d24c014","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-12-24 02:01:23.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-36-696136-3/ns5_manila : FAILURE in 2h 20m 58s","accounts_in_message":[],"_revision_number":3},{"id":"7fa330904e4c85de7cafe46b891a3fbc0ca4f0a7","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-01-29 13:11:42.000000000","message":"Topic changed from bug/1850545 to limit-share-replicas-per-share","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b55f214921b9db874befe1f2ded119e10dc1e0c9","revisions":{"1faf9727ecf9746cdb63368167ba36d2708d887f":{"kind":"REWORK","_number":1,"created":"2019-11-26 17:12:29.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/36/696136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/696136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/696136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/696136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/696136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/696136/1"}}},"commit":{"parents":[{"commit":"572be43bb03a67376b52a11c70e587eb160b68a3","subject":"[NetApp] Fix share replica failing for \u0027transfer in progress\u0027 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/572be43bb03a67376b52a11c70e587eb160b68a3"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-11-15 18:49:02.000000000","tz":0},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2019-11-26 17:11:08.000000000","tz":-120},"subject":"[WIP] Fix unlimited share replica per share","message":"[WIP] Fix unlimited share replica per share\n\nThis patch fixes a bug that makes the user able to create unlimited\nshare replicas from a given share. Now, we have some validations\nin the share type and the administrator is able to specify how many\nshare replicas a given share type is able to support.\n\nChange-Id: I4bd02918ff08711dfadbd129b70b79d843ef7634\nCloses-Bug: #1850545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1faf9727ecf9746cdb63368167ba36d2708d887f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1faf9727ecf9746cdb63368167ba36d2708d887f"}]},"branch":"refs/heads/master"},"a05dc20741b0f0c9246ab9b39c7af2eb2140aa90":{"kind":"REWORK","_number":2,"created":"2019-12-06 20:14:35.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/36/696136/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/696136/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/696136/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/696136/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/696136/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/696136/2"}}},"commit":{"parents":[{"commit":"572be43bb03a67376b52a11c70e587eb160b68a3","subject":"[NetApp] Fix share replica failing for \u0027transfer in progress\u0027 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/572be43bb03a67376b52a11c70e587eb160b68a3"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-11-15 18:49:02.000000000","tz":0},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-12-06 20:12:38.000000000","tz":0},"subject":"[WIP] Fix unlimited share replica per share","message":"[WIP] Fix unlimited share replica per share\n\nThis patch fixes a bug that makes the user able to create unlimited\nshare replicas from a given share. Now, we have some validations\nin the share type and the administrator is able to specify how many\nshare replicas a given share type is able to support.\n\nChange-Id: I4bd02918ff08711dfadbd129b70b79d843ef7634\nCloses-Bug: #1850545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a05dc20741b0f0c9246ab9b39c7af2eb2140aa90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a05dc20741b0f0c9246ab9b39c7af2eb2140aa90"}]},"branch":"refs/heads/master"},"b55f214921b9db874befe1f2ded119e10dc1e0c9":{"kind":"REWORK","_number":3,"created":"2019-12-13 13:09:08.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/36/696136/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/696136/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/696136/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/696136/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/696136/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/696136/3"}}},"commit":{"parents":[{"commit":"62efe7b88598fedd46f010d04aa79b8acacc4332","subject":"Merge \"VNX/Powermax: Make it work under python3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/62efe7b88598fedd46f010d04aa79b8acacc4332"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-11-15 18:49:02.000000000","tz":0},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-12-12 17:17:46.000000000","tz":0},"subject":"[WIP] Fix unlimited share replica per share","message":"[WIP] Fix unlimited share replica per share\n\nThis patch fixes a bug that makes the user able to create unlimited\nshare replicas from a given share. Now, we have some validations\nin the share type and the administrator is able to specify how many\nshare replicas a given share type is able to support.\n\nChange-Id: I4bd02918ff08711dfadbd129b70b79d843ef7634\nDepends-On: I2c690b8740d3888c3737c880704c6d28bd7df47f\nCloses-Bug:~#1850545\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b55f214921b9db874befe1f2ded119e10dc1e0c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b55f214921b9db874befe1f2ded119e10dc1e0c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
