)]}'
{"id":"openstack%2Fmanila~697111","triplet_id":"openstack%2Fmanila~master~I6ae48e2f174e0197fdeeaca5fd5a96a23c8bb182","project":"openstack/manila","branch":"master","topic":"bug/1843953","hashtags":[],"change_id":"I6ae48e2f174e0197fdeeaca5fd5a96a23c8bb182","subject":"Update admin guide for share network subnets change","status":"MERGED","created":"2019-12-03 15:25:57.000000000","updated":"2020-06-17 12:47:34.000000000","submitted":"2020-06-04 17:08:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":52,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697111-1591290537362-c9d3cef6","meta_rev_id":"3fa8973ce0c1ad828c561f7bf02360089efd4aec","_number":697111,"virtual_id_number":697111,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-04 17:56:51.000000000","post_submit":true,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-04 16:37:20.000000000","_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 17:08:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-04 18:26:55.000000000","post_submit":true,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-06-04 16:37:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2020-06-04 16:37:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-03 15:56:17.000000000","updated_by":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-03 16:20:14.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2019-12-03 16:26:39.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-06-03 16:00:33.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 12:36:12.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-06-04 13:57:18.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 14:42:56.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 16:37:41.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-06-04 17:08:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 17:56:51.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 18:26:55.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"}],"messages":[{"id":"9d07913f76c455b983fb5a34deb61c8f858e02f5","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2019-12-03 15:25:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25c95bb0cb95da2db96556d0902de03d4d416aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 15:37:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a104f663efe476ba4ffc54f6525b26a : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec571e21595d4de98a6b3430b7b938d4 : SUCCESS in 5m 02s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5fdb35a1f42f4c4aa406741d607b769b : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"84c262e9317de6b4ece63e86090fa0b3415dd5f6","author":{"_account_id":22248,"name":"HPE STORAGE CI","display_name":"HPE AlletraMP CI","email":"hp.cinder.blr.ci@groups.ext.hpe.com","username":"hp-storage-blr-ci","status":"Alletra MP Openstack CI","tags":["SERVICE_USER"]},"date":"2019-12-03 15:56:17.000000000","message":"Patch Set 1:\n\nBuild failed. Add comment \u0027run-HPE Storage CI\u0027 to run tests again. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/HPE_Storage_CI\n\n- manila-master-client-pip-ssa03-dsvm-neutron http://54.201.44.218/11/697111/1/check/manila-master-client-pip-ssa03-dsvm-neutron/c9d622c : FAILURE in 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d267828c041b9e1a1ee8d4811cf6f340120f8e4d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2019-12-03 16:17:05.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/11/697111/1/check/DellEMC_Unity_Manila_dhss_true/d4a6d9d/DellEMC_Unity_Manila_dhss_true/53 : SUCCESS in 50m 58s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/11/697111/1/check/DellEMC_Unity_Manila_dhss_false/6f489d8/DellEMC_Unity_Manila_dhss_false/54 : SUCCESS in 36m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"c601d99450d4b9d54fde0755647aebf0786ac8e7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2019-12-03 16:20:14.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/820/697111/1/check/infortrend-manila-ci/08f6505006ee4bd5b9dda8dc1700703a9760c364 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"cbda21f039746f9d341f95793571b548fd3e634b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2019-12-03 16:26:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/11/697111/1/check/dsvm-tempest-infinibox-nfs/6fd7a41 : SUCCESS in 1h 00m 17s","accounts_in_message":[],"_revision_number":1},{"id":"10338a311086938566074dbeceecdb3e24ca95a4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2019-12-03 16:48:40.000000000","message":"Patch Set 1:\n\n* nexenta-manila_ns5_nfs http://104.254.242.106/refs-changes-11-697111-1/manila_ns5_nfs : FAILURE in 69m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5b5d4ad262f3885afb7b3479e5d0981ffb8dac6c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2019-12-03 17:30:54.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/11/697111/1/check/DellEMC_PowerMAX_Manila/5c01348/DellEMC_PowerMAX_Manila/62 : SUCCESS in 2h 04m 46s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"0bb28eabe3fa111955b152a09e0a9dab53a20454","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2019-12-03 17:56:52.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/11/697111/1/check/DellEMC_VNX_Manila/7c37afb/DellEMC_VNX_Manila/32 : SUCCESS in 2h 18m 42s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"64c7749fb2e9eb1bbf1416d05a36873ca066b26c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-02 00:14:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e4c2d0eb8f2e64d28fbd712589f043df8c145956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 00:28:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e54e8d2365be466d901a5a4a90a2eccb : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/644ac09125c34bcea829ccf990f81baa : SUCCESS in 5m 38s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/03162dda4d1f40c0a24caeb7c17a2a8a : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7022b6f0a26ae640d1fdde450337f14e0a6962ef","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-05-02 06:27:44.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/11/697111/1/check/DellEMC_Unity_Manila_dhss_true/9a47624/DellEMC_Unity_Manila_dhss_true/468 : SUCCESS in 3h 01m 23s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/11/697111/1/check/DellEMC_Unity_Manila_dhss_false/2755592/DellEMC_Unity_Manila_dhss_false/501 : SUCCESS in 1h 25m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"fe0abb84ebf151cf39fef848e7114339e5834779","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-05 16:35:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(24 comments)\n\nThanks for your patience with this documentation, Carlos! Looks really good, some comments inline..","accounts_in_message":[],"_revision_number":1},{"id":"8324842a67e799cc8e021786e7be5ec59279e719","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-06-03 13:59:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"50fdd3e3ab66f8b9902f3bc2580609e5150d65f1","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-06-03 14:00:17.000000000","message":"Patch Set 2:\n\n(24 comments)\n\nThanks for reviewing this change, Goutham. I have addressed the answers and changes regarding your comments. Please check them inline.","accounts_in_message":[],"_revision_number":2},{"id":"93cd1d2076912e8e02b6b15882905c4923bf385f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 14:15:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a75b62b24d8046ffb13622c0530c97e9 : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/708f4818ba19408d86d3698719e8add0 : SUCCESS in 5m 25s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/925f5f0322734f9cbdc0c36b1a96d6a5 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":2},{"id":"89653efbf45ae72f2f75bd8de00e1e12a4add4a3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-03 15:15:56.000000000","message":"Patch Set 2:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-11-697111-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"cf8c58155ddbcfd3f8cd8e8c8ff24c0f503ea66e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-06-03 15:25:05.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/11/697111/2/check/DellEMC_PowerMAX_Manila/5b2632e/DellEMC_PowerMAX_Manila/51 : SUCCESS in 1h 25m 42s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"f4531212cc1a8c66988e77da87aa7a471d001cf0","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-06-03 16:00:33.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/11/697111/2/check/DellEMC_VNX_Manila/e55c04d/DellEMC_VNX_Manila/39 : SUCCESS in 1h 05m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"63e5ccea21c13d0e719d926ffe462d19f5e4e694","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-03 16:29:11.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-11-697111-2/ns5_manila : FAILURE in 2h 29m 42s","accounts_in_message":[],"_revision_number":2},{"id":"27d8d01e30a2c68323d3864bd075c82ea9bcce4b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-03 16:48:40.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/11/697111/2/check/DellEMC_Unity_Manila_dhss_true/954e4c3/DellEMC_Unity_Manila_dhss_true/44 : SUCCESS in 1h 51m 28s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/11/697111/2/check/DellEMC_Unity_Manila_dhss_false/1966899/DellEMC_Unity_Manila_dhss_false/44 : SUCCESS in 51m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"a3e6110e855bb54f908e78b6baec8c121121dcee","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-03 19:48:20.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)\n\nLooks good to me, thanks Carlos! If you make another patch, please address the nit inline..","accounts_in_message":[],"_revision_number":2},{"id":"a1483ecb747efdcf8d345389acdcdf4c0edac55c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-06-04 12:35:14.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"dc06a8441a9505a1cc934c327b1a56e4467ed186","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-06-04 12:36:02.000000000","message":"Patch Set 3:\n\nThanks again for reviewing it, Goutham!\nJust changed as you suggested :)","accounts_in_message":[],"_revision_number":3},{"id":"5d7ac59073c0bea63cb0f9454db0fd55bea0c495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:51:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acc3354f72ae44c888249029f96be398 : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e41d3688cfc40cc84777c20f9623d63 : SUCCESS in 6m 35s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b67b567d09214c9a81fcf5f42a09e506 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":3},{"id":"3c54ae0a66ec51da748c5759b55ba2031467e9fe","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-04 13:57:18.000000000","message":"Patch Set 3:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-11-697111-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"6ed6c5d16c71485d82ba20d53084b2b1a433e3c6","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-06-04 13:57:36.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/11/697111/3/check/DellEMC_PowerMAX_Manila/aada054/DellEMC_PowerMAX_Manila/53 : SUCCESS in 1h 22m 12s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"57de5ad4f51dae2df9d3d1b8ecdded2938c8e80b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-04 14:27:39.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/11/697111/3/check/DellEMC_Unity_Manila_dhss_true/4e8fa96/DellEMC_Unity_Manila_dhss_true/49 : FAILURE in 1h 52m 15s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/11/697111/3/check/DellEMC_Unity_Manila_dhss_false/d8a1179/DellEMC_Unity_Manila_dhss_false/49 : SUCCESS in 46m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"6d3a9dfb0fd26fa86b803499a9b4b129e55525f4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-04 14:42:56.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-11-697111-3/ns5_manila : FAILURE in 2h 7m 23s","accounts_in_message":[],"_revision_number":3},{"id":"6363c71d7145de12ec7d07ed246afe0aa4664ac3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-04 16:31:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nminor slip up..","accounts_in_message":[],"_revision_number":3},{"id":"9fbff57ba4a57f0c2d5fafbe762f0a46d089da95","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-06-04 16:37:11.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"69c2d282cba7b431c3f4e27f09ccd2701436cbd6","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-06-04 16:37:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e109f64037c19c0cb9d881f7a8f330f62d6381f7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-04 16:37:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"1845490e01353a19e57453b8d2c4b4874db77e96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 16:54:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86bc26ce9619437280ac3a72920f3dd1 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42d209b64690424cbd38a0112226ecdb : SUCCESS in 7m 59s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8273d9c7e96540f2a97be84e046058a5 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":4},{"id":"273186850de200def6775ea0bf4cd9d85a772408","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 16:54:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"30020721d7f333770acf0e249ff158a33c4fbe29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:08:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ab53e34755f64d5637c437de272aa8d3e2ed592a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:08:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac496ce54427440eb9f6d1d70ed6adc2 : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa536978d7ad4c488a134cae8b1f3977 : SUCCESS in 7m 58s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4ed383d286f54beba2950cedc234b1f5 : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":4},{"id":"792ec421ab8e78087fdf9e82881111ed4199dddf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:10:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6684b5df32d442babca04a50980358c : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":4},{"id":"3f455e39d718b0ae63ab8b9e796888d1cd45b54d","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-06-04 17:56:51.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/11/697111/4/check/DellEMC_PowerMAX_Manila/5c04661/DellEMC_PowerMAX_Manila/55 : SUCCESS in 1h 19m 29s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"bf36b0f91f6e3634807fa4f56081d7aacd334a60","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-04 17:59:52.000000000","message":"Patch Set 4:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-11-697111-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"6fdda69b5e55af59cd7dc1b94db030beb1e23a68","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-06-04 18:26:55.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/11/697111/4/check/DellEMC_Unity_Manila_dhss_true/bf02e58/DellEMC_Unity_Manila_dhss_true/51 : SUCCESS in 1h 49m 32s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/11/697111/4/check/DellEMC_Unity_Manila_dhss_false/de18712/DellEMC_Unity_Manila_dhss_false/51 : SUCCESS in 52m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"2e979d79014a39645104bf5b12c7d269d3cd03fd","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-06-04 18:41:37.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 43d0241b9ce6f3d4c65655265c9ca857d5a658eb","accounts_in_message":[],"_revision_number":4},{"id":"ce2b320f72f2aa483001be8da7b44070b73fc441","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-04 19:45:15.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-11-697111-4/ns5_manila : FAILURE in 2h 43m 4s","accounts_in_message":[],"_revision_number":4},{"id":"a4c677dc353c45d25963f69d9bce2c441ef1faaf","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-17 12:47:34.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-11-697111-4/ns5_manila : FAILURE in 3h 6m 25s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fee462c2f0bd33bc4b160c63ee8264b9dc2e60f3","revisions":{"08f6505006ee4bd5b9dda8dc1700703a9760c364":{"kind":"REWORK","_number":1,"created":"2019-12-03 15:25:57.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/11/697111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/11/697111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/11/697111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/11/697111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/11/697111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/11/697111/1"}}},"commit":{"parents":[{"commit":"572be43bb03a67376b52a11c70e587eb160b68a3","subject":"[NetApp] Fix share replica failing for \u0027transfer in progress\u0027 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/572be43bb03a67376b52a11c70e587eb160b68a3"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-12-03 01:19:52.000000000","tz":0},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-12-03 13:31:49.000000000","tz":0},"subject":"Update admin guide for share network subnets change","message":"Update admin guide for share network subnets change\n\nThis patch updates the admin guide to make it up to date with the\nintroduced changes for share networks in Train release.\n\nChange-Id: I6ae48e2f174e0197fdeeaca5fd5a96a23c8bb182\nPartial-Bug: #1843953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/08f6505006ee4bd5b9dda8dc1700703a9760c364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/08f6505006ee4bd5b9dda8dc1700703a9760c364"}]},"branch":"refs/heads/master"},"0cdaed3efe9f8ae2b74a73cc6595103da09b72c0":{"kind":"REWORK","_number":2,"created":"2020-06-03 13:59:12.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/11/697111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/11/697111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/11/697111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/11/697111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/11/697111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/11/697111/2"}}},"commit":{"parents":[{"commit":"572be43bb03a67376b52a11c70e587eb160b68a3","subject":"[NetApp] Fix share replica failing for \u0027transfer in progress\u0027 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/572be43bb03a67376b52a11c70e587eb160b68a3"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-12-03 01:19:52.000000000","tz":0},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2020-06-03 13:58:40.000000000","tz":0},"subject":"Update admin guide for share network subnets change","message":"Update admin guide for share network subnets change\n\nThis patch updates the admin guide to make it up to date with the\nintroduced changes for share networks in Train release.\n\nChange-Id: I6ae48e2f174e0197fdeeaca5fd5a96a23c8bb182\nPartial-Bug: #1843953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0cdaed3efe9f8ae2b74a73cc6595103da09b72c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0cdaed3efe9f8ae2b74a73cc6595103da09b72c0"}]},"branch":"refs/heads/master"},"4980877dfeac426eec341e6803d1ef148e1b9262":{"kind":"REWORK","_number":3,"created":"2020-06-04 12:35:14.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/11/697111/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/11/697111/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/11/697111/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/11/697111/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/11/697111/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/11/697111/3"}}},"commit":{"parents":[{"commit":"572be43bb03a67376b52a11c70e587eb160b68a3","subject":"[NetApp] Fix share replica failing for \u0027transfer in progress\u0027 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/572be43bb03a67376b52a11c70e587eb160b68a3"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-12-03 01:19:52.000000000","tz":0},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2020-06-04 12:34:51.000000000","tz":0},"subject":"Update admin guide for share network subnets change","message":"Update admin guide for share network subnets change\n\nThis patch updates the admin guide to make it up to date with the\nintroduced changes for share networks in Train release.\n\nChange-Id: I6ae48e2f174e0197fdeeaca5fd5a96a23c8bb182\nPartial-Bug: #1843953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4980877dfeac426eec341e6803d1ef148e1b9262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4980877dfeac426eec341e6803d1ef148e1b9262"}]},"branch":"refs/heads/master"},"fee462c2f0bd33bc4b160c63ee8264b9dc2e60f3":{"kind":"REWORK","_number":4,"created":"2020-06-04 16:37:11.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/11/697111/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/11/697111/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/11/697111/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/11/697111/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/11/697111/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/11/697111/4"}}},"commit":{"parents":[{"commit":"572be43bb03a67376b52a11c70e587eb160b68a3","subject":"[NetApp] Fix share replica failing for \u0027transfer in progress\u0027 error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/572be43bb03a67376b52a11c70e587eb160b68a3"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2019-12-03 01:19:52.000000000","tz":0},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2020-06-04 16:37:02.000000000","tz":0},"subject":"Update admin guide for share network subnets change","message":"Update admin guide for share network subnets change\n\nThis patch updates the admin guide to make it up to date with the\nintroduced changes for share networks in Train release.\n\nChange-Id: I6ae48e2f174e0197fdeeaca5fd5a96a23c8bb182\nPartial-Bug: #1843953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fee462c2f0bd33bc4b160c63ee8264b9dc2e60f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fee462c2f0bd33bc4b160c63ee8264b9dc2e60f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
