)]}'
{"id":"openstack%2Fmanila~706749","triplet_id":"openstack%2Fmanila~stable%2Ftrain~Ib79ae3f08843414a552c89cd9a254b5e562123b8","project":"openstack/manila","branch":"stable/train","topic":"manila/generic-improvements","hashtags":[],"change_id":"Ib79ae3f08843414a552c89cd9a254b5e562123b8","subject":"generic: Refactor network functions to l3_init","status":"MERGED","created":"2020-02-10 08:36:54.000000000","updated":"2020-02-15 04:48:23.000000000","submitted":"2020-02-10 13:12:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706749-1581340336768-e685ee51","meta_rev_id":"fd4fa1c12e5e77d1b0646a100349a94fcd954dc5","_number":706749,"virtual_id_number":706749,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-15 04:48:23.000000000","post_submit":true,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 13:12:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":2,"date":"2020-02-10 11:46:55.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":1,"date":"2020-02-10 11:46:55.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 11:46:55.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-02-10 13:12:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-15 04:48:23.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"}],"messages":[{"id":"fec6017908eaac3943791d0d15f17be8c4c4400a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-02-10 08:36:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"865d179e3a4a08ebb52ad59c27bab24c08a80d1a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-02-10 08:40:26.000000000","message":"Topic changed from manila/generic-improvements-train to manila/generic-improvements","accounts_in_message":[],"_revision_number":1},{"id":"a01e9cfd3cde8de38c5c8e620ddbdc85c286dd13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 10:44:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4275e9c657c24727a3afc2423ad9a59f : SUCCESS in 12m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ff6c214e5d7473ba23c56ea7f8a95c8 : SUCCESS in 17m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94148990ecd64f928a7d68cfb4a70a13 : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d4cb15fb6b24488916cc41d059c52f6 : SUCCESS in 7m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4aeb925cc264476a846cc361d412c7b : SUCCESS in 9m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a4a6271009a45a3ace7ee0337869648 : SUCCESS in 10m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9b81dd2f0e2415abcaa9415be0b97d1 : SUCCESS in 8m 27s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e9d904e2824f4374aa1bc7ca27ecca0f : SUCCESS in 7m 19s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/c9705d1db6414082aabb8870b554f88b : FAILURE in 1h 22m 43s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/62ee9be74e9e4d9fb392a4c4eaf77417 : SUCCESS in 57m 48s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/813839e5046d4df0bc66d6450fd7d7c5 : FAILURE in 1h 04m 06s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/cc249d083f024153a329fc23f4edbf13 : SUCCESS in 56m 52s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/636652611d53477da595c613e06a9a0e : SUCCESS in 1h 02m 15s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa212c06c2df4c029717f314c53301d2 : FAILURE in 1h 43m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/7d844284a95e4c3fb50c81a13598d92b : SUCCESS in 28m 45s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/98022eda11f043698062203fc6fba058 : FAILURE in 1h 20m 53s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5c69a66dcf464773b507bd4e33d5c54f : SUCCESS in 1h 20m 04s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/7a8915d101b64632813eef6077dc8777 : SUCCESS in 48m 35s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/4287be5cef1946bda678d97ef94f44df : SUCCESS in 52m 18s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/1935dee4a99d41e5ab69e7bc4d54e7c1 : SUCCESS in 1h 28m 12s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/17f86ba4078c4c4fa745407e2ee060b7 : SUCCESS in 1h 10m 41s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/26f0dbf4bd214da3948f9f108e21f5ec : SUCCESS in 31m 59s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/56b48e46d3784ecfb4272833c6e03048 : SUCCESS in 42m 03s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/dee49ae3f76b4bc19403eb9b452b0cf8 : SUCCESS in 43m 31s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4858163801f94c858150a1e81bfd7fe0 : SUCCESS in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da39e10073b78490279a6acdada3c34aaad8ab65","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-02-10 11:46:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f017931255ee1350b4d1bb4892c4808c09443e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 11:47:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd3b98ddad76ddabdc2ca4065c1d42e8daba3d5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:12:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3f9b06f281c4420b4c055a8877097d6 : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60c141c3bff946a0b4ca3f081ef20fe7 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5af290bd9764dc892b61562083db6e3 : SUCCESS in 8m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c3134edc0cb4839967cc3fff2bcfe1a : SUCCESS in 12m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f91f186b705475b87ab984f52c84c99 : SUCCESS in 8m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6fb1abae6a142b4ba47ecef08d527e4 : SUCCESS in 8m 30s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f1305518b7af4ee3afcdadd8363b0359 : SUCCESS in 9m 14s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/2ae929d984664572b3bf67c725f39d89 : SUCCESS in 51m 36s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/5c549d1e5e664fa1b380639c322b2165 : SUCCESS in 1h 04m 54s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/f0756aaf6b45435f89088951ebdbe928 : SUCCESS in 1h 20m 20s","accounts_in_message":[],"_revision_number":1},{"id":"07ef8fd568f4d0ad7dd6d8727397ad0c2d56be7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:12:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b8f2bd94da8d2ed8d78d4c783e68c4e3ef1b24c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:16:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3081f6f18f645dc83d0570ed7af395b : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"fd4fa1c12e5e77d1b0646a100349a94fcd954dc5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-02-15 04:48:23.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/49/706749/1/check/dsvm-tempest-infinibox-nfs/7901872 : EXCEPTION","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64b7712ab23e5159e9d1cf4c6fef5b90ee821bf3","revisions":{"64b7712ab23e5159e9d1cf4c6fef5b90ee821bf3":{"kind":"REWORK","_number":1,"created":"2020-02-10 08:36:54.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/49/706749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/706749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/706749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/706749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/706749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/706749/1"}}},"commit":{"parents":[{"commit":"9ff0d6d9fb2c52b22f6948fa90793a7e083ed1ec","subject":"Use StrOpt for instance type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9ff0d6d9fb2c52b22f6948fa90793a7e083ed1ec"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-30 15:38:10.000000000","tz":60},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-02-10 08:32:33.000000000","tz":60},"subject":"generic: Refactor network functions to l3_init","message":"generic: Refactor network functions to l3_init\n\nThere\u0027s a few functions which are used inside service_instance\nwhich do networking that can live inside the linux interface\ndriver code.\n\nThis patch refactors them into there as this helps allow for\nthe noop driver.\n\nChange-Id: Ib79ae3f08843414a552c89cd9a254b5e562123b8\n(cherry picked from commit e8adbf3cd4aa7a3a010a20a8fc728107e52acd4b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/64b7712ab23e5159e9d1cf4c6fef5b90ee821bf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/64b7712ab23e5159e9d1cf4c6fef5b90ee821bf3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
