)]}'
{"id":"openstack%2Fmanila~706752","triplet_id":"openstack%2Fmanila~stable%2Ftrain~I2c3d6e4234bf7185b8da3c8e1701069c3a165ffc","project":"openstack/manila","branch":"stable/train","topic":"manila/generic-improvements","hashtags":[],"change_id":"I2c3d6e4234bf7185b8da3c8e1701069c3a165ffc","subject":"Introduce noop interface driver","status":"MERGED","created":"2020-02-10 08:36:54.000000000","updated":"2020-02-15 04:46:46.000000000","submitted":"2020-02-10 15:18:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706752-1581347888907-4f5f8f67","meta_rev_id":"acc42e7fb7dda6b6a1740b905b5715eaaf2c0100","_number":706752,"virtual_id_number":706752,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-15 04:46:46.000000000","post_submit":true,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 15:18:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":2,"date":"2020-02-10 11:50:03.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":1,"date":"2020-02-10 11:50:03.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 11:50:03.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-02-10 15:18:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-15 04:46:46.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"}],"messages":[{"id":"87e4fd468bb38afa75c5b9b058fc317efd639582","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-02-10 08:36:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"172771d9d7fef94e6fe1579b0c36b65885631e0b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-02-10 08:40:38.000000000","message":"Topic changed from manila/generic-improvements-train to manila/generic-improvements","accounts_in_message":[],"_revision_number":1},{"id":"e8c027e012df6f4551d851814854143f42d9c14c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 10:37:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef5b6a44b7845869ab071da923d11d0 : SUCCESS in 9m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d7be5ae81ca4697885452ba3009a6d5 : SUCCESS in 15m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddf7a7daf93649fdb2ba840098a3d271 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a4cecd11c17432ead0c2589a094d14d : SUCCESS in 7m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c474f37574ea4f198b1438eb8a86fdb4 : SUCCESS in 10m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e531124945164227a8f2b480a1b9c76f : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0dbc743147d442b5ae8b19d7b3604115 : SUCCESS in 8m 32s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3792ff8f91244d41ae9ee08b7f0dafe1 : SUCCESS in 8m 04s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/40563b335aac462083f720a28bd829be : FAILURE in 1h 21m 38s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1a506416c2f04f909db3144441d39dfa : SUCCESS in 48m 31s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/6962296bd5964937bda9789797466b21 : SUCCESS in 1h 12m 21s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/5e61e2554a574bf99d01bfbcd9b986da : SUCCESS in 1h 00m 59s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/5df7d6c325834b81b590cad06ccc977b : SUCCESS in 1h 13m 11s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d15cbc950cdb4c20bb62e933fe11427a : FAILURE in 1h 47m 26s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/90d354def55641d6bb9e9d6eafc5f3a4 : SUCCESS in 35m 15s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b51d9b7803054e86b809012279622d21 : SUCCESS in 1h 31m 02s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/79273cb62007485fb6e2160704d055ce : SUCCESS in 1h 27m 08s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/31c3cd0f055a4555b869ea414f617bdf : SUCCESS in 47m 06s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/5599dc590d034bc3802781bdc53a777a : SUCCESS in 44m 13s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/a673d8f290f9439e81fa6d7b0b3f699b : SUCCESS in 1h 07m 36s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/3f39858a6cb24dafbbd7f0e3ebc3b8a1 : SUCCESS in 1h 20m 36s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/9eaa6c9a667342d58d6bc0d61ffec22a : SUCCESS in 31m 54s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9bf66415ee454312843a760d62faa9eb : SUCCESS in 41m 53s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2d4db192c78443acbafc98b9f3c8a511 : SUCCESS in 46m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/75ac35dd140c458ebf9e15065b36236c : SUCCESS in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f397c24a22ef73e6e81f0e1431d46ca037435d9e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-02-10 11:50:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6426da60790e3c8f33c568a67ef636336feb24e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:53:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70bb8a3e74f06ad46c0a941d4ae88bf8a6ead298","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 15:18:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcf68124ab234ea4ba5f411f6b681204 : SUCCESS in 11m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae815300d80e4eafa8498f5dff836d9a : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9116dd5a2e247bd9b0cfc428e665ecb : SUCCESS in 7m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e75c440af573457bb6486e6845565e9f : SUCCESS in 14m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82ea3b7bd0f44ef9b15ef676f3fc75d0 : SUCCESS in 8m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f2d52cc97df46e5a6a025d991d92d73 : SUCCESS in 9m 56s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6a7ff2b157f945c1b3ba5cae0b5a7176 : SUCCESS in 6m 13s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ec083b142f0845fea8f4c35244f6fe5d : SUCCESS in 45m 12s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/3067869022e74c89a045fca33dc7cacf : SUCCESS in 1h 13m 04s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/9ee2803c80a24f80ac0346818eaa1535 : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":1},{"id":"69b6aa78ef6077ae8ac2154cb18ffef8e8fbea07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 15:18:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"acc42e7fb7dda6b6a1740b905b5715eaaf2c0100","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-02-15 04:46:46.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/52/706752/1/check/dsvm-tempest-infinibox-nfs/fe7a507 : EXCEPTION","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b9ba1d3a78ebe32ea6953a24a4c7cd630b83979","revisions":{"3b9ba1d3a78ebe32ea6953a24a4c7cd630b83979":{"kind":"REWORK","_number":1,"created":"2020-02-10 08:36:54.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/52/706752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/706752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/706752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/706752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/706752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/706752/1"}}},"commit":{"parents":[{"commit":"ccc781d0847ae0e5c4903108a93c148675e06b43","subject":"Refactor route clearing to linux interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ccc781d0847ae0e5c4903108a93c148675e06b43"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-29 14:02:26.000000000","tz":60},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-02-10 08:33:05.000000000","tz":60},"subject":"Introduce noop interface driver","message":"Introduce noop interface driver\n\nIn a Manila deployment using the Generic driver, it is possible\nthat the administrator has deployed manila-share on a machine that\nalready has access to the admin network used by Manila by being on\nthe same VLAN for example.\n\nThis means that it\u0027s not necessary for it to attempt to plug the ports\nand do all the wiring due to the fact that it\u0027s already reachable,\ntherefore this patch introduces a new `interface_driver` which is\ndoes nothing in order to avoid plugging/unplugging ports and avoiding\nthe requirement of running neutron-(openvswitch|linuxbridge)-agent\non the same node as manila-share.\n\nThe tests are pretty trivial because there\u0027s not much logic, but we\nat least make sure that we don\u0027t call any ip_lib functions which means\nwe\u0027re not touching anything in host networking.\n\nChange-Id: I2c3d6e4234bf7185b8da3c8e1701069c3a165ffc\n(cherry picked from commit 6aec28f11f2ae7504d776b932800433263fd82da)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3b9ba1d3a78ebe32ea6953a24a4c7cd630b83979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3b9ba1d3a78ebe32ea6953a24a4c7cd630b83979"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
