)]}'
{"id":"openstack%2Fmanila~710639","triplet_id":"openstack%2Fmanila~master~I5f358266739f1c42343d5a0c5ec8109c8fcaac4d","project":"openstack/manila","branch":"master","topic":"bug/1655427","hashtags":[],"change_id":"I5f358266739f1c42343d5a0c5ec8109c8fcaac4d","subject":"Enforce policy checks for share export locations","status":"MERGED","created":"2020-03-01 12:12:57.000000000","updated":"2020-03-24 15:15:58.000000000","submitted":"2020-03-24 08:32:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710639-1585038779000-795c03e7","meta_rev_id":"ef731de2bdf4cdf4b9c1375b6ec50d54c9de7661","_number":710639,"virtual_id_number":710639,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-03-23 18:37:53.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-24 08:32:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-24 06:58:45.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-24 06:56:38.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-03-24 05:55:43.000000000","_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"date":"2020-03-23 19:18:48.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"date":"2020-03-24 08:16:58.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-03-23 17:49:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2020-03-23 17:27:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-23 17:35:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2020-03-24 01:27:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-03 10:28:56.000000000","updated_by":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"reviewer":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"state":"REVIEWER"},{"updated":"2020-03-04 02:14:28.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-04 20:45:50.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-17 09:19:31.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2020-03-23 17:35:57.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-03-23 17:49:33.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-03-23 18:37:53.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-03-23 19:18:48.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-24 01:27:21.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-03-24 05:55:43.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2020-03-24 06:56:38.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-24 06:58:45.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-24 08:16:58.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-03-24 08:32:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd596acf0f36db6af8f48bcf00b377cf34ed59b6","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-01 12:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b32141ae3301fa403cb31acbbf438cf59217c258","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-01 12:13:37.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/965/710639/1/check/infortrend-manila-ci/03ab67c0db809ce0e6436343df46ae5333d71735 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"483db8d13ccca170ac27b8cf24c36584c14096b8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-01 13:15:12.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/710639/1/check/DellEMC_Unity_Manila_dhss_true/0fe19e4/DellEMC_Unity_Manila_dhss_true/173 : FAILURE in 1h 02m 04s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/710639/1/check/DellEMC_Unity_Manila_dhss_false/69292e1/DellEMC_Unity_Manila_dhss_false/179 : FAILURE in 52m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"be34efb36e28a669712e58b28fbf60b1640f6476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 13:57:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c22aa5dd8e149eba2bb39000b675a25 : SUCCESS in 9m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/266341ee11fc4fc6b325bb6f74be4ea6 : SUCCESS in 15m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4036c0281363469cabaf7a142355738c : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82020ad2aefc4194bbd34310df813ec7 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e401d800eac496fa773218fe07f76a1 : SUCCESS in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c19ab5335b04cea9a2b127cb9c246dc : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea805b2cdb7b47f59dfcb5bcb31b561d : SUCCESS in 7m 51s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1c96cb9961604d8282b79f0a82493875 : SUCCESS in 5m 55s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/748c5dd79f1949719588f0cb4046f057 : FAILURE in 1h 39m 04s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/d4824a9bc4b94d8da567f494cfd8f098 : FAILURE in 49m 33s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/f2fff254a2f744e488405c9d5c2fd960 : SUCCESS in 1h 15m 35s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/b0b530021f8b4286af6ad152487e23a8 : SUCCESS in 55m 15s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/c26478ef00074909bbb678e9e20f3e25 : SUCCESS in 1h 01m 49s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7d6e86211ddd4320b8d674700a285662 : FAILURE in 1h 33m 16s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/d3300848229d4b6ca48feca2ce72cc74 : FAILURE in 24m 42s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cb5ae70a95fa4af6b7f352b0a83d9669 : FAILURE in 1h 17m 11s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a7f7fe46a4cc4a81931eedb19b2b9284 : FAILURE in 1h 07m 43s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/6897a98d92fd4720b52d117722a94691 : FAILURE in 1h 07m 56s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/92d2c7c8a3734752b1f138a6b3ab4590 : FAILURE in 42m 59s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/87ae4ddf34704f6dabd3f92a28ec9319 : SUCCESS in 1h 22m 58s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/3980a5c92cc2403482ac60e9d742931f : SUCCESS in 1h 05m 58s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8dac08d8ecbd45abbbaf10da5452097d : FAILURE in 25m 04s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6b948aa996d546d8b8272febe84bf744 : SUCCESS in 43m 32s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/986bbdcdd74c47f99738cb20f63c0922 : SUCCESS in 42m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4ee076f668c145a9a83926bac0c9624a : SUCCESS in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8335956446a4526577e0c7e92e6c198e352cc272","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-01 14:07:54.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-39-710639-1/ns5_manila : FAILURE in 1h 54m 31s","accounts_in_message":[],"_revision_number":1},{"id":"edc8563e6e467237cad3f646d4068e2b8c5553af","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-01 14:10:10.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/39/710639/1/check/DellEMC_PowerMAX_Manila/8cfa2bd/DellEMC_PowerMAX_Manila/197 : FAILURE in 1h 56m 50s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"a1bee0855b948ced19d21515a3442663d445da83","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-01 20:38:44.000000000","message":"Topic set to bug/1655427","accounts_in_message":[],"_revision_number":1},{"id":"986929930e46b4faa5b799c376efc8140fa70b5f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-01 20:38:46.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"02399fc061335d5a520d763bb438e30429fd65c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-01 20:39:27.000000000","message":"Patch Set 2:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/966/710639/2/check/infortrend-manila-ci/42540e0f0eee5fb2b60c1d37de131f8fd54aff6a : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"3380f92a4df62e6a627a3fc272a18754d9fcc538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 20:42:58.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"11bd44c742c44a0e42f45ad6f995fe51e00c64bf","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-01 20:45:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8d64bbb628c1da41fa39da141e3ac449a7282196","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-01 20:46:03.000000000","message":"Patch Set 2:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/967/710639/2/check/infortrend-manila-ci/42540e0f0eee5fb2b60c1d37de131f8fd54aff6a : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"d079144eec57e55560a7e91509c121ec4099920f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-01 20:50:14.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e3f25a6d27469ccbfe7fa90bf0f75b8e7284f986","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-01 20:50:53.000000000","message":"Patch Set 3:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/968/710639/3/check/infortrend-manila-ci/2af4ca8e1c60e3d786a56177f4b1be872befa932 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"0af3c2b3af60f0e3ef34a75b102d96cd8aac5d41","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-01 21:54:38.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/710639/3/check/DellEMC_Unity_Manila_dhss_true/65101d2/DellEMC_Unity_Manila_dhss_true/175 : FAILURE in 1h 04m 07s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/710639/3/check/DellEMC_Unity_Manila_dhss_false/9693d67/DellEMC_Unity_Manila_dhss_false/181 : FAILURE in 48m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"5e509bae82ea86cc886e06510f45384e2bca71fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 22:24:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac34001153b14ae9a98d93bd46ec3d43 : SUCCESS in 12m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37f7119a00c14e849c9e4c6270fe2f2b : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17aef7d560de4ba985b6f7b59f17b299 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bddd2393f2af467291bd33ff6a07db0f : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ed4e5516a704c62a31e4e22644046c0 : SUCCESS in 7m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6e488d50cfd4d70b44160692b6bb8fb : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7b0587cb63f41b4b293fb6a737bc5b6 : SUCCESS in 8m 16s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/06ce771df62b45c29bfa62b0ef9a68b2 : SUCCESS in 5m 35s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/6795ddfbdf334dd7abc617e72d2a18a8 : FAILURE in 1h 33m 21s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/49e7d66fdee5452a9470fad9779bf154 : SUCCESS in 46m 45s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/e8cab36a34b946b0a9a3f01cb63afea5 : FAILURE in 1h 12m 37s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/9e66cd6ba3d940abb695ee0bcd855de4 : SUCCESS in 52m 21s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/9c04b3a1e1ac443fb3e49427e8fc77fd : SUCCESS in 59m 19s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/879397e834d3456eb4a03e3ce38fa0c3 : FAILURE in 1h 15m 35s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/2c0d61d57dc246548ab1f01982f762eb : SUCCESS in 26m 45s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9e6476c70a7e45e5a24b5278ca75f5d8 : SUCCESS in 1h 17m 57s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4b1e501425c846dcbcbebc6e32227719 : SUCCESS in 1h 11m 16s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/74487d0cd84143e3b15d744c6eb8ad8b : SUCCESS in 1h 07m 22s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/423914725d9444c283ec316af35ba5c0 : SUCCESS in 41m 40s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/d03758ae66254b9cb4e8e4b93fb45c7b : SUCCESS in 1h 05m 30s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/209dc63b53674b46807c7db170020879 : SUCCESS in 1h 00m 20s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c8a645e0ff704a8da4e3021004f47fa6 : FAILURE in 21m 22s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b4461b8c865b4ec5bb2be488f2d7cbbf : SUCCESS in 42m 32s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/91342002323d4df69d49017e0ea392eb : SUCCESS in 40m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/910e8deb993a4f74a0303b19be573ea1 : SUCCESS in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c77768195bfae2113a43a87ebaa8b24338813562","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-01 22:26:53.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-39-710639-2/ns5_manila : FAILURE in 1h 47m 45s","accounts_in_message":[],"_revision_number":2},{"id":"8390c0548788c0c9f7187822b81dd51471d188e7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-01 22:36:48.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-39-710639-3/ns5_manila : FAILURE in 1h 46m 14s","accounts_in_message":[],"_revision_number":3},{"id":"c214d95ea2e562272c694dfbcc3e4de0354e4895","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-01 22:44:42.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/39/710639/3/check/DellEMC_PowerMAX_Manila/a2e48d3/DellEMC_PowerMAX_Manila/199 : FAILURE in 1h 54m 09s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"7499e938fc6c6dabba17e340ef299a3226709410","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-01 23:29:17.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/710639/3/check/DellEMC_Unity_Manila_dhss_true/8c24110/DellEMC_Unity_Manila_dhss_true/176 : FAILURE in 1h 04m 30s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/710639/3/check/DellEMC_Unity_Manila_dhss_false/ec76bc7/DellEMC_Unity_Manila_dhss_false/182 : FAILURE in 50m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d554e1f7145652b538d42a44bc50acf47c8eec82","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-02 02:11:58.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-ss http://logs.openstack.netapp.com/logs/39/710639/3/upstream-check/manila-cDOT-ss/e013830/ : FAILURE in 1m 03s\n- manila-cDOT-no-ss http://logs.openstack.netapp.com/logs/39/710639/3/upstream-check/manila-cDOT-no-ss/ed22dd1/ : SUCCESS in 1h 31m 03s\n- manila-cDOT-no-ss-IPv6 http://logs.openstack.netapp.com/logs/39/710639/3/upstream-check/manila-cDOT-no-ss-IPv6/1a7eb00/ : FAILURE in 1h 57m 24s (non-voting)\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"09727e322201efd8f19fd284d9f49225f1ba82d8","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2020-03-02 13:43:57.000000000","message":"Patch Set 3:\n\nrun-Quobyte Ci","accounts_in_message":[],"_revision_number":3},{"id":"503473041ed89eb34daa2e7b357abbd1c75f42ea","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2020-03-02 14:49:00.000000000","message":"Patch Set 3:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":3},{"id":"bd6886867d63c317fac3f7ab92d86ff26366dc65","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-03 01:23:53.000000000","message":"Patch Set 3:\n\nNeeds a release note","accounts_in_message":[],"_revision_number":3},{"id":"ca85d45e97eac07e54636d411742754930f69949","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-03 03:38:12.000000000","message":"Patch Set 3:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/39/710639/3/check/dsvm-tempest-infinibox-nfs/69c3fab : UNSTABLE in 52m 05s","accounts_in_message":[],"_revision_number":3},{"id":"8c19383c5cc0c5b182067407e72d8efde6815a08","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-03 07:12:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b93d7998a364d7741232bcccb3c6988448490174","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-03 07:12:45.000000000","message":"Patch Set 4:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/970/710639/4/check/infortrend-manila-ci/0c5e66700493d6c65de3b50deaca3d00bd8113c4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"75ba1e9439950275e774b2709b5fb4069f75daad","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-03 07:41:27.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/39/710639/4/check/DellEMC_PowerMAX_Manila/9f7417e/DellEMC_PowerMAX_Manila/202 : FAILURE in 29m 12s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"e65243a171ebbf3e36ef25bee24bc05d4dc3c084","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-03 07:41:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b06e06e426b39e0d04ad12dba87467cdc962cbbf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-03 08:01:27.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/710639/4/check/DellEMC_Unity_Manila_dhss_true/58d8046/DellEMC_Unity_Manila_dhss_true/179 : FAILURE in 49m 13s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/710639/4/check/DellEMC_Unity_Manila_dhss_false/b4cc37b/DellEMC_Unity_Manila_dhss_false/185 : FAILURE in 40m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"04893644fb41e8dc9f734b3f191f9e1df53c5006","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-03 08:09:35.000000000","message":"Patch Set 4:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/39/710639/4/check/dsvm-tempest-infinibox-nfs/6e7782f : UNSTABLE in 56m 57s","accounts_in_message":[],"_revision_number":4},{"id":"a5714e05c50229f4d3cb8cd39c7ec20cccb7f910","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-03 08:27:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9898959c9ccc29586a38f6cd1af41ccade2092df","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-03 09:01:04.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-39-710639-4/ns5_manila : FAILURE in 1h 48m 43s","accounts_in_message":[],"_revision_number":4},{"id":"26b2ee621eb07005d738c141f36d2dc4fe4d4544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 09:20:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2732af2c2a7f4030aa182e623994b41c : SUCCESS in 13m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adaf4a1feda44f4cbf11f5d941edfb99 : SUCCESS in 16m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9e47f1df5124ea1a3ce0572a21cac81 : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc450cdafc6e47cd815f986137ee2e93 : SUCCESS in 8m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcd09a4b8f014ef7990492e74a4610c0 : SUCCESS in 15m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d71097a37514651a78595821760ff0e : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6dab0bcf28e4375addf74945327de76 : SUCCESS in 12m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81c508e214654cb399816e3d667a9ed1 : SUCCESS in 11m 18s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d7dd48866c584ce5af5b5242169ba2c8 : SUCCESS in 8m 08s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/5a2ceb326ef9425a9f0725a2a5fecd5c : SUCCESS in 1h 38m 20s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/bd756b220b2c4718b816de3a59572556 : SUCCESS in 52m 41s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/91bd79e3eada4c78b1fc81b8b4b045e9 : FAILURE in 1h 21m 39s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/a9d3eabececa471b9935e5b486423617 : SUCCESS in 54m 20s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/6122678cd98f4e12af70344b202c06e1 : SUCCESS in 1h 06m 08s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9af3b4eae8af45ef9821501463961b55 : FAILURE in 1h 45m 52s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/02ced1cda6cf4ac0b05171238bcbb6a9 : SUCCESS in 29m 27s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2eaf3de5edc340d2b5808779bc4e069e : SUCCESS in 1h 33m 05s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9f13039c98714c2ea142366b04946550 : FAILURE in 1h 58m 01s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/eb31cbe7dd814098ac64cd1f16e12bd4 : SUCCESS in 1h 27m 19s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f5679960550243a694d85c2475404740 : SUCCESS in 48m 44s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/c92451dfa1de4c76a46a84dab302384f : SUCCESS in 1h 06m 19s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/5e0588d0e2054da48352e89d56779815 : SUCCESS in 1h 07m 30s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/add5c9d9270541cbb34dd476a5d66cd4 : FAILURE in 28m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c542d6ce04a24431a7ac52b3047f4d6c : SUCCESS in 40m 45s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d545ad3f2b86446087292b56dc1243f3 : SUCCESS in 43m 24s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b9bfbb06b7d046869038c911e586618e : SUCCESS in 7m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"35cae32cba49eaf5b5063046591a2dbc8f71659e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-03-03 09:50:34.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/39/710639/4/check/DellEMC_VNX_Manila/49547be/DellEMC_VNX_Manila/104 : FAILURE in 29m 51s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"cd464ca71ffb83a22e524ce983bc9f7bb3eef6a8","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-03 09:52:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"42242cb04f51b691c142c28407a7628539972d10","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-03 09:53:13.000000000","message":"Patch Set 5:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/972/710639/5/check/infortrend-manila-ci/6db1850e7f9a6f9fcf760e89f49372ff57d0eb10 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"c1a50c3863b595a295af21856b8f323fa531e84a","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-03 10:16:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2d4342fad15db0c232d59c14ecf5439d81c87288","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2020-03-03 10:28:30.000000000","message":"Patch Set 3:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":3},{"id":"0c32edbb43b3f21c3014ea0f6b24891b6f5f1785","author":{"_account_id":13915,"name":"Silvan Kaiser","email":"silvan@quobyte.com","username":"kaisers"},"date":"2020-03-03 10:28:56.000000000","message":"Patch Set 5:\n\nrun-Quobyte CI","accounts_in_message":[],"_revision_number":5},{"id":"05d9724994cb31cbc4986a868954da0a5d61bb8a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-03 10:42:59.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/710639/5/check/DellEMC_Unity_Manila_dhss_true/5aa83b0/DellEMC_Unity_Manila_dhss_true/182 : FAILURE in 50m 05s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/710639/5/check/DellEMC_Unity_Manila_dhss_false/e64d595/DellEMC_Unity_Manila_dhss_false/188 : FAILURE in 39m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"267d28e55ebe997d47a24fc53aa6f7d0facc2c89","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-03 10:46:43.000000000","message":"Patch Set 5:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/39/710639/5/check/dsvm-tempest-infinibox-nfs/9bd589d : UNSTABLE in 53m 35s","accounts_in_message":[],"_revision_number":5},{"id":"24d6068f32166feacb38c96460306c49d180bab4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-03 11:09:30.000000000","message":"Patch Set 4:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-39-710639-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"5312c6ff143afadcdba88787d6059a1dafc6b08b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-03 11:35:19.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/39/710639/5/check/DellEMC_PowerMAX_Manila/c475cb5/DellEMC_PowerMAX_Manila/204 : FAILURE in 1h 42m 25s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"c56bfb20668f7093f874be778b27b0c04c678d11","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-03 12:01:52.000000000","message":"Patch Set 5:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-39-710639-5/ns5_manila : FAILURE in 2h 8m 57s","accounts_in_message":[],"_revision_number":5},{"id":"4b45dbe83e686f7764af64c857eb668ed7c28f15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 12:02:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8be1ce209434f3f9a485ef208a501a1 : SUCCESS in 15m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c35adf23f2954f7d96c244addcb1b473 : SUCCESS in 13m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86bb5070d02441cd8e85d5f048ccba61 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1250e124c0bf421b8e250be1cdfa91d7 : SUCCESS in 9m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afe97edf2f4a4f568fc2c4026e6ce99b : SUCCESS in 8m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/679eb68d07b149398f09392e3eba8b61 : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2feccddb2fd84e839d29d470e02f6d28 : SUCCESS in 9m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4670f993ab2a45aeb6e3f507eb2611cf : SUCCESS in 10m 15s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ee1ca193acef4f24ae09dfad767158dc : SUCCESS in 6m 22s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/f9283fe6aa124e239e1d8207a23f8588 : SUCCESS in 1h 40m 25s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/b34f2e619a4e4ed795ca933856f4cef7 : SUCCESS in 52m 35s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/8d40e65ad2db4c679f41001d832543df : SUCCESS in 1h 12m 22s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/80bd42dd96d248e2815fb9c49ed181bc : SUCCESS in 1h 00m 46s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/65c44aa54efc40499f3c5f910b2a76a1 : FAILURE in 1h 12m 13s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e450e6d34394e59bdc6358c206d7010 : FAILURE in 1h 54m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/9cc3e9ab66974102b96a99faf9867973 : SUCCESS in 29m 03s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/71415418acd940ffa05bd08a2b2f7204 : FAILURE in 1h 26m 34s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a46021c9f05c45149230f7692b426b33 : SUCCESS in 1h 10m 08s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/ec646875bce543f99ffa1645b7569b56 : SUCCESS in 1h 05m 13s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/27c441f90c57442eb0eb8e6058bf3985 : SUCCESS in 47m 11s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/918c422605c440b19b5d0d80510c561e : SUCCESS in 1h 09m 04s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/ac8008536d4d4745bf97e8b32f444c0b : SUCCESS in 1h 24m 25s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/cebfb63f10d24d60ba513f67413c81e9 : FAILURE in 23m 34s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/af090dd3de284871884964a44c50573b : SUCCESS in 38m 57s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/59033c5c07e6462fafe38b65f12d6b18 : SUCCESS in 45m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ea0247818f524f14be440fbad7f9257d : SUCCESS in 7m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c13441aceaee9e949da04f61001f360cddd96e20","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-03 13:01:29.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/710639/5/check/DellEMC_Unity_Manila_dhss_true/5611b80/DellEMC_Unity_Manila_dhss_true/184 : FAILURE in 59m 07s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/710639/5/check/DellEMC_Unity_Manila_dhss_false/0016770/DellEMC_Unity_Manila_dhss_false/190 : FAILURE in 41m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"6aea6bab832994016858312bc7f8e7334123ab1a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-03-03 13:53:33.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/39/710639/5/check/DellEMC_VNX_Manila/ef4d5b1/DellEMC_VNX_Manila/106 : SUCCESS in 1h 51m 05s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"72c2f5fb9594f02e97dc01fe645c0da475266740","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-03 14:33:34.000000000","message":"Patch Set 3:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-39-710639-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"894415499e05de99991460804a4e29df822cfc4c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-03 14:35:18.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-ss http://logs.openstack.netapp.com/logs/39/710639/5/upstream-check/manila-cDOT-ss/6abd29b/ : SUCCESS in 2h 32m 10s\n- manila-cDOT-no-ss http://logs.openstack.netapp.com/logs/39/710639/5/upstream-check/manila-cDOT-no-ss/6e199ef/ : SUCCESS in 1h 22m 53s\n- manila-cDOT-no-ss-IPv6 http://logs.openstack.netapp.com/logs/39/710639/5/upstream-check/manila-cDOT-no-ss-IPv6/ae72fc9/ : FAILURE in 5m 59s (non-voting)\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"d0bbece9ba18a78bc98bfd286756d6d73d017675","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-03 14:58:16.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/39/710639/5/check/DellEMC_PowerMAX_Manila/7b02a6a/DellEMC_PowerMAX_Manila/206 : FAILURE in 1h 43m 12s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"668c38f2234ab7e95bb39463710e477a6660b6d1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-03 15:54:50.000000000","message":"Patch Set 5:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-39-710639-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"fa9d673887a1f3eabc2a56ec139e7892f62dfe92","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-03 17:16:46.000000000","message":"Patch Set 5:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-39-710639-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"838ee9a3e6f8c1cfd55eeda453965a74254c1cc1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-03 17:52:28.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a1aac033cb6fd25b8c241189badd7535afbbba7e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-03 22:11:29.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"db2c837f533a15de733c837b09d2f7c66d246d6f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-03 22:12:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7077b346a8eeae32d858afa0b8099e509315196f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-03 22:12:55.000000000","message":"Patch Set 6:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/973/710639/6/check/infortrend-manila-ci/55a7566a2ee192d68c0b650bb0b49c90bc861d97 : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"2b92df981e1653f61934d8e27b8b7dd9612b7b95","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-03 23:00:02.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/710639/6/check/DellEMC_Unity_Manila_dhss_true/b165cc3/DellEMC_Unity_Manila_dhss_true/185 : FAILURE in 47m 33s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/710639/6/check/DellEMC_Unity_Manila_dhss_false/cac4a7f/DellEMC_Unity_Manila_dhss_false/191 : FAILURE in 41m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"e16ba17d4f1c14df973d8afd5a33f18ccc2522ef","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-03 23:05:44.000000000","message":"Patch Set 6:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/39/710639/6/check/dsvm-tempest-infinibox-nfs/fdda934 : UNSTABLE in 52m 46s","accounts_in_message":[],"_revision_number":6},{"id":"6e2cd9f035460d0e9f0ded77ded17587e651b67e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 23:57:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07c78a72836740cab5ab34e95740e2df : SUCCESS in 10m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec6b9ee92d254d55a4c5e61472792d3b : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a710322387548688322e4262d8c08c8 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55667d72d1584fcb86bc2ef864d67127 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aebb5a37b340486b8fcea016d01581f0 : SUCCESS in 7m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6cefc34ca7bc4aa89e14570bffa050c0 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/259e0801c1224cfd881b9ef0fd4f24e7 : SUCCESS in 9m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f427a738446411f866e207f02cb792f : SUCCESS in 10m 19s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/fa84e0067051491aa3835caaa2fee8f7 : SUCCESS in 6m 39s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/940bd426d2d042218aa0a2d267896390 : SUCCESS in 1h 39m 10s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/7e9958bd0a9146a5bbbf3319c4eb155f : SUCCESS in 51m 41s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/0bc1b74fda2f4aa28127cc2a87ddb5d1 : FAILURE in 1h 09m 59s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/7d5678aa6d014a3ea83821bb9c5e243f : SUCCESS in 56m 03s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/43c423c5ff1b4f6a847e7afcbc273ed7 : SUCCESS in 1h 08m 02s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7553d27cbb5a494bb583f930bc8163e1 : FAILURE in 1h 36m 07s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/45aa69c98b2b4962bf5253704ca11ef4 : SUCCESS in 28m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2756a6f1743a4a04aa3262a7fb5942b9 : SUCCESS in 1h 16m 37s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5c674c3b45d648ed9628f10f45b94637 : SUCCESS in 1h 20m 54s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/c5bdc86f74274198bf28a46a7839f8a7 : SUCCESS in 1h 07m 19s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/b62de0c55ec84aafbb7bc1742d4bf23e : SUCCESS in 45m 31s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/a691d4ba93a94382a0a97f6fd3b5e117 : SUCCESS in 1h 02m 21s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/d901f086f1184f9e86828ef25c6c45ee : SUCCESS in 56m 12s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5f50fc314ff2406dbf0e62da38e11d8f : FAILURE in 27m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f21a86f5a8d646cba1c851ab1f32c79d : SUCCESS in 36m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/027a3921e5cd4a89a7bb0e9416038550 : SUCCESS in 37m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0999b743eb804bd59d59a968008cf8c1 : SUCCESS in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"720b7d264b56bc341917d04696f005c088cbefc6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-03 23:59:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"04131999c4a3707a84d554d320867c3ef68b8c40","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-03 23:59:50.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/39/710639/6/check/DellEMC_PowerMAX_Manila/09c0e11/DellEMC_PowerMAX_Manila/207 : FAILURE in 1h 47m 21s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"02a74aac1f95a18dfee54054f94a1e93b6d72f93","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-04 00:21:47.000000000","message":"Patch Set 6:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-39-710639-6/ns5_manila : FAILURE in 2h 9m 12s","accounts_in_message":[],"_revision_number":6},{"id":"959dd336d77e4409d9068d5e2f93324c2987c84a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-04 00:53:50.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/710639/6/check/DellEMC_Unity_Manila_dhss_true/8a541d5/DellEMC_Unity_Manila_dhss_true/186 : FAILURE in 56m 32s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/710639/6/check/DellEMC_Unity_Manila_dhss_false/6d4d5c4/DellEMC_Unity_Manila_dhss_false/192 : FAILURE in 50m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"01903007bb616e88dc80ca34f627587bc28358b6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-04 01:58:24.000000000","message":"Patch Set 6:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-39-710639-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"4a6ee9f868d08bb3965991c0746aa5e35d513a9a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-03-04 02:02:46.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/39/710639/6/check/DellEMC_VNX_Manila/448aa1c/DellEMC_VNX_Manila/107 : SUCCESS in 2h 05m 22s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"2aae402b1daf1810579820e84e89b7c665d317ec","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-04 02:14:28.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-cDOT-ss http://logs.openstack.netapp.com/logs/39/710639/6/upstream-check/manila-cDOT-ss/a5cf5c5/ : POST_FAILURE in 3m 07s\n- manila-cDOT-no-ss http://logs.openstack.netapp.com/logs/39/710639/6/upstream-check/manila-cDOT-no-ss/b1e1ecc/ : SUCCESS in 1h 31m 36s\n- manila-cDOT-no-ss-IPv6 http://logs.openstack.netapp.com/logs/39/710639/6/upstream-check/manila-cDOT-no-ss-IPv6/c3bb173/ : FAILURE in 2h 16m 39s (non-voting)\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"2bf00ffc36aa7050a39e7e1ea8032802bc990f1f","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-04 09:17:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f7aab46b7f27247425976ab064e941eed0d3a170","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-04 09:18:08.000000000","message":"Patch Set 7:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/975/710639/7/check/infortrend-manila-ci/c424ca53e0a9efe7e5b2ef10264404d69d0b3fb5 : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"49baebfbb2cbe45bf89ef781b4b7a3abee6bf46b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-04 09:50:54.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/39/710639/7/check/DellEMC_PowerMAX_Manila/29d65c2/DellEMC_PowerMAX_Manila/209 : FAILURE in 33m 15s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"e153434ad4a4eaf10bad9834d60060efb90dd4b3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-04 10:15:27.000000000","message":"Patch Set 7:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/39/710639/7/check/dsvm-tempest-infinibox-nfs/dd4de72 : UNSTABLE in 57m 26s","accounts_in_message":[],"_revision_number":7},{"id":"ad67a0616e159e976630174b631d998c7882b73b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 11:14:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2197640d249f4173affcc41ef0d58539 : SUCCESS in 10m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c632e1fab88241b582cb5eab9e3a822b : SUCCESS in 17m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/119a493fc4874ee585263f1c986f73bd : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d5c6b6c6a2a404d99dc1f848e094e3d : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b6ab75db9dd4dc4af46898af4052e46 : SUCCESS in 7m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5df6a513eca43cc92a59d0e03da713c : SUCCESS in 16m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f2e74067857438798c4f391820da3b5 : SUCCESS in 15m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba7e6dc640b0400fb837a36442a9a02b : SUCCESS in 9m 03s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/250f66a22e9a448a88eb293946b14dc4 : SUCCESS in 6m 10s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/193902ca307243cb92d60b8046e5f317 : SUCCESS in 1h 24m 25s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/d2758b0d2eca4538b572219f64320462 : SUCCESS in 1h 08m 55s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/18f2a04b4dce401ab3d35cbe407086e0 : FAILURE in 1h 17m 57s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ede283061f104269b6ab8784d102c95b : SUCCESS in 1h 11m 07s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/db8cb0bafeb240e1b87ef32bcb690100 : FAILURE in 1h 08m 55s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96a9037e14584d648105aa8c3c106fd8 : FAILURE in 1h 50m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/78a9dc478c734cc290d115ea5e60198d : SUCCESS in 22m 19s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/af0a760abc074a9f8af0de1ed2739812 : SUCCESS in 1h 20m 29s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5f278e3e50ec454e8038b2610c9fc351 : SUCCESS in 1h 05m 34s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/bec6d4bfb9584405af46dbef2fe45425 : SUCCESS in 1h 07m 40s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/6fc418a8147147c594361e37062054be : SUCCESS in 38m 59s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/e1cf449174d1406b887192d167707a8f : SUCCESS in 1h 12m 25s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/9b8048e1e5684d7d822e969cf3e9304e : SUCCESS in 1h 12m 12s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/9e36270f37f049ee8380feef86b1de7a : FAILURE in 21m 51s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2cbce5c797b24d55a989e4541be7864f : SUCCESS in 39m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1faf4ef87d524adeb8336d5b26f170ce : SUCCESS in 43m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/28d42f38ad78459eae9e17b3074142ff : SUCCESS in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6f797a84f05b9cce95241bc992eb937d848fcec2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-04 11:34:51.000000000","message":"Patch Set 7:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-39-710639-7/ns5_manila : FAILURE in 2h 17m 6s","accounts_in_message":[],"_revision_number":7},{"id":"57c29910a211966375a4860b148a70c006fd1edf","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-03-04 11:53:35.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/39/710639/7/check/DellEMC_VNX_Manila/c8c6ea5/DellEMC_VNX_Manila/109 : FAILURE in 14m 14s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"5ed973df2bd841d13f2e08f5802db9031c6d5f7c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-04 13:06:00.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/39/710639/7/check/DellEMC_PowerMAX_Manila/244606d/DellEMC_PowerMAX_Manila/210 : FAILURE in 1h 51m 21s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"cb5cb969196154a9c34e0eeb59fe28a8ddf19089","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-04 13:24:45.000000000","message":"Patch Set 7:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-39-710639-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"a4056501210da1b6fcc30e03776174287fe94d4f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-04 16:41:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b6471860f9c5e9fb7b486fb253135652ec9bbb7c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-04 16:42:01.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"2de58d1693cd664ec15f3914361174dff7f51bba","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-04 16:42:41.000000000","message":"Patch Set 8:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/976/710639/8/check/infortrend-manila-ci/5720d4f4ef9d9e1a76fe5ef20c725280eb23cbbb : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"6cbd7894ef09abecb59955f010769a081c7b94f9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-04 16:43:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"89b923d890a909cfc4d08fed434ee6b77f749abd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-04 17:38:27.000000000","message":"Patch Set 8:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/39/710639/8/check/dsvm-tempest-infinibox-nfs/0c38f22 : UNSTABLE in 55m 49s","accounts_in_message":[],"_revision_number":8},{"id":"382c9b3965729d73a73e300fdc5acc666b0bb922","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-04 18:27:49.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/39/710639/8/check/DellEMC_PowerMAX_Manila/5e3a750/DellEMC_PowerMAX_Manila/211 : FAILURE in 1h 45m 36s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"47b97d0f7f5cfaf6f8f950b311d42a823991a6c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 18:54:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e3fa710e7fb478597771dbb3ea0471c : SUCCESS in 13m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99e531624d9f4752bc767f9b61411a74 : SUCCESS in 14m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ee34ff81a1249839dd245f2a0542c30 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aab46aad770d4cdfb9ab1c9ea56fe45e : SUCCESS in 13m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5caee2b149494521863826b9edebd1f4 : SUCCESS in 12m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fdccf5c944bc40359acc4e2753de5638 : SUCCESS in 11m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faa5acf278374cfca4c906cd304b66f1 : SUCCESS in 11m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64cca772d09a41a890ad0caf92f981a0 : SUCCESS in 10m 02s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/595fd8429bb5472dbcbc2cae2deef52c : SUCCESS in 8m 25s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/58a2588c6b734c3b8ac7c3c84d3a618c : FAILURE in 1h 31m 49s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/c5631ddb73334cbcbfc94b718b90732a : SUCCESS in 49m 44s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/8071cf0a7a024358adca7aeb6c176a53 : SUCCESS in 1h 07m 45s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/89dd6308e7c143bea6392637a61f90e3 : SUCCESS in 53m 33s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e0bfc225ce244aa99c6926b381f73f89 : SUCCESS in 1h 13m 34s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a57d7ad9f0364c2db392a6a6b3f0c40a : FAILURE in 1h 56m 25s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/d79f9359717d4e61ad528310e309feeb : SUCCESS in 27m 05s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ea5160b6ce75466dba017b9b0d9a2223 : SUCCESS in 1h 18m 31s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b5d7c25c66cb42e59fa51d4326fa06ac : SUCCESS in 1h 28m 50s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/7f9826a3cafe401ca629ad1946cf5253 : FAILURE in 1h 56m 44s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/7618a08ae61642a497fb038a9c943366 : SUCCESS in 47m 03s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/15c35991731a418da37d0f625dca7bb0 : SUCCESS in 1h 03m 04s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/24041fe9174d4459a0156b58d44b3ec7 : SUCCESS in 1h 01m 20s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/cc9fb6e1823f46b08fc04bdaec5ff3cc : FAILURE in 22m 27s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d1e4d32e185144b2b67876ff8f14d36d : SUCCESS in 46m 29s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/32cf42f3cec940889e3d9d73d37ff323 : SUCCESS in 47m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8ffd70c05bb64af5a3e2d801831ea150 : SUCCESS in 6m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"02403aaa231398b48c2045db5b1685e06920a38c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-04 18:54:45.000000000","message":"Patch Set 8:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-39-710639-8/ns5_manila : FAILURE in 2h 12m 19s","accounts_in_message":[],"_revision_number":8},{"id":"3ce7569293e2e40217d687e37013d84e754cdef5","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-04 20:21:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b76607ac80f4d9a33360a1c0aca3ddb96d13e6b5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-04 20:44:02.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/39/710639/8/check/DellEMC_PowerMAX_Manila/97cdc33/DellEMC_PowerMAX_Manila/212 : FAILURE in 1h 49m 24s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"bbc9f232e4775b3bf067c476b61d687b2a1d0da2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-03-04 20:45:50.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/39/710639/8/check/DellEMC_VNX_Manila/2dd60e3/DellEMC_VNX_Manila/110 : SUCCESS in 1h 51m 05s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"0addfbd14536bee5bc47fa235c8a82876bd4fd84","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-04 20:53:20.000000000","message":"Patch Set 8:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-39-710639-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"7956a23f5b127ab2479ffcbeeabbeca8d56b6713","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-03-05 11:05:48.000000000","message":"Patch Set 8: Code-Review+2\n\nThanks Tom, LGTM.","accounts_in_message":[],"_revision_number":8},{"id":"57b082d8a57e5066dbdf972df78b0996c4318b93","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-03-09 06:08:31.000000000","message":"Patch Set 8:\n\nhi, Tom and Goutham Pacha Ravi.\nAs for this bug, I think the most direct way to fix it is modify the database layer\u0027s API logic. Please look at the following code.\nmanila/db/sqlalchemy/api.py:1426 ---\u003e def share_instance_get();\n\n@require_context\ndef share_instance_get(context, share_instance_id, session\u003dNone,\n                       with_share_data\u003dFalse):\n    if session is None:\n        session \u003d get_session()\n    result \u003d model_query(\n        context, models.ShareInstance, session\u003dsession,\n    ).filter_by(\n        id\u003dshare_instance_id,\n    ).options(\n        joinedload(\u0027export_locations\u0027),\n        joinedload(\u0027share_type\u0027),\n    ).first()\n    if result is None:\n        raise exception.NotFound()\n\n    parent_share \u003d share_get(context, result[\u0027share_id\u0027], session\u003dsession)\n    if not context.is_admin:\n        if not parent_share[\u0027is_public\u0027] \\\n                and context.project_id !\u003d parent_share[\u0027project_id\u0027]:\n            raise exception.NotFound()\n    if with_share_data:\n        result.set_share_data(parent_share)\n\n    return result\n\nin this function, we determine whether the share instance belong to the project of context. how do you think about it?","accounts_in_message":[],"_revision_number":8},{"id":"c3c1bcc98ad8c7e8c9582e55d4a0a9938f2dc46c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-09 21:02:42.000000000","message":"Patch Set 8:\n\n\u003e hi, Tom and Goutham Pacha Ravi.\n\u003e As for this bug, I think the most direct way to fix it is modify the database layer\u0027s API logic. Please look at the following code.\n\u003e manila/db/sqlalchemy/api.py:1426 ---\u003e def share_instance_get();\n\nHi Haixin, yes that fix would work as well. However, it may not be any more direct than the change Tom\u0027s making in this patch. In general, we should probably secure gathering resources via the db_api by setting \"project_only\" [1] to True by default. We will explore that route on the master branch - I don\u0027t wish to make a sweeping change like that in the fear that our testing may not be adequate.\n\n[1] https://opendev.org/openstack/manila/src/commit/ef6b9a0c72a0e47d7345286911a3d882ee1f1909/manila/db/sqlalchemy/api.py#L238","accounts_in_message":[],"_revision_number":8},{"id":"faf18ad1dc72a6d69d6d02fd4833639bf14b957e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-16 20:02:29.000000000","message":"Patch Set 8:\n\n@haixin, @gouthamr: Would it make sense to merge and backport this, then to merge Haixin\u0027s patch on master branch?","accounts_in_message":[],"_revision_number":8},{"id":"cac24d77a03a48267adbd71b84635dcc961ccc69","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-03-17 07:07:10.000000000","message":"Patch Set 8:\n\nhi, Tom and Goutham Pacha Ravi.\ni have try to solve this bug with \"project_only\" set to True.\nhere is the link:https://review.opendev.org/#/c/713376/\nwe can discuss about this bug.","accounts_in_message":[],"_revision_number":8},{"id":"cc0d8814e0f923e9dadd730de59b995597ff21c2","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-03-17 09:19:31.000000000","message":"Patch Set 8: Code-Review+1\n\nas gouthamr said \" I don\u0027t wish to make a sweeping change like that in the fear that our testing may not be adequate.\" i agree merge this patch, my patch is Just a Just a sample, for reference onlysample, for reference only,Because there\u0027s a lot of room for optimization.","accounts_in_message":[],"_revision_number":8},{"id":"d912dd3cd1b71ee6bbde0919d9dc434dafc86dcc","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-03-17 12:10:17.000000000","message":"Patch Set 8: Code-Review+2\n\nLGTM, Tom. Thanks!","accounts_in_message":[],"_revision_number":8},{"id":"581edca771b2de76d6740a0ab6717156675470b0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-23 17:27:15.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"7991b2efd934dcda1fe7389ba45cae978b939523","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-23 17:27:47.000000000","message":"Patch Set 9: Code-Review+2\n\nRemoved the Depends-On since the tempest side change has merged.","accounts_in_message":[],"_revision_number":9},{"id":"baa875b6d1751e923df7f6ce7a7d305fb2fb3f23","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-23 17:27:57.000000000","message":"Patch Set 9:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/1016/710639/9/check/infortrend-manila-ci/84daeb481d852d6531df11e842df1a70672d938c : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"634514d97ce4195592e6863daf6e44077405911f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-23 17:28:52.000000000","message":"Patch Set 9:\n\n\u003e Removed the Depends-On since the tempest side change has merged.\n\nJust to clarify, we preserved the tests instead of skipping/dropping them by changing the test setup. Please see: https://review.opendev.org/#/c/714181/ for more details.","accounts_in_message":[],"_revision_number":9},{"id":"1c45531f6340a8c1eb6404e36fe8fab967b2cdf1","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-03-23 17:35:57.000000000","message":"Patch Set 9: Code-Review+2\n\nOnly the commit message was updated since my last review, everything still looks great. Thanks Tom!","accounts_in_message":[],"_revision_number":9},{"id":"b976bb96123090b865f6f4dc8f6f43fa27509b04","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-03-23 17:49:33.000000000","message":"Patch Set 9: Code-Review+2\n\nLGTM, Tom! Thanks for working on this!","accounts_in_message":[],"_revision_number":9},{"id":"9665567f304d57d755f4cd30c2b0f7af0c832c88","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-23 18:13:35.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/710639/9/check/DellEMC_Unity_Manila_dhss_true/8e19881/DellEMC_Unity_Manila_dhss_true/243 : FAILURE in 46m 09s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/710639/9/check/DellEMC_Unity_Manila_dhss_false/30f415a/DellEMC_Unity_Manila_dhss_false/258 : SUCCESS in 45m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"b26ddf5f6b7e5f39033cabf16913aae1df46648f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-23 18:37:53.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/39/710639/9/check/dsvm-tempest-infinibox-nfs/ce0a0a3 : SUCCESS in 1h 10m 04s","accounts_in_message":[],"_revision_number":9},{"id":"726c0c085dd26cfb0be149632692b6095afc5eca","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-23 18:51:29.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/39/710639/9/check/DellEMC_PowerMAX_Manila/2f9e73e/DellEMC_PowerMAX_Manila/255 : FAILURE in 1h 24m 02s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":9},{"id":"568217313c10d20b6a3fbe1758ca5db8e2626f9d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-23 19:18:48.000000000","message":"Patch Set 9:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-39-710639-9/ns5_manila : SUCCESS in 1h 51m 16s","accounts_in_message":[],"_revision_number":9},{"id":"0dd04095e1286f26ccd994c646b2d966907951ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 19:21:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc1bb35894b74e8dbe5ff3c75ff88b5d : FAILURE in 7m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/394bb26f4c3b45928a95277ea6feb1f6 : SUCCESS in 14m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95ae31941f524b26a51a4093f41cf93a : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44f760fc95324ff889fa090703b7ff36 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e379c9af5ac48e4bedd9eb8d8813f2a : SUCCESS in 10m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fa6af5cdb3a426eb5cd8d070148b5f1 : SUCCESS in 9m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/943905876f0148838ecc94e18e56e0ba : SUCCESS in 8m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7e891edbee449a7b1bee34c8939f63c : SUCCESS in 10m 47s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/27bc4ff8f11a4079a835fb75a0415340 : SUCCESS in 5m 16s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/0393a3eca2304108bce212722e2b3c11 : SUCCESS in 1h 50m 26s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/0cb9ca09905d4f6f8fc76f6f4625decc : SUCCESS in 48m 27s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/eb0f97b0f22d468f9ff040134441dbdb : FAILURE in 1h 12m 27s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/f3a1b902db8c415cbd840f16b8982cc2 : SUCCESS in 51m 08s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/f02fa3464f7b4a03b6090a2f26ed2c69 : FAILURE in 1h 11m 38s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/acbcf391656e4366b29c2f2a17447765 : FAILURE in 1h 51m 40s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/6469e46ebaca4cbabd773358f085022d : SUCCESS in 32m 45s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c598f65c3b844dd2a9b9c98722339c75 : SUCCESS in 1h 08m 51s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/75cd6666fd6d4bb281118f1dcc6b73be : SUCCESS in 1h 03m 30s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/e263a184a42b431a94739ae57dfd609d : SUCCESS in 1h 09m 04s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/73322803edd74560bc6a723ca43653e6 : SUCCESS in 46m 45s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/7ca193848f0c48b2873ca9ee8e7d8ef2 : SUCCESS in 59m 03s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/06e763810ecf4b2f992b4244cb2d6d2b : FAILURE in 58m 41s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/48bfebd7c0d14e1f8c0de27142b57ccb : FAILURE in 24m 16s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4152e0516cb24259be6e3c6cf8568999 : SUCCESS in 38m 47s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/10f8e0c49f7e43c387b6afe4bb35268b : SUCCESS in 41m 20s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ae1a320f0ced473fa0908c074aafc3dc : SUCCESS in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7ccdf14f5201dfe6bd425f663973c2052baf2d3f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-24 01:27:21.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"1d80e266faa5282aa6b6e323e76edfd79f764171","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-24 01:28:36.000000000","message":"Patch Set 9:\n\nThe docs job blocker has merged: https://review.opendev.org/#/c/714207/\n\nWorkflowed this change for expediency","accounts_in_message":[],"_revision_number":9},{"id":"a0ec228c2f940a401f213affb7e77ae23bc26e97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 03:14:35.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d27f4358f5f4adf86d9cce047058c82 : SUCCESS in 8m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d2eba5fbb5d4f2da1b6799abd356e6f : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c311f633c0dc4891b7bf7838d9ab1e9d : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0147548b1af34559982e4016f94e2963 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/860c80bd4daf4f7cbe153dafaa074878 : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fdff501cf2c44a2e9c5f583883f82545 : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acd18c90ea1a4a0f8a6232c0b9408427 : SUCCESS in 6m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5acb90947b18426499c7389aa09eacae : SUCCESS in 8m 53s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a35065dbca6b44c3a552568b86bd8e65 : SUCCESS in 4m 58s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/4d321e841db54f82a09a37dd39d3e776 : FAILURE in 1h 26m 23s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/ae0cd8a7165f4f52b02df6a7b9a5d213 : SUCCESS in 54m 35s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/f53434bcc0c248cf9f1e5a0482cda9de : FAILURE in 1h 04m 01s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/b150b00c5720458585114a5ef70d46ca : SUCCESS in 45m 36s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/1a24cfd0c33e4da2b9b0913500bb0259 : FAILURE in 1h 05m 41s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f6354aca27e4f668a43d9a0e6a00e0a : FAILURE in 1h 41m 50s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/93d66c8818e74725ad925ced1adac913 : SUCCESS in 21m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/28fd9d06c317436c950a25ddcdd7daba : SUCCESS in 1h 20m 30s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1d0cfcdf37a34257a0c44a86297d8ffb : SUCCESS in 1h 11m 23s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/8440b946773d4a96855bf5bfe4736734 : FAILURE in 1h 10m 09s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d3e0f1d7b5c64caa9daff76e9663d959 : SUCCESS in 45m 07s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/cb31925b16414e73b21d1f9162d1ba98 : SUCCESS in 1h 03m 07s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/c05e6109e84c43d1bf14ace8885ffacc : SUCCESS in 1h 06m 40s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b87cb29f068047d69e94c40029179dd3 : FAILURE in 19m 33s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7a5be2a9f2694746bd74a7ac5361ca12 : SUCCESS in 35m 54s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c850e9e1d40c4627a0fedd00b2de07df : SUCCESS in 37m 12s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1e15308c753e4d32b8f5da745ed8222b : SUCCESS in 4m 16s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0a045fc12d82ea82310fd01c2c47719164f17ee9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 03:14:47.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"968187571dd63e9810b7c990e30e34601153c308","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-24 05:11:45.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/710639/9/check/DellEMC_Unity_Manila_dhss_true/a41b196/DellEMC_Unity_Manila_dhss_true/247 : FAILURE in 46m 47s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/710639/9/check/DellEMC_Unity_Manila_dhss_false/7824428/DellEMC_Unity_Manila_dhss_false/262 : SUCCESS in 50m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"403abdfad626f82b8332c748547b74f18143f3e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 05:27:59.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bccbc9c4bcd545d69c5b7ad7b736c9f8 : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/264e1b48803b4a82a0b2abb1ccdc1aee : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/232c30a73e804fba915c3343a8cbf7c4 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fdc63a904c241678cf760a2a43b6dfb : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f05023cfe8854703be53a44e70d511cf : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ce274f43a56464fa6abb3e44d305a04 : SUCCESS in 10m 42s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9736709e52a24490832c629da6b89d5b : SUCCESS in 4m 41s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/cf66043b777944f6ab70218bcee179cc : SUCCESS in 43m 20s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/b68fdeba12c941aab87013d8cd98a81b : RETRY_LIMIT in 47m 30s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/e4369f3aed0c4f4880be31c236ca9f5d : SUCCESS in 1h 05m 14s","accounts_in_message":[],"_revision_number":9},{"id":"f4ef2f1789cd53e45ee9a241a0a50199405e00ab","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-24 05:55:01.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"5d4664b10e790ba72421318fa91f6062ce7479d6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-24 05:55:43.000000000","message":"Patch Set 9:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/1021/710639/9/check/infortrend-manila-ci/84daeb481d852d6531df11e842df1a70672d938c : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"fc1c2c1f39d7d5ad03601962216fa2c1d55d07d5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-24 06:56:38.000000000","message":"Patch Set 9:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-39-710639-9 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":9},{"id":"123074175339d73192eac8b01302e9ce19b5080b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-24 06:58:45.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/39/710639/9/check/DellEMC_PowerMAX_Manila/93eff52/DellEMC_PowerMAX_Manila/259 : FAILURE in 1h 20m 50s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":9},{"id":"c147c6990c7236b80cf7053606b5788e94db3128","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 07:24:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a83d011b0014418189dc0a89ab7d15d7 : SUCCESS in 9m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83bbf5beada14d2f975b2a035cd409d5 : SUCCESS in 12m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9951aa2613f042a980d682f7ef212790 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a16521758db4b8c9e1b6998a3f3ae70 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12fc363ab0e94cfc8b023a7319e34a7c : SUCCESS in 7m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7fbe466f16554ce9ae0a5d2dd7afcf41 : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/571fc210585943bb9163659264f3864c : SUCCESS in 7m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35640b109b934cc587876947aaefd193 : SUCCESS in 9m 09s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0ff3f5d9d485415d912d121c0441feb4 : SUCCESS in 5m 38s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/52d2af34514142348d92fd075574c697 : SUCCESS in 1h 28m 16s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/e121fe5ddfaf482c9558e9d1b6a4366d : SUCCESS in 59m 44s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/2ac801dc843c4e0b9170a4ab633fd325 : SUCCESS in 1h 11m 45s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/6fb161b79af543b69ce34cc11ab81abf : SUCCESS in 1h 01m 13s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/693f4a6972fe40cfaa0f5eec129c5749 : SUCCESS in 1h 02m 50s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb6bfbbde1f348a29606ee3ae95fc2ff : FAILURE in 1h 19m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/bbb7ae953dac41e98e232d7a307e0c82 : SUCCESS in 28m 12s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6fd15c8d5ad04231ba1a802017724cfc : SUCCESS in 1h 16m 28s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9f75a76f22ad406d804eb6dc6339c2a1 : SUCCESS in 1h 11m 29s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/b0d534bb0d66491cbb69d466b3ee39e2 : FAILURE in 1h 10m 30s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d65585155eb74f2da6c5668b402242e8 : SUCCESS in 48m 15s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/5a163b2058144fd4a99cd5fc9b68a7bc : SUCCESS in 1h 06m 44s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/abcd7be381294d14b26c5fa715d204be : SUCCESS in 1h 03m 08s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a3863f2e598a45829938e993736884ca : FAILURE in 28m 20s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/343bde11cb3649a3aae858237f095926 : SUCCESS in 42m 34s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a3c47ef581c94d659a3af272daaea1f5 : SUCCESS in 43m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/72bafe2caf984232bafd4af283170f8e : SUCCESS in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"69383c5bbac4a5e522a8b891828cdd69fd990672","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 07:24:27.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"1fc1eebbfb697467865f78f96da933087f16fec5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-24 08:16:58.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/710639/9/check/DellEMC_Unity_Manila_dhss_true/fab868f/DellEMC_Unity_Manila_dhss_true/250 : FAILURE in 46m 23s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/710639/9/check/DellEMC_Unity_Manila_dhss_false/027895e/DellEMC_Unity_Manila_dhss_false/265 : SUCCESS in 52m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"6e0d2c184297a88c4165095b170b040b60f68039","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 08:32:58.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdd9d9b2d483450eb3cd056101f693ff : SUCCESS in 8m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4efde7821854c10b6f8f4a0a8a855e0 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1406025e4ef6476a91ee1f8cc5ca0b1a : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/012171059317475190f2d1efcb414cfe : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcad729fb652406881bc1508dd3461db : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a58687f72355436f879160ad6e2625b4 : SUCCESS in 8m 59s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ef813e2b257e40a389313cd944f3fdf8 : SUCCESS in 4m 21s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e95c32e2898f4563bc6e488ea649c01d : SUCCESS in 42m 12s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/c809ce6fdd684d0ab8e3507d8c97f272 : SUCCESS in 1h 02m 55s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/256950ea96a94f5f954081e60f69abaa : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":9},{"id":"57ae8ed6f26105011004e9867047060bb576aca5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 08:32:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"56abac00973c6963dd06278f4f37eac1b8c0c98a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 08:34:58.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/585dae0d9b22470d97f3a0a8a529570d : SUCCESS in 1m 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/059e4b44c09f44308a9c7b4602235a7f : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":9},{"id":"746295a3b4cea0485766204c595e68cbd48ca2f3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-24 09:35:09.000000000","message":"Patch Set 9:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-39-710639-9 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":9},{"id":"ef731de2bdf4cdf4b9c1375b6ec50d54c9de7661","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-24 15:15:58.000000000","message":"Patch Set 9: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 02fd716bf83849873f0bccb78b851196e6acf2b7","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"84daeb481d852d6531df11e842df1a70672d938c","revisions":{"03ab67c0db809ce0e6436343df46ae5333d71735":{"kind":"REWORK","_number":1,"created":"2020-03-01 12:12:57.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/39/710639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/710639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/710639/1"}}},"commit":{"parents":[{"commit":"a43f0666c8f127a8ec1175211a6bc3e9cf239004","subject":"Merge \"[Unity]: Failed to delete cifs share if wrong access set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a43f0666c8f127a8ec1175211a6bc3e9cf239004"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 12:12:08.000000000","tz":60},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 12:12:08.000000000","tz":60},"subject":"WIP - expecting tempest failures","message":"WIP - expecting tempest failures\n\nChange-Id: I5f358266739f1c42343d5a0c5ec8109c8fcaac4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/03ab67c0db809ce0e6436343df46ae5333d71735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/03ab67c0db809ce0e6436343df46ae5333d71735"}]},"branch":"refs/heads/master"},"42540e0f0eee5fb2b60c1d37de131f8fd54aff6a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-01 20:38:46.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/39/710639/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/710639/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/710639/2"}}},"commit":{"parents":[{"commit":"a43f0666c8f127a8ec1175211a6bc3e9cf239004","subject":"Merge \"[Unity]: Failed to delete cifs share if wrong access set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a43f0666c8f127a8ec1175211a6bc3e9cf239004"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 12:12:08.000000000","tz":60},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 20:38:09.000000000","tz":0},"subject":"Enforce policy checks for share export locations","message":"Enforce policy checks for share export locations\n\nCloses-bug: #1655427\nDepends-on: https://review.opendev.org/710660\nChange-Id: I5f358266739f1c42343d5a0c5ec8109c8fcaac4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/42540e0f0eee5fb2b60c1d37de131f8fd54aff6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/42540e0f0eee5fb2b60c1d37de131f8fd54aff6a"}]},"branch":"refs/heads/master"},"2af4ca8e1c60e3d786a56177f4b1be872befa932":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-03-01 20:50:14.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/39/710639/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/710639/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/710639/3"}}},"commit":{"parents":[{"commit":"a43f0666c8f127a8ec1175211a6bc3e9cf239004","subject":"Merge \"[Unity]: Failed to delete cifs share if wrong access set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a43f0666c8f127a8ec1175211a6bc3e9cf239004"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 12:12:08.000000000","tz":60},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 20:50:04.000000000","tz":0},"subject":"Enforce policy checks for share export locations","message":"Enforce policy checks for share export locations\n\nCloses-bug: #1655427\nDepends-on: I886a6d0c377f677fa12b837f67301f1a6ed94072\nChange-Id: I5f358266739f1c42343d5a0c5ec8109c8fcaac4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2af4ca8e1c60e3d786a56177f4b1be872befa932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2af4ca8e1c60e3d786a56177f4b1be872befa932"}]},"branch":"refs/heads/master"},"0c5e66700493d6c65de3b50deaca3d00bd8113c4":{"kind":"REWORK","_number":4,"created":"2020-03-03 07:12:03.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/39/710639/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/710639/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/710639/4"}}},"commit":{"parents":[{"commit":"a43f0666c8f127a8ec1175211a6bc3e9cf239004","subject":"Merge \"[Unity]: Failed to delete cifs share if wrong access set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a43f0666c8f127a8ec1175211a6bc3e9cf239004"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 12:12:08.000000000","tz":60},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-03 07:11:29.000000000","tz":60},"subject":"Enforce policy checks for share export locations","message":"Enforce policy checks for share export locations\n\nCloses-bug: #1654598\nDepends-on: I886a6d0c377f677fa12b837f67301f1a6ed94072\nChange-Id: I5f358266739f1c42343d5a0c5ec8109c8fcaac4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0c5e66700493d6c65de3b50deaca3d00bd8113c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0c5e66700493d6c65de3b50deaca3d00bd8113c4"}]},"branch":"refs/heads/master"},"6db1850e7f9a6f9fcf760e89f49372ff57d0eb10":{"kind":"REWORK","_number":5,"created":"2020-03-03 09:52:36.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/39/710639/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/710639/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/710639/5"}}},"commit":{"parents":[{"commit":"a43f0666c8f127a8ec1175211a6bc3e9cf239004","subject":"Merge \"[Unity]: Failed to delete cifs share if wrong access set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a43f0666c8f127a8ec1175211a6bc3e9cf239004"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 12:12:08.000000000","tz":60},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-03 09:52:26.000000000","tz":0},"subject":"Enforce policy checks for share export locations","message":"Enforce policy checks for share export locations\n\nCloses-bug: #1654598\nDepends-on: I886a6d0c377f677fa12b837f67301f1a6ed94072\nChange-Id: I5f358266739f1c42343d5a0c5ec8109c8fcaac4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6db1850e7f9a6f9fcf760e89f49372ff57d0eb10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6db1850e7f9a6f9fcf760e89f49372ff57d0eb10"}]},"branch":"refs/heads/master"},"55a7566a2ee192d68c0b650bb0b49c90bc861d97":{"kind":"REWORK","_number":6,"created":"2020-03-03 22:12:17.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/39/710639/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/710639/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/710639/6"}}},"commit":{"parents":[{"commit":"a43f0666c8f127a8ec1175211a6bc3e9cf239004","subject":"Merge \"[Unity]: Failed to delete cifs share if wrong access set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a43f0666c8f127a8ec1175211a6bc3e9cf239004"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 12:12:08.000000000","tz":60},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-03 22:12:08.000000000","tz":60},"subject":"Enforce policy checks for share export locations","message":"Enforce policy checks for share export locations\n\nCloses-bug: #1654598\nDepends-on: I886a6d0c377f677fa12b837f67301f1a6ed94072\nChange-Id: I5f358266739f1c42343d5a0c5ec8109c8fcaac4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/55a7566a2ee192d68c0b650bb0b49c90bc861d97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/55a7566a2ee192d68c0b650bb0b49c90bc861d97"}]},"branch":"refs/heads/master"},"c424ca53e0a9efe7e5b2ef10264404d69d0b3fb5":{"kind":"REWORK","_number":7,"created":"2020-03-04 09:17:28.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/39/710639/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/710639/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/710639/7"}}},"commit":{"parents":[{"commit":"a43f0666c8f127a8ec1175211a6bc3e9cf239004","subject":"Merge \"[Unity]: Failed to delete cifs share if wrong access set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a43f0666c8f127a8ec1175211a6bc3e9cf239004"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 12:12:08.000000000","tz":60},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-04 09:16:53.000000000","tz":60},"subject":"Enforce policy checks for share export locations","message":"Enforce policy checks for share export locations\n\nCloses-bug: #1654598\nDepends-on: I886a6d0c377f677fa12b837f67301f1a6ed94072\nChange-Id: I5f358266739f1c42343d5a0c5ec8109c8fcaac4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c424ca53e0a9efe7e5b2ef10264404d69d0b3fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c424ca53e0a9efe7e5b2ef10264404d69d0b3fb5"}]},"branch":"refs/heads/master"},"5720d4f4ef9d9e1a76fe5ef20c725280eb23cbbb":{"kind":"REWORK","_number":8,"created":"2020-03-04 16:42:01.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/39/710639/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/710639/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/710639/8"}}},"commit":{"parents":[{"commit":"a43f0666c8f127a8ec1175211a6bc3e9cf239004","subject":"Merge \"[Unity]: Failed to delete cifs share if wrong access set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a43f0666c8f127a8ec1175211a6bc3e9cf239004"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 12:12:08.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-04 16:42:00.000000000","tz":0},"subject":"Enforce policy checks for share export locations","message":"Enforce policy checks for share export locations\n\nCloses-bug: #1654598\nDepends-on: I886a6d0c377f677fa12b837f67301f1a6ed94072\nChange-Id: I5f358266739f1c42343d5a0c5ec8109c8fcaac4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5720d4f4ef9d9e1a76fe5ef20c725280eb23cbbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5720d4f4ef9d9e1a76fe5ef20c725280eb23cbbb"}]},"branch":"refs/heads/master"},"84daeb481d852d6531df11e842df1a70672d938c":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2020-03-23 17:27:15.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/39/710639/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/710639/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/710639/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/710639/9"}}},"commit":{"parents":[{"commit":"a43f0666c8f127a8ec1175211a6bc3e9cf239004","subject":"Merge \"[Unity]: Failed to delete cifs share if wrong access set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a43f0666c8f127a8ec1175211a6bc3e9cf239004"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-03-01 12:12:08.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-23 17:27:13.000000000","tz":0},"subject":"Enforce policy checks for share export locations","message":"Enforce policy checks for share export locations\n\nCloses-bug: #1654598\nChange-Id: I5f358266739f1c42343d5a0c5ec8109c8fcaac4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84daeb481d852d6531df11e842df1a70672d938c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/84daeb481d852d6531df11e842df1a70672d938c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
