)]}'
{"id":"openstack%2Fmanila~712166","triplet_id":"openstack%2Fmanila~stable%2Fqueens~Id67a939a475c4ac06d546b7e095bd10f1a6d2619","project":"openstack/manila","branch":"stable/queens","topic":"bug/1861485","hashtags":[],"change_id":"Id67a939a475c4ac06d546b7e095bd10f1a6d2619","subject":"share_networks: enable project_only API only","status":"MERGED","created":"2020-03-10 19:35:13.000000000","updated":"2020-03-11 07:22:41.000000000","submitted":"2020-03-11 07:21:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712166-1583911272234-74ea99eb","meta_rev_id":"f4f676ebb2a5e3ebf16cca51cb77c600e15ce8bd","_number":712166,"virtual_id_number":712166,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-03-11 01:58:21.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-03-11 04:49:00.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-11 07:21:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"recommended":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2020-03-11 03:20:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-03-11 04:14:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2020-03-11 05:46:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-11 01:58:21.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-03-11 03:20:49.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-03-11 04:49:00.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-11 05:46:38.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-03-11 07:21:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d25745e7b53b3e84aff81bd195cbe48f9d53d50","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-03-10 19:35:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"222d06b6b41ea3fe50e02721b956f90c9fe0a02c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-10 21:07:56.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/66/712166/1/check/dsvm-tempest-infinibox-nfs/0afdc5b : FAILURE in 22m 29s","accounts_in_message":[],"_revision_number":1},{"id":"8ac527fc64cdb2a18000f2d18138e3be6e0da133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 21:15:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8896578c1d62487ea5620fb7c81117d9 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/066e77f520bb4a8bbef4580d4b96cb27 : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c50d02c47b1a4eb9a7d3b9485dd41850 : FAILURE in 7m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/072bc093be2e4be9bcfd3c29572a6240 : FAILURE in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ab98721205846d39fc179ae3b51cb96 : SUCCESS in 10m 36s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/da0a17e4af27418e9410feb936cada87 : SUCCESS in 5m 55s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/ed1ca0f887484084914e12573822b30d : SUCCESS in 1h 07m 54s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/7150647c817a461fb545411da250f84e : SUCCESS in 43m 10s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/24ea8e5eede945df96705e92b2206778 : SUCCESS in 1h 11m 05s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/97c5db3e62ac4373ba6031f2a82dd011 : FAILURE in 43m 05s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/abc29d083ee6418a90b91c0bc3fb05a8 : FAILURE in 1h 03m 21s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a86ebede80f14553a300521dc34537b3 : SUCCESS in 1h 17m 10s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/f1df81570d004709948499b952d0008a : FAILURE in 8m 28s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/5c087ffe58bd433591ce98b17ca463a5 : FAILURE in 8m 18s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/0fbfec7fbf4c49e891d5106c0d4a8658 : SUCCESS in 37m 16s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/6e0d2cd8166c4c61a920f2c5e805b7a3 : SUCCESS in 1h 32m 41s\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/7c83d61380034842afa3f076b12cdfae : SUCCESS in 31m 35s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/cb0f2459f4a5435fa86f3d7cb4bfb7f7 : FAILURE in 24m 53s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/9d37704f49854a1fbdaa632e93fc3881 : FAILURE in 19m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf7159dffe3144b2aa76782645b6e4fc : FAILURE in 18m 49s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/916ef060c3334931a3a93e6542fb4f08 : SUCCESS in 11m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dae1832ff2b86ed9b754b4c51b6eebd23367fc44","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-11 01:34:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"88bc3cf83e43ac3cb3cac3105c83e18e15e94660","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-11 01:58:21.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/66/712166/2/check/dsvm-tempest-infinibox-nfs/b195128 : FAILURE in 23m 47s","accounts_in_message":[],"_revision_number":2},{"id":"349065985ac1bfcdeae7a24d94f8460110a8a155","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 03:16:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d000ec1d2fc240b0af0044666792c72f : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4ccaf5fe6ea43ff881bc88ea8038935 : SUCCESS in 7m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7fbc110a663444181318238fa1de98b : SUCCESS in 9m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/20a3c228f7ee43b59ed8299f750d3640 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f71a1f2d0fd45c2afc9b89fa360af05 : SUCCESS in 8m 37s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d32dc687dade43b0bb8f9dece019abbb : SUCCESS in 4m 56s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/5c0033dd714a457ab8aeb98d7500e2aa : SUCCESS in 1h 07m 47s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/e9b6dedc08be47729f50baaccf1c4ea0 : SUCCESS in 38m 36s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/cec7ebede27549c9ac054aeaf4fbef2e : FAILURE in 59m 14s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/61fcbf8f6ce54a4694f352f1855b3bc2 : SUCCESS in 46m 27s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/bea9795503e749f4a4b1f7ff1c618e04 : SUCCESS in 55m 31s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c5e9c5cac454474af0fbafc695540cc : SUCCESS in 1h 11m 06s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/564d5ba12e1041bab03721a69a17f82e : FAILURE in 6m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/943ce0147c354d4eb89406763fe0665a : FAILURE in 5m 30s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/4654a580ca09455aaac1783ec2e05154 : SUCCESS in 34m 30s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/27abc6ffea7c4df49baee2e72c84859d : SUCCESS in 1h 36m 50s\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/0a4976f03be74427bebf5022947c011e : SUCCESS in 22m 43s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/3482f72df5434caca81e2fcffc4d37c2 : FAILURE in 15m 37s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/c2f00a311b9b43679e7334d1567814cf : FAILURE in 19m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a9e74625282498f930fd2a9b7103455 : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1a7270760d35436bbec5a8f634bd9049 : SUCCESS in 9m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2a5de2089606e7e3116217a03bdb255c477b0f3e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-03-11 03:20:49.000000000","message":"Patch Set 2: Code-Review+1\n\nClean cherry-peak, LGTM, thanks.","accounts_in_message":[],"_revision_number":2},{"id":"83b83e73f0c4d94691bf5409477cc44d1f122ca8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-11 04:14:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow-1\n\nLet\u0027s wait for the patch on stable/rocky to merge [1] \n\n[1] https://review.opendev.org/#/c/712165/","accounts_in_message":[],"_revision_number":2},{"id":"bbce4298f428a13d761d01856710b774b2dca8a0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-11 04:49:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss http://logs.openstack.netapp.com/logs/66/712166/2/upstream-check/manila-cDOT-ss/5648567/ : FAILURE in 1h 07m 34s\n- manila-cDOT-no-ss http://logs.openstack.netapp.com/logs/66/712166/2/upstream-check/manila-cDOT-no-ss/0868800/ : SUCCESS in 1h 32m 18s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"675287a3773b4807e42e213c8c43989876fb63f8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-11 05:46:38.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"933b82c1a0626734deee18cc54c6159758e17816","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 05:46:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7fa6002d5542778237ca6bc18cb9296befa02cb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 07:21:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f02fe04145584de987e27736d1b16d51 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37617bd7873a49caa1d486db3297c41d : SUCCESS in 6m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb6fc1d7bee44edda7b43dafa63780b8 : SUCCESS in 7m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ba38c3f8833b47f1a9ca77eb9c0db0d3 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f17316bd49914a95939ba9a174c67fed : SUCCESS in 7m 53s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cfbc6d076074494dbb5d758aa58f4737 : SUCCESS in 7m 09s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e40899ae6267494cbd24997c2046b1fc : SUCCESS in 37m 32s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/c7f23af02ac74d3dbd667ce81507bb0d : SUCCESS in 1h 33m 29s","accounts_in_message":[],"_revision_number":2},{"id":"3ae05283e776cd2f180f74b125f9b52493833b57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 07:21:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"25297dfc2faa92c1dd2c85b805731e1ac9f8ea8e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 07:22:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bd6deea3f88410b9b49e2c2ae7a0d58 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cf4963c5918572fe2d415d6268f5c746546cfada","revisions":{"c3b92b030aa3c0eda2549947df9b752c7393849e":{"kind":"REWORK","_number":1,"created":"2020-03-10 19:35:13.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/66/712166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/66/712166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/712166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/712166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/712166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/66/712166/1"}}},"commit":{"parents":[{"commit":"f4bc383de1ab81a642496d7bb9373fd45f2c26c3","subject":"Fix [Unity] verification and convert mgmt ipv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4bc383de1ab81a642496d7bb9373fd45f2c26c3"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-31 15:13:24.000000000","tz":60},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-03-10 19:34:57.000000000","tz":-240},"subject":"share_networks: enable project_only API only","message":"share_networks: enable project_only API only\n\nAt the moment, the share_network database API which the web\nAPI layer interacts with directly does not have any checking\nfor project_id which means that a user has the ability to run\noperations against any other share_network if they have the ID.\n\nThis patch implements the usage of project_only in the database\nquery which ensures that administrators still have the behaviour\nof getting any share network they want, but users can only pull\nup those which are part of their context/authenticated project.\n\nThis patch also adjusts a few other tests due to the fact that\nthe existing tests would run a lot of inserts with a different\nproject_id than the context, which is not allowed in this new\nAPI behaviour.  Therefore, the instances that involved projects\ndifferent than the context were converted to elevated ones.\n\nThere was also an instance where they were being created with a\nproject_id that did not match the fake context, therefore the\ncontext was adjusted accordingly as well.\n\nCloses-Bug: #1861485\nChange-Id: Id67a939a475c4ac06d546b7e095bd10f1a6d2619\n(cherry picked from commit 947315f0903c823b0fdd9d99c60078814587272c)\n(cherry picked from commit 496e6e1d2a074ab85f434fe2a88a6c0159696419)\n(cherry picked from commit 039ab2b020e4ca13fa723b6cb931f921944894ee)\n(cherry picked from commit 496bb1473ea1ad8143bfb65d1727166de543affc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c3b92b030aa3c0eda2549947df9b752c7393849e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c3b92b030aa3c0eda2549947df9b752c7393849e"}]},"branch":"refs/heads/stable/queens"},"cf4963c5918572fe2d415d6268f5c746546cfada":{"kind":"REWORK","_number":2,"created":"2020-03-11 01:34:01.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/66/712166/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/66/712166/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/712166/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/712166/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/712166/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/66/712166/2"}}},"commit":{"parents":[{"commit":"f4bc383de1ab81a642496d7bb9373fd45f2c26c3","subject":"Fix [Unity] verification and convert mgmt ipv6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4bc383de1ab81a642496d7bb9373fd45f2c26c3"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-01-31 15:13:24.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-11 01:33:42.000000000","tz":-420},"subject":"share_networks: enable project_only API only","message":"share_networks: enable project_only API only\n\nAt the moment, the share_network database API which the web\nAPI layer interacts with directly does not have any checking\nfor project_id which means that a user has the ability to run\noperations against any other share_network if they have the ID.\n\nThis patch implements the usage of project_only in the database\nquery which ensures that administrators still have the behaviour\nof getting any share network they want, but users can only pull\nup those which are part of their context/authenticated project.\n\nThis patch also adjusts a few other tests due to the fact that\nthe existing tests would run a lot of inserts with a different\nproject_id than the context, which is not allowed in this new\nAPI behaviour.  Therefore, the instances that involved projects\ndifferent than the context were converted to elevated ones.\n\nThere was also an instance where they were being created with a\nproject_id that did not match the fake context, therefore the\ncontext was adjusted accordingly as well.\n\nCloses-Bug: #1861485\nChange-Id: Id67a939a475c4ac06d546b7e095bd10f1a6d2619\n(cherry picked from commit 947315f0903c823b0fdd9d99c60078814587272c)\n(cherry picked from commit 496e6e1d2a074ab85f434fe2a88a6c0159696419)\n(cherry picked from commit 17b29e2b50d41db13f29eae86437ef91f6432c8d)\n(cherry picked from commit d16b5f0ea41419bdf4eb4e66ce4551571ee6acec)\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cf4963c5918572fe2d415d6268f5c746546cfada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cf4963c5918572fe2d415d6268f5c746546cfada"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
