)]}'
{"id":"openstack%2Fmanila~713073","triplet_id":"openstack%2Fmanila~master~I57c2e15fc22eb2bf69ebee25a92ba10746505e73","project":"openstack/manila","branch":"master","topic":"bug/1706137","hashtags":[],"change_id":"I57c2e15fc22eb2bf69ebee25a92ba10746505e73","subject":"[NetApp] cDOT to set valid QoS during migration","status":"MERGED","created":"2020-03-13 22:14:24.000000000","updated":"2020-03-30 18:50:53.000000000","submitted":"2020-03-30 18:38:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713073-1585593506145-e8e41dea","meta_rev_id":"24b4298cb9c60f54a1d519ffcad73e238483f2db","_number":713073,"virtual_id_number":713073,"owner":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":29831,"name":"Naresh Kumar Gunjalli","email":"NareshKumar.Gunjalli@netapp.com","username":"nareshkumarg"},{"value":0,"date":"2020-03-18 10:36:00.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-03-18 16:44:57.000000000","_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-30 18:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-14 01:38:22.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-14 01:44:48.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-03-13 22:15:03.000000000","_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"date":"2020-03-14 00:06:46.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-14 03:14:16.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-14 01:02:43.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"recommended":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2020-03-20 11:44:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2020-03-18 17:06:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":29831,"name":"Naresh Kumar Gunjalli","email":"NareshKumar.Gunjalli@netapp.com","username":"nareshkumarg"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-24 15:14:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-03-13 22:14:24.000000000","_account_id":29831,"name":"Naresh Kumar Gunjalli","email":"NareshKumar.Gunjalli@netapp.com","username":"nareshkumarg"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},{"value":1,"date":"2020-03-30 17:28:06.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":29831,"name":"Naresh Kumar Gunjalli","email":"NareshKumar.Gunjalli@netapp.com","username":"nareshkumarg"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 22:14:24.000000000","updated_by":{"_account_id":29831,"name":"Naresh Kumar Gunjalli","email":"NareshKumar.Gunjalli@netapp.com","username":"nareshkumarg"},"reviewer":{"_account_id":29831,"name":"Naresh Kumar Gunjalli","email":"NareshKumar.Gunjalli@netapp.com","username":"nareshkumarg"},"state":"REVIEWER"},{"updated":"2020-03-13 22:15:03.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2020-03-14 00:06:46.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-14 01:02:43.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-03-14 01:38:22.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-14 01:44:48.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-14 03:14:16.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-18 10:36:00.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-03-18 17:06:38.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-03-20 11:44:30.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-03-24 15:14:34.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-03-30 17:28:06.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-03-30 18:38:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37e152f0895fb20eb1900aaa2e1ddad25bce36bc","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-03-13 22:14:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e89d10e635b344f54a9e52435d6f7d3416b0403c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-13 22:15:03.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/997/713073/1/check/infortrend-manila-ci/14b75b1fdd753c6d90c301cd4576706024a9356f : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d0afad07b2861a43f1bd9eb01c0266bacb4f1849","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-13 23:03:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/73/713073/1/check/DellEMC_Unity_Manila_dhss_true/00a8782/DellEMC_Unity_Manila_dhss_true/219 : FAILURE in 48m 47s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/73/713073/1/check/DellEMC_Unity_Manila_dhss_false/89a3410/DellEMC_Unity_Manila_dhss_false/234 : SUCCESS in 48m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"5957175ac70234e72d5a95222c8d7095ec8f875b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-13 23:40:48.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/73/713073/1/check/DellEMC_PowerMAX_Manila/0ca6e5b/DellEMC_PowerMAX_Manila/238 : FAILURE in 1h 26m 11s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"8df98e332d958dbcf918a5eb814cb2bc25bac4b4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-14 00:06:46.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-73-713073-1/ns5_manila : SUCCESS in 1h 51m 57s","accounts_in_message":[],"_revision_number":1},{"id":"fe01deeacf1f42dd92fad64912d97a5d465c44a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 00:11:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45fd9d7925974210a664bbff702768f7 : SUCCESS in 9m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b81dde91419d42c2ad56626d88840b45 : SUCCESS in 17m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b2e4b54c71445b9a07977ad6272e847 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d68b70792484bd292316465e8cbd730 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/261df290b61b42288517c3f42e5a59e1 : SUCCESS in 11m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1ce0f47b0e74074ac8a9428d8363c2b : SUCCESS in 9m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1c55daeb484474aaadd56ab914e0383 : SUCCESS in 9m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73a2480fa8f64b9ea837219736eae14d : SUCCESS in 9m 20s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9de8055a06de46da9e420ac85efc1341 : SUCCESS in 7m 20s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/d806d3947f01424a85efbb26cedc038a : SUCCESS in 1h 35m 04s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/bbc43c410a0c49aba51b61bab5f024c9 : SUCCESS in 49m 58s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/f74f29ca212d4ed0bd856eddafd1a4f0 : SUCCESS in 1h 03m 15s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/9bc577807db943f9a1b453cb280d96a7 : SUCCESS in 58m 22s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/32103075245347498b0f3f09a2d57a47 : SUCCESS in 1h 13m 37s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e04ff40b2a52462680f6ea97fa701f13 : FAILURE in 1h 51m 30s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/3de52d07008b4f4a96a404317bff852d : SUCCESS in 29m 36s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/546ff1e913384d998524ee422001890a : FAILURE in 1h 51m 24s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ede123e3f74b4f29b47259cdb50be0b7 : SUCCESS in 1h 09m 00s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/15e4e4dc99a443a2bf787ec4236699ff : SUCCESS in 1h 11m 29s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/99de16f133ba423c9d67f8cee594e6f8 : SUCCESS in 41m 11s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/5d58483c73404151bc24ecd0cbdf7a8f : SUCCESS in 59m 34s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/15bbbab3a4dc4988b87cf5bdacea0e90 : SUCCESS in 1h 07m 58s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/db3f1154b89b408ca2c2082a90d79cd8 : FAILURE in 21m 16s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3ae360c825004e4da657613da0109316 : SUCCESS in 36m 32s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7e0de00de4e24ba7bbb20213720ef97d : SUCCESS in 38m 33s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/afea96cef70247f188f079398bcfc258 : SUCCESS in 7m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96d62da39d801e85715ee94c940a005ca706b21c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-03-14 01:02:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/73/713073/1/check/DellEMC_Unity_Manila_dhss_true/32ffc25/DellEMC_Unity_Manila_dhss_true/220 : FAILURE in 51m 00s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/73/713073/1/check/DellEMC_Unity_Manila_dhss_false/43e4977/DellEMC_Unity_Manila_dhss_false/235 : SUCCESS in 49m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"fd97d336869369c3c75ef675f4597a9794a595ab","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-14 01:28:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nThanks for your contribution, LGTM - I\u0027m waiting on the NetApp CI.\n\nplease address the nit in the test file if you\u0027re going to make another patch.","accounts_in_message":[],"_revision_number":1},{"id":"6fe8f6946503acb38b2cfcb0e676f765f104f89c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-14 01:38:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/73/713073/1/check/DellEMC_PowerMAX_Manila/3c84a87/DellEMC_PowerMAX_Manila/239 : FAILURE in 1h 26m 38s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"3836eb05e6f9d905fdf96ad999e933f8cafe4c8e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-14 01:44:48.000000000","message":"Patch Set 1:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-73-713073-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"fbdf996038ae663dae1f5f6d4040d92eea53ae99","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-14 03:14:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-ss http://logs.openstack.netapp.com/logs/73/713073/1/upstream-check/manila-cDOT-ss/3aea7eb/ : SUCCESS in 2h 03m 40s\n- manila-cDOT-no-ss http://logs.openstack.netapp.com/logs/73/713073/1/upstream-check/manila-cDOT-no-ss/c939c57/ : SUCCESS in 1h 24m 39s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f4f48887fe807c33bbd40e2640d20f332474d711","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-18 10:36:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/73/713073/1/check/dsvm-tempest-infinibox-nfs/ee20255 : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b41710efbff6ade2d5ec2fc8942c356224857815","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-03-18 16:44:57.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n \u003e \n \u003e Thanks for your contribution, LGTM - I\u0027m waiting on the NetApp CI.\n \u003e \n \u003e please address the nit in the test file if you\u0027re going to make\n \u003e another patch.\n\nHi Goutham, thank you for your review :) I will consider this in the next patch.","accounts_in_message":[],"_revision_number":1},{"id":"dd19c2368f549371449f20469162ae5bc3ab69c2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-18 17:06:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"af649a60279b418e8e22ee922f170f93becd9f08","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-03-20 11:44:30.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM, Daniel! Thanks for working on this issue.","accounts_in_message":[],"_revision_number":1},{"id":"f2fa0849aa37cb5e7786e2ee22e0fb98392cfa95","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-03-24 15:14:34.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks Naresh for working on this fix.\nThanks Daniel for the uploading and tests.\nLGTM!","accounts_in_message":[],"_revision_number":1},{"id":"66a473108d250d8e6304ca36ec057d2664bf383e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-30 17:28:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"129d0d02f047e23631d3cd96e1a0e9f13660feb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 17:28:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5493ea87f79a2e87cf7e664851fef04725e42af7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 18:38:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f0655abad994c838160efbbffbf1843 : SUCCESS in 11m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f67473504174e35a45a3dd3766d94b1 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7790b4645474703a8afe6f3bc52b04b : SUCCESS in 8m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21d59c14c82b4e67b35ce5c85a5e6818 : SUCCESS in 11m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/146f80e59726470dbbef7497a09e945b : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/453217e3e49541199eb9fb7b5c26a44b : SUCCESS in 10m 21s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/69ae6674958d499895df0323368b6f17 : SUCCESS in 5m 28s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/8b53c9bcb90040eea62bfcec2e1cb8f8 : SUCCESS in 44m 36s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/9876203a72d44efa9ebc7e3aa9874de1 : SUCCESS in 54m 42s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/23556573008f49d7a337e5814dcab36c : SUCCESS in 1h 06m 17s","accounts_in_message":[],"_revision_number":1},{"id":"0f148bbe000b85c21346184f6eaa3cb1d824a125","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 18:38:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1d6ad7aac28f257c852e1b16f2662297ef45b08e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 18:41:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb059e7204b34dc19615806ba0ff30d4 : SUCCESS in 2m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/357da6cc56dd451d8bce7847876fb7b0 : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":1},{"id":"be06c3daff9f01f97ab7324cd281707910b1d150","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-03-30 18:50:53.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 1fddc0794f89dc9fb8381e622855d76839136c5d","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14b75b1fdd753c6d90c301cd4576706024a9356f","revisions":{"14b75b1fdd753c6d90c301cd4576706024a9356f":{"kind":"REWORK","_number":1,"created":"2020-03-13 22:14:24.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/73/713073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/73/713073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/73/713073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/73/713073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/73/713073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/73/713073/1"}}},"commit":{"parents":[{"commit":"bf15efe3184bbd4b7176db2447dbab157545c71e","subject":"Update devstack repository URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bf15efe3184bbd4b7176db2447dbab157545c71e"}]}],"author":{"name":"Naresh Kumar Gunjalli","email":"NareshKumar.Gunjalli@netapp.com","date":"2019-06-17 11:15:21.000000000","tz":-240},"committer":{"name":"Daniel Tapia","email":"danielarthurt@gmail.com","date":"2020-03-13 22:14:22.000000000","tz":-180},"subject":"[NetApp] cDOT to set valid QoS during migration","message":"[NetApp] cDOT to set valid QoS during migration\n\nNetApp cDOT driver is now fixed to remove the QoS Policy on the\nbackend volume when a share is migrated from an extra-spec which had\nQoS defined to another extra-spec which has no QoS defined in it.\n\nChange-Id: I57c2e15fc22eb2bf69ebee25a92ba10746505e73\nCloses-Bug: #1706137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/14b75b1fdd753c6d90c301cd4576706024a9356f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/14b75b1fdd753c6d90c301cd4576706024a9356f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
