)]}'
{"id":"openstack%2Fmanila~714949","triplet_id":"openstack%2Fmanila~stable%2Frocky~I82854d0f9075e6fc7d33c6459d71d7f0d502ff3c","project":"openstack/manila","branch":"stable/rocky","topic":"bug/1868905-stable/rocky","hashtags":[],"change_id":"I82854d0f9075e6fc7d33c6459d71d7f0d502ff3c","subject":"[stable-only] Pin neutron-tempest-plugin to 0.9.0","status":"MERGED","created":"2020-03-25 14:12:18.000000000","updated":"2020-03-25 21:12:19.000000000","submitted":"2020-03-25 21:12:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714949-1585170739790-a663f06a","meta_rev_id":"caaaf10235714b93893f56363291ec63b4200c87","_number":714949,"virtual_id_number":714949,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2020-03-25 14:39:04.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-03-25 17:38:51.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 21:12:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"recommended":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2020-03-25 16:50:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2020-03-25 16:16:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-25 19:13:03.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-25 19:13:03.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 14:39:04.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-03-25 16:16:25.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-03-25 16:50:58.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-03-25 17:38:51.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-25 19:13:03.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-03-25 21:12:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4f9f5ff3169d25f2e179b6e4330e25bdfcbc3ba","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-03-25 14:12:18.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"7a9309dd6fcf7bb936b1f5c93fd97c5e055d1c19","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-03-25 14:39:04.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/49/714949/1/check/dsvm-tempest-infinibox-nfs/b9112a8 : FAILURE in 26m 20s","accounts_in_message":[],"_revision_number":1},{"id":"efa737d4251d3da8269880e8cc2e98c657931b4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 16:11:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3f71f60f8e78452fa7fc714d315dc221 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e07c876e8ec041b1a85e44c3faefcce7 : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f0b2e20ae7440589664a11caa9cf9ae : SUCCESS in 12m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8fdc1ff8b1224b3e85cf9d91319a11c2 : SUCCESS in 7m 10s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8cc0f4c801624a80ad87d2c0b61a7766 : SUCCESS in 5m 28s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/aa7e2ab520d742dbad92aa45f0f07991 : FAILURE in 1h 39m 22s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/3df4e002a131417a85604ac47323a151 : SUCCESS in 48m 41s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/e68d9fce889b4eb6912eb86f99781d60 : SUCCESS in 1h 13m 04s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/cf2408dcc42b4988bf61c4bc7bd04ae5 : SUCCESS in 1h 04m 38s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/293c73ef4f774953961227870dc796eb : FAILURE in 57m 11s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edc04ef295ce4c60b9bbc61009213df0 : SUCCESS in 1h 37m 47s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/fe414ecda048460893090604f9402a83 : SUCCESS in 29m 20s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d21521e4788d48a1b2e0853c0e0f42e5 : FAILURE in 21m 22s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3c4412c79484453bb232f0692e3ff8e5 : SUCCESS in 36m 17s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/2e87dc966df442a3b46e4ce311aeadc1 : SUCCESS in 1h 52m 43s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4b6cdb0b14524b95a696d1f46a512c75 : SUCCESS in 24m 57s (non-voting)\n- manila-rally-multibackend https://zuul.opendev.org/t/openstack/build/3b5cfae7fe494724b968b430be271039 : FAILURE in 20m 18s (non-voting)\n- manila-rally-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/ae87bbfb880f4d308cc1b9f917630d54 : FAILURE in 16m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b5b1be174ce4ddb83933df8ef0847b1 : SUCCESS in 12m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6850f4a6441945e0a27ea3ed6424ba13 : FAILURE in 14m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/999f1a94691d4c37ae352ce8213f200e : SUCCESS in 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30726bf5ccce816f64cca080279e84c0832f2e3e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-03-25 16:16:25.000000000","message":"Patch Set 1: Code-Review+1\n\nClean cherry-pick, LGTM, tks.","accounts_in_message":[],"_revision_number":1},{"id":"971a1b85879364a3949a44c3deb7f86caf1a796f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-03-25 16:50:58.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM, Goutham!","accounts_in_message":[],"_revision_number":1},{"id":"808eb2933ce6aa656913117911a1d03885d462c1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-03-25 17:38:51.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/49/714949/1/upstream-check/manila-cDOT-ss/5b26b59/ : FAILURE in 1h 11m 43s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/49/714949/1/upstream-check/manila-cDOT-no-ss/413b783/ : FAILURE in 1h 26m 59s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"fd3fb21a3d7d67c1990c7cc4ea091aa877e683a0","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-03-25 19:13:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30b4f4e4817d8638ef8865a08d5ca99e228bd432","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 19:13:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10aab40bdc8be020796cde7086865c02b1b6b63d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:12:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"658cc2a43ae97a41d17c080fae2fbce9403fa06d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:12:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d38e3fbf507a41b383706f6e3c0fa340 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec79ee37895b47ce934b339b96e84c8f : SUCCESS in 7m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/37f49ac7215844949e45ef3c18ef82f6 : SUCCESS in 8m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5f079fbee76d4eaf857959222dd3b99b : SUCCESS in 13m 55s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c300b534bbe34eea8dbef89fabe8bd01 : SUCCESS in 6m 59s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c46ee732518b4922bd91f81cd5b831b8 : SUCCESS in 43m 41s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/3f69a0a5064f441788cf69a30522c2b9 : SUCCESS in 1h 53m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ead2e32827a47e6aedecfcde429dd3a : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c856f4ffa245c14c2233d95c94c10fb3577d519c","revisions":{"c856f4ffa245c14c2233d95c94c10fb3577d519c":{"kind":"REWORK","_number":1,"created":"2020-03-25 14:12:18.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/49/714949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/714949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/714949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/714949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/714949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/714949/1"}}},"commit":{"parents":[{"commit":"d16b5f0ea41419bdf4eb4e66ce4551571ee6acec","subject":"share_networks: enable project_only API only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d16b5f0ea41419bdf4eb4e66ce4551571ee6acec"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-25 01:58:45.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-03-25 14:12:18.000000000","tz":0},"subject":"[stable-only] Pin neutron-tempest-plugin to 0.9.0","message":"[stable-only] Pin neutron-tempest-plugin to 0.9.0\n\nmanila\u0027s IPv6 jobs setup bgp with the help of\nneutron-dynamic-routing. The jobs enable\nneutron-dynamic-routing using its devstack\nplugin to install it alongside openstack/neutron.\n\nneutron-tempest-plugin is a \"branchless\" tempest\nplugin to the neutron project. We don\u0027t need to\ninstall or run this tempest plugin in manila test\njobs. However, it is imported by test code within\nthe neutron-dynamic-routing project, forcing us to\ninclude it in the list of requirements for the IPv6\njobs.\n\nneutron-tempest-plugin dropped support for python3\nin ussuri. python3 only code was added to\nneutron-tempest-plugin causing failures in the\nvoting ipv6 lvm test job\n(manila-tempest-minimal-dsvm-lvm).\n\nSo, pin the version of neutron-tempest-plugin we\u0027re\npulling in to a tagged git ref when the plugin last\nsupported python2.\n\nChange-Id: I82854d0f9075e6fc7d33c6459d71d7f0d502ff3c\nCloses-Bug: #1868905\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 9b2753084559bf8fe415770e71e18a1477760ddb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c856f4ffa245c14c2233d95c94c10fb3577d519c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c856f4ffa245c14c2233d95c94c10fb3577d519c"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
