)]}'
{"id":"openstack%2Fmanila~715939","triplet_id":"openstack%2Fmanila~master~I7aefa9a3a3aa86fdbfc990ca46c79655265b1cde","project":"openstack/manila","branch":"master","topic":"bug/netapp_migration_complete","hashtags":[],"change_id":"I7aefa9a3a3aa86fdbfc990ca46c79655265b1cde","subject":"netapp: fix migration provisioning options","status":"ABANDONED","created":"2020-03-30 13:08:04.000000000","updated":"2020-03-30 15:17:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e3350ac973fce20855b777ded57e62a0e9d95181","_number":715939,"virtual_id_number":715939,"owner":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"actions":{},"labels":{"Verified":{"all":[{"date":"2020-03-30 13:08:16.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"tag":"autogenerated:jenkins-gerrit-trigger","date":"2020-03-30 13:08:49.000000000","_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 13:08:16.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-30 13:08:49.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"}],"messages":[{"id":"7d5fc6e70b2670698714965a704e915550de82ad","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2020-03-30 13:08:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"433ea6ebedadb5b4bdaf85816577d248755a2da7","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-03-30 13:08:16.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/39/715939/1/check/DellEMC_PowerMAX_Manila/36448a0/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"ebdc8e0810e1c251ad91483dc58325f8b82868bd","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-03-30 13:08:49.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/1044/715939/1/check/infortrend-manila-ci/6aa6149d9607f9c8e7cfba07a54d1dbd44ab918d : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"be2de5c31782ba6f2237e575aaa2f9d8510b95bf","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2020-03-30 13:21:23.000000000","message":"Abandoned\n\nin favor of https://review.opendev.org/#/c/712512","accounts_in_message":[],"_revision_number":1},{"id":"463feb2353d76b66b92b5185edeb6907a4a12d55","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-03-30 15:17:37.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-39-715939-1/ns5_manila : SUCCESS in 2h 9m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6aa6149d9607f9c8e7cfba07a54d1dbd44ab918d","revisions":{"6aa6149d9607f9c8e7cfba07a54d1dbd44ab918d":{"kind":"REWORK","_number":1,"created":"2020-03-30 13:08:04.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/39/715939/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/715939/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/715939/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/715939/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/715939/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/715939/1"}}},"commit":{"parents":[{"commit":"a4d4159820d07d5fcfde7e6254a73668bfe743cb","subject":"Cleanup Python 2.7 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a4d4159820d07d5fcfde7e6254a73668bfe743cb"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2020-03-30 13:00:10.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2020-03-30 13:08:02.000000000","tz":120},"subject":"netapp: fix migration provisioning options","message":"netapp: fix migration provisioning options\n\nremapping of NetApp-specific specs and validation was missing\n\nChange-Id: I7aefa9a3a3aa86fdbfc990ca46c79655265b1cde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6aa6149d9607f9c8e7cfba07a54d1dbd44ab918d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6aa6149d9607f9c8e7cfba07a54d1dbd44ab918d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
