)]}'
{"id":"openstack%2Fmanila~716552","triplet_id":"openstack%2Fmanila~master~I3cb344680ec39c3b73817e9a9b51e2bf593bea36","project":"openstack/manila","branch":"master","topic":"bug1870105","hashtags":[],"change_id":"I3cb344680ec39c3b73817e9a9b51e2bf593bea36","subject":"Fixes pool capabilities reported on CephFS driver","status":"ABANDONED","created":"2020-04-01 11:18:33.000000000","updated":"2020-04-21 23:48:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8248fa65786727f756387ff708ae81f0c8b1d335","_number":716552,"virtual_id_number":716552,"owner":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-04-02 06:28:44.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"date":"2020-04-01 22:05:20.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"date":"2020-04-01 13:31:40.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"tag":"autogenerated:jenkins-gerrit-trigger","date":"2020-04-01 11:19:17.000000000","_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"date":"2020-04-01 21:01:13.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"date":"2020-04-01 16:28:22.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"date":"2020-04-01 19:30:35.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-01 13:04:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-04-01 15:42:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":-1,"date":"2020-04-01 17:21:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"date":"2020-04-01 11:23:39.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-01 11:19:17.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2020-04-01 11:23:39.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-04-01 13:04:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-01 13:31:40.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-01 15:42:02.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-01 16:28:22.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-01 19:30:35.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-04-01 21:01:13.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-01 22:05:20.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-02 06:28:44.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8f12bde0047b24f70d1a2e2afc726220202e53c","author":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"date":"2020-04-01 11:18:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afa4f12dc768aa9d8f29cd2e9d957291be4f3529","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-04-01 11:19:17.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/1067/716552/1/check/infortrend-manila-ci/ab58145d105f4f2b862f595e874fd4e2cb8e9884 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"96ed57f3377d020d9e40b846544244262681559b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 13:04:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2376dc1c614045bb8b5c9f4e274804fa : SUCCESS in 10m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d44a3c94e5524938b956efc7dd5e34b3 : SUCCESS in 17m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/018ea9cb163e497c851408b8f1afb65c : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2369e6de27f04c338e6f7f108199ba0f : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c777f224a161482ca4f189260319c79f : SUCCESS in 9m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/747b847968fc45f4841ba493a52c6256 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b31fb82051364e3a80c6384201cfd081 : SUCCESS in 9m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e451c7dcc4954de59116db01739e4d30 : SUCCESS in 9m 11s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/7a921ebf3c2340faa8c58d5626c3e9e0 : SUCCESS in 5m 46s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/c3c3543c3293453f95e5e87342ba728d : SUCCESS in 1h 30m 51s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/5ad750957d334195b7e002376b9963cf : SUCCESS in 52m 39s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/2c16aaec33ab4f209207d73412a313fa : SUCCESS in 1h 19m 15s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/d072930af5d940d4ad43b03d12f5f7d3 : SUCCESS in 1h 01m 53s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/4ed67e7b0f7c4390845ec889197588ba : SUCCESS in 1h 01m 02s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb00bf50eb674efb9db718be88939f64 : FAILURE in 1h 39m 52s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/baae0e6cba4749708db89b82627a5cc2 : SUCCESS in 24m 02s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4887ed9007ca4061a25c332c399669ec : SUCCESS in 1h 09m 24s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/92f13a436ede4924a62c1ca1ead89a0b : SUCCESS in 1h 00m 48s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/40344f6ae0574942ae9991f5648cd2b3 : SUCCESS in 1h 01m 54s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/526bcb7e72e14c4cbef72989d04be2be : SUCCESS in 43m 30s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/4daad6e289de4785966b2ba616a54d51 : SUCCESS in 54m 56s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/911ed6236256477d8b46d87d13fbd241 : SUCCESS in 1h 11m 08s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e270e96384cd45e9b95d2d86895d3ccf : FAILURE in 27m 33s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6440eab035ff4ac997ac75be13e30675 : SUCCESS in 34m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fe054bff77bb43ce9e60aa974ced6ffd : SUCCESS in 42m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81de895d4fb14b7ead39c521ebffb87c : SUCCESS in 4m 54s (non-voting)\n\nWarning:\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Job openstack-tox-cover: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path stdin in the commit\u0027\n  Comments left for invalid file stdin","accounts_in_message":[],"_revision_number":1},{"id":"80a14a723c960a5bcb90c8babae7711614367fd3","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-04-01 13:31:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/52/716552/1/check/DellEMC_VNX_Manila/3c8064a/DellEMC_VNX_Manila/137 : FAILURE in 26m 43s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"cb52831464fb439a2d1f5c585cf45aa689dc515f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-01 15:42:02.000000000","message":"Patch Set 1: Code-Review-1\n\nHi Jose, thank you for noticing this!\n\ndrivers sending capabilities as a list is allowed to be done - we need to account for that in the scheduler logic.. This stems from the fact that the backend pool can allow both thin and thickly provisioned shares, and the only way to indicate that seems to be by sending a list of capabilities:\n\n https://opendev.org/openstack/manila/blame/commit/a149c283d7701760e5b8040f6daea3965ae11d3a/doc/source/admin/capabilities_and_extra_specs.rst#L121-L136","accounts_in_message":[],"_revision_number":1},{"id":"39773e4a877a8cad68c6f2660035c20f22e9b186","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-01 16:11:27.000000000","message":"Patch Set 1:\n\nJose: We\u0027ve a helper method to determine if a backend pool supports thin provisioning at all, given that they can take thin/vs/thick shares sometimes:\n\nhttps://opendev.org/openstack/manila/src/commit/4249e94c6bb7f5efb1a2095c925398277046936d/manila/scheduler/utils.py#L121-L127\n\nSo, we have an option to revert https://review.opendev.org/#/c/716265/ and redo the fix - that would be my personal preference because I think this is worth back porting, and it\u0027d be easier to back port as a single change. What do you think?","accounts_in_message":[],"_revision_number":1},{"id":"f7713a947074e24f3c9749041d3ac0037c55d88f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-01 16:28:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/52/716552/1/upstream-check/manila-cDOT-ss/c3e0918/ : SUCCESS in 2h 02m 00s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/52/716552/1/upstream-check/manila-cDOT-no-ss/fcce0d7/ : FAILURE in 2m 51s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"aed00dbd1c383b45207cd5723a4f7cb2d116c850","author":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"date":"2020-04-01 17:21:06.000000000","message":"Patch Set 1: Code-Review-1\n\nIt\u0027s fine to do a revert, I can use that helper function for thin_proviosiong","accounts_in_message":[],"_revision_number":1},{"id":"6265e0f093cc2fbb0fcfe8d6dbb7feb78c9e5d18","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-04-01 19:30:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/52/716552/1/check/DellEMC_Unity_Manila_dhss_true/bbb04d1/DellEMC_Unity_Manila_dhss_true/295 : FAILURE in 59m 58s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/52/716552/1/check/DellEMC_Unity_Manila_dhss_false/77c6b81/DellEMC_Unity_Manila_dhss_false/310 : FAILURE in 59m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6bc892ad5bff1fcda19266ace921b7e793571d22","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-01 21:01:13.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-52-716552-1/ns5_manila : SUCCESS in 7h 47m 7s","accounts_in_message":[],"_revision_number":1},{"id":"c8c46f37e0bb207baaa1f78ad5ffa7ccfff7dd8c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-04-01 22:05:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/52/716552/1/check/DellEMC_PowerMAX_Manila/43f6c90/DellEMC_PowerMAX_Manila/270 : FAILURE in 1h 39m 32s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"d50ce3c2c6073dfcf2102c1d387ec27d96ffc934","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-02 06:28:44.000000000","message":"Patch Set 1:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-52-716552-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"12eb06c30ef5df920960fcf237e03a1bc56eda6b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-21 23:48:02.000000000","message":"Abandoned\n\nNo longer required, https://bugs.launchpad.net/manila/+bug/1870105 has been marked Invalid.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab58145d105f4f2b862f595e874fd4e2cb8e9884","revisions":{"ab58145d105f4f2b862f595e874fd4e2cb8e9884":{"kind":"REWORK","_number":1,"created":"2020-04-01 11:18:33.000000000","uploader":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"ref":"refs/changes/52/716552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/716552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/716552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/716552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/716552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/716552/1"}}},"commit":{"parents":[{"commit":"cd9292b931a0f74079512cd43652129a9c595ee8","subject":"Remove provisioned calculation on non thin provision backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cd9292b931a0f74079512cd43652129a9c595ee8"}]}],"author":{"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","date":"2020-04-01 11:11:28.000000000","tz":120},"committer":{"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","date":"2020-04-01 11:17:10.000000000","tz":120},"subject":"Fixes pool capabilities reported on CephFS driver","message":"Fixes pool capabilities reported on CephFS driver\n\nDedupe, compression and thin_provisioning capabilities reported by\nthe driver are wrongly interpreted as they are submitted as lists.\nThis commit changes its report and send the information as booleans\n\nCloses-Bug: #1870105\n\nChange-Id: I3cb344680ec39c3b73817e9a9b51e2bf593bea36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ab58145d105f4f2b862f595e874fd4e2cb8e9884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ab58145d105f4f2b862f595e874fd4e2cb8e9884"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
