)]}'
{"id":"openstack%2Fmanila~718016","triplet_id":"openstack%2Fmanila~master~Icb8947991723a2a1cc334a72276a35cf50fc1b7d","project":"openstack/manila","branch":"master","topic":"bug/1869712","hashtags":[],"change_id":"Icb8947991723a2a1cc334a72276a35cf50fc1b7d","subject":"Remove provisioned calculation on non thin provision backends","status":"MERGED","created":"2020-04-07 10:07:10.000000000","updated":"2020-04-08 11:39:11.000000000","submitted":"2020-04-08 08:12:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718016-1586333542477-75094fdf","meta_rev_id":"4333874064197e2095876d9e2f7594a2c1aa33bf","_number":718016,"virtual_id_number":718016,"owner":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-08 01:26:28.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-08 01:42:15.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-07 23:48:13.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-04-07 19:11:46.000000000","_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"tag":"autogenerated:jenkins-gerrit-trigger","value":0,"date":"2020-04-07 21:17:13.000000000","_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"date":"2020-04-07 21:44:21.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-04-08 01:30:49.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 08:12:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-07 23:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-07 19:59:32.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-07 23:00:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 19:59:32.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-04-07 21:17:13.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2020-04-07 21:44:21.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-07 23:00:12.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-07 23:48:13.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-08 01:26:28.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-08 01:30:49.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-08 01:42:15.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-08 08:12:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6bc2fa947ec3289038e35f77eea4e18a014cc05","author":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"date":"2020-04-07 10:07:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27f4d64a9d3e9da67502dd6e6e87f40a8d12e000","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-04-07 10:08:01.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/1122/718016/1/check/infortrend-manila-ci/2de4f7ebaf2862053aeeaefeca72467fc9a55c59 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"54f6adca90866031573c643eef197f55bc3f0c24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 11:51:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae45c0cd387447e296e690bf26413d36 : SUCCESS in 10m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5398cd7194964da1a0cfc537b980d38d : SUCCESS in 16m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3803b65bd9e8466a8400e5e23787d875 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10ffa9687300420fba5251bdcafee90b : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d55ecc70b63c46cc882979c096cdacff : SUCCESS in 8m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/204bef726d4b4c87a7934c27a700a331 : SUCCESS in 11m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7753170d23214f42a63d1853bd33ff48 : SUCCESS in 9m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e1b51e86b6043b199e2a61e77b90593 : SUCCESS in 11m 15s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ebf979ec5d8a40638ee40008297d56cc : SUCCESS in 7m 29s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/e0e04cb226ad40dba228c674badcc6cc : FAILURE in 1h 34m 47s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/c07b362200f14b5a8c0603a0e852d296 : SUCCESS in 54m 56s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/164bc4a78d574c8b99624a3a54c35fdd : SUCCESS in 1h 22m 44s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/2311e990429f4436ac499152ac3e910c : FAILURE in 1h 02m 51s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/2ce8f2ee0d6946eb9ffff079ba5e717f : FAILURE in 47m 25s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd150989dcda4ee5bb1e719514814fef : FAILURE in 1h 35m 53s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/dc9536f0cfc741bab66aafd17878ccba : SUCCESS in 27m 17s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5c22e76efdc34fbe98aa8b5a5c26365a : SUCCESS in 1h 20m 44s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a67ca598ec3c4f03977838a43be671ff : SUCCESS in 1h 04m 50s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/8b518f5dcffb47dca437b34dabf76118 : FAILURE in 1h 05m 43s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/72337bf842424975a0e82f93a114d1c8 : SUCCESS in 42m 32s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/5e163ed2128a47b194fbb92165b83dbd : SUCCESS in 1h 06m 17s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/75551035bccd4266b4acfef37d1efe71 : SUCCESS in 1h 01m 48s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d4948073edda44629d1f030c77135d85 : FAILURE in 22m 33s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/312a36db8cd5486d9c987650058ff979 : SUCCESS in 35m 55s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f2e688e7f8cf4a1b9edc929eafe8af64 : SUCCESS in 41m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cba9e8b2c21a40c1b5aaae397b4e3b5b : SUCCESS in 7m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19953e6856f470e386520f76c058f1fbdda72236","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-07 12:01:40.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks right.","accounts_in_message":[],"_revision_number":1},{"id":"a3d1dde821d390a848221d310ab2c8f32200b63b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-07 12:03:03.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-16-718016-1/ns5_manila : SUCCESS in 1h 55m 32s","accounts_in_message":[],"_revision_number":1},{"id":"186719dfecd50932af7500fd2f9e0d846138a4cc","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-04-07 12:03:42.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/16/718016/1/check/DellEMC_PowerMAX_Manila/4b1c200/DellEMC_PowerMAX_Manila/303 : FAILURE in 1h 56m 20s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"f4c8e1d70acc4382be43521cde5e094ce184b31a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-04-07 12:14:08.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/16/718016/1/check/DellEMC_VNX_Manila/87c22f4/DellEMC_VNX_Manila/164 : FAILURE in 22m 37s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"bbf820f49cfa74a46c264f3f85647bd1e0f92e39","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-07 13:43:54.000000000","message":"Patch Set 1:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-16-718016-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"46033626774b2c513a1de7d9750b0646c3af9400","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-07 17:39:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nHi Jose, \n\nThanks for re-submitting this fix, please see my comment inline..","accounts_in_message":[],"_revision_number":1},{"id":"37aecdf24bef89f480b4a3f57d0359da41815676","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-07 17:41:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"428d623c3fd74d340df67e8dd5a015f42e201bac","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-07 19:07:21.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f2bb86849df48a0c2bb85dec7b1449548110016","author":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"date":"2020-04-07 19:10:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0331b4140929f7249b730a2d1ad5f918ab32e8f6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-04-07 19:11:36.000000000","message":"Patch Set 2:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/1124/718016/2/check/infortrend-manila-ci/2c30fed486320ec5ba88f21ae3df14b14da6eb35 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"daa2782ae8942cc1627c6ffbd33dd574ec3b1393","author":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"date":"2020-04-07 19:11:46.000000000","message":"Patch Set 2:\n\nJust uploaded version adapted with your comments","accounts_in_message":[],"_revision_number":2},{"id":"d5bc6c7a918cd033be5ddfa9ca91cb1ac9ff21cc","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-07 19:59:32.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks.","accounts_in_message":[],"_revision_number":2},{"id":"a4379248b744284feb8b1b31cc15faf42016b495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 21:02:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc6c4d8704945219d2642bf216c40b6 : SUCCESS in 12m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb7f4803b6bf441ab5a4789e9e51cf55 : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49351458a57c46c28114440787cbecc5 : SUCCESS in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/470f469edb7942c9aa30333372e0dc51 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d44ddfcaa7a040c8bd554da2c351789f : SUCCESS in 10m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/495e2477e20740b69a2ecf0c18238d02 : SUCCESS in 14m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fb3360b87c0473f8012b64fcee0ecc2 : SUCCESS in 8m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd4b86262c864de49f0554b4c875ffb3 : SUCCESS in 10m 21s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0f8320cf22e74c3e9a0d75d798b9f578 : SUCCESS in 5m 24s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/3dd35884bb1344928413005fdfb43a48 : FAILURE in 1h 21m 49s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/f7e44fa6c5e64996b3323b9a3d5eb01f : SUCCESS in 49m 42s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/02ebe966d5ee413c87e18b1ef91f1cbb : SUCCESS in 1h 10m 11s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/0282523256234e22a880f92bab5b6368 : SUCCESS in 51m 30s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/714ff0855b8d471b9e03629e3fa7b698 : FAILURE in 46m 08s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5e0ae7dc6ea4c9ba62935db8d9c454d : FAILURE in 1h 47m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/c37819f786a24abbaed99cae215ff2c4 : SUCCESS in 33m 15s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b18f79916b0746ce8c5f99682651dd72 : SUCCESS in 1h 13m 39s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9cbc396b42d744ada4c0fbe3b17c4d3d : SUCCESS in 1h 04m 17s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/b076617f5ebd42cdb77c4b01d9fd1df8 : SUCCESS in 1h 04m 41s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/bd5c26d9907f4766bd331910b28686c3 : FAILURE in 50m 18s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/15cd7940d24e418cb5f079e13fec9118 : SUCCESS in 1h 08m 01s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/5ea47b9cf4a44acbac694b2b0264149d : SUCCESS in 57m 39s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/60b4d2c56acd4a4d897b1e097f86f7f9 : FAILURE in 20m 12s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/95c2406939dc489eb1410440945b5ff9 : SUCCESS in 45m 48s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ab69ed8a778a401eb1e067a665d110c7 : SUCCESS in 49m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9b741ceaa55431f90cb29c29b7b5afb : SUCCESS in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e34bfac9107a86f09a5c213dd26af118c85c3388","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-04-07 21:16:22.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/16/718016/2/check/DellEMC_PowerMAX_Manila/8ba7796/DellEMC_PowerMAX_Manila/305 : FAILURE in 2h 05m 19s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"0a777bc7ffcc852912170ed69fadb025c7a90622","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-07 21:16:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f6205e502e4ee99ed6f14c75a340ef6d7795efa4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-04-07 21:17:13.000000000","message":"Patch Set 2:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/1126/718016/2/check/infortrend-manila-ci/2c30fed486320ec5ba88f21ae3df14b14da6eb35 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"820e588b3c84fc45f8daa811389ff182532aefe5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-07 21:44:21.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-16-718016-2/ns5_manila : SUCCESS in 2h 33m 10s","accounts_in_message":[],"_revision_number":2},{"id":"8e7012fc7116d6fa6ac7e7913ec61e7df2b329d5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-07 23:00:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks, Jose!","accounts_in_message":[],"_revision_number":2},{"id":"0979fb042d41018ba304604abd49ae81e92114d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 23:25:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f788d2c51fa4eb29d7f35715744de44 : SUCCESS in 10m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24ad57e10ebc4790a71671b0fdbd2f8b : SUCCESS in 15m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac571a8f7ba242a5b7e1c94dc91f0d7f : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dcee592f3c644ab84817b1f471ce962 : SUCCESS in 8m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c670632728954374ae0de9c18a4f3293 : SUCCESS in 7m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b81ae059509a414fa202f076dfea8c54 : SUCCESS in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ecef762e3a749a892d0fa2f0c17e2ff : SUCCESS in 10m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e6424bdce964baca33bd7946cfe5253 : SUCCESS in 11m 56s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/722eac3096aa4fe4b781020a3bab33ca : SUCCESS in 8m 27s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/6f2cdf694f264b6c88d63c573052e89a : FAILURE in 1h 15m 26s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/939ac09780d34410aae98c9a936b8e5c : SUCCESS in 59m 59s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/a5c8f74e01a947ceb16c5ff81b2ea5d3 : FAILURE in 1h 11m 24s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/c6c47bd6061b4ca28e8667a3c7723182 : SUCCESS in 52m 11s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/51585888c2554636a7d9c2f6319bac4d : FAILURE in 49m 44s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/473c2faff85247039cf6abe361122bfd : FAILURE in 1h 54m 32s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/a5984308141a4282ab5d2de2f08b6d08 : FAILURE in 33m 00s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/12829421bee04062adf2d85659397f8f : SUCCESS in 1h 24m 20s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f1439ef9e6724fabac22f7e0e86fa9c1 : SUCCESS in 1h 10m 35s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/986bc39e28c244309ceeff8688e182d7 : FAILURE in 1h 13m 44s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/21b05f5fd1c94e4c9b14dce5e7fa3961 : SUCCESS in 49m 12s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/d67880cde5804a30a5feb4560fe366bc : SUCCESS in 1h 30m 00s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/d2b13dd2d71f46cf9e64dbed34d02442 : SUCCESS in 1h 27m 02s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2a443a2c8a2f4b809b00d8d270ad4ed2 : FAILURE in 24m 23s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0a718c4d1a694e43a186699adff160bb : SUCCESS in 37m 45s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7a215786959243bc8a6033c4c533d504 : SUCCESS in 39m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b0c737505ef46f08ef37c7cc453644b : SUCCESS in 7m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be3fe5d379d1bff7edeb45e958629837ca7b16bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 23:25:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2b3c13bffbbb52eb64a449be9e25e8ec6f686a38","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-07 23:42:06.000000000","message":"Topic changed from bug1869712 to bug/1869712","accounts_in_message":[],"_revision_number":2},{"id":"db58562873265cca6b76fa266d562f26e77c729e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-04-07 23:48:13.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/16/718016/2/check/DellEMC_VNX_Manila/46d5153/DellEMC_VNX_Manila/166 : FAILURE in 22m 12s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"31dcad529170ec6ef9406a2664f84b18d3f550c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 00:51:17.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d87ae73f8ae24b449c1b26d9d6fec709 : SUCCESS in 12m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84ec3be81c8b4c9790de5bddbe3985da : SUCCESS in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6a5443ae5a944e1b8612adf30f8c18b : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c8ec8bc1fc64fd5b530f94464092501 : SUCCESS in 13m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1213d06ac1c74d01a78a08c32518d063 : SUCCESS in 14m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3b594efb67747358d0a26b4529793bb : SUCCESS in 10m 23s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d3896ae9fd44451584b7ee18bc0b0e92 : SUCCESS in 7m 18s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/745d4ba2c5054bfa891527d104fbf822 : SUCCESS in 41m 37s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/ae4ad868b4664226b8daef14f5570572 : FAILURE in 1h 21m 17s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/0e088040aa604a0387c2fed1cee6e1f4 : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":2},{"id":"b6ee1b23035b7046748db9a8b2cbdba5c1204df7","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-04-08 01:26:28.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/16/718016/2/check/DellEMC_PowerMAX_Manila/6cf5663/DellEMC_PowerMAX_Manila/307 : FAILURE in 1h 53m 12s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"257b2e7e712d4d00da7d03485bf1aca537e5c876","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-08 01:30:49.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/16/718016/2/upstream-check/manila-cDOT-ss/8298083/ : SUCCESS in 2h 04m 42s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/16/718016/2/upstream-check/manila-cDOT-no-ss/b610eee/ : FAILURE in 3m 00s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"5be5aeaaec00c3b95de7fcff57c10794fbc59c4e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-08 01:42:15.000000000","message":"Patch Set 2:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-16-718016-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"3ef1d717f9ea03be5cd72f82d14a3125d6bd396c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-08 04:37:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"39ef3f01a6f099d54b549f8ee97f628213361014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 06:43:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d122a6c31042ec8e2ebea2bc805744 : SUCCESS in 10m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e987670c3c24c4aa0b404737c70bba7 : SUCCESS in 13m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1620b83515c4e478e160a8a106ab433 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4321a32b5df44e719ec48544903efdb1 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e7805a1b388443ebe56388d43a51b83 : SUCCESS in 9m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/938fbd19bcd44030a77a9234c5b29d96 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bcb9783dda1448692470ddb5709da87 : SUCCESS in 10m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7668477df286400dacbd503638d54018 : SUCCESS in 9m 28s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/bfbaeb8b404d4a11a924f329eabe0814 : SUCCESS in 4m 55s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/d1a28aa82d4e4602bcfe07dcbac3b2cf : FAILURE in 1h 10m 13s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/2b7e78d193704eb89cdffda69dff333a : SUCCESS in 45m 41s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/63fa29f93fdd496c9ef522ed809e51f3 : FAILURE in 1h 20m 14s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/efb5bb88c4f0435fb4cdb20efa884fa3 : SUCCESS in 1h 02m 21s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/5a0c38e1d0de46c6b0a104db8e2dbf81 : FAILURE in 41m 24s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5169fdfe9f5448b5977906c61e13870b : FAILURE in 1h 57m 17s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/d7a452bb61944b019c20aca8b0163e26 : SUCCESS in 27m 41s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0ce6801f82c14465b59c458192da48b5 : SUCCESS in 1h 15m 42s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/08936bbfe11243f0a1d7139beecd020b : SUCCESS in 1h 15m 03s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/552619a6596b4e59a65d545fdcb9ae93 : SUCCESS in 1h 01m 22s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/b03993f73a5e4d86b9603b7a8626f69c : SUCCESS in 48m 12s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/85dae32c7d7f4178bec54c187d7473bd : SUCCESS in 1h 07m 48s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/0fb391feca9748658bacfc9cc75df0e7 : SUCCESS in 1h 28m 37s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5a2f9a9803d34ed39d51377e68d746c1 : FAILURE in 19m 18s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9150d26f7cfd4707b08acd7057467742 : SUCCESS in 40m 37s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f5159639ffac4429a1e8db45dd20c087 : SUCCESS in 41m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/75849c4d8c9c4625952fa7bce198eb37 : SUCCESS in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c866c5f374ccb0d42803e2b89821618140ccfb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 06:44:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f43cc40a1aef1a9c2e719ccf05f022322c72567f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 08:12:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"114cb220412330dfab8a3d5ca6e7e053d904eaea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 08:12:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6acee13f6dd44523ab7037ea0b59a689 : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7045f4ea3db346df8d6845952b042678 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fe737bd4f02493fb510a9d1f400ee5e : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9249fe129eaa4ec88bc81c275e6f84e6 : SUCCESS in 8m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6e3e7a2ceda4f549c8873cc91cb55dd : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5412f51758f4e20aeb17c43f3f8de72 : SUCCESS in 9m 02s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/31b63f9a02b84458b756d70523772434 : SUCCESS in 6m 55s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c73c50c0252b49d086c429bd5bae7269 : SUCCESS in 45m 21s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/0920e3bc28634c7a8797c08f69175450 : SUCCESS in 1h 21m 37s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/0e0e81352392472f8268ab15540b1a56 : SUCCESS in 1h 09m 00s","accounts_in_message":[],"_revision_number":2},{"id":"839567dbcc062e58d17bea584e201e89e1427bd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 08:16:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f9dcf3988ae4835bf4d20ae7fd2a13a : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cfcde645f6344ddb661930da060d46b : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2},{"id":"a79a1f714fe05154ffed92c546bb4057b54ba6a2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-08 08:37:36.000000000","message":"Patch Set 2:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-16-718016-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"4333874064197e2095876d9e2f7594a2c1aa33bf","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-08 11:39:11.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 89aa8939aa53e52254e4f98da0d80a76ef12b9e7","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c30fed486320ec5ba88f21ae3df14b14da6eb35","revisions":{"2de4f7ebaf2862053aeeaefeca72467fc9a55c59":{"kind":"REWORK","_number":1,"created":"2020-04-07 10:07:10.000000000","uploader":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"ref":"refs/changes/16/718016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/16/718016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/16/718016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/16/718016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/16/718016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/16/718016/1"}}},"commit":{"parents":[{"commit":"7c6bd92b66cf02d8aa7b133c1c97ac1ded7587f6","subject":"Merge \"[NetApp] Fix driver to honor standard extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7c6bd92b66cf02d8aa7b133c1c97ac1ded7587f6"}]}],"author":{"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","date":"2020-04-07 09:24:00.000000000","tz":120},"committer":{"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","date":"2020-04-07 09:37:49.000000000","tz":120},"subject":"Remove provisioned calculation on non thin provision backends","message":"Remove provisioned calculation on non thin provision backends\n\nOn those backends, there is no need to calculate provisioned_capacity_gb\nas it is not used during the scheduling. This calculation was not\nscaling properly on big environments as it implies many database\nqueries.\n\nCloses-Bug: #1869712\n\nChange-Id: Icb8947991723a2a1cc334a72276a35cf50fc1b7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2de4f7ebaf2862053aeeaefeca72467fc9a55c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2de4f7ebaf2862053aeeaefeca72467fc9a55c59"}]},"branch":"refs/heads/master"},"2c30fed486320ec5ba88f21ae3df14b14da6eb35":{"kind":"REWORK","_number":2,"created":"2020-04-07 19:10:52.000000000","uploader":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"ref":"refs/changes/16/718016/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/16/718016/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/16/718016/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/16/718016/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/16/718016/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/16/718016/2"}}},"commit":{"parents":[{"commit":"7c6bd92b66cf02d8aa7b133c1c97ac1ded7587f6","subject":"Merge \"[NetApp] Fix driver to honor standard extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7c6bd92b66cf02d8aa7b133c1c97ac1ded7587f6"}]}],"author":{"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","date":"2020-04-07 09:24:00.000000000","tz":120},"committer":{"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","date":"2020-04-07 19:09:40.000000000","tz":120},"subject":"Remove provisioned calculation on non thin provision backends","message":"Remove provisioned calculation on non thin provision backends\n\nOn those backends, there is no need to calculate provisioned_capacity_gb\nas it is not used during the scheduling. This calculation was not\nscaling properly on big environments as it implies many database\nqueries.\n\nCloses-Bug: #1869712\n\nChange-Id: Icb8947991723a2a1cc334a72276a35cf50fc1b7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2c30fed486320ec5ba88f21ae3df14b14da6eb35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2c30fed486320ec5ba88f21ae3df14b14da6eb35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
