)]}'
{"id":"openstack%2Fmanila~718233","triplet_id":"openstack%2Fmanila~master~I76e11fef27c9723f019cfdfdc6ea86878db78776","project":"openstack/manila","branch":"master","topic":"bug/1688620","hashtags":[],"change_id":"I76e11fef27c9723f019cfdfdc6ea86878db78776","subject":"[NetApp] Fix falsely report migration cancelation success","status":"MERGED","created":"2020-04-07 21:38:16.000000000","updated":"2020-05-22 18:50:58.000000000","submitted":"2020-05-22 17:52:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718233-1590169930346-f8705a57","meta_rev_id":"b4861f6fbdd59e47e356a7f11c19343c99626b95","_number":718233,"virtual_id_number":718233,"owner":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-05-04 16:05:32.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-22 17:52:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-04 18:10:53.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-22 18:11:15.000000000","post_submit":true,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-22 17:52:28.000000000","post_submit":true,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-22 14:17:26.000000000","_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"date":"2020-05-04 16:25:33.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-22 16:52:12.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-04 17:47:13.000000000","_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-05-04 16:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":2,"date":"2020-05-21 22:07:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":1,"date":"2020-05-21 22:07:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 21:38:59.000000000","updated_by":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"reviewer":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"state":"REVIEWER"},{"updated":"2020-04-08 08:50:07.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2020-05-04 16:05:32.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-05-04 16:08:36.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-05-04 16:25:33.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-04 17:47:13.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-05-04 18:10:53.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-21 22:07:19.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-05-22 14:17:26.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-05-22 16:52:12.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-22 17:52:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-22 17:52:28.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-22 18:11:15.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b798e6dcb32c0ceef8c4ccbee72f691eeae6d6d2","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-07 21:38:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e9e95783861d5a4868880d3c5a38b62d7d67597","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":15831,"name":"Infortrend CI","email":"ryan.chiang@infortrend.com","username":"InfortrendCI"},"date":"2020-04-07 21:38:59.000000000","message":"Patch Set 1:\n\nBuild failed.  \nFor information on how to proceed, see\n        https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures.\n\nFor 3rd party ci contact info: \n        https://wiki.openstack.org/wiki/ThirdPartySystems/Infortrend_Storage_CI\n\n\n\n- infortrend-manila-ci http://openstack.infortrend.com/1127/718233/1/check/infortrend-manila-ci/e7e1b4410a33e0469d972ed9c13e55a953afa760 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"1b685415b4aebaa2eb7f71ce901d42493a1e74fd","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-04-07 23:33:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/33/718233/1/check/DellEMC_PowerMAX_Manila/cb5eb44/DellEMC_PowerMAX_Manila/306 : FAILURE in 1h 54m 43s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"87336fd1fd28df59b5757707fac41f9d702a4b62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 23:40:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39aad70a99c14a38a667081baafc2151 : SUCCESS in 13m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd4b0cb5fb844fe699baa5512f04c82d : SUCCESS in 15m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/332d21c50d5f42128dda1281c9e1752b : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e01b3d5fad548fa826f584f6b29bba7 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f2e9bae5735444ea5c74ec27dde26cf : FAILURE in 22m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2fddc39a7d3f4a629dbf407adb5be1f8 : SUCCESS in 9m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ca858770c73464186938da39be4a246 : SUCCESS in 9m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0601d633a8f74c259a117164a6d57498 : SUCCESS in 12m 17s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cfcde10a67a543079a075ef03e454e5a : SUCCESS in 7m 04s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/82642a7ae2a2425493a35dee70f26c06 : FAILURE in 1h 19m 12s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/dfbcfb3d315b4be9a4f7192aa7039133 : SUCCESS in 48m 19s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/bdd570fff6e64b12b61ad5f17785ff48 : SUCCESS in 1h 05m 36s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/0cc324b245294e7f84437ebd991853db : SUCCESS in 1h 04m 20s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/7a22dc0ea99b47a39888dd23e41b614d : FAILURE in 47m 59s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/574d6a96ddc34396bc3860b8c5cf3fd9 : FAILURE in 1h 51m 27s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/649591719bdc4651ad6e169d56dd1130 : SUCCESS in 24m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2632b01c42da4bdcb1141d471db50bb9 : FAILURE in 1h 37m 56s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f7d6119774464587a00ef75d39e3012a : SUCCESS in 1h 31m 28s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/baf87bad065c47c9987f627133843179 : FAILURE in 1h 23m 49s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a9ca13b28de84a909e216bdc0f9c57c6 : SUCCESS in 49m 34s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/c9cdd175c3e14a8194d2fd9189e1a9bf : SUCCESS in 1h 33m 22s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/7eccf342876440adad9b262fb733b93d : SUCCESS in 1h 06m 56s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0d5810c5a4c74f78943498e335280682 : FAILURE in 26m 58s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/02166b4c5d9c45edb3bfe336eabe156b : SUCCESS in 36m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/135f6d1f9fa44c2c95d83a558489021b : SUCCESS in 42m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dfe9b873ca404d01a717c1c405ec595e : SUCCESS in 7m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7665e7014cca07eb4745eeecf35869bfd7e3a165","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-08 00:03:14.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-33-718233-1/ns5_manila : SUCCESS in 2h 24m 30s","accounts_in_message":[],"_revision_number":1},{"id":"dc48eb290ab7b4de7c6e8e029fa5a14c409d01b4","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-04-08 12:30:51.000000000","message":"Patch Set 1:\n\nrecheck\n\npy36 job failed due to the db intermitent issue","accounts_in_message":[],"_revision_number":1},{"id":"1d20cd44ef1b00225e40f89e0d98e98627987081","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-08 13:24:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ff85ce3823949387ec985992ba9b65bf2b15943c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 14:54:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7796252579854a14a295459b6b7f2ad1 : SUCCESS in 15m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d144dafb98104cfb9ed34498c136acb9 : SUCCESS in 15m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/129e2b2935774f6c8f06e14e481fead5 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b809a6b764fe49cca98eaab702b65761 : SUCCESS in 8m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd8a4dc95c434e9287e9891cc6dcf6e4 : SUCCESS in 9m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/155133b47a814cf3903e64dbe23a0b34 : SUCCESS in 10m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10b07214883746efa247045fb3f65932 : SUCCESS in 16m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3911644afe0d4434ba27c45b435108d5 : SUCCESS in 13m 05s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d0ad362e1ecb424d967bac202d65c9c4 : SUCCESS in 9m 20s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/844b36137a0a4676891dccc99bfe870d : FAILURE in 1h 16m 33s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/76d480adb74349fb8ff0732e3e47f118 : SUCCESS in 50m 05s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/2a094b8a2afd4501b3c8c9a91b620774 : SUCCESS in 1h 11m 03s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/e61afb3c9321442b8c6b14cafb0a8aa4 : SUCCESS in 58m 16s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/ed6f61320e57405c811382d4c1e4ad06 : FAILURE in 50m 09s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4104a9151b17479598f51444b13c6cda : FAILURE in 1h 48m 47s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/88030b2a14014c738940a485855ac66f : SUCCESS in 30m 16s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/27d302c39fe34ffea8c83095195832a9 : FAILURE in 1h 05m 36s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/36922c68506c451397ed846a710e1a00 : FAILURE in 1h 12m 07s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/4803e2196f954d138e91feaa8320d010 : FAILURE in 1h 15m 56s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/5599895f5ee64b49a4d69e860a548e6f : SUCCESS in 45m 11s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/2f03ae352511440a995989b4f1f0bfd7 : SUCCESS in 1h 11m 21s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/34e275f623ed4b9f952cc4b61954b510 : SUCCESS in 1h 01m 53s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/654d346c98e64768a42884f2b93c6d85 : FAILURE in 25m 18s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f51cb5c412b643d2a48e8efd2e22857b : SUCCESS in 43m 36s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/069bc2f27e4c46fbbd704cf05dc7ee3f : SUCCESS in 40m 25s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/441522a917b04572a372cadb7ff6b469 : SUCCESS in 8m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ebbf554a2d5fcac618d3aac5ad04b3aba8d3fca","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-04-08 15:25:15.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/33/718233/1/check/DellEMC_Unity_Manila_dhss_true/5b02e28/DellEMC_Unity_Manila_dhss_true/336 : ABORTED in 32s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/33/718233/1/check/DellEMC_Unity_Manila_dhss_false/f1af7f6/DellEMC_Unity_Manila_dhss_false/355 : ABORTED in 30m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"a3f13ba28d84fa02d2c92bef5b012b80d70c706a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-08 16:58:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/33/718233/1/upstream-check/manila-cDOT-ss/ad19475/ : FAILURE in 2h 03m 43s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/33/718233/1/upstream-check/manila-cDOT-no-ss/99c9a95/ : SUCCESS in 1h 30m 57s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"24c46ee2e60a71a6f57d71ec1fc6445d5a6439fb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-08 17:02:36.000000000","message":"Patch Set 1:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-33-718233-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"59c00bc933dbcba7bf3ac2a59cdfc4d415fbe1e8","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-04-08 21:37:45.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/33/718233/1/check/DellEMC_PowerMAX_Manila/2047e77/DellEMC_PowerMAX_Manila/316 : FAILURE in 2h 08m 53s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"faf208b141b122a0c6223c5dff2ba85180361d37","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-04-09 07:54:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/33/718233/1/check/DellEMC_VNX_Manila/b31ce82/DellEMC_VNX_Manila/169 : FAILURE in 1h 10m 23s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"bdd694390e8633f0f2a3f2a1c9b569c02142a6c4","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-28 20:36:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c583c5c19d400b9491a0c09c782e8f7bca2c3723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 22:38:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65b13c0ea27482f8915133a6688d535 : SUCCESS in 10m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01bd577a808a4d9c8ddd63acb2e470f2 : SUCCESS in 16m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/714454f2197342faa804f7ee37580203 : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e0f352676174bc6bf0dd19524361938 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c247ecf12e84db5bb4fa4838cba511f : SUCCESS in 12m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4f189eeb811492794a881fdfca0efb6 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be248d0115cc4297b1430d3c396fab1b : SUCCESS in 9m 40s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/76c415a86f4441e1b08425e57212d932 : SUCCESS in 6m 11s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/079f4892037a424aaf0f10b183a40e19 : FAILURE in 1h 07m 05s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/72338c1ba1b24739a1545d2494670995 : SUCCESS in 56m 45s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/6d6fcfd5077e4cb7821995edc5e5dc73 : FAILURE in 1h 21m 08s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/46d906cc00494ae5a8094bfeb5d58408 : SUCCESS in 1h 04m 42s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/112714ef7440401c8d733afb65ba1349 : FAILURE in 59m 32s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b8a2a4ce0024a3f9068eac88cfabd05 : FAILURE in 1h 54m 07s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/e7113968922d4b4383feb95a35a5f70d : SUCCESS in 26m 43s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/76403b2d4a584c63983c7c1a81382fb5 : SUCCESS in 1h 03m 34s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d707986a168f4f539822b82589a79664 : SUCCESS in 1h 30m 53s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/a5a7168712414100a86025f99b20249c : FAILURE in 1h 34m 42s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d3f7aaea3afa4e2fbd18327ec5a08f82 : SUCCESS in 53m 52s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/b69b4b2d08764a6488a6f0d2761d01af : SUCCESS in 1h 04m 36s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3dd6372f46014f37ab0b44513e4b4f24 : SUCCESS in 34m 03s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ea66776ae15548e88b5c199bc8c5f47d : SUCCESS in 42m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fe08296629974a4eb70d5b1d17649939 : SUCCESS in 50m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4522ac7799ca417aa9ec1389d759b5e7 : SUCCESS in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"509e44e12bc8f6c12d59512ed87273a3d659354f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-28 22:39:58.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-33-718233-2/ns5_manila : SUCCESS in 2h 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d796b49b0b3db15a34176fe170f5a973cbb26525","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-04-28 22:44:10.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/33/718233/2/check/DellEMC_PowerMAX_Manila/b43962f/DellEMC_PowerMAX_Manila/423 : SUCCESS in 2h 07m 33s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"a249f008c508ddaabf0890674a740ed89a3147ce","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-04-28 23:26:17.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/33/718233/2/check/DellEMC_Unity_Manila_dhss_true/22c85c0/DellEMC_Unity_Manila_dhss_true/449 : SUCCESS in 2h 49m 39s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/33/718233/2/check/DellEMC_Unity_Manila_dhss_false/a8b1305/DellEMC_Unity_Manila_dhss_false/474 : SUCCESS in 1h 21m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"61f1aff4586700b026d26146f0872ed33904c457","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-04-28 23:56:10.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/33/718233/2/check/DellEMC_VNX_Manila/233a54d/DellEMC_VNX_Manila/220 : SUCCESS in 1h 17m 41s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"ccf18dd752cb68e08669a207a79b4b66a9e311da","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-29 00:01:46.000000000","message":"Patch Set 2:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-33-718233-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"d6f46e14d5237c0c029bed4970a8ecabc2054262","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-29 00:36:33.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/33/718233/2/upstream-check/manila-cDOT-ss/d618062/ : FAILURE in 1h 57m 56s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/33/718233/2/upstream-check/manila-cDOT-no-ss/75b572a/ : FAILURE in 1h 26m 18s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b7163661af2d1fae353fb089155c2b30621f6491","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-04-29 12:53:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\nThanks for working on this, Daniel!\nI have added some comments for you, please check them inline.","accounts_in_message":[],"_revision_number":2},{"id":"be631ef9642aed002f04aa54aff354b502242892","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-29 23:06:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ad844e684f8c204f99278fa1d1a13040d1d360af","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-04-29 23:08:47.000000000","message":"Patch Set 2:\n\n(4 comments)\n\nHi Carlos, thank you for your review, I\u0027ve done the corrections.","accounts_in_message":[],"_revision_number":2},{"id":"a3bffac3e157420c2d5cb7e1fce8f8be5fa5e1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 00:59:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcffd5cc57b04f2aa98522a8fc87eee6 : SUCCESS in 8m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9f789a7b74e468cbde86f29dc5fbac2 : SUCCESS in 13m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7eb601267fcd46eb8d55afbf92e00aae : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c04b7578bdc74717849adbc3b32b9f00 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1495d121b0ee4d57987730c688a25c8c : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3acc1206f425426d80dc8757e83016d4 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db006df1c9dd49dea2dd8f09b15d4bab : SUCCESS in 9m 08s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4d1a9b71832e41fa9cba11f2d46e7415 : SUCCESS in 4m 51s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/2f9399cf6fc84e8880e3acfc0023d3fa : FAILURE in 1h 12m 35s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/98d99b439b4546bb89d387bfa69d46ba : SUCCESS in 1h 03m 10s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/d9364613944a40eaab8509d8ab1feee2 : FAILURE in 1h 22m 48s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/bc36ceef965e47f79b000aefeea7a40e : FAILURE in 50m 23s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/ef423cb7241649a1b2719a7e31d0b9c3 : FAILURE in 47m 56s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3765b9a6001c48f4b06cf2073a78337d : FAILURE in 1h 50m 06s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/95ba6a6fd396472ea712bdc5447f2b78 : SUCCESS in 25m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2c444eea78b84dbd95a704ee49a4c350 : FAILURE in 34m 04s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4dfeda4cd3d24c0b87cb9ec072f79577 : SUCCESS in 1h 19m 11s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/d40b8d4b123143ed8cd07d8283fe6523 : FAILURE in 1h 09m 32s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e3ccfefbfa4544d1b7b44ce01eb3f6b1 : SUCCESS in 52m 24s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/98705d787d7e4c199d6abb60bf2cf9a9 : SUCCESS in 1h 00m 39s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/31a69c835b214c28b40b8039c702c063 : SUCCESS in 24m 06s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7d8cf6c5068a43b1bcc07f3abce6bf1f : SUCCESS in 43m 31s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b95a9f4743d5476a895d126441577df9 : SUCCESS in 38m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/12c49a4128734abe96302a4b63537e21 : SUCCESS in 6m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6e5c71f214b0852dc2408b13fc1dde248624ccdc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-04-30 01:10:52.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-33-718233-3/ns5_manila : SUCCESS in 2h 3m 34s","accounts_in_message":[],"_revision_number":3},{"id":"d079ffa934cff5d48e93ec6c2eb64db285ac3e66","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-30 03:41:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/33/718233/3/upstream-check/manila-cDOT-ss/bd7ac0e/ : FAILURE in 2h 19m 48s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/33/718233/3/upstream-check/manila-cDOT-no-ss/4119b0c/ : SUCCESS in 1h 35m 56s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"a5898dfe275b431bb3f4eb8cb68dd697a523d5e6","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-04-30 04:04:28.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/33/718233/3/check/DellEMC_VNX_Manila/03fada2/DellEMC_VNX_Manila/233 : SUCCESS in 1h 13m 36s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"da8435ea43eb6c2b26ae279b689c85bd38ffb0e4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-30 04:27:07.000000000","message":"Patch Set 3:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-33-718233-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"35d5f8a5ec656e477325a8834e4582c7ca09ceb6","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-05-04 10:59:31.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)\n\nHey, Daniel! Thanks for answering to my comments. LGTM in an overall :)","accounts_in_message":[],"_revision_number":3},{"id":"3e5101db1827f2d40fa18194e49926a7a7ec56ed","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-05-04 14:21:28.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"b9452ebd089358dd21b99bb7e8bf708cd2fcef53","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-05-04 14:24:45.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f73d6baed3800bd933aae21b2e785a6b17e76b9c","author":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"date":"2020-05-04 14:28:56.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"a040125f1399469ebf8d1db67c7d860ffa6c0e42","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-05-04 16:05:32.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/33/718233/6/check/dsvm-tempest-infinibox-nfs/8cd9faa : SUCCESS in 1h 13m 21s","accounts_in_message":[],"_revision_number":6},{"id":"d385f6e4921f667b7e5f59975004c12e9b0f0816","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-05-04 16:08:36.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"1071e574c91b7a684f03958b6a00d03cd38b1968","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:11:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c95c1c3fb854992952982322ac7a508 : SUCCESS in 11m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7eef0264eadc4563b835c292543e0e50 : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4617913fba04583a34479c619e86e51 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c76e01003842618d30a3d6eb3de379 : SUCCESS in 11m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a29d9bfb110642fd897e531eb0dc15c9 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/417ad2b6b96e4fa192a1e1441a643cf2 : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/203b8253d8d043a08c45ef2d04bbc56a : SUCCESS in 9m 40s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d4822b6fa0654666994c2a93d1adca9e : SUCCESS in 7m 46s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/987a1d8ba0474680ac30b4cd346a134b : FAILURE in 1h 08m 27s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/629ae4b920b2453d86f83edd78452652 : FAILURE in 59m 34s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/41f98667488042ca87474834b9b8021c : SUCCESS in 1h 12m 02s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/2a59e4c61d7c45d6b7774ef93a9d58bd : SUCCESS in 55m 47s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/c0effc2270614632a654762801ff86e0 : FAILURE in 1h 12m 17s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/884bf8aa08ba4f07a87e03705afaaa34 : FAILURE in 1h 40m 21s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/9c0772c24ce843839637c33f3c8c1b9f : SUCCESS in 31m 54s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/367912eaffde4ef0a6de22562dd12566 : SUCCESS in 1h 19m 56s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/51b1611e4b194bd8abe3dbd3e86eab66 : SUCCESS in 1h 09m 24s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/0869b9b703e04c1f8fca7abad592e7da : FAILURE in 1h 01m 03s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/747bb692762b48c4bf58db326b0273d1 : SUCCESS in 45m 27s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/ee7b5cef7215444f804606b17f8055ee : SUCCESS in 57m 53s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5861134a08cf48f581831593e1977b84 : SUCCESS in 25m 32s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6f0566320dfd4f31b53a815991c09e4e : SUCCESS in 41m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1aa6d13e44044cc2ac97982aed5a83b5 : SUCCESS in 40m 11s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6c0af94b74af4a9f8febc2e167e1cedb : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3776e4df04899cb532b80e560d5eef6898765f6e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-05-04 16:25:33.000000000","message":"Patch Set 6:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-33-718233-6/ns5_manila : SUCCESS in 1h 56m 18s","accounts_in_message":[],"_revision_number":6},{"id":"7d9d6f806dfdcd721b69d7531f318bf9f80e593f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-05-04 16:25:42.000000000","message":"Patch Set 5:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-33-718233-5/ns5_manila : SUCCESS in 2h 39s","accounts_in_message":[],"_revision_number":5},{"id":"60e266b5ecf3c253b8092c0ebbe7ba854eecf2e6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-05-04 16:28:54.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-33-718233-4/ns5_manila : SUCCESS in 2h 7m 1s","accounts_in_message":[],"_revision_number":4},{"id":"4b5e6548b37baaee5b0b51da56212b5d8b46e89e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-05-04 17:47:13.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/33/718233/6/check/DellEMC_Unity_Manila_dhss_true/7da1f89/DellEMC_Unity_Manila_dhss_true/474 : SUCCESS in 3h 18m 02s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/33/718233/6/check/DellEMC_Unity_Manila_dhss_false/f4cbd9c/DellEMC_Unity_Manila_dhss_false/507 : FAILURE in 42m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"dec747b51f9dcd6cd400b53ed198f35489db352e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-05-04 18:10:53.000000000","message":"Patch Set 6:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-33-718233-6 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":6},{"id":"25769ca43c4ca8aa2613fdd26f58fb234e5030b6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-04 18:18:41.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/33/718233/6/upstream-check/manila-cDOT-ss/7f9acb4/ : SUCCESS in 2h 06m 33s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/33/718233/6/upstream-check/manila-cDOT-no-ss/c571316/ : FAILURE in 1h 33m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"a2a02a0d8968a60d45ebe8b91aa0c33a600555a4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-21 16:30:42.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)\n\nrun netapp-ci","accounts_in_message":[],"_revision_number":6},{"id":"2e947af01e3c6d3343ecbd77157440129df6b958","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-05-21 17:17:32.000000000","message":"Patch Set 6:\n\nrun-NetApp CI","accounts_in_message":[],"_revision_number":6},{"id":"59ccf7c8dcb96b855b1cadc1860a75388fddf4c7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-21 20:09:11.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/33/718233/6/upstream-check/manila-cDOT-ss/88dd789/ : SUCCESS in 2h 09m 51s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/33/718233/6/upstream-check/manila-cDOT-no-ss/4c25ed4/ : SUCCESS in 1h 26m 08s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"9cf4e9adcf72b3c0ccd21778980d7e275a1882d7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-21 22:07:19.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ae5efdaf42fee1c7f816edc941d29d878f2d10b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 22:07:31.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f4c82abcdc77c34f4964c975b5511786167fc770","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 23:17:35.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d7063c630140c9a6793b6368569bbc : SUCCESS in 9m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22569d939e0c4efd9661cb29a4b68394 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/188183f985eb4d46b445ecac3da7417c : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b494e4252ead421d82e6bf5ccd15535d : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f60d2279d66457dabca1adab28be4bb : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7af1d9cd3424264ba3d73de17a8ef2a : SUCCESS in 5m 59s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d351099c07a84ebb8b1e7a987fc6ffc4 : SUCCESS in 6m 16s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/30f465a5a5b240f9991aa8cb5836f9f2 : FAILURE in 47m 07s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/295b67e714f347dbba619162b3357bd2 : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":6},{"id":"f2c5a8b1d4f4021b037fb8c40810c8ddc772ce6d","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-05-22 14:17:26.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"73868c5031a99bda296ff167914555f7d023d55b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 16:42:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3c9b4a26717411ba1a2e3b6b2be65ed : SUCCESS in 12m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0c5377bd6f542d9a839c85fa1bd561d : SUCCESS in 14m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3230a954762467c9517e824d14dc6a3 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd01e49a90eb408abdf4b4b3b16d7012 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78b9a9acd58d4076a10a0869fb6251f2 : SUCCESS in 10m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3195ce13b69c4076b2d75923a388f6c7 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a80fc75cbe34f23ad7f408acc56a144 : SUCCESS in 7m 27s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/155fb94c969445f59d5c0f4454a0749a : SUCCESS in 7m 20s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/3d565648d91b4551ad4c7f23da4ccc29 : SUCCESS in 1h 19m 26s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/6b5606b4895b4627af73468a2a29421d : SUCCESS in 53m 33s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/74e408ddad7f47a38b6654e1e8093ac4 : FAILURE in 1h 07m 07s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/28ab98d3cd3146f1a7f5fa92299d4ddf : FAILURE in 1h 03m 52s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/59e6a9601b584061807a19944a0b7ce6 : FAILURE in 48m 59s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1fcc91d653949799d2f893b6653a484 : SUCCESS in 1h 28m 35s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/f51567f38b354aa0b5bbd735a6c449c1 : SUCCESS in 27m 36s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b62cf58318d74c0dbe543889da7d42f7 : FAILURE in 1h 39m 01s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/22a73270be0243cc92593be81ad63809 : SUCCESS in 1h 01m 53s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/30ba73a1691a43e9b13bc4c6d84e282c : FAILURE in 1h 39m 17s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/20693644fcda460f99113749e33a3c90 : SUCCESS in 49m 30s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/115ff064833d4d1d86b4646161df0e78 : SUCCESS in 1h 09m 02s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/16ba77e5b8be4a5fa7ae4811fec66d9c : SUCCESS in 32m 35s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0f4f8104dcca4be3817485d61887a964 : SUCCESS in 33m 58s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8e1726dfa3c94b688e31df6d3be7cd8a : SUCCESS in 42m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cb5389e1f850400f8745eed4e9b5352f : SUCCESS in 7m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"77db7359a0c158f62f9bc9c003a67f049c4337a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 16:42:40.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2e94a53572e7eb24532f94d00da82de2cae1137c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-22 16:52:12.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/33/718233/6/upstream-check/manila-cDOT-ss/48ea128/ : FAILURE in 9m 17s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/33/718233/6/upstream-check/manila-cDOT-no-ss/0130ef1/ : FAILURE in 8m 19s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"297cc572ce04e046815dc7a0e8b274665b2870ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 17:52:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"6a0e35666c4e4f6998176ff31024c4280ab0b3c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 17:52:10.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/703013b5b1324e4faf50c714ff805885 : SUCCESS in 10m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bdb2e6f996744df8ec2e6d5e113230f : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67438852dce04291921aefefb8176455 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f399b03c62845ffa4d54f65530dc651 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32ac52b75f104379b7a0d9a864743e4f : SUCCESS in 14m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8897ed44ad7c4d988d6cd27ab9a01e3f : SUCCESS in 6m 56s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/042f254f03aa42428241dfb6e23c1db7 : SUCCESS in 5m 38s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c2dee247e2dc4222a9b4f922983436aa : SUCCESS in 43m 31s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/104e762635f740ed8ed7e38e5197edd9 : SUCCESS in 1h 06m 45s","accounts_in_message":[],"_revision_number":6},{"id":"8d509522c9f3c11b241d151cf5de8def12c524f2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-05-22 17:52:28.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/33/718233/6/check/DellEMC_VNX_Manila/1f27aa2/DellEMC_VNX_Manila/7 : SUCCESS in 1h 09m 44s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"2bf75d3a2355519f51c5fec59bdc47600e4b01bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 17:53:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e1fa1df9b0e4d0bb52b9fe7cd6c7949 : SUCCESS in 1m 30s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbb33c4bba094a1197cddfa921308859 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":6},{"id":"5e96d45e52d64890cd577b3276a9d651c783964a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-05-22 17:55:10.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 9efbc09648871639f5d7b23e6c8de3de835ffc03","accounts_in_message":[],"_revision_number":6},{"id":"a15cdc43707f8ff6b926df08698f98f86db06ca0","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-05-22 18:11:15.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/33/718233/6/check/DellEMC_PowerMAX_Manila/2f283c8/DellEMC_PowerMAX_Manila/15 : SUCCESS in 1h 28m 36s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"eb9196aa3da9a93ef6e74f8578f5eac2f6629ebe","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-05-22 18:50:58.000000000","message":"Patch Set 6:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-33-718233-6 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"0ee414082318d22b3ed19acad6a479cb105c30e5","revisions":{"e7e1b4410a33e0469d972ed9c13e55a953afa760":{"kind":"REWORK","_number":1,"created":"2020-04-07 21:38:16.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/33/718233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/33/718233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/33/718233/1"}}},"commit":{"parents":[{"commit":"7c6bd92b66cf02d8aa7b133c1c97ac1ded7587f6","subject":"Merge \"[NetApp] Fix driver to honor standard extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7c6bd92b66cf02d8aa7b133c1c97ac1ded7587f6"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-06 13:26:13.000000000","tz":0},"committer":{"name":"Daniel Tapia","email":"danielarthurt@gmail.com","date":"2020-04-07 21:38:14.000000000","tz":-180},"subject":"[NetApp] Fix falsely report migration cancelation success","message":"[NetApp] Fix falsely report migration cancelation success\n\nNetApp ONTAP share delete operation can fail sometimes when is triggered\nimmediately after migration cancelation on a overloaded NetApp backend.\nMigration cancelation invokes \"abort_volume_move\" which is an asynchronous\nAPI. If share delete operation is requested immediately after call the\nformer API, it fails because the \"abort_volume_move\" is still in progress.\nNow NetApp cDOT driver checks, for a period of time, if the\nvolume-move-abort has ended before report migration cancelation success.\n\nChange-Id: I76e11fef27c9723f019cfdfdc6ea86878db78776\nCloses-Bug: #1688620\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e7e1b4410a33e0469d972ed9c13e55a953afa760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e7e1b4410a33e0469d972ed9c13e55a953afa760"}]},"branch":"refs/heads/master"},"dcca320c056fa92507371827887f1430a5a86f23":{"kind":"REWORK","_number":2,"created":"2020-04-28 20:36:24.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/33/718233/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/33/718233/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/33/718233/2"}}},"commit":{"parents":[{"commit":"67fe72f56330fc233760cef41deb8f991cac5630","subject":"Merge \"[doc] Fix tempest doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/67fe72f56330fc233760cef41deb8f991cac5630"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-06 13:26:13.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-28 20:36:23.000000000","tz":0},"subject":"[NetApp] Fix falsely report migration cancelation success","message":"[NetApp] Fix falsely report migration cancelation success\n\nNetApp ONTAP share delete operation can fail sometimes when is triggered\nimmediately after migration cancelation on a overloaded NetApp backend.\nMigration cancelation invokes \"abort_volume_move\" which is an asynchronous\nAPI. If share delete operation is requested immediately after call the\nformer API, it fails because the \"abort_volume_move\" is still in progress.\nNow NetApp cDOT driver checks, for a period of time, if the\nvolume-move-abort has ended before report migration cancelation success.\n\nChange-Id: I76e11fef27c9723f019cfdfdc6ea86878db78776\nCloses-Bug: #1688620\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dcca320c056fa92507371827887f1430a5a86f23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dcca320c056fa92507371827887f1430a5a86f23"}]},"branch":"refs/heads/master"},"5f64aaff713a4bf2ecddc49b4f2ad10f2c9a63ba":{"kind":"REWORK","_number":3,"created":"2020-04-29 23:06:50.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/33/718233/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/33/718233/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/33/718233/3"}}},"commit":{"parents":[{"commit":"67fe72f56330fc233760cef41deb8f991cac5630","subject":"Merge \"[doc] Fix tempest doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/67fe72f56330fc233760cef41deb8f991cac5630"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-06 13:26:13.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-29 23:06:11.000000000","tz":0},"subject":"[NetApp] Fix falsely report migration cancelation success","message":"[NetApp] Fix falsely report migration cancelation success\n\nNetApp ONTAP share delete operation can fail sometimes when is triggered\nimmediately after migration cancelation on a overloaded NetApp backend.\nMigration cancelation invokes \"abort_volume_move\" which is an asynchronous\nAPI. If share delete operation is requested immediately after call the\nformer API, it fails because the \"abort_volume_move\" is still in progress.\nNow NetApp cDOT driver checks, for a period of time, if the\n``volume-move-abort`` operationhas ended before report migration\ncancelation success.\n\nChange-Id: I76e11fef27c9723f019cfdfdc6ea86878db78776\nCloses-Bug: #1688620\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5f64aaff713a4bf2ecddc49b4f2ad10f2c9a63ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5f64aaff713a4bf2ecddc49b4f2ad10f2c9a63ba"}]},"branch":"refs/heads/master"},"87a6eb5cabcf1d7cddfe0eabf70276d8c43c29ca":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-05-04 14:21:28.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/33/718233/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/33/718233/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/33/718233/4"}}},"commit":{"parents":[{"commit":"67fe72f56330fc233760cef41deb8f991cac5630","subject":"Merge \"[doc] Fix tempest doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/67fe72f56330fc233760cef41deb8f991cac5630"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-06 13:26:13.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-05-04 14:21:08.000000000","tz":0},"subject":"[NetApp] Fix falsely report migration cancelation success","message":"[NetApp] Fix falsely report migration cancelation success\n\nNetApp ONTAP share delete operation can fail sometimes when is triggered\nimmediately after migration cancelation on a overloaded NetApp backend.\nMigration cancelation invokes \"abort_volume_move\" which is an asynchronous\nAPI. If share delete operation is requested immediately after call the\nformer API, it fails because the \"abort_volume_move\" is still in progress.\nNow NetApp cDOT driver checks, for a period of time, if the\n``volume-move-abort`` operationhas ended before report migration\ncancelation success.\n\nChange-Id: I76e11fef27c9723f019cfdfdc6ea86878db78776\nCloses-Bug: #1688620\nDepends-On: I4ab2cab3ce1a729dabc7779cdef691c7141e870b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/87a6eb5cabcf1d7cddfe0eabf70276d8c43c29ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/87a6eb5cabcf1d7cddfe0eabf70276d8c43c29ca"}]},"branch":"refs/heads/master"},"bfec33f0e840f90cbf93084f8abe0a0e06bd58b0":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-05-04 14:24:45.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/33/718233/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/33/718233/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/33/718233/5"}}},"commit":{"parents":[{"commit":"67fe72f56330fc233760cef41deb8f991cac5630","subject":"Merge \"[doc] Fix tempest doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/67fe72f56330fc233760cef41deb8f991cac5630"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-06 13:26:13.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-05-04 14:24:16.000000000","tz":0},"subject":"[NetApp] Fix falsely report migration cancelation success","message":"[NetApp] Fix falsely report migration cancelation success\n\nNetApp ONTAP share delete operation can fail sometimes when is triggered\nimmediately after migration cancelation on a overloaded NetApp backend.\nMigration cancelation invokes \"abort_volume_move\" which is an asynchronous\nAPI. If share delete operation is requested immediately after call the\nformer API, it fails because the \"abort_volume_move\" is still in progress.\nNow NetApp cDOT driver checks, for a period of time, if the\n``volume-move-abort`` operationhas ended before report migration\ncancelation success.\n\nChange-Id: I76e11fef27c9723f019cfdfdc6ea86878db78776\nCloses-Bug: #1688620\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bfec33f0e840f90cbf93084f8abe0a0e06bd58b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bfec33f0e840f90cbf93084f8abe0a0e06bd58b0"}]},"branch":"refs/heads/master"},"0ee414082318d22b3ed19acad6a479cb105c30e5":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-05-04 14:28:56.000000000","uploader":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"ref":"refs/changes/33/718233/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/33/718233/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/33/718233/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/33/718233/6"}}},"commit":{"parents":[{"commit":"67fe72f56330fc233760cef41deb8f991cac5630","subject":"Merge \"[doc] Fix tempest doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/67fe72f56330fc233760cef41deb8f991cac5630"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-04-06 13:26:13.000000000","tz":0},"committer":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-05-04 14:28:35.000000000","tz":0},"subject":"[NetApp] Fix falsely report migration cancelation success","message":"[NetApp] Fix falsely report migration cancelation success\n\nNetApp ONTAP share delete operation can fail sometimes when is triggered\nimmediately after migration cancelation on a overloaded NetApp backend.\nMigration cancelation invokes \"abort_volume_move\" which is an asynchronous\nAPI. If share delete operation is requested immediately after call the\nformer API, it fails because the \"abort_volume_move\" is still in progress.\nNow NetApp cDOT driver checks, for a period of time, if the\n``volume-move-abort`` operation has ended before report migration\ncancelation success.\n\nChange-Id: I76e11fef27c9723f019cfdfdc6ea86878db78776\nCloses-Bug: #1688620\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0ee414082318d22b3ed19acad6a479cb105c30e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0ee414082318d22b3ed19acad6a479cb105c30e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
