)]}'
{"id":"openstack%2Fmanila~718401","triplet_id":"openstack%2Fmanila~stable%2Ftrain~Icb8947991723a2a1cc334a72276a35cf50fc1b7d","project":"openstack/manila","branch":"stable/train","topic":"bug/1869712-stable/train","hashtags":[],"change_id":"Icb8947991723a2a1cc334a72276a35cf50fc1b7d","subject":"Remove provisioned calculation on non thin provision backends","status":"MERGED","created":"2020-04-08 11:39:11.000000000","updated":"2020-04-15 00:10:18.000000000","submitted":"2020-04-15 00:07:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718401-1586909277261-5f04ccdf","meta_rev_id":"b77076b1a8a9b67bbaca0d98a8989a1639c19b9b","_number":718401,"virtual_id_number":718401,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"date":"2020-04-08 14:00:31.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-04-08 20:45:58.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-15 00:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"recommended":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2020-04-12 22:05:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-04-08 23:41:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2020-04-14 22:42:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-04-08 11:39:11.000000000","_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 11:39:11.000000000","updated_by":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"reviewer":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"state":"REVIEWER"},{"updated":"2020-04-08 20:45:58.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-12 22:05:27.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-04-14 22:42:39.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-15 00:07:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6751073e6ee73c6537718511643a37d7e58be9c","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-08 11:39:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5202db8ebccc3d1e5e11230b811f746e7110f120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 13:38:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6ee4854ce14d46b71901381edeb50d : SUCCESS in 10m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a2bee028eb14b49b162494030e35140 : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7af4b1a8be2a46588086033a2b29a6f9 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2303acbdc7594f27ade9c5d5b89d4ef7 : SUCCESS in 6m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1aa6435ba870470a84fcc7d0728caeb5 : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e411c47a07d345b1986fcec0871b4e01 : SUCCESS in 19m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/791f67b6f12941da9659ff4168c2f01c : SUCCESS in 13m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2333c514b674dcf8025cf8c1a18120b : SUCCESS in 9m 30s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cf7b6061182948b59d84ca07e620f605 : SUCCESS in 8m 14s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/553ad8fb37464ec2a616147c783ae0cb : FAILURE in 1h 44m 14s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/34e4e6d1a91240148ce0efcfeb293c7a : SUCCESS in 57m 59s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/7a2c8e5ff973424cb5c5d101f17ac5f7 : FAILURE in 1h 18m 32s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/2caf642956864845847bb33d52de20c6 : SUCCESS in 1h 03m 27s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/1bbfd37432ba44c2aad1301fd952186d : FAILURE in 46m 45s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aebca767318147adbd31e623d1b2e18e : FAILURE in 1h 52m 20s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/95dc5d06c6874a278ec9f131705890e3 : SUCCESS in 26m 30s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/87a481f0f2b34e34bd014fe0e87c35fa : SUCCESS in 1h 25m 08s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1b8fdc7140e546a3b817cef39beda05a : SUCCESS in 1h 23m 17s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/5c1caa66e3814098a917fcea43e0356c : SUCCESS in 47m 14s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/b9a1197e063845239e15b5d7c0989305 : FAILURE in 49m 34s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/6532e8f4de904d629ba12e15347f1423 : SUCCESS in 1h 05m 10s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/1c092cee4fc14f3aa1a88eadd21b5d35 : SUCCESS in 1h 07m 14s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4b444a0d36c54e2b81a72e768c875e10 : SUCCESS in 26m 30s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/bde2eb3b6bbd4f73930572cce388d402 : FAILURE in 33m 52s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/9b7200f8ec3b48cea7981a8a69f69fc1 : FAILURE in 47m 23s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4f8581c31f5a4a1a9f0d5e23fa7be671 : SUCCESS in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b203900b50e8a0c1bd99bf05f23a06a48a9ce78","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-08 14:00:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"59ec44cc388e43b6781f41eea5ad3e5a8894bc7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 16:00:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d6237c0400541a0bb915df3734c5821 : SUCCESS in 9m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b59a3732b5d420598b63ff0463ca28a : SUCCESS in 17m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bc8654c512e4007818ba54943d1faad : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8daf263b7d8f446b883ad926dc57dabb : SUCCESS in 7m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9be2f351837d410b8405c00676b02856 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d7936fcfe00488ba938496c8f4db285 : SUCCESS in 9m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56739b0444b7469690cc12f80c5121aa : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e56dbea96238477883ad207654843537 : SUCCESS in 10m 45s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/aedbecb3f19d4d4b906f9d7cc12bf43f : SUCCESS in 5m 50s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/cb0e99bf30804c88b16c9ef22454c340 : FAILURE in 1h 25m 05s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/7d39d88d65d24bed948d934e0c80c0a7 : SUCCESS in 53m 27s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/5dd9440c8c544cd184cfdbd853773b13 : SUCCESS in 1h 15m 41s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/3a6f401ae7c443519e95389ef1b58003 : FAILURE in 57m 08s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/ac85c90be9ae4f1caff152db1d17151c : FAILURE in 45m 27s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9aa15fc8497e4f3e83f9ae1ed7b6610c : FAILURE in 1h 45m 21s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/dc4bfcd9984e4778a7c034016832bdb2 : SUCCESS in 26m 38s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2a17ca4fdba24ada92da72a8641ef364 : SUCCESS in 1h 33m 45s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cd2d1bfc2d364f36bfe4f106624abc18 : SUCCESS in 1h 06m 30s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ea970be100fd4650a124016cc0a3b4a1 : SUCCESS in 41m 30s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/81de39530dc547c1aea250b4ae9f8666 : SUCCESS in 49m 08s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/a529f9a019674e4bb787001a46feb13f : SUCCESS in 1h 08m 31s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/cf0e8e9550324a04a5ab4439ee9bc1c8 : SUCCESS in 1h 01m 24s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bacc9415d8454c45ae1166e21ce15727 : SUCCESS in 25m 46s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2b98fa0b67854b009f33f0828a42815e : FAILURE in 37m 27s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1e148145ad964bec8655085e8944347e : FAILURE in 50m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a8136140fcd24a2eab2f158ede6aa3ad : SUCCESS in 6m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dfc105453fca874e16befd7df34422914d66722b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-08 20:45:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/01/718401/1/upstream-check/manila-cDOT-ss/4fab3fa/ : FAILURE in 1h 56m 26s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/01/718401/1/upstream-check/manila-cDOT-no-ss/3ea4beb/ : SUCCESS in 1h 34m 46s\n\nComment \u0027run-netapp-ci\u0027 or \u0027netapp-ci-recheck\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"aa1036e0e53d7bd5263253e801a349c84065face","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-08 23:41:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ce9861b2c3f5ca21ba17398a3a4b1f61793e2c48","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-12 22:05:27.000000000","message":"Patch Set 1: Code-Review+1\n\nclean cherry-pick, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"21ad5f28a9f91deb5e7e43101583205ea07b0bbd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-14 22:42:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e024a7271770146f36f7066601b7e8b00050e8ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 22:42:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2c622d0b0b64977c7d4f8a54e4af1534164e929","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-14 23:27:13.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit a9f38d8bb4e0b8d83a9a41afd320ed5d554fe73c","accounts_in_message":[],"_revision_number":1},{"id":"7de2a879e038755894e292b3792c94a70506e4a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 00:07:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f64a0c0c9f864c8eaba0056c590ffadb : SUCCESS in 12m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae2310748ac743debc6e34879e1cdf40 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e86b67097d8948fb8f5c6dd97711b8a7 : SUCCESS in 8m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b64384efd620462486e77dd71d9b4ee7 : SUCCESS in 14m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bb6bf003dfa4663b810066c279a24b5 : SUCCESS in 15m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e33166823ce4161aee397d0b8a4ec97 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca9268f45f344619a7270205e44416fb : SUCCESS in 9m 37s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e23546123ea54f798a5d69c7be948801 : SUCCESS in 5m 15s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/33ffc2afe61a42a0baa8ef4b2f6fcb55 : SUCCESS in 45m 28s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/9dabe895d74f477e8bbeb5e533254bb9 : SUCCESS in 1h 20m 02s","accounts_in_message":[],"_revision_number":1},{"id":"eeca3fcfdc35c9d1d0b32b28b154b3e0f9787e42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 00:07:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4770f11e12483bba5e3ddb757ff4dae886ec8d80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 00:10:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/becd16eb18654d72a64565d30c00cea5 : SUCCESS in 1m 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b70c2732e134e508a115fbcce8e79e5 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"89aa8939aa53e52254e4f98da0d80a76ef12b9e7","revisions":{"89aa8939aa53e52254e4f98da0d80a76ef12b9e7":{"kind":"REWORK","_number":1,"created":"2020-04-08 11:39:11.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/01/718401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/01/718401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/01/718401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/01/718401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/01/718401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/01/718401/1"}}},"commit":{"parents":[{"commit":"678ea79f85e53ec21a0fb74758875b3a7ebe1aa7","subject":"[NetApp] Fix driver to honor standard extra specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/678ea79f85e53ec21a0fb74758875b3a7ebe1aa7"}]}],"author":{"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","date":"2020-04-07 09:24:00.000000000","tz":120},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-04-08 11:39:11.000000000","tz":0},"subject":"Remove provisioned calculation on non thin provision backends","message":"Remove provisioned calculation on non thin provision backends\n\nOn those backends, there is no need to calculate provisioned_capacity_gb\nas it is not used during the scheduling. This calculation was not\nscaling properly on big environments as it implies many database\nqueries.\n\nCloses-Bug: #1869712\n\nChange-Id: Icb8947991723a2a1cc334a72276a35cf50fc1b7d\n(cherry picked from commit 2c30fed486320ec5ba88f21ae3df14b14da6eb35)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/89aa8939aa53e52254e4f98da0d80a76ef12b9e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/89aa8939aa53e52254e4f98da0d80a76ef12b9e7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
