)]}'
{"id":"openstack%2Fmanila~720206","triplet_id":"openstack%2Fmanila~stable%2Fstein~Id86a93f1923a493cefe73e2d90d5094005b42ae4","project":"openstack/manila","branch":"stable/stein","topic":"bug/1699836","hashtags":[],"change_id":"Id86a93f1923a493cefe73e2d90d5094005b42ae4","subject":"Prevent share type deletion if linked to group types","status":"MERGED","created":"2020-04-15 14:24:41.000000000","updated":"2020-04-27 19:05:37.000000000","submitted":"2020-04-27 18:28:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720206-1588012087429-71cafbf4","meta_rev_id":"2220500333610f4e0b6f1e9770173683aaae47d9","_number":720206,"virtual_id_number":720206,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2020-04-22 23:09:08.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-04-15 15:07:34.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-04-23 05:24:11.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 18:28:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"recommended":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2020-04-27 14:29:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2020-04-27 16:06:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-27 16:48:36.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-27 16:48:36.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-15 15:07:34.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-04-23 05:24:11.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-27 14:29:55.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-04-27 16:06:40.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-04-27 16:48:36.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-04-27 18:28:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7769e0bf0ea50b2c26704d344ee452224bf6847a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-15 14:24:41.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"6fdce98de919f52c4da48b161ed2d442544548dd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-15 14:24:51.000000000","message":"Topic changed from bug/1699836-stable/stein to bug/1699836","accounts_in_message":[],"_revision_number":1},{"id":"d12017cbf3d7196fc09ac368b8fee0b1b313b962","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-15 14:45:03.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"59d5aab970e4ca633c33380886377edd161747d9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-15 14:45:18.000000000","message":"Topic changed from bug/1870751 to bug/1699836","accounts_in_message":[],"_revision_number":2},{"id":"0cc4eebfb3ff74d9c8fe163fc1aecb1c909b3c7b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-04-15 15:07:34.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/06/720206/2/check/dsvm-tempest-infinibox-nfs/e519367 : FAILURE in 22m 02s","accounts_in_message":[],"_revision_number":2},{"id":"bb4c74407ffe1aaa0c5f83c94431b87b98c33749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 16:21:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/857b0e496b88406f9b765410292ef1ec : SUCCESS in 8m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb8b6885482b4aa99316f6970583b2ff : SUCCESS in 16m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07f31ce1e38b42c29395bf95eced191c : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f96c96d52944c7ab34d4410365316ae : SUCCESS in 8m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64557e5127b24b86974fec6ca2713a8c : SUCCESS in 10m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b0cb78ca9284fb1b16e94998ffd5e6b : SUCCESS in 7m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1c8482689bd4419ac7c427e5387222f : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e6296dfb97a4bf68b3bf4e8f3f431a3 : SUCCESS in 10m 32s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/38d9bbbc7ace4536b7509204dbff234c : SUCCESS in 6m 48s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/0893666b5bf94e5e81721dae832a2c31 : FAILURE in 1h 31m 58s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1e872924d40641049c9f6c8bcc7fa5c8 : FAILURE in 23m 16s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/a77dacd6e4684715a2268079dda9e716 : FAILURE in 20m 44s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ef66f89d2e994040b71682b070223ffb : FAILURE in 20m 57s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/65b512f95d9c4a7fb97659f556b0c772 : FAILURE in 47m 07s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f01369936b0425aa990cfb502c29ea3 : FAILURE in 1h 30m 10s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/b06dcda9afcf411f93c1fdf3053b664d : SUCCESS in 31m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d3027b6031314a3b9949dc324d46c65a : FAILURE in 27m 20s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/2b537772368b4b84987bea3ac2f824d4 : SUCCESS in 48m 48s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/41d05c0f9d4144309e6e16fe8795902f : FAILURE in 28m 59s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2e6b45c46bed47a0ba06027ce5f33d2c : SUCCESS in 32m 34s (non-voting)\n- manila-rally-multibackend https://zuul.opendev.org/t/openstack/build/67ba44cbdcf34c468f5b81f80918e33a : FAILURE in 26m 34s (non-voting)\n- manila-rally-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/bc4c0f21589a434f893b960be4f25184 : FAILURE in 22m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/06359431fb2c4f07aed5da45897902fa : SUCCESS in 7m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8085f84bffa48b59671b9eaf2b8f6ee0c701c354","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-15 19:28:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"68c23b42fd290566f3fb7b5101a0c598431fbad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 21:28:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/930873b3e0484c60849c048746474903 : SUCCESS in 7m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b2c2fd9a5f34ff8b30e5c0380c993ce : SUCCESS in 17m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/408e993a4e324190a53c37ed0d8fe55f : FAILURE in 15m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b90a8b59683f43519f4e1a871d6fe43f : SUCCESS in 9m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6f4960dcca249b789657f1cc936e278 : SUCCESS in 11m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8161ded18c0d4d5f893e2c99fab446f1 : SUCCESS in 7m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84c0da68885b46189d294137d0414fcd : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffe12edbd3b648b392ed5d81f41ec9a5 : SUCCESS in 10m 48s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/33b30f902d884d6aba388eba4dc16b9e : SUCCESS in 7m 03s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/160ca2e7811c4c6bb3a437afc784c250 : FAILURE in 1h 39m 13s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/38825882ee844fb980e1293bd5bd876a : FAILURE in 22m 28s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/73b1883d4e0148cb9578f948438597d2 : FAILURE in 23m 22s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/03151ed88fd545c697029dd1d26de2e8 : FAILURE in 21m 32s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/c263e3e5c7224b8dbb5e71ae579d0041 : FAILURE in 44m 47s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80b69d6162584ba984ed8d78b2020160 : FAILURE in 1h 56m 32s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/fd95d982d8c04836ab3b595bb0b47462 : SUCCESS in 29m 29s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9598fb32637942659e8c5ed315bebf07 : FAILURE in 20m 37s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ea7a532b9273477699e24b320e248d03 : SUCCESS in 41m 53s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/960f14b868b9488dac1dfccdf1c77574 : SUCCESS in 1h 32m 13s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/77848c96328c482fbff2d72822f0804f : SUCCESS in 26m 07s (non-voting)\n- manila-rally-multibackend https://zuul.opendev.org/t/openstack/build/949bc151ac3d4f1c8edc5e064e8e69df : FAILURE in 19m 07s (non-voting)\n- manila-rally-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/3a26136b97144e35ac64f6c13331f898 : FAILURE in 19m 36s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/053217b957b54116ac865b7860328981 : SUCCESS in 6m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee15204950c9a2538b8f36ef20d810ae702f63dd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-22 23:09:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6fd02f053f4e5f1036e8efb1db1db06a50e6f735","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 01:20:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab2b0430c314277b7766fed5bb744df : SUCCESS in 6m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/445ffe79e8984d4d8fffc55e72c7540d : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64a156a20fab447bbfcd723a9cd916e3 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d563334a72a44156a63a48df30b3bcb7 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/363abbb9054749499a0e69265b00ef7a : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/148895fb21cf4837aa4fe3a824364823 : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed2df1a82f1b40c785546f64325521a3 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/030c3a684b984046996d70b1a6e23267 : SUCCESS in 8m 50s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/fa6f46ca28b24dc196a02086bd9eabac : SUCCESS in 5m 23s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/2a513b2ef9d04c92893163f18fdcaf24 : FAILURE in 11m 09s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/42f803cf2acb4293b246f39893d9f2a8 : FAILURE in 25m 36s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/1f333246944c4941991b99c53ea05099 : FAILURE in 24m 51s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/9db6c4bfda3a4690ae907fc1249b6eba : FAILURE in 25m 00s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/44dcfab2be7d4ac6a4050a4353332e8c : FAILURE in 38m 05s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e5666b799204d4da51b6829e0f50e4f : FAILURE in 1h 46m 04s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/7c3b854aeff541ff9319655b2bdcd578 : SUCCESS in 29m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/681709a27be44b75be1d9b23065a43c6 : FAILURE in 22m 32s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a65c9d6b2d194a61b659ed5cfd7c43e8 : SUCCESS in 34m 11s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/1b6f80451d104ec48e78d5e1bbffc0cf : SUCCESS in 1h 37m 21s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/af05787ae12a459391adf9f70b6946c9 : SUCCESS in 25m 12s (non-voting)\n- manila-rally-multibackend https://zuul.opendev.org/t/openstack/build/75b9295ecd5245fcafb3960703d6f91c : FAILURE in 16m 25s (non-voting)\n- manila-rally-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/27c1398fff2b439cb2278e11644cbfbc : FAILURE in 19m 43s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/51aadb0d6b244d0192efa400a04b0a2f : SUCCESS in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bfc40ddc8a39c24aa0ce36159c9670c85267ab22","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-23 05:24:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/06/720206/2/upstream-check/manila-cDOT-ss/9abe24e/ : SUCCESS in 1h 16m 41s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/06/720206/2/upstream-check/manila-cDOT-no-ss/bd77ecd/ : FAILURE in 5m 23s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f522fc93701301ccc6ecdf9f2f4bf462422be009","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-04-27 14:29:55.000000000","message":"Patch Set 2: Code-Review+1\n\nClean cherry-pick. LGTM :)","accounts_in_message":[],"_revision_number":2},{"id":"6d2514309a5a0e5c96c0815716a40ff7c25a593f","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-27 16:06:40.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM, thanks","accounts_in_message":[],"_revision_number":2},{"id":"e4aa307d4c96e37a2dd5564f41abb896eecf7351","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-04-27 16:48:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8e28f43bd0d386ba0a66a639399d0756e5f29ba9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 16:51:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d0ff50ce58a8768b2e2d96e61892a5350592f7a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 18:28:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5058a97771bf2171d01b68f3827c682e1aa7dd1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 18:28:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab14887301e84762b19c6e1c5fdafdbb : SUCCESS in 8m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fda8f9497e245a6b1e96f9c6935b02f : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5bfc5ccdc974057950fd8cd4a202230 : SUCCESS in 7m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52e89f2d018447ba9aaaf5780f153525 : SUCCESS in 10m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d485aa530594965aabc4c1a24165715 : SUCCESS in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd595534d8bf4d4cb1ec7f1839f71fe9 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9730f639ab6b4cc9b733c7f62225fc15 : SUCCESS in 11m 17s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/06b14eb74c9e45878d80628edade2b56 : SUCCESS in 9m 21s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/2be0417ad8334535911589b5f14b729e : SUCCESS in 44m 44s\n- manila-tempest-minimal-dsvm-lvm https://zuul.opendev.org/t/openstack/build/a81a8cbd6ae8410da20079000b0bcf44 : SUCCESS in 1h 29m 26s","accounts_in_message":[],"_revision_number":2},{"id":"36a85550481c46fc6e50be0244afbaec8a8e217b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 18:31:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aac19071170e449d8f1b0b1029e6cba1 : SUCCESS in 2m 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b74f17d70dd4411d83355921afbbbf3e : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2},{"id":"2220500333610f4e0b6f1e9770173683aaae47d9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-27 19:05:37.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit f08c8be4370e3ec23edb5dd7b7a76325e94dc650","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"30473c34810665dda1830d516951694ada09d198","revisions":{"2c8600a6f6b6dcac76d2c1c8bab71625683317e1":{"kind":"REWORK","_number":1,"created":"2020-04-15 14:24:41.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/720206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/06/720206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/720206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/720206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/720206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/06/720206/1"}}},"commit":{"parents":[{"commit":"df30ceb0ea200c75122e8cdfa78d66ef97e0e895","subject":"Increase MANILA_SERVICE_VM_FLAVOR_DISK","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/df30ceb0ea200c75122e8cdfa78d66ef97e0e895"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-04 23:57:22.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-15 14:24:41.000000000","tz":0},"subject":"Prevent share type deletion if linked to group types","message":"Prevent share type deletion if linked to group types\n\nShare types can be associated with share group\ntypes. This relationship needs to be dropped in\ncase share types are to be deleted, or we\u0027ll be\nleaving behind dead references in the database.\nShare groups cannot exist without share group\ntypes, and hence we don\u0027t need to check on existing\ngroups when deleting share types; checking for\nexisting group types will suffice.\n\nChange-Id: Id86a93f1923a493cefe73e2d90d5094005b42ae4\nCloses-Bug: #1699836\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 470bab7c712b41fb22d2fd48ae6e43505525db31)\n(cherry picked from commit 18430ceafe2d5b46d496cf6ef33d5f1825cb82b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2c8600a6f6b6dcac76d2c1c8bab71625683317e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2c8600a6f6b6dcac76d2c1c8bab71625683317e1"}]},"branch":"refs/heads/stable/stein"},"30473c34810665dda1830d516951694ada09d198":{"kind":"NO_CHANGE","_number":2,"created":"2020-04-15 14:45:03.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/06/720206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/06/720206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/720206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/720206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/720206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/06/720206/2"}}},"commit":{"parents":[{"commit":"df30ceb0ea200c75122e8cdfa78d66ef97e0e895","subject":"Increase MANILA_SERVICE_VM_FLAVOR_DISK","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/df30ceb0ea200c75122e8cdfa78d66ef97e0e895"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-04 23:57:22.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-15 14:44:13.000000000","tz":-420},"subject":"Prevent share type deletion if linked to group types","message":"Prevent share type deletion if linked to group types\n\nShare types can be associated with share group\ntypes. This relationship needs to be dropped in\ncase share types are to be deleted, or we\u0027ll be\nleaving behind dead references in the database.\nShare groups cannot exist without share group\ntypes, and hence we don\u0027t need to check on existing\ngroups when deleting share types; checking for\nexisting group types will suffice.\n\nChange-Id: Id86a93f1923a493cefe73e2d90d5094005b42ae4\nCloses-Bug: #1699836\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 470bab7c712b41fb22d2fd48ae6e43505525db31)\n(cherry picked from commit 18430ceafe2d5b46d496cf6ef33d5f1825cb82b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/30473c34810665dda1830d516951694ada09d198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/30473c34810665dda1830d516951694ada09d198"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
