)]}'
{"id":"openstack%2Fmanila~723656","triplet_id":"openstack%2Fmanila~stable%2Fqueens~Icb8947991723a2a1cc334a72276a35cf50fc1b7d","project":"openstack/manila","branch":"stable/queens","topic":"bug/1869712","hashtags":[],"change_id":"Icb8947991723a2a1cc334a72276a35cf50fc1b7d","subject":"Remove provisioned calculation on non thin provision backends","status":"MERGED","created":"2020-04-27 19:02:20.000000000","updated":"2020-05-02 01:56:23.000000000","submitted":"2020-05-02 01:54:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723656-1588384466614-39316be2","meta_rev_id":"0a412616f5f0d01e1897a17723bbfc1f226959ab","_number":723656,"virtual_id_number":723656,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"date":"2020-04-27 22:41:40.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-02 01:54:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2020-05-02 00:16:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-28 01:20:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2020-05-02 00:22:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-04-27 19:02:20.000000000","_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 19:02:20.000000000","updated_by":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"reviewer":{"_account_id":5575,"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","username":"jose-castro-leon"},"state":"REVIEWER"},{"updated":"2020-04-27 22:41:40.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-28 01:20:16.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-05-02 00:16:20.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-05-02 01:54:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed4a0d0c3a6e73228ec0a4c7de9c84975b4345da","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-27 19:02:20.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"7606eccf8ce99fe4826ae175b9199cb55fca891a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-27 19:05:54.000000000","message":"Topic changed from bug/1869712-stable/rocky-stable/queens to bug/1869712","accounts_in_message":[],"_revision_number":1},{"id":"8a04e51e5482dd915cc65e94ddc54cc2b20ac1c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 20:51:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d7ddfab4447a4a9e8308169514b0bfe0 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef367a983184444bbda74be1ad3400cf : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b7b8c3d4f6c0428783687803adfbf9d2 : SUCCESS in 9m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/22816bbf371c4a9b893933e1051b648f : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01c486ad81e84e3fa4abe5031cb9d5bc : SUCCESS in 11m 25s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6fc316fef5414e82a9141d842232c622 : SUCCESS in 7m 29s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/a8284ad982f041c3a2c81cbeba4b2e31 : SUCCESS in 1h 38m 30s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/f6dbec2480d74c15a2d71e24839ee129 : SUCCESS in 46m 45s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/0607d5bfc1f2494a9ab1e4adb2053da7 : FAILURE in 1h 17m 07s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/36ba563982674e6689e3437595af2b62 : SUCCESS in 54m 44s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/7fb3274314534621b0362275ef936044 : FAILURE in 43m 05s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27789e1b00454500af7e518ba6d4f163 : SUCCESS in 1h 08m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/c50b47f12cb84d428438d3a177117a63 : FAILURE in 8m 37s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/327d82ae7746482e83d40319616710a6 : FAILURE in 7m 54s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a07d46fd98d64dda9644851d83ef011a : SUCCESS in 45m 57s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/42485802f01a40358fd8a74de32f4290 : SUCCESS in 1h 39m 46s\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/603daf9df79e4b2083f5005701d05af1 : FAILURE in 8m 47s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/a615eec20863456682ac353e0abcfeca : FAILURE in 15m 54s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/41dd5b16eee843ecb02e5d716f1b60fc : FAILURE in 18m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d66319b8f6241908ec1571883ad54f7 : SUCCESS in 13m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a2203be5f0c54b299378e3506a179f28 : SUCCESS in 10m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5ee519ddf5de171f9cba1d7bf10e939b5b2d6ce","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-27 22:41:40.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/56/723656/1/upstream-check/manila-cDOT-ss/5fcbbbf/ : FAILURE in 1h 02m 34s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/56/723656/1/upstream-check/manila-cDOT-no-ss/4095198/ : FAILURE in 1h 26m 36s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e4f6c48f97ed3b101aa190dd95de4243556b8786","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-04-28 01:20:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c2bba9b2e2ed94f887bf0a77160b911ed62b176b","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-28 02:53:42.000000000","message":"Patch Set 1: Code-Review+1\n\nclean cherry-peak, LGTM, thanks","accounts_in_message":[],"_revision_number":1},{"id":"36d42f95bc8d6beaf5fd13051cba587e5f8a9058","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-28 18:05:23.000000000","message":"Patch Set 1: Workflow-1\n\nstable/rocky needs to merge: https://review.opendev.org/#/c/723615/2","accounts_in_message":[],"_revision_number":1},{"id":"0b145ce79af7750b99215e069245f64499e21307","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-02 00:04:24.000000000","message":"Patch Set 1: -Workflow\n\nready now","accounts_in_message":[],"_revision_number":1},{"id":"8cd7c287b32b90b509cb6d413297e954f2602ebf","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-05-02 00:16:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"afa454a5f1b494d38eead25c137745779d8baee7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-02 00:22:35.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"027056698b5733aae6f6a9f0c84cae042e4dd3e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 00:22:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ac71d8d24643e7810c82dfc7eff895a10a75e8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 01:54:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1223838b6da16ec1773e2df84c2a42f2dadb8d93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 01:54:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0b939926eb6641a38e826082de2ce582 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad559447104b4234893d37e5eedc7849 : SUCCESS in 6m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44d75cf774884cc4aa1adeeccee77c58 : SUCCESS in 6m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/07cae4aa21694e61910b2ecbfc418b41 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6284c6ed2784771aaaf01acdce8d580 : SUCCESS in 8m 47s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d6f6af8260a64a71b90b675be4448bae : SUCCESS in 4m 28s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/834866a592d5464a9d4acf5e521b942b : SUCCESS in 39m 06s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/3eaa6d04dedb420c8346de00837ac6c7 : SUCCESS in 1h 13m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0a412616f5f0d01e1897a17723bbfc1f226959ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 01:56:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05aa344d626240a793e8bf0885a06c58 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd50f6be5a5b1847354790e2fe1f426185e4abff","revisions":{"dd50f6be5a5b1847354790e2fe1f426185e4abff":{"kind":"REWORK","_number":1,"created":"2020-04-27 19:02:20.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/56/723656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/56/723656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/723656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/723656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/723656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/56/723656/1"}}},"commit":{"parents":[{"commit":"cd8c0045505275998cd083cd16e0f32335f8036a","subject":"Merge \"If only .pyc exist, the extension API will be disabled\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cd8c0045505275998cd083cd16e0f32335f8036a"}]}],"author":{"name":"Jose Castro Leon","email":"jose.castro.leon@cern.ch","date":"2020-04-07 09:24:00.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-27 19:02:20.000000000","tz":0},"subject":"Remove provisioned calculation on non thin provision backends","message":"Remove provisioned calculation on non thin provision backends\n\nOn those backends, there is no need to calculate provisioned_capacity_gb\nas it is not used during the scheduling. This calculation was not\nscaling properly on big environments as it implies many database\nqueries.\n\nCloses-Bug: #1869712\n\nChange-Id: Icb8947991723a2a1cc334a72276a35cf50fc1b7d\n(cherry picked from commit 2c30fed486320ec5ba88f21ae3df14b14da6eb35)\n(cherry picked from commit 89aa8939aa53e52254e4f98da0d80a76ef12b9e7)\n(cherry picked from commit 920cd8e6df8a9b553a77a3a029c9d282c73e1634)\n(cherry picked from commit 49deb1ed312ad9b89e573d84fedf31b0e7894dd2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dd50f6be5a5b1847354790e2fe1f426185e4abff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dd50f6be5a5b1847354790e2fe1f426185e4abff"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
