)]}'
{"id":"openstack%2Fmanila~724342","triplet_id":"openstack%2Fmanila~stable%2Frocky~I01a549a0e299ae0d0d9b603a9d0797f16f4b68b3","project":"openstack/manila","branch":"stable/rocky","topic":"bug/1858328","hashtags":[],"change_id":"I01a549a0e299ae0d0d9b603a9d0797f16f4b68b3","subject":"[NetApp] Fix share shrink error status","status":"MERGED","created":"2020-04-29 15:04:00.000000000","updated":"2020-05-01 14:28:19.000000000","submitted":"2020-04-30 20:36:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724342-1588278979834-56a7dbf5","meta_rev_id":"5e36eead9909c89808fa00f5683ba1259f91b401","_number":724342,"virtual_id_number":724342,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2020-04-30 17:21:29.000000000","_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},{"value":0,"date":"2020-04-29 18:12:16.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 20:36:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"recommended":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2020-04-30 19:19:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-04-30 19:17:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2020-04-30 19:19:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-04-29 15:04:00.000000000","_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-30 17:21:19.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 15:04:00.000000000","updated_by":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"reviewer":{"_account_id":31528,"name":"Daniel Tapia","email":"danielarthurt@gmail.com","username":"danielarthurt"},"state":"REVIEWER"},{"updated":"2020-04-29 18:12:16.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-30 17:21:19.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-04-30 19:19:10.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-04-30 19:19:39.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-04-30 20:36:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eda86e92d40e7ec797709e8815cd2d7052756e15","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-29 15:04:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"4536c1f02f554f61ba77abbc564cdbf1a00f0414","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-29 15:09:48.000000000","message":"Topic changed from bug/1858328-stable/rocky to bug/1858328","accounts_in_message":[],"_revision_number":1},{"id":"a7102271b07d7ccbb1e1c9575cc5c39992b0b50e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 16:42:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4b235f99739c4aa594743eed92a04f59 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1315fd735d54508817e7d984386170c : SUCCESS in 7m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d50738d00174141a7e92848a87f6e83 : SUCCESS in 7m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/36e5e43808564b398a76f615fba54205 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb89ef18d1344d559d5bb90220bc4afa : SUCCESS in 10m 52s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/7a5576ee0f204b27a6b08f150d27beff : SUCCESS in 7m 54s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/9bc86d8918604de39f66d41bdfd0bec8 : FAILURE in 1h 32m 40s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/70aebb6642aa454cae50399ebe627f5e : SUCCESS in 48m 40s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/d084b07621cc4437bf4a6a5a482e12b2 : SUCCESS in 1h 18m 19s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/6988901517694944974c23c5ae6f3001 : SUCCESS in 53m 20s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/d3911178fae54ec590b1751f018941fa : FAILURE in 54m 19s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e4a943f4cf94dfa946f210b1f661dc6 : FAILURE in 1h 35m 15s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/2bc9ac66e46e4fe6bbaacd9d512b2288 : FAILURE in 35m 24s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/760a4c779b7e427ead31b0194672b248 : FAILURE in 21m 52s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e73d87b29d2748f29bcabe48d7b61620 : SUCCESS in 48m 26s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/b1c709f2b2ec49c6bbbf3153de6b28da : SUCCESS in 1h 11m 27s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0af012dce3ed471f8fc347a17d37e208 : SUCCESS in 25m 37s (non-voting)\n- manila-rally-multibackend https://zuul.opendev.org/t/openstack/build/222f31c9131142e8817c6e6efa71fec5 : FAILURE in 24m 41s (non-voting)\n- manila-rally-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/6bfe6e097da14435ba993faa6133a649 : FAILURE in 25m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49c676480e1a487fb191b90ad572dffb : SUCCESS in 9m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5013c8e6b27c4751994c7e0784a58580 : SUCCESS in 16m 18s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6a326b517cc04dd2bc09ce8aab89d7da : SUCCESS in 33m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7827e4a712cf322d422fa25119b51287feac1449","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-04-29 18:12:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/42/724342/1/upstream-check/manila-cDOT-ss/94691f5/ : SUCCESS in 1h 10m 46s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/42/724342/1/upstream-check/manila-cDOT-no-ss/7c1ea8f/ : SUCCESS in 1h 29m 00s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"1cd4a7dd652a14d109467b758fc2c3a21ffeaa04","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-04-30 17:21:29.000000000","message":"Patch Set 1:\n\nThis one is good to go!","accounts_in_message":[],"_revision_number":1},{"id":"3eccb18ebce8e1aa1419673ac9af8cbac0c6f1ce","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-30 19:17:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4fe63225b3e54ecb10e7307c705254c09ac3e11f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-04-30 19:19:10.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM. Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"3f7ca81a72693751748c61d1b7112ebc9cbb3d95","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-04-30 19:19:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"14e2ce68e6f60e6a77362c60ccbddcc73522f851","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 19:19:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cee1316ce6c688492f148c7438e5e6528a672b5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 20:36:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d08b6abe5e850646fc701186d281e7821d27eb62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 20:36:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3e328c2d2feb438ca8210f3fbba783b5 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e21c3daa271d423f922f23becf5ae480 : SUCCESS in 6m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2df59f7881504048a6d8a333864c9e50 : SUCCESS in 8m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2175fe97d6de4ebab422f0ff97aa8cfb : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e87f995fbc0a4c0083d526fb793e1d39 : SUCCESS in 8m 11s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e8d6daf57a8a402db2e8205160f43694 : SUCCESS in 5m 21s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ce1f5a4267524a1b829a2fe15d8a192f : SUCCESS in 41m 47s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/a893fce007fe4a84bf47e2db16a38926 : SUCCESS in 1h 15m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5068502105d4b8fbdfbed90f1304746 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0dc86106003dfd22657d29987adda937eac8f2cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 20:38:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af15ca76d0e440b48f1191bd5c42fa5b : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5e36eead9909c89808fa00f5683ba1259f91b401","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-05-01 14:28:19.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit d5a16f96a4160f55e43d93fa9f225ddef395d1d1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"421cd52014a860815dd4df06cc0100a59a6d2f81","revisions":{"421cd52014a860815dd4df06cc0100a59a6d2f81":{"kind":"REWORK","_number":1,"created":"2020-04-29 15:04:00.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/42/724342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/42/724342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/724342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/724342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/724342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/42/724342/1"}}},"commit":{"parents":[{"commit":"b99b57825e9b140e3d6c163a810e7cd49adb6e3d","subject":"fix bug in quota checking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b99b57825e9b140e3d6c163a810e7cd49adb6e3d"}]}],"author":{"name":"danielarthurt","email":"danielarthurt@gmail.com","date":"2020-02-14 13:28:19.000000000","tz":0},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-04-29 15:04:00.000000000","tz":0},"subject":"[NetApp] Fix share shrink error status","message":"[NetApp] Fix share shrink error status\n\nThis patch fixes the error raised by NetApp driver when the user try\nto shrink a share to a size smaller than the current used space. Now\nthe driver raises \"shrinking_possible_data_loss\" instead of\n\"shrinking_error\".\n\nChange-Id: I01a549a0e299ae0d0d9b603a9d0797f16f4b68b3\nCloses-Bug: #1858328\n(cherry picked from commit c169902d41ef2b70ce1c1181408bc0916a7ea319)\n(cherry picked from commit 97a913209825f975ccfc2d76c91f4215d02622ad)\n(cherry picked from commit d7853ae6ed258077eab90fb29055a055e085a5fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/421cd52014a860815dd4df06cc0100a59a6d2f81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/421cd52014a860815dd4df06cc0100a59a6d2f81"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
