)]}'
{"id":"openstack%2Fmanila~724894","triplet_id":"openstack%2Fmanila~stable%2Fqueens~Id135a96f2d1b7306156bab0ddedcb1b838a74f0a","project":"openstack/manila","branch":"stable/queens","topic":"bug/1870751","hashtags":[],"change_id":"Id135a96f2d1b7306156bab0ddedcb1b838a74f0a","subject":"Delete type access list when deleting types","status":"MERGED","created":"2020-05-01 16:56:56.000000000","updated":"2020-05-06 22:55:56.000000000","submitted":"2020-05-06 22:54:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724894-1588805673650-d95910ec","meta_rev_id":"8e0796b4d0603e582eb342b650949348f8616e0e","_number":724894,"virtual_id_number":724894,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2020-05-04 23:21:21.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-05-01 17:34:08.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-05-05 02:17:57.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-06 22:54:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-05-06 20:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2020-05-05 21:44:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2020-05-06 20:52:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-05 21:38:11.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-01 17:34:08.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-05-05 02:17:57.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-05 21:38:11.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-05-05 21:44:41.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-05-06 20:52:44.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-05-06 22:54:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"33644c2ff2c7961bd5f68512b9deda59bfa7c46d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-01 16:56:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1320f0126beab393fea42dd8d9c8b0f9b53389e2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-05-01 17:34:08.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/94/724894/1/check/dsvm-tempest-infinibox-nfs/4c6f468 : FAILURE in 22m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ed33c55f8ceb7b8300e258e471e82aee29037f50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 18:23:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4be9423bebb648b393af1bc4998138f4 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51d863ab331c4fd089bbf3bed00c70ad : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c10b7b4abdf40bc952e86a3c9fbdd0b : FAILURE in 15m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/046ac2beeb3742918efdab5c66190298 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ac20a4f551f456ca72ecc246e70f23b : SUCCESS in 10m 43s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f356c4ddc0134a76bcdb303841f9a4d3 : SUCCESS in 5m 34s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/e2e01f824efc4405be2755f4e57d1bc4 : FAILURE in 1h 16m 59s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1f0ec7608b144b8a998bf0abae0a93f4 : SUCCESS in 41m 45s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/e25139f8a8ce48a5b28963cc877a709b : SUCCESS in 1h 00m 50s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/83132052b84b44cd988d4f20901de410 : SUCCESS in 1h 00m 07s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/ea8506cac0fa41f584b066be2d1ad979 : FAILURE in 48m 01s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f64dd71d0f3f44599d2813fcd622cfc2 : SUCCESS in 1h 09m 37s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/e97a619abe06430bbc0e1a7dcdd33d9b : FAILURE in 7m 41s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/66a392baf5cc4a44be81267a2701baac : FAILURE in 6m 21s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/2417060acd454b3db036aca849a1921d : SUCCESS in 50m 26s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/38995fe3d0a74c4aa23b5cd42eae3fe4 : SUCCESS in 1h 17m 56s\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/21efa9286a334bbab3ea24b8c596d1fa : SUCCESS in 31m 50s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/c1ef22451dae4ca0aee38e2424d2d0f2 : FAILURE in 18m 05s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/30673110d6ab4da8aeb28dc2ec060339 : FAILURE in 16m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9d48b2b4d26494aaa5429f76741e5e7 : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e90af6d93d224a10a5a6eb96632691c0 : SUCCESS in 10m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2c0cf8e8c96242e00f336f96b84b0d0b8742d0b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-02 00:05:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c28a787d9bb8e168cd2a97746218750f9e689951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 01:33:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8f6580603f474ce792c111beba0743f2 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/729288840ff54a819510fcace14c5442 : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b108aa202fca46d89d1f752aab14bc3f : SUCCESS in 8m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4da0dee4f4664d62bb66e2a95f4fbdc1 : FAILURE in 13m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c79da8b40d234ac5b73131d52cb83812 : SUCCESS in 9m 29s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/93ce8e8edb414d13b383dd16da853195 : SUCCESS in 5m 04s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/9c4f6d3155894c6ba206541deff12931 : SUCCESS in 1h 26m 35s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/179dc72dcc3142b787a881674bc5b1aa : SUCCESS in 44m 16s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/ec92c254a42649fcaeea4c5469494cb6 : SUCCESS in 1h 02m 56s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/4303ae7a4b6c439a817ab72d79ef8eb9 : SUCCESS in 53m 32s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/8c44c98ae6644fdbbf0d85ac4cc06f4d : FAILURE in 42m 58s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59469dcb5a6d48e3930880d92dde9ed8 : SUCCESS in 1h 14m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/733ceb1e70804b4c83b3ff8ed0213ed4 : FAILURE in 5m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/6a14101899e0448284f0d5e110ba5dec : FAILURE in 5m 47s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/8b60dc30ea8f4208ac730c2b84e5b9c0 : SUCCESS in 35m 22s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/f0a6aa2f493445e990b2afd3dc22e0e8 : SUCCESS in 1h 16m 28s\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/f00150d0d77e407380342e047654eb0d : SUCCESS in 26m 29s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/8836905b89f149289991e0464a8852c9 : FAILURE in 16m 52s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/5285067881d34009a6660a5f16a9c50f : FAILURE in 19m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/055164a6a7e5417aa0c82db0217f599d : SUCCESS in 13m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/13daeb2001e54293add64339d7acdd75 : SUCCESS in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"177ed0dfd10d3c0b59c50f80e8e2f4e2f14b6091","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-02 01:53:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b0a7f7edb1e7b4ec82e91996b5f760d02238536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 03:46:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/589691f9f2874e418fed27a4f9ff61b0 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67328dd53f9e4916aa94a3d2ba3978cb : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ecc811b1fb724c7ead72ccd1e93a473a : FAILURE in 13m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9285b0ee483a4324a2bd009b796cf4d4 : FAILURE in 13m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54d539b3fc4c4c20aa70f9d48e6b00dd : SUCCESS in 10m 10s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a90029e8082d43bd8c16ab4d52e3ef3d : SUCCESS in 5m 32s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/37b34af169064554937fd3322003a3bb : SUCCESS in 1h 26m 10s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/cf7ab7e351b64a7b9f633603a833629c : SUCCESS in 42m 27s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/3cead1d6e5a74c18b32c336a709477c4 : FAILURE in 1h 10m 56s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ac97622a66864cf9b95572a2e1cbbe2a : SUCCESS in 45m 03s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e9979a49fae146189edee4941e630471 : FAILURE in 53m 49s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e32a892166c24be8b29c71b2e0accfa4 : FAILURE in 1h 26m 50s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/284db0e6bdd54bcfbbb4dda65630e2fa : FAILURE in 8m 05s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/704ee8d19380429793a2b1f35221e79a : FAILURE in 8m 57s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3b18df4862de44688090c2448f48a846 : SUCCESS in 43m 52s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/4ee3d410e5f14dc89e671ef07295be6f : FAILURE in 1h 22m 56s\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/84c29d8acb304a1a9733929ad046b69b : SUCCESS in 30m 07s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/14da0c420cac48198bc2fc22c19f36fc : FAILURE in 25m 59s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/687fb5d7d59e4d83be7287e14a77fc06 : FAILURE in 25m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88e058de1f7b49f3896ba43b003758c2 : FAILURE in 18m 55s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2a9c8e06c734477c958e781ea49f05d0 : SUCCESS in 10m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53d4327364a1bf62c54cbde9f0eecc8ac89656ee","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-04 17:29:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f3cdc9c586b61a1a180703ab180dd4d7b5428e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 19:05:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9ac3008ef3c6421e97acf0d5cce7dfb2 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db7930c59fe24dd688f17866f99efa77 : SUCCESS in 8m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/393f59a52e1a44c9a0c105332ae0de94 : SUCCESS in 7m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/744ceca2735b4a7aa7c93c0b49bc4637 : FAILURE in 14m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25f47f34c2074ad58c3c5fac1a080fb6 : SUCCESS in 10m 22s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9687a5bf59df45858db4364b0cc19205 : SUCCESS in 8m 29s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/0c095064ac514ff7934097d8e98c9b82 : SUCCESS in 1h 19m 01s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/b1c8cea116434af0bdbe0f46a9221bef : SUCCESS in 34m 48s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/e8fa7a28c7354228a2a29e59e0259955 : SUCCESS in 1h 06m 18s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/6e2fd7684dec466bb7633c381f28c0fd : SUCCESS in 1h 01m 12s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/3a62049bfbbb41279a345fa34c203114 : FAILURE in 48m 05s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27b7fd50a03b46de9641799306704f8c : FAILURE in 1h 13m 17s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/d0a0f31316b14186bfe91cd0886d5fb8 : FAILURE in 7m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/a7f3498626c349d4a8611cdff6865138 : FAILURE in 7m 00s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e238e1ff12ef4dc48e072525e8b28085 : SUCCESS in 43m 18s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/7a06c58e07d641cb8610bf8ba1cabfb4 : SUCCESS in 1h 18m 57s\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/9581cf44ce164772a25f5fcbc6e46999 : SUCCESS in 26m 06s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/2c615471b2234a768a4b6db5e5950720 : FAILURE in 21m 03s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/1a1d9d56d59a43a8bb19c2a564490cf8 : FAILURE in 18m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c27517dd1ff4e6fbf0ae1d635d3c415 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4776690841e7450f9f7272515eca9034 : SUCCESS in 10m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c2ada937879deeaf8dbd045950e51ca90b553d8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-04 21:00:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b4750dbf7a6e4eaef3843fba9bee33d48d81732f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 22:36:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7da6750814594435bf48cdd9382faddf : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92ff3563529d4919998a2783a6fc771b : SUCCESS in 9m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66202e58d7464d2890aa7ab98f4f56c2 : FAILURE in 14m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/85dadc597ea34e54aef67eb417305a24 : SUCCESS in 13m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/588fa90850c04341ad720c721bb2fc08 : SUCCESS in 11m 12s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1bcf9215ee864fac92a51091bba1c42e : SUCCESS in 8m 41s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/fd5491dcbbad43d289d6f6ccffde22e2 : SUCCESS in 1h 27m 02s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/0a6cce07d5a44cdcb344c4d905cd7cb1 : SUCCESS in 39m 01s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/742f8184b4b3424a9862b3a089c39200 : FAILURE in 1h 13m 20s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/85c6759f27ac4861b93aa442078304c1 : SUCCESS in 57m 13s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e491a466a9994e19b43f1fe0dcf2d20d : FAILURE in 44m 24s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aea6218c95f14dccb5b615ecdf556ba3 : FAILURE in 1h 33m 24s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/a20ce1c5dafe4d0f94d873a66667c77d : FAILURE in 7m 12s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/f36f8efd894d45abae9e8f17973eb716 : FAILURE in 9m 34s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f3b9993ad8684b369eadb11b262fb1e5 : SUCCESS in 41m 53s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/eb5b326af2df4e37949bd4ecfb58bcf8 : FAILURE in 1h 19m 27s\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/39b69258f2cd48e8af05c12e75401de0 : SUCCESS in 25m 19s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/4bff5b809d8442aa97e61181298886dc : FAILURE in 24m 40s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/314626261e724211bcd7e9d51d8b94e4 : FAILURE in 24m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbc771d3527d421e9311a1704136d85b : SUCCESS in 18m 47s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9bd1db3634ae4d02936e19fd6bd4c6f2 : SUCCESS in 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4394dd9eff4612e7bb2a27d127e695749d8290b9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-04 23:21:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9f31a55f3c0061e29557511a3af95db3f265bf2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 00:47:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d7b155f682044223875940c9e96293c5 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b69cffb3e80f4ee9aef0c93401df5220 : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f92d0972f0a9423599aae63221a2a157 : SUCCESS in 6m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f102c708fe5c491483d75f065daab91c : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ba0e3e789214a3d95181cbd69a9065d : SUCCESS in 9m 13s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cfcfb816fad5444885cce7bdd10ac674 : SUCCESS in 5m 36s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/a62e9bf9a4cb48e091b08a7a2c682160 : FAILURE in 1h 20m 28s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/3cf28d70f7b742c2a390e9c449ff3a04 : SUCCESS in 39m 00s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/fdbdacb7d6864a4da9341ef867ab20e5 : FAILURE in 1h 06m 35s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/cf47a70cf87f4c51b8f63a835a688058 : SUCCESS in 48m 32s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/47ddacfe3b2a4adf8c98e10b3b4175cc : FAILURE in 48m 07s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ffd660b99074d488b665fc33809ba20 : SUCCESS in 1h 06m 36s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/50600b5cc3474507bd2403bce7fb249a : FAILURE in 6m 05s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/58e0662847c74bac96d665f3839f9199 : FAILURE in 4m 54s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/50dcfbd4d7094712b86d32270fbaf5c9 : SUCCESS in 38m 49s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/b4a75946f4e74189ad78051e929402c6 : SUCCESS in 1h 18m 25s\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/d1838ee395bb4dc9ac2573983cf7c5a4 : SUCCESS in 25m 07s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/15284a6a7bbd45a99dcfb973d0b1a63b : FAILURE in 15m 44s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/6fbce814bd1e45b7ac3a2d4e9a373dc6 : FAILURE in 22m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9373357d6524daab1915db10b807772 : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/63a12c72c5b74e34b375afd379d24d49 : SUCCESS in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11c1f8f329adfbcaa6ec18cf8b1bf79b143012b3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-05 02:17:57.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/94/724894/1/upstream-check/manila-cDOT-ss/eb7c581/ : SUCCESS in 1h 08m 21s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/94/724894/1/upstream-check/manila-cDOT-no-ss/483d42a/ : SUCCESS in 1h 29m 54s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"11dd0a1ee7ca7ae7753546c3ec06c8aa55476aa7","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-05-05 21:44:41.000000000","message":"Patch Set 1: Code-Review+2\n\nClean cherry-pick, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"16cbb690df209ea2ee173f921db90c8ac6db5d16","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-05-06 20:52:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM. Thanks, Goutham!","accounts_in_message":[],"_revision_number":1},{"id":"5cb1bb1cf6f1a940ca4d15b4ae1daf043915cd6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 20:52:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a4fa45659a7abde124ebda650aebbf1caf02e2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 22:54:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b010d02a913d4368b4809b30ea033f9b : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ff5a31605844293bd20123a4b961328 : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f53570ce5d0a4b41af2411705d10233e : SUCCESS in 6m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d32c7d815ef540e7b505ce2174574e14 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed167f72bbe94ebbabfb11eb9ec282bd : SUCCESS in 9m 26s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/63829c43e39e4808bbe73932fbcaa66b : SUCCESS in 5m 25s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/c4e492fecd2f4b2da93a2359c9de7aa7 : SUCCESS in 37m 29s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/6da28c110b954042a6189140bbf5e1ce : SUCCESS in 1h 12m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0097e24fe8b058c4fa6d63cb6282361143873cde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 22:54:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3fd78a76ef9d3e1ab2a2b11ca032c8a772138a13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 22:55:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29c9965d6e6047dab3000f3e1225d7c6 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cae5ea83effd0bf8f930b6fdf37cb60f6a4c60ca","revisions":{"cae5ea83effd0bf8f930b6fdf37cb60f6a4c60ca":{"kind":"REWORK","_number":1,"created":"2020-05-01 16:56:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/94/724894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/724894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/724894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/724894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/724894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/724894/1"}}},"commit":{"parents":[{"commit":"1f91b438d710547b9b8864c9d58695643f434c7e","subject":"Prevent share type deletion if linked to group types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1f91b438d710547b9b8864c9d58695643f434c7e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-04-06 15:52:22.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-01 16:56:21.000000000","tz":-420},"subject":"Delete type access list when deleting types","message":"Delete type access list when deleting types\n\nCurrently we\u0027re in a foobar situation with two database\nmodels ShareTypeProjects and ShareGroupTypeProjects\nwhere we keep stuff around even when the concerned\nShareType or ShareGroupType objects are deleted.\n\nSo, delete type access rules when types are being deleted.\n\nChange-Id: Id135a96f2d1b7306156bab0ddedcb1b838a74f0a\nCloses-Bug: #1870751\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 5f80a24ff2a9e68a34e9257a1e533b5dfc8c65da)\n(cherry picked from commit da38cf6194fe748edc82bcbc51bd7e8127fb15c4)\n(cherry picked from commit 995b8c3150e00ba514d86f67388bdb43637c8a15)\n(cherry picked from commit d01708ef7a500929a2f92ea99b2ba14cf789a30f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cae5ea83effd0bf8f930b6fdf37cb60f6a4c60ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cae5ea83effd0bf8f930b6fdf37cb60f6a4c60ca"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
