)]}'
{"id":"openstack%2Fmanila~726348","triplet_id":"openstack%2Fmanila~master~I60617764313ee8ebefc71c72730eefaab64253b0","project":"openstack/manila","branch":"master","topic":"bug/1863021","hashtags":[],"change_id":"I60617764313ee8ebefc71c72730eefaab64253b0","subject":"Monkey patch original current_thread _active","status":"ABANDONED","created":"2020-05-08 11:21:47.000000000","updated":"2020-05-15 22:12:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d96b5dc68ff5356e4991ab0180942c754f661dcc","_number":726348,"virtual_id_number":726348,"owner":{"_account_id":24501,"name":"sue","display_name":"suzhengwei","email":"sugar-2008@163.com","username":"suzhengwei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"date":"2020-05-08 13:29:51.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-08 13:08:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"date":"2020-05-08 15:10:52.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"date":"2020-05-08 14:12:26.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"date":"2020-05-08 13:31:21.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2020-05-09 10:06:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2020-05-09 10:06:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"date":"2020-05-09 10:06:22.000000000","_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"date":"2020-05-09 10:06:22.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-05-09 10:06:22.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-09 10:06:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"date":"2020-05-09 10:06:22.000000000","_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"date":"2020-05-09 10:06:22.000000000","_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},{"date":"2020-05-09 10:06:22.000000000","_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 13:08:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-08 13:29:51.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-05-08 13:31:21.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-08 14:12:26.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-08 15:10:52.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-09 10:06:22.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-05-09 10:06:22.000000000","updated_by":{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},"reviewer":{"_account_id":6491,"name":"xing-yang","email":"xingyang105@gmail.com","username":"xing-yang"},"state":"REVIEWER"},{"updated":"2020-05-09 10:06:22.000000000","updated_by":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"reviewer":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"state":"REVIEWER"},{"updated":"2020-05-09 10:06:22.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-05-09 10:06:22.000000000","updated_by":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"reviewer":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"state":"REVIEWER"},{"updated":"2020-05-09 10:06:22.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-05-09 10:06:22.000000000","updated_by":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"reviewer":{"_account_id":26968,"name":"Amit O","email":"oamit.work@gmail.com","username":"amito"},"state":"REVIEWER"},{"updated":"2020-05-09 10:06:22.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-05-09 10:06:22.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"85a610f900858ee378f84423f750236089cb4b05","author":{"_account_id":24501,"name":"sue","display_name":"suzhengwei","email":"sugar-2008@163.com","username":"suzhengwei"},"date":"2020-05-08 11:21:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bee1115230fb18a4e34f4efeaacd1c4ae78420b","author":{"_account_id":24501,"name":"sue","display_name":"suzhengwei","email":"sugar-2008@163.com","username":"suzhengwei"},"date":"2020-05-08 11:22:42.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1edeb06523cf82b6cda8526c491080694f952a98","author":{"_account_id":24501,"name":"sue","display_name":"suzhengwei","email":"sugar-2008@163.com","username":"suzhengwei"},"date":"2020-05-08 11:23:32.000000000","message":"Topic set to bug/1863021","accounts_in_message":[],"_revision_number":2},{"id":"8b39226b80f968f7e840a9aff489f499c026285e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 13:08:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb904e35cff4da6bcaa67239e7bc41a : SUCCESS in 12m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f14d18d771f49d386556fe5a9b378c6 : SUCCESS in 18m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ab8e5ed110d46d699d3d37dbd888c41 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc68a4fb73f1486d808acdf1ff975e9c : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06c171752dc54d139f538f89480e8edf : SUCCESS in 11m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e951794fccab48f796d11adc902b6c1d : SUCCESS in 12m 09s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/89eabf840adf47e188c143e3992b8b88 : SUCCESS in 5m 10s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/97f396fd0f1f4e55a6378a61618e92cd : FAILURE in 1h 23m 27s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/ab5c62227d434f658223aa770666810f : SUCCESS in 55m 19s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/64cc93abe00b4dd3992a5708af65cbec : FAILURE in 1h 03m 47s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/fd447c31835949eebc244f01e54f5699 : SUCCESS in 54m 35s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/9abb7b5d408a43d5bf2da5c8cb8b9eed : FAILURE in 42m 30s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98405e574abd4189a16110a4b13a287b : FAILURE in 1h 38m 23s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/cbe893b8e08f44c0bb85a4ca98cfcb64 : SUCCESS in 31m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/94fecc3ec482403587cfdb6f7f72e545 : SUCCESS in 1h 15m 58s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/de22bcf9e88f47aeb052ef7df29559f3 : SUCCESS in 1h 09m 07s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/65847ad8b7634824800a38a7d222edfb : FAILURE in 1h 11m 24s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/abe4f5a9739c424bb1da1ddbd1af8fb5 : SUCCESS in 49m 39s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/8b7cee5a8d454ed5968a20509cf3349f : SUCCESS in 1h 07m 47s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/39b51a51b6094c6eb9e430450ab34352 : SUCCESS in 28m 25s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/36d8b33f21ed476abe42d06b4a81fcb4 : SUCCESS in 42m 46s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1f6851d0e96a424ab85e67c0ea7de186 : SUCCESS in 45m 22s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/77e776e62cdc41cbb711eea7087f7617 : FAILURE in 8m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a898895b128ca02751c032d0a03ea6bd673c884","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-05-08 13:29:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/48/726348/2/check/dsvm-tempest-infinibox-nfs/172de21 : SUCCESS in 1h 16m 15s","accounts_in_message":[],"_revision_number":2},{"id":"85dd468a38e19ac6296b42e4fe2f5d922f631af4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-05-08 13:31:21.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-48-726348-2/ns5_manila : SUCCESS in 2h 8m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4fd8cc730db8458e1091b976cf40923b043a9875","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-05-08 14:00:29.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-48-726348-1/ns5_manila : SUCCESS in 2h 38m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ca54aa268ca782d8b7041fb07b861f20769f9bdf","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-05-08 14:12:26.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/48/726348/2/check/DellEMC_VNX_Manila/8ac3022/DellEMC_VNX_Manila/236 : SUCCESS in 1h 03m 23s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"47b21114fd1ae781da2b057ef12dbdc3b3915b1b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-05-08 15:10:52.000000000","message":"Patch Set 2:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-48-726348-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"d96b5dc68ff5356e4991ab0180942c754f661dcc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-05-15 22:12:01.000000000","message":"Abandoned\n\nThis has been proposed here, thank you: https://review.opendev.org/#/c/725393/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e0613c61c7d8b48041863948b58cf29fc2e980fc","revisions":{"f9bbb8f71f002de3bf2e39c73550bc9fea5d354e":{"kind":"REWORK","_number":1,"created":"2020-05-08 11:21:47.000000000","uploader":{"_account_id":24501,"name":"sue","display_name":"suzhengwei","email":"sugar-2008@163.com","username":"suzhengwei"},"ref":"refs/changes/48/726348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/48/726348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/48/726348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/48/726348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/48/726348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/48/726348/1"}}},"commit":{"parents":[{"commit":"79784bac4219c2d2fa3a5d0e8b2ba867aac0e672","subject":"Merge \"[devstack][ci] Set public network ID in tempest.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/79784bac4219c2d2fa3a5d0e8b2ba867aac0e672"}]}],"author":{"name":"suzhengwei","email":"sugar-2008@163.com","date":"2020-05-08 11:19:42.000000000","tz":480},"committer":{"name":"suzhengwei","email":"sugar-2008@163.com","date":"2020-05-08 11:21:41.000000000","tz":480},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nChange-Id: I60617764313ee8ebefc71c72730eefaab64253b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f9bbb8f71f002de3bf2e39c73550bc9fea5d354e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f9bbb8f71f002de3bf2e39c73550bc9fea5d354e"}]},"branch":"refs/heads/master"},"e0613c61c7d8b48041863948b58cf29fc2e980fc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-08 11:22:42.000000000","uploader":{"_account_id":24501,"name":"sue","display_name":"suzhengwei","email":"sugar-2008@163.com","username":"suzhengwei"},"ref":"refs/changes/48/726348/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/48/726348/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/48/726348/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/48/726348/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/48/726348/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/48/726348/2"}}},"commit":{"parents":[{"commit":"79784bac4219c2d2fa3a5d0e8b2ba867aac0e672","subject":"Merge \"[devstack][ci] Set public network ID in tempest.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/79784bac4219c2d2fa3a5d0e8b2ba867aac0e672"}]}],"author":{"name":"suzhengwei","email":"sugar-2008@163.com","date":"2020-05-08 11:19:42.000000000","tz":480},"committer":{"name":"suzhengwei","email":"sugar-2008@163.com","date":"2020-05-08 11:22:38.000000000","tz":0},"subject":"Monkey patch original current_thread _active","message":"Monkey patch original current_thread _active\n\nMonkey patch the original current_thread to use the up-to-date _active\nglobal variable. This solution is based on that documented at:\nhttps://github.com/eventlet/eventlet/issues/592\n\nCloses-Bug: #1863021\nChange-Id: I60617764313ee8ebefc71c72730eefaab64253b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e0613c61c7d8b48041863948b58cf29fc2e980fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e0613c61c7d8b48041863948b58cf29fc2e980fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
