)]}'
{"id":"openstack%2Fmanila~731504","triplet_id":"openstack%2Fmanila~master~I32307d5349c8ead5c0f7846f245d49f36dc3bf19","project":"openstack/manila","branch":"master","topic":"bug/1881112","hashtags":[],"change_id":"I32307d5349c8ead5c0f7846f245d49f36dc3bf19","subject":"add manila-manage service cleanup","status":"MERGED","created":"2020-05-28 11:48:17.000000000","updated":"2020-06-18 19:18:33.000000000","submitted":"2020-06-18 19:15:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731504-1592507748821-d8d80f32","meta_rev_id":"833b69793c682f9d2548bf60b17e538b97305a86","_number":731504,"virtual_id_number":731504,"owner":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-18 11:32:05.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-18 10:34:59.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-18 11:13:14.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-06-18 09:18:42.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-06-18 10:13:55.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-18 10:25:11.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 19:15:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-18 17:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-06-18 17:50:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-18 17:54:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 09:18:42.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-06-18 10:13:55.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-18 10:25:11.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-18 10:34:59.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-18 11:13:14.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-18 11:32:05.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-18 17:50:28.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-06-18 17:54:29.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-06-18 19:15:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cea73516ef0d5d6bf00887400c558f139ecf021e","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2020-05-28 11:48:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4467381cc578e8972f86b3c68ec3c117a1273b2d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-05-28 13:05:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/04/731504/1/check/dsvm-tempest-infinibox-nfs/c234859 : SUCCESS in 1h 17m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3719bd21b00f814674f7a448cbfc10391db5c68f","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-05-28 13:20:28.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/04/731504/1/check/DellEMC_PowerMAX_Manila/5422e2e/DellEMC_PowerMAX_Manila/35 : SUCCESS in 1h 32m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"9cae4091526dfb77e630e3a0f3a4480b006f197f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:36:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe919eea9a14196a8eafab214587077 : SUCCESS in 10m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2d8fa45204a4ae4aad406a7a2dd4abd : FAILURE in 19m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/982e3bfc30524e4a8fcc6a325501193d : SUCCESS in 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f0bd4215c88467da2c1fb99f657c5a6 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c94b3ce38504050bf9cdee5bbd65131 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cd045c294ad4e968d24a132d3b2fda0 : SUCCESS in 7m 57s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5067d99889d54b4796cc8c511662bdcd : SUCCESS in 5m 42s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/84c8cf2c7e0648f487990e8e7175b741 : FAILURE in 1h 07m 11s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/9f8290614a4f415faa3bee0f185d1ab7 : SUCCESS in 46m 28s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/099df94a9f7f4a78a9d4b3b54173868e : SUCCESS in 1h 12m 18s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/6f94b104f7434be7917fb24153221b80 : SUCCESS in 51m 01s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/cf29e2b8170f48a8a12f2ce2faf1075d : FAILURE in 31m 01s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a3e8c766099406a8ac05c11ca506007 : SUCCESS in 1h 41m 16s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/0e21aee83ea144b5b026e208ae68cdc3 : SUCCESS in 31m 25s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c79e658fefae4186bfaee7ba2dc797e3 : SUCCESS in 49m 58s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/131dc05a53174f1e9a7cc51514a17915 : SUCCESS in 1h 06m 42s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/3f5f9607aeb048ccb2a058d1360957a9 : FAILURE in 1h 25m 47s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f95de2cbcfb3484cb1116a3f00b43256 : SUCCESS in 49m 11s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/d955410ac25f477bb60e37c668a95f7b : SUCCESS in 1h 05m 51s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/cd6675641eff4b3d83679d52677cfefd : SUCCESS in 26m 47s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9f9e1768da51427ca3faeeaf539e2576 : SUCCESS in 37m 00s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4df762dfa10440339e340407829915b6 : SUCCESS in 38m 01s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7cddb025427f4e9880904a93eb9fd9c0 : SUCCESS in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af86d96267f592d5c15508672dc13233ff8c5578","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-05-28 13:56:11.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-04-731504-1/ns5_manila : SUCCESS in 2h 7m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f0a2eca8e56f9877316fa7f10f2cc88fd40832bb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-05-28 14:48:16.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/04/731504/1/check/DellEMC_VNX_Manila/d6e2831/DellEMC_VNX_Manila/20 : SUCCESS in 1h 11m 04s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"35d8ef388eb6204da1d5912fec9faebab9e2eda4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-05-28 15:09:11.000000000","message":"Patch Set 1:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-04-731504-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"0315a066221ea744f347c374600d3c5ce9b4152f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-05-28 15:54:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/04/731504/1/upstream-check/manila-cDOT-ss/e0284ea/ : SUCCESS in 2h 16m 51s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/04/731504/1/upstream-check/manila-cDOT-no-ss/0b11077/ : SUCCESS in 1h 38m 31s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"1ffcd09c4f161b843c80ccd5bdd4549c88db956e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-05-28 16:33:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/04/731504/1/check/DellEMC_PowerMAX_Manila/3860fcc/DellEMC_PowerMAX_Manila/37 : FAILURE in 1h 39m 57s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"9c69277709d85438c94422be3138bbdcf7dfe628","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-06-09 12:54:06.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks Maurice!","accounts_in_message":[],"_revision_number":1},{"id":"3a677481dd1b403cdb223d2382bde2679f4999ca","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-16 04:35:57.000000000","message":"Patch Set 1: Code-Review-1\n\nThank you so much for the fix... code looks good, but, this could use a unit test in manila/tests/cmd/test_manage.py and a release note: https://docs.openstack.org/manila/latest/contributor/adding_release_notes.html","accounts_in_message":[],"_revision_number":1},{"id":"aacae4458ec419bacc1f7d6ac8c463c0835781ec","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2020-06-17 17:12:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92c4a98bbeafd2d3122f6157a8b51dfdb367b3aa","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-17 18:25:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/04/731504/2/check/dsvm-tempest-infinibox-nfs/239a0b4 : SUCCESS in 1h 12m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b596ef9b8483da8a74f7f3313c628283dd541025","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-06-17 18:33:42.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/04/731504/2/check/DellEMC_PowerMAX_Manila/2eb7010/DellEMC_PowerMAX_Manila/72 : SUCCESS in 1h 21m 09s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"eb60a0dc4a6897fe9ab9ad811118981347f6558c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 18:54:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/817771879e5249a9b6aef4eed24ca686 : SUCCESS in 10m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc5296190b464c82a6aecdbb6c78bfa5 : SUCCESS in 17m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5087618fdfd0460d8eb9b39a44f1560c : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/012210e2d0574b30b3a31bb289dfd508 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43f8ed71c5a8455d8a8f24a1754c7ccd : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9483b7183aa442abb6b216eee7b9d309 : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7fa82caa56345f392e432e52153995a : FAILURE in 7m 06s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cfe1475a7dbf4fc99f0f14d100fdb15a : SUCCESS in 6m 59s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/b66c173d07534562bfe1f5189dd9465b : SUCCESS in 1h 23m 57s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/ef7e560e2aa14447914bac92b54458c8 : SUCCESS in 52m 55s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/df8cf3f6962549c798e16c7f749b9988 : FAILURE in 1h 12m 28s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/b2ae3a06a887423db996cea7ee1ffc68 : SUCCESS in 45m 45s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/4dffb227c19e4d83b701f7eaf242e9d2 : FAILURE in 46m 55s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b03ff139b874ca3893c037ebc722842 : SUCCESS in 1h 38m 24s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/9e3f6b98cc6f49d7b720347195866924 : SUCCESS in 29m 14s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f18e85bc894846418e5df7ab6a36d275 : SUCCESS in 1h 26m 27s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/54b65f3bbb644757b43e3b1e2e43a92d : FAILURE in 1h 07m 26s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/6c4fe6815af94b9cb7a74f7b94833712 : FAILURE in 1h 06m 27s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/41db9486fb6c4be0b941fd1889a3c29a : SUCCESS in 44m 01s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/3ffab4da5fe4495f8249f7317f34921e : SUCCESS in 1h 04m 02s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/230ea52e07d8450290072a70fcea6416 : SUCCESS in 31m 44s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/361af504650f4b9ab1e36326555eb735 : SUCCESS in 37m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/691b9c35f3584471bc930a568ca4c12c : SUCCESS in 37m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2ea27654708842c9a498d4b04a96ab75 : SUCCESS in 6m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"81e4dcc222c6a7089ba02e49080908e55d071d09","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-17 19:23:29.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nThanks, looks like the release note needs a minor fix.","accounts_in_message":[],"_revision_number":2},{"id":"0178c98ceed436ade65026f9335914dd20d23106","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-17 19:47:16.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-04-731504-2/ns5_manila : SUCCESS in 2h 34m 30s","accounts_in_message":[],"_revision_number":2},{"id":"0f12627d5c054ef828552109d584ebaf8cce77dd","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2020-06-18 08:07:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"df799ab72c7cf14dd8cacab1603a3d4e69de16ae","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-18 09:18:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/04/731504/3/check/dsvm-tempest-infinibox-nfs/65a00eb : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":3},{"id":"2b148243510b543fd86bc8e06626f84a674e87aa","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-06-18 09:32:25.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/04/731504/3/check/DellEMC_PowerMAX_Manila/5b7123f/DellEMC_PowerMAX_Manila/74 : SUCCESS in 1h 25m 04s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"3562e87a450ac2e81a4a8827d120e0234c5ec1cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 10:09:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93d174812af7406693b3211c9b11a23a : SUCCESS in 12m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81f90ef4cc90458f8398a0f89357299e : SUCCESS in 14m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43a68659be97406a9c06be3c431b8318 : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f24517d223b44b799e34c2e23658a1f7 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a501bbba49744e0b11af83fdbe7acf2 : SUCCESS in 9m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7355bdc4bf74430a826a223abb51ff11 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3802323c30f4e549f139f86d28666ae : SUCCESS in 7m 35s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b2323f2306024225aef5b3cc69e9d621 : SUCCESS in 8m 49s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/3cbedf15b0d24218a96b076b1d27e9c0 : SUCCESS in 1h 55m 24s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/4aaa8d41b19f4381b19d017d5b399e28 : SUCCESS in 56m 44s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/9869174ab98b4e56a879deadfa8ff1a5 : FAILURE in 1h 08m 35s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/35057350055b46d99869aeddc858b0c7 : SUCCESS in 49m 51s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/82e205ceb5064e02a5e05e4395a9c87a : FAILURE in 49m 00s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f65b2d0abfb5450eb077005b98111caf : SUCCESS in 1h 35m 58s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/3c6150e4a7dc425591092a663ad6cf38 : SUCCESS in 42m 25s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c5cd850ce2d741af95edafa41376748a : SUCCESS in 1h 08m 19s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/630cbcfae76c487c8c7836cf195175b0 : SUCCESS in 1h 02m 07s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/c588c8b073b9401fa365c6e92bf8863a : FAILURE in 1h 10m 40s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/188fd1eb44a047c59925307f12287627 : SUCCESS in 45m 50s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/25ddec2b247e4517afee18b2c1447f9d : SUCCESS in 57m 58s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f0565bfa24b847a1b420f278626d6f1e : SUCCESS in 31m 43s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/24b59310592c41708e67be7f8c88400f : SUCCESS in 42m 46s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/9794bcf242534389aa65a8bda86282a1 : SUCCESS in 43m 58s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/186377bcc55646a6a7f5ae78bc12b9e2 : SUCCESS in 8m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"55bc63c829b8d54e52c4b3864108f3c4398ca729","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-06-18 10:13:55.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-04-731504-3/ns5_manila : SUCCESS in 2h 6m 23s","accounts_in_message":[],"_revision_number":3},{"id":"2dca3719c76949e3ffd51856abb0f68a8ed61466","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-18 10:25:11.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/04/731504/3/upstream-check/manila-cDOT-ss/991e0ba/ : FAILURE in 15m 39s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/04/731504/3/upstream-check/manila-cDOT-no-ss/1621470/ : FAILURE in 15m 37s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"7356d9c7c92648f6d6700dd2a587d43a9cafa8cf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-18 10:34:59.000000000","message":"Patch Set 3:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-04-731504-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"ffed373e2127084cbb77923acd4e5c6d8cd29148","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-06-18 11:13:14.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/04/731504/3/check/DellEMC_VNX_Manila/4c169e1/DellEMC_VNX_Manila/50 : FAILURE in 1h 03m 51s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"d46aa201abd21eba727f0dc5ce2893228e8e3f46","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-06-18 11:32:05.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/04/731504/3/check/DellEMC_PowerMAX_Manila/b5d8a67/DellEMC_PowerMAX_Manila/75 : SUCCESS in 1h 22m 51s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"da1f2e248a50cf9ab81559f34dc97190f0b9f5c8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-18 17:50:28.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks for the changes, Maurice!","accounts_in_message":[],"_revision_number":3},{"id":"ed556c7e52e5416b7e827b58398e50998e5f9f8b","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-06-18 17:54:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM, thanks Maurice!","accounts_in_message":[],"_revision_number":3},{"id":"d89e57cef9a63117af4e969aeb47ec0a5e591d42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 17:54:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8c466bf63f13cd5cd17e8777743d4008462963fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 19:15:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"10d48695805c15bec89c6eb235da6533b291d2eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 19:15:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7589d37de3064a7da5d6507dfbfd2df0 : SUCCESS in 14m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2cca8a767bd4cb29d797f4bbe91008d : SUCCESS in 14m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f487669316b348d293846917a3bf6fda : SUCCESS in 8m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a6320bf29db4e79863e1d3b019a20ba : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/336efdb882f54943af7c046d0b5f5e43 : SUCCESS in 15m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4bfae59b58448d2af683bdd045f3d28 : SUCCESS in 11m 43s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2226fe01dd4d495a930ac4494497bcaf : SUCCESS in 7m 28s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/66973564c3cf44d1ba8ef6d76cebd890 : SUCCESS in 1h 05m 33s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/84bd594e8dfc46c2a31b0c254f6d8d10 : SUCCESS in 1h 13m 24s","accounts_in_message":[],"_revision_number":3},{"id":"e83358741723a228f5a34cdcf5dddec7269c20c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 19:18:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c95b6ac35b84953a3d6405d5e81dba2 : SUCCESS in 2m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/383ab7cde18946a1ba33d4aba0ed92fe : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c8a8494c78a4a96c2009e5ada6297d1aba788f92","revisions":{"0d6992aacc3299fb6cb5c1331d88e8e1c08dbfe6":{"kind":"REWORK","_number":1,"created":"2020-05-28 11:48:17.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/04/731504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/04/731504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/731504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/731504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/731504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/04/731504/1"}}},"commit":{"parents":[{"commit":"82f01bd17f053979cb488c58df6d472975d11b8c","subject":"Merge \"[Unity] Update release note for filter function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/82f01bd17f053979cb488c58df6d472975d11b8c"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2020-05-26 18:34:27.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2020-05-28 11:47:51.000000000","tz":120},"subject":"add manila-manage service cleanup","message":"add manila-manage service cleanup\n\nChange-Id: I32307d5349c8ead5c0f7846f245d49f36dc3bf19\nCloses-Bug: #1881112\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0d6992aacc3299fb6cb5c1331d88e8e1c08dbfe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0d6992aacc3299fb6cb5c1331d88e8e1c08dbfe6"}]},"branch":"refs/heads/master"},"1060d2de3af9d3601f31ff748e6cd32599350642":{"kind":"REWORK","_number":2,"created":"2020-06-17 17:12:22.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/04/731504/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/04/731504/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/731504/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/731504/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/731504/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/04/731504/2"}}},"commit":{"parents":[{"commit":"82f01bd17f053979cb488c58df6d472975d11b8c","subject":"Merge \"[Unity] Update release note for filter function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/82f01bd17f053979cb488c58df6d472975d11b8c"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2020-05-26 18:34:27.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2020-06-17 17:12:07.000000000","tz":120},"subject":"add manila-manage service cleanup","message":"add manila-manage service cleanup\n\nChange-Id: I32307d5349c8ead5c0f7846f245d49f36dc3bf19\nCloses-Bug: #1881112\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1060d2de3af9d3601f31ff748e6cd32599350642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1060d2de3af9d3601f31ff748e6cd32599350642"}]},"branch":"refs/heads/master"},"c8a8494c78a4a96c2009e5ada6297d1aba788f92":{"kind":"REWORK","_number":3,"created":"2020-06-18 08:07:10.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/04/731504/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/04/731504/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/731504/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/731504/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/731504/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/04/731504/3"}}},"commit":{"parents":[{"commit":"82f01bd17f053979cb488c58df6d472975d11b8c","subject":"Merge \"[Unity] Update release note for filter function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/82f01bd17f053979cb488c58df6d472975d11b8c"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2020-05-26 18:34:27.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2020-06-18 08:06:54.000000000","tz":120},"subject":"add manila-manage service cleanup","message":"add manila-manage service cleanup\n\nChange-Id: I32307d5349c8ead5c0f7846f245d49f36dc3bf19\nCloses-Bug: #1881112\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c8a8494c78a4a96c2009e5ada6297d1aba788f92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c8a8494c78a4a96c2009e5ada6297d1aba788f92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
