)]}'
{"id":"openstack%2Fmanila~734697","triplet_id":"openstack%2Fmanila~stable%2Fussuri~I91fa2222743da06557ceb52fcbae0f8928d0752c","project":"openstack/manila","branch":"stable/ussuri","topic":"bug/1881921","hashtags":[],"change_id":"I91fa2222743da06557ceb52fcbae0f8928d0752c","subject":"Add documentation for share replica quotas","status":"MERGED","created":"2020-06-09 20:35:08.000000000","updated":"2020-06-10 00:23:23.000000000","submitted":"2020-06-10 00:21:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734697-1591748494197-0a846ac3","meta_rev_id":"0a099be961530c23d56904aa5efc3f0b3a0ec354","_number":734697,"virtual_id_number":734697,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 00:21:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2020-06-10 00:11:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2020-06-10 00:11:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 00:11:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-06-10 00:21:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61b5294d4d5a44a565b7ad3b0411f8eab579f459","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-06-09 20:35:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6a58f019c0023dcab554c018df8470ffbc9d9221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 20:46:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a55cc371ef407a9c13a9b3084394e7 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8790301b0e8d470289658aef962ea528 : SUCCESS in 6m 57s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c2920db28ab44799b542b26d5d7fa139 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0154c01c26b958cbbc67d88737d2b7e14e3f171b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-10 00:11:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"85606fc1a720099bcde4f3db226dd4901c5f4ac1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-06-10 00:11:56.000000000","message":"Topic changed from bug/1881921-stable/ussuri to bug/1881921","accounts_in_message":[],"_revision_number":1},{"id":"a74f0818599a1202b99f25854baad063f7990b00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 00:12:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ade461370daadab1ef45024c18d6f2d1da7b3d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 00:21:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/010f50fd3cc946b6bc550cfbea390505 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f20bb4af520e46768a947f045c82ec35 : SUCCESS in 5m 52s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8f4f385feb80448f8fb8f9f3992ef6c6 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"2f7bd7568d4ee0553297c75240e1bac7bb7a0177","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 00:21:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0a099be961530c23d56904aa5efc3f0b3a0ec354","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 00:23:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bca8d0b1330549c2b4dbd3220d7e129e : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1bebf39d4575539a0fe7d00af3c7926568e1cd6","revisions":{"c1bebf39d4575539a0fe7d00af3c7926568e1cd6":{"kind":"REWORK","_number":1,"created":"2020-06-09 20:35:08.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/97/734697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/97/734697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/734697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/734697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/734697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/97/734697/1"}}},"commit":{"parents":[{"commit":"5d1c30676e8deb906eddc59852a9b1b60d3ee7d6","subject":"[devstack] Allow cephfs daemon port access","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5d1c30676e8deb906eddc59852a9b1b60d3ee7d6"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2020-04-15 19:59:41.000000000","tz":0},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2020-06-09 20:35:08.000000000","tz":0},"subject":"Add documentation for share replica quotas","message":"Add documentation for share replica quotas\n\nUpdate documentation to fit into the new manila quota keys:\n- `share_replicas`\n- `replica_gigabytes`\n\nCloses-Bug: #1881921\nChange-Id: I91fa2222743da06557ceb52fcbae0f8928d0752c\n(cherry picked from commit 7bc4886a03d9490517448e25be531246a7e76b2e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c1bebf39d4575539a0fe7d00af3c7926568e1cd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c1bebf39d4575539a0fe7d00af3c7926568e1cd6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
