)]}'
{"id":"openstack%2Fmanila~737514","triplet_id":"openstack%2Fmanila~stable%2Frocky~Ib71a6cec939288498e48736f129fbfdacaabe9da","project":"openstack/manila","branch":"stable/rocky","topic":"bug/1882590","hashtags":[],"change_id":"Ib71a6cec939288498e48736f129fbfdacaabe9da","subject":"[NetApp] Fix svm scoped account","status":"MERGED","created":"2020-06-23 12:20:52.000000000","updated":"2020-07-25 23:26:07.000000000","submitted":"2020-07-25 23:24:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737514-1595719498641-1d1e633c","meta_rev_id":"361567e24f526eb36b0c4ddf72b3304ce26671b2","_number":737514,"virtual_id_number":737514,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-25 20:42:22.000000000","_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-06-23 12:52:15.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-07-25 00:49:16.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-25 23:24:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-07-25 00:38:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2020-07-25 00:38:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 12:52:15.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-07-25 00:38:41.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-07-25 00:49:16.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-25 23:24:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0196896fec59c605bb58d263555aabb7c82841e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-06-23 12:20:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"eb0fbf6b9663a593782d81fc80a3be77b0387a8e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-06-23 12:21:26.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit bbf4d7255636a571252d05497bbab59efc039540","accounts_in_message":[],"_revision_number":1},{"id":"e13a4afbda4b20ce81565267374f03720a04b935","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-23 12:52:15.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/14/737514/1/check/dsvm-tempest-infinibox-nfs/f377265 : FAILURE in 30m 56s","accounts_in_message":[],"_revision_number":1},{"id":"32af3de77af9cd226c8413ecb340ff4d508ad0d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 13:02:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d0d842e6777d4c5296a1769359df1633 : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aca7eacb28be4c32a4f4fd2085a9f76b : SUCCESS in 9m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a123e667761409a87e4e1ded85c0374 : POST_FAILURE in 13m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/10e1f10f8611401ab25cfc92726ab670 : POST_FAILURE in 14m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1ddbfe69c1b47249bae2dcfb9984837 : POST_FAILURE in 8m 24s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6256c53b13064b9baf3fe76be86aaa65 : SUCCESS in 8m 11s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/53e3864fafaf4f398f3ae612b5cd8037 : POST_FAILURE in 14m 01s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/39a08ce6039c4dfb9f890c1fc6f99531 : FAILURE in 16m 22s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/a6bc94dd21734884b8760c4b67ee7205 : POST_FAILURE in 17m 30s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/8b2e1618d0e245fd9951775282772416 : FAILURE in 15m 02s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/0ac356d9ab59455587dcd52753b3bd20 : POST_FAILURE in 13m 38s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ccd57014af046498b384ee355c6bf0a : POST_FAILURE in 20m 53s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/55f3559e0d4b43a2ac11b826beafd217 : POST_FAILURE in 22m 52s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2efdffe92f8841d685e283284f50d7cd : POST_FAILURE in 18m 30s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/6a945dfca3f443d7a8037d71eb3ee03e : FAILURE in 19m 02s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/9158255b344c4716b297901a8ea62bd9 : POST_FAILURE in 15m 38s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/32a76b90e6af43c088d68cfda082eb48 : POST_FAILURE in 17m 27s (non-voting)\n- manila-rally-multibackend https://zuul.opendev.org/t/openstack/build/36f15145f83145019f581cae720c01be : FAILURE in 17m 59s (non-voting)\n- manila-rally-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/283534a8ddcd471ca1af3e9b7d09bf71 : POST_FAILURE in 28m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afdfd933a28141a28a7e090522b869e1 : SUCCESS in 13m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4df2419ff9db4efb9fbe498023e78406 : POST_FAILURE in 16m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fdc9cd5d6fbf4d7a83c3b01c2f7db253 : SUCCESS in 32m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0e9013a0b36bd8e278d1ce36afedf86361cc9b4","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-01 16:57:18.000000000","message":"Topic changed from bug/1882590-stable/rocky to bug/1882590","accounts_in_message":[],"_revision_number":1},{"id":"205d0ddbf8677f21593d9d26f9bb52f674c8f0b9","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-01 17:00:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7e23b4dd2792ba188595862ea22d9ad73b62b8da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 18:57:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ca1e667c6d464260bc1306a5352c5d63 : POST_FAILURE in 39m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c24abf208c0493e978f729d9bec90d6 : POST_FAILURE in 38m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b22a38843d6e413dbffff04705219893 : POST_FAILURE in 37m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fbf7f2bed5084a6cb78f16f154192720 : SUCCESS in 15m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cfc481abc24465db6f96cfc05644181 : SUCCESS in 15m 03s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9f39620668bc4075b95e11f38f2084f0 : SUCCESS in 10m 40s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/bef4cc2b9f4b4579a04d7ec62fe9879e : SUCCESS in 1h 44m 02s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/0992daed609f40b19e312db648a637fd : FAILURE in 28m 07s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/589ed30cca9f47dc8ec8da4d498f7ff4 : SUCCESS in 1h 03m 31s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/31caa1cd98574696ace42fd0bbe8f1b2 : SUCCESS in 49m 44s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/b0dafa174c4943e1bfc3e4b11b4b550d : FAILURE in 47m 20s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b4488d83197f44e1a5b3cafa68a33b16 : SUCCESS in 1h 32m 33s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/1463493469e74d36b421086f3f241429 : SUCCESS in 29m 45s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/665aef4dcc7a44adab2a315c3d2646f7 : POST_FAILURE in 35m 20s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/3e12659b774640d584b449ad1956c8be : SUCCESS in 34m 23s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/df5e475a3d1246da99c923878e87f1cb : POST_FAILURE in 42m 36s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/de0c7c5e359f470788a697ff5d79c297 : FAILURE in 17m 56s (non-voting)\n- manila-rally-multibackend https://zuul.opendev.org/t/openstack/build/4059e5a0c9db4f11af413433b4e49be7 : FAILURE in 21m 54s (non-voting)\n- manila-rally-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/54c5dd52612e47e68c3d95f995b78fe1 : POST_FAILURE in 41m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a93b4c77ef54b56b703ce247169f1c4 : SUCCESS in 14m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cf2b40e088a48f3990902fc2a346f9e : SUCCESS in 38m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fe8cadadaee343428c8b8f5ada3c3900 : RETRY_LIMIT in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a54b2814bb706f8f7509d71223cbbe4c1102bacd","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-07 14:45:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"78d10c4c3cd238693b26ea07dfade5ff58630ad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 17:06:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f79b782faf7d4aac80cf9ebf646e1108 : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f32076d5abc467ea5778e6e340ff269 : SUCCESS in 9m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/462e267e695c4ad7881113c1055dca07 : SUCCESS in 11m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5320bad371914a9ca3a544b905ed49a4 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9b22b8cc1244ec69bea6715d88ddcd6 : SUCCESS in 8m 26s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/68abcee556f347ff8c7c3b21e13da067 : SUCCESS in 7m 10s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/79235c609aff4d6abcf6135b9b0e5f2d : FAILURE in 1h 55m 58s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/2d87c80c2435436cb49652f110bca85b : SUCCESS in 56m 49s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/19ac4c67b0574b2aa9c6fe55b80195c5 : FAILURE in 1h 15m 37s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/d4f15c98d2124bce90be9456376350fe : FAILURE in 1h 01m 08s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/60c77ebff6c94730b366ec555eb4f524 : FAILURE in 51m 33s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b421021f943a47f4b7c6bbeff2000ec0 : SUCCESS in 1h 29m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/ff29a7cfd6f94b5eafa8879063a2a4b3 : FAILURE in 17m 24s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c6b21e2da2b1462b9572487ffe64460f : FAILURE in 18m 25s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e7241fa2cee94daf8a1a733a76ea091a : SUCCESS in 47m 44s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/b12ed41dc30a4d74b25b846487427f37 : FAILURE in 26m 31s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d707eb317b874be3bdb3d8c86b3acf12 : SUCCESS in 32m 41s (non-voting)\n- manila-rally-multibackend https://zuul.opendev.org/t/openstack/build/98f501be41304fb69babde61ed4159f2 : FAILURE in 23m 13s (non-voting)\n- manila-rally-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/631069fffd11460eb23d3925b0409f84 : FAILURE in 28m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/487a7e5b9c24403688bd80daacbfad59 : SUCCESS in 10m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62fbab58e4f14abcae070f4a3408ef96 : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f3f47a44ed5448909c424ddd9c84ed64 : SUCCESS in 33m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19a24e89201430d144d61b63ff07e1da507798a5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-24 20:14:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a7b3d5ea07333d2688aa9bccdb7629460f249b6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 22:03:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/446e3e1b96d742f08c1228c7739503dc : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e82c75ef9b4839a329457b96e87a2e : SUCCESS in 7m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4e275a4137a44186aae2aaf5b549b8f8 : SUCCESS in 12m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/90078744144549d18d49626fe55939fc : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebc1017d45254c5da8b96e68ee220103 : SUCCESS in 5m 56s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4d54761a47d44392b575cee3e489964b : SUCCESS in 6m 07s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/327019073a7545d4aa5239a05e903a2e : FAILURE in 1h 43m 53s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1a4cdc8bea5e45bc8cc2583d7b177ed3 : SUCCESS in 46m 57s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/f4dbc4b930cb41578c58dbf294223850 : SUCCESS in 1h 18m 32s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/b3292a4acd4d48ffb44dbd7c04d4fb54 : SUCCESS in 57m 33s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/91155b75f17c4e0fa0e34d3318402873 : FAILURE in 49m 04s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41e84c472ab64ace84da39dd005afc08 : SUCCESS in 1h 36m 05s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/c658109542864005825c434f720813a8 : SUCCESS in 32m 17s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/421199e108d44de59731ceb57cbfabad : FAILURE in 20m 03s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/29f56466e6b346858ec6af9aedf5f862 : SUCCESS in 45m 50s\n- manila-tempest-dsvm-lvm-em-only https://zuul.opendev.org/t/openstack/build/5ad74e77ad2c4b03800b83e2fbf6161b : SUCCESS in 1h 25m 58s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/42aa1b07ee6b43c2801d9cf938399741 : FAILURE in 22m 29s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f4bb1b3e948841de996291ba85d45db2 : SUCCESS in 24m 27s (non-voting)\n- manila-rally-multibackend https://zuul.opendev.org/t/openstack/build/6d5aa010b52c46d1a8c94f9453b8ad13 : FAILURE in 18m 41s (non-voting)\n- manila-rally-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/73a14c350a09460f8a820d724367eac8 : FAILURE in 24m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adc59318502044299e6d67e8e5dfe8cc : SUCCESS in 8m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf4eef2244dd49d5a8de77fab031bac2 : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/daa0f7a07ff64d98aa8f3f074887ef56 : SUCCESS in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d57a5f4951a9079f1a5f0170ece792234633bef","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-25 00:38:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean cherry-pick, isolated to the driver","accounts_in_message":[],"_revision_number":1},{"id":"c0a073bd86d7bf1c4bf6bf504538a0fe58048eaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 00:38:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bad31b7dd6828ce13f2510f93b17ccd3d11c66f3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-25 00:49:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/14/737514/1/upstream-check/manila-cDOT-ss/362429d/ : FAILURE in 2h 33m 23s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/14/737514/1/upstream-check/manila-cDOT-no-ss/7bc1993/ : SUCCESS in 59m 43s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d1a2eccb859bcbd77eb7e778e5480fdbe9c29e50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 01:50:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b8828b56f77f4093994a5dbdc6aedcbb : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f0f9e25b5b848a897d3d4e66d68a6db : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d362d0a9511849c89bd4983d683e998f : SUCCESS in 7m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c974a9431b8a4d3f9227375cfe8e2c37 : FAILURE in 13m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34742a7b907d411983ec118a5c21a141 : SUCCESS in 5m 19s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/404419cdaae946f18cb9f56c85f2e3f8 : SUCCESS in 7m 49s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/dfbcc6d041124e318a5ff9bee3c1f7de : SUCCESS in 37m 17s\n- manila-tempest-dsvm-lvm-em-only https://zuul.opendev.org/t/openstack/build/9684d0c5c87e475db2d88dc2e0924169 : SUCCESS in 1h 09m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96a2cf35fc774e15a12ab85976728095 : FAILURE in 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d3a861cb70d2e6c77c2676cde19f44c724a7ef76","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-25 20:42:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f4233724376a1b296f72dbc86d312ba18cc0a5b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 22:35:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4bc22cf410814fb9bf1297a71bc48f23 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b13bdb88814398829dc6ac7487c63e : SUCCESS in 7m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa27eeaea3384363a53b0555d297742c : SUCCESS in 7m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/abe31572e3c0410c9418faff315bb001 : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44a51480d6b241cb825358f136626618 : SUCCESS in 6m 27s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9fc1add5c0864a5688d9b819c59ea8b6 : SUCCESS in 5m 26s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/c654d5dce940480f9f377f6707232bda : SUCCESS in 1h 47m 13s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/61c995bd098f49feb09bb865ee16c98f : SUCCESS in 1h 25m 42s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/f63606efdd5b49f2a36a676e12877307 : SUCCESS in 59m 51s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/49f1423e96e949b2975bdf59d7586b16 : SUCCESS in 46m 10s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/6371b023f37e4b01b482473443044229 : FAILURE in 47m 01s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f6f80dce9c84267976a067c8736b5f3 : SUCCESS in 1h 22m 51s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/4e30bf02a58540eaa34917f4218d1b51 : SUCCESS in 31m 07s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1c93c56cd5d740a3a26b5a01385f3e45 : FAILURE in 13m 48s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/b22ce105c75b473ca34ad7d7b1a62220 : SUCCESS in 37m 39s\n- manila-tempest-dsvm-lvm-em-only https://zuul.opendev.org/t/openstack/build/a86e6080b781413e89122d1cc1165726 : SUCCESS in 1h 14m 04s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/932a383f850f4f488e53eb25915ef7d8 : FAILURE in 19m 33s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f8bc06b9897d4996b3e022de73843fd6 : SUCCESS in 20m 36s (non-voting)\n- manila-rally-multibackend https://zuul.opendev.org/t/openstack/build/082f4ad2b6d94a3cb0e9e1a4496ca005 : FAILURE in 22m 44s (non-voting)\n- manila-rally-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/bdd6b65a996e415fbde6b369a5ae7a2d : FAILURE in 22m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f8547f4d7264ab09309139d6ef05c8f : SUCCESS in 8m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d86e215d4a0a4778afad869f36b51758 : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/50cebd27f0bd4d098a5c0fe4651c8e35 : SUCCESS in 37m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d937b6b78ffb3eb00b93763257ae19462a39e3c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 22:35:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce3fa707c572f1fedb902828d7245ab2fd139154","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 23:24:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"788f18004735f9ed645aa01a7315538109680b52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 23:24:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3a89dd661a094d56bb35a7407341b999 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c31c8f29236a4a5f8d127d4c5b3d7c2e : SUCCESS in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf20d41506b646ea880d748453bb4409 : SUCCESS in 8m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f2565de9f08140faae1d9c0028af2eec : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72b72a0867da4bb28a1f64c6c26f1e31 : SUCCESS in 5m 29s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5ff07191fe714ffcbb734677652a3633 : SUCCESS in 5m 52s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/fceefd50dee8460799c488adfa90c318 : SUCCESS in 30m 58s\n- manila-tempest-dsvm-lvm-em-only https://zuul.opendev.org/t/openstack/build/02ef7ed853484a9aafee3b3083ae70ac : SUCCESS in 45m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c2cdcc9234247888c2f6d1615e4fdce : SUCCESS in 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"361567e24f526eb36b0c4ddf72b3304ce26671b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 23:26:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/767d7a7eedd2486498f3551bbd4f52d2 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9bc395407fcc897c689911842c4bd277c5728fc","revisions":{"a9bc395407fcc897c689911842c4bd277c5728fc":{"kind":"REWORK","_number":1,"created":"2020-06-23 12:20:52.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/14/737514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/14/737514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/14/737514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/14/737514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/14/737514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/14/737514/1"}}},"commit":{"parents":[{"commit":"892f1d60ae8dd0831779ba9d57ccf1e5b1e89caa","subject":"Merge \"[Unity]: Failed to delete cifs share if wrong access set\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/892f1d60ae8dd0831779ba9d57ccf1e5b1e89caa"}]}],"author":{"name":"Felipe Rodrigues","email":"felipe_futty@hotmail.com","date":"2020-06-05 15:28:01.000000000","tz":0},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-06-23 12:20:52.000000000","tz":0},"subject":"[NetApp] Fix svm scoped account","message":"[NetApp] Fix svm scoped account\n\nWhen the NetApp backend starts, it needs to know whether the\n`revert_to_snapshot` support exist. So, it was retrieving the\nlicenses and checking if the `SnapRestore` is included. Using a\nscoped account, the backend cannot retrieve that information,\nthough. So, this patch solves it by sending a fake operation\n`revert_to_snapshot` that must fail. Analyzing the given error, it\nsets the backend support field.\n\nCloses-Bug: #1882590\nChange-Id: Ib71a6cec939288498e48736f129fbfdacaabe9da\n(cherry picked from commit 6f58af1ae39b4151db5feaa72deecafb9365ca16)\n(cherry picked from commit b5e541a6284200d28c7e6298ec46cb1f2f1eb9b9)\n(cherry picked from commit b252cb801d547c9a9717d705303c3206a079c27f)\n(cherry picked from commit 036908dd59684e9f5f44e305535cdc81dc563f2c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a9bc395407fcc897c689911842c4bd277c5728fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a9bc395407fcc897c689911842c4bd277c5728fc"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
