)]}'
{"id":"openstack%2Fmanila~737516","triplet_id":"openstack%2Fmanila~stable%2Fqueens~Ib71a6cec939288498e48736f129fbfdacaabe9da","project":"openstack/manila","branch":"stable/queens","topic":"bug/1882590","hashtags":[],"change_id":"Ib71a6cec939288498e48736f129fbfdacaabe9da","subject":"[NetApp] Fix svm scoped account","status":"MERGED","created":"2020-06-23 12:21:26.000000000","updated":"2020-07-28 03:50:32.000000000","submitted":"2020-07-28 03:46:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737516-1595908005860-6bbcf8b7","meta_rev_id":"73df06fbee15f046e661123b35a4e5939a7dea2a","_number":737516,"virtual_id_number":737516,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-06-23 13:00:56.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-07-28 03:50:32.000000000","post_submit":true,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 03:46:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2020-07-27 17:06:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2020-07-28 02:16:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 13:00:56.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-07-28 02:16:04.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-07-28 03:46:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-28 03:50:32.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"233a59b9f19043f45d243de7699e7d69da4a2a31","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-06-23 12:21:26.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"e5d6af8bfe34cdf49c945c046c6fda7b128a688c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 12:55:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bed1650a78864b4dad79134c0fb8e16d : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe86e1e7f4c64b94999cb997e265e743 : SUCCESS in 8m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f117432cc8954622a40286fb4290b272 : SUCCESS in 11m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e8b4445498744453988f77bc4ae95204 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/938d786ceb7e44d898a515e092e2eef2 : POST_FAILURE in 5m 41s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4a450d0a53e647398619b2c85e18ae76 : SUCCESS in 6m 41s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/aee310ab04ce423880f1000346c6fc56 : POST_FAILURE in 20m 03s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/d40b6eaa15814cc59feb510ea3483e8d : FAILURE in 23m 07s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/9dd1f9cdfc5f4b0f88b5627ff49cad85 : FAILURE in 20m 21s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/06762518665d41749d1a07be7e7bf7da : FAILURE in 18m 18s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/f20659d2bd2d4263ad39002720755be2 : POST_FAILURE in 20m 08s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc71e94d2468423f8666993603319e15 : FAILURE in 21m 39s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/01fa820ec62741388153a694056aeba7 : FAILURE in 8m 55s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/22368e097a9a4a67830b259b66aca183 : FAILURE in 9m 16s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/30d8a823b4a7427eb0a6b6f6d33924d4 : POST_FAILURE in 16m 41s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/71affb2376024fccb9a5717ac7114c16 : FAILURE in 26m 14s\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/eb3dd12e0c074a7890c4ae9279b8a510 : POST_FAILURE in 21m 17s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/ab455c6b957a4ef4bdc7289387f77fa2 : FAILURE in 19m 27s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/a7faeb4aa1ed4806bdbace4dca846626 : POST_FAILURE in 18m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3005a50b9764e42a9c356dd97320d4a : POST_FAILURE in 15m 07s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f3cf191829641a295a56557a3637612 : SUCCESS in 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d9c30067283e0ab1e036cc228745c8ce5dc403c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-23 13:00:56.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/16/737516/1/check/dsvm-tempest-infinibox-nfs/ae10199 : FAILURE in 24m 08s","accounts_in_message":[],"_revision_number":1},{"id":"141f9e337cbffeb4995408014c488faf0d852428","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-07 14:46:14.000000000","message":"Topic changed from bug/1882590-stable/rocky-stable/queens to bug/1882590","accounts_in_message":[],"_revision_number":1},{"id":"dd4874d81ee3339917d262fc9d095dbe8b5c9a4e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-27 17:06:44.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76bc47a70c532ff0923378fd13478c30bec81310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 18:34:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4a929832be93471a927055fb05309f88 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/384f0444b3bc4932a365a44df9dafc34 : SUCCESS in 8m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f970e9d57cd4edb820342952a77112e : SUCCESS in 12m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/264cd4da4a8a4b08a5ab90ce4b3970ae : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21c9694b3b524dfb8ece4168265fbfa0 : SUCCESS in 7m 23s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e0d4aa51b421403a86d6af8cc02081db : SUCCESS in 6m 18s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/386b0a7d19834d91a4f0eb6300ca123d : SUCCESS in 1h 04m 14s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/12ee1bec71e0485ebc0c39fc3513216b : SUCCESS in 46m 58s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/3c0b8964a783454cb54b12200aa87221 : SUCCESS in 1h 22m 05s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/7eb796eb596649019c2920ecf0455adc : SUCCESS in 53m 37s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/4586a0b9dcd8450e9cf6f4ec06dc3ca1 : FAILURE in 50m 35s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d372d95bb8294f1b8549c995bb90f5ef : SUCCESS in 1h 10m 59s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/cf4d6dc3cac1410bb12ed164060ec205 : FAILURE in 6m 45s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/691488d906554dd8a60d1984815c69d4 : FAILURE in 9m 18s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a26959374cec4373a9fa9193557995ac : SUCCESS in 40m 25s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/ea5a66726e194e1ba619e71bafa70f94 : FAILURE in 22m 53s\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/9a8fde92e348453091d7e9daa04416fb : SUCCESS in 28m 51s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/c4fdcc4892a94e9c8684d20c2491d067 : FAILURE in 21m 49s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/4487396512404372bad445f2943a1085 : FAILURE in 22m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cef51f67bb1d4e82a048d59f3967f9f5 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/de38de124b7a450198b0ebf41c2bd203 : SUCCESS in 7m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66e8d3a3ac2334a864bd0755ddc6c3b0a02bf732","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-27 19:07:34.000000000","message":"Patch Set 1:\n\nThis needs https://review.opendev.org/742987/, since the voting job\u0027s broken.","accounts_in_message":[],"_revision_number":1},{"id":"f64ab326e0e6b2314a6068b85a713d4f2784f9d8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-27 22:18:34.000000000","message":"Patch Set 1:\n\nrecheck \n\nhttps://review.opendev.org/742987 fixed the gate","accounts_in_message":[],"_revision_number":1},{"id":"4d88cc7f27a617d87f8d8ba546100719c0935aa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 00:09:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/592f26a7e5cf41de9af8a78fbd5323ce : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5331255f78439182b1ffd87efcf813 : SUCCESS in 7m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27e671f5ab944f28b8a2f5448adc6eed : SUCCESS in 8m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/76dd5d0013274fc28ad4419fb1e9383b : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fcb4f4eae924318875d53a159635151 : SUCCESS in 5m 52s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ce7fbdc4a0ae4e1786ed7a360f37e0ed : SUCCESS in 6m 29s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/95739226d8c74e849abbd2165d07e915 : SUCCESS in 1h 29m 59s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/46587ab2a2f146b89c2f18aac3d1aa08 : SUCCESS in 37m 12s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/141b104b908f43ba99a50115bfb98a50 : FAILURE in 1h 17m 56s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/c417dade9c9b462a98bf9ebcb9fa896b : SUCCESS in 54m 11s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/1bafab01f93d4a31b0ed83ceb1370e60 : FAILURE in 52m 21s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a848f7be57984b9fa6a61fb7660bd022 : SUCCESS in 1h 14m 40s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/984b1524e11b43229728731aae7c93f8 : FAILURE in 10m 02s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/4a064989438942cca0de2f48d20b3075 : FAILURE in 7m 19s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/8187ae42a0f9464190b708919823b176 : SUCCESS in 45m 52s\n- manila-tempest-dsvm-lvm-em-only https://zuul.opendev.org/t/openstack/build/d193bc110f6a4bd1b7cb7c819d6189e2 : SUCCESS in 1h 11m 16s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/074c7a420d2f4ce78a2cf7dea733ddef : FAILURE in 25m 32s (non-voting)\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/8905e746804e4653a90660c456829040 : SUCCESS in 28m 49s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/6fb1baf5932e42cc83883338749e83c7 : FAILURE in 16m 09s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/41c86276dc5f47f888dc31bded50c863 : FAILURE in 20m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7efb798f21f5437786448c2b42081b04 : SUCCESS in 19m 00s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cc7d209e66ab43789371acb687528dcc : SUCCESS in 9m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"560d6869efb2adfc7cdb8621904082015096c251","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-28 02:16:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"537096d23600bd29061501599ff5153639997f48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 02:16:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5bb0828b63df4b1c00c7fc185cb328f3c59eafd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 03:46:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4dba3b937f9e2689591dccc78f483fd3902c71d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 03:46:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f5f6c9ce99ec4f95b53c55fd1555bbaf : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9608d4bfe604f0da86bdf2687a7b2e8 : SUCCESS in 7m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0fd77d5e2bc425d93ec3d0843d6ec77 : SUCCESS in 6m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9b27801482e14d0eb663a0e24d3afeba : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/158fcb16bcf04a788483797be6e60e4c : SUCCESS in 6m 43s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3c88f810c48045c19cb6069f5a33e25a : SUCCESS in 6m 42s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ba2a15bede5e4c62a146b0d3d5d480a4 : SUCCESS in 38m 49s\n- manila-tempest-dsvm-lvm-em-only https://zuul.opendev.org/t/openstack/build/c0fecefd6ce84dd39a3d4526cc961258 : SUCCESS in 1h 25m 59s","accounts_in_message":[],"_revision_number":1},{"id":"858565516792778e5b9e16c05e5a38e3d813211a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 03:47:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6e99d64bcc949569ce8d0467cfdf66d : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1},{"id":"73df06fbee15f046e661123b35a4e5939a7dea2a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-28 03:50:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/16/737516/1/upstream-check/manila-cDOT-ss/460902f/ : SUCCESS in 2h 39m 09s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/16/737516/1/upstream-check/manila-cDOT-no-ss/671f3cb/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bbf4d7255636a571252d05497bbab59efc039540","revisions":{"bbf4d7255636a571252d05497bbab59efc039540":{"kind":"REWORK","_number":1,"created":"2020-06-23 12:21:26.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/16/737516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/16/737516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/16/737516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/16/737516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/16/737516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/16/737516/1"}}},"commit":{"parents":[{"commit":"0d5a1873f2427424b181b98a816f376d92ce61ac","subject":"Merge \"[devstack][ci] Modify firewall in ds-plugin\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0d5a1873f2427424b181b98a816f376d92ce61ac"}]}],"author":{"name":"Felipe Rodrigues","email":"felipe_futty@hotmail.com","date":"2020-06-05 15:28:01.000000000","tz":0},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-06-23 12:21:26.000000000","tz":0},"subject":"[NetApp] Fix svm scoped account","message":"[NetApp] Fix svm scoped account\n\nWhen the NetApp backend starts, it needs to know whether the\n`revert_to_snapshot` support exist. So, it was retrieving the\nlicenses and checking if the `SnapRestore` is included. Using a\nscoped account, the backend cannot retrieve that information,\nthough. So, this patch solves it by sending a fake operation\n`revert_to_snapshot` that must fail. Analyzing the given error, it\nsets the backend support field.\n\nCloses-Bug: #1882590\nChange-Id: Ib71a6cec939288498e48736f129fbfdacaabe9da\n(cherry picked from commit 6f58af1ae39b4151db5feaa72deecafb9365ca16)\n(cherry picked from commit b5e541a6284200d28c7e6298ec46cb1f2f1eb9b9)\n(cherry picked from commit b252cb801d547c9a9717d705303c3206a079c27f)\n(cherry picked from commit 036908dd59684e9f5f44e305535cdc81dc563f2c)\n(cherry picked from commit a9bc395407fcc897c689911842c4bd277c5728fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bbf4d7255636a571252d05497bbab59efc039540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bbf4d7255636a571252d05497bbab59efc039540"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
