)]}'
{"id":"openstack%2Fmanila~737697","triplet_id":"openstack%2Fmanila~stable%2Ftrain~Id9d248857c9a02eca486185d41b1fc22520d0919","project":"openstack/manila","branch":"stable/train","topic":"bug/1841035","hashtags":[],"change_id":"Id9d248857c9a02eca486185d41b1fc22520d0919","subject":"[Unity]: Failed to enabled ACE for share","status":"MERGED","created":"2020-06-24 07:43:56.000000000","updated":"2020-07-01 18:47:50.000000000","submitted":"2020-07-01 18:43:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737697-1593629023376-b4231495","meta_rev_id":"40e9182a1638db18f9113a0e4b23945b3739a183","_number":737697,"virtual_id_number":737697,"owner":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-06-24 08:14:16.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-06-24 10:45:04.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-01 18:43:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2020-06-25 13:18:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2020-07-01 16:35:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-30 01:32:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2020-07-01 16:35:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 08:14:16.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-06-24 10:45:04.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-25 13:18:17.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-06-30 01:32:01.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-07-01 16:35:41.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-07-01 18:43:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0553accd6dee073f6d72bc96698fc67330d22df2","author":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"date":"2020-06-24 07:43:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79c856efeac229f4ae7090de1dcbfdb566a6216f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-06-24 08:14:16.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/97/737697/1/check/dsvm-tempest-infinibox-nfs/481b6ce : FAILURE in 29m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e505411b46b43d5cf873d010cb1ae617383163fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 09:46:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79821d2b1822477a9dfc7102dcf3ed8f : SUCCESS in 9m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a516b41a21f749cfa2c7b94cbc522bdb : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91ad3c60de1d4ac0b6bfe137d000b27f : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fc2a8127fe742abb0a5b9eb7c4437f2 : SUCCESS in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e9f150336b84b89a6b2078224e0020b : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9aff380e93404cfa94f221996ff7ebe0 : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17fff3cacb9f4370a84f99de0217c839 : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/098b715f134e4d648638f5ca764f8b4a : SUCCESS in 7m 36s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3d447f17660a4279b20042b7341c8fa1 : SUCCESS in 6m 25s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/e35b244f407b46c19d3df0220d59aa7f : SUCCESS in 1h 23m 57s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/4411961bf8454284ab83806234651b36 : SUCCESS in 57m 58s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/23c2d49127874f359a487c1b5886ffc1 : FAILURE in 1h 38m 48s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/154e5291f3fb44f2addd332a5abcb57f : SUCCESS in 1h 12m 54s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/3d3a17e6773745fa874d59eda08313b7 : FAILURE in 49m 56s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aad6db74f3904a7399f8c9b375708445 : FAILURE in 1h 56m 06s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/7c3ebcb8adff475985693d8531c08d8d : SUCCESS in 30m 18s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/57df97758f5a454bb2f85516fe47190c : SUCCESS in 55m 01s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6e9d4ca4da3d4301832ebe021d585c8a : FAILURE in 56m 06s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/bce82a9e03b84de081504a32ab5b98df : SUCCESS in 46m 53s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/a4bae5ce8e8d4f7cad92158fb62a9525 : SUCCESS in 51m 14s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/259cd760cf6d4e798f0ea38d4a9472de : SUCCESS in 1h 08m 58s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/55d90907d68f41ada4cbe87a23f798f1 : SUCCESS in 18m 20s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/22cef543b6d44b028ffbc1ada264a440 : FAILURE in 40m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/180ca1cbb8684607991b122920ed9c4c : FAILURE in 44m 28s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/239f4385e663490dabc4b85e235752dc : SUCCESS in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8eb0ecae632b77635c1e402c6d43cba98861fa55","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-06-24 10:45:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/97/737697/1/upstream-check/manila-cDOT-ss/608d01a/ : FAILURE in 57m 49s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/97/737697/1/upstream-check/manila-cDOT-no-ss/6820f70/ : FAILURE in 1m 02s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"fb670a6c18edea8b9560628d146aa333ea5532bd","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-06-25 13:18:17.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM. Thanks Yong :)","accounts_in_message":[],"_revision_number":1},{"id":"a7d737ded3597f4f1d82eb744e81becb2c4b71a9","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-06-30 01:32:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"86eedc6467d929866ba53bffd95dfac953f63103","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-01 16:35:41.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1d7af7c51b1e68c48dd0e884d3654b287bea2b1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-01 16:35:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a6eb6fa01df050dae9cb2471133afad86ae8a7d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 16:35:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"794914f758372976cc6b3c0d680cbefb39ef1930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 18:43:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da6c69cfe3af46b78ff5803778223ee1 : SUCCESS in 14m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4082855aea7f4d8881d3cbc69e2403e4 : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60bd2ca2c16d466bbad79db340160326 : SUCCESS in 8m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27e71cee01bf405caed39c1fb0828e2d : SUCCESS in 15m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0ad99bb19fb437bb80024625619e93b : SUCCESS in 13m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14f36a9504f24f26aa8b19d415de6a0d : SUCCESS in 13m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/125376943fe34c44a91f4cf41f4c5398 : SUCCESS in 8m 50s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b2d8ef38bef4434a936e7bccd5d0909e : SUCCESS in 9m 17s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d7dacce05f6240fb9851ab070c36cc62 : SUCCESS in 48m 21s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/acdd7369f2ea47aa86bca70dc0b6a475 : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":1},{"id":"810a953b0f58e5cf929c7b5bb8d7c1d8c4645513","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 18:43:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e96ec41f717c9642a1303f3acc9be491f4b93922","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 18:47:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36be30175d394477bdd76695174700f5 : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/085a97543f454ea4b8e6b1421e16b0d5 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a0036a2dda18cc6ba2bde9e934eaed615fe5c22","revisions":{"9a0036a2dda18cc6ba2bde9e934eaed615fe5c22":{"kind":"REWORK","_number":1,"created":"2020-06-24 07:43:56.000000000","uploader":{"_account_id":26114,"name":"Yong Huang","email":"yong.huang@emc.com","username":"yong.huang"},"ref":"refs/changes/97/737697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/97/737697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/737697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/737697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/737697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/97/737697/1"}}},"commit":{"parents":[{"commit":"0fb1e94e67ab14ac8a0978f6c6cd791393829843","subject":"Merge \"CIFS extension failing because of volume in use\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0fb1e94e67ab14ac8a0978f6c6cd791393829843"}]}],"author":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-04-26 16:47:22.000000000","tz":0},"committer":{"name":"Yong Huang","email":"yong.huang@emc.com","date":"2020-06-24 07:43:07.000000000","tz":480},"subject":"[Unity]: Failed to enabled ACE for share","message":"[Unity]: Failed to enabled ACE for share\n\nIt is no need to enable ACE when creating CIFS share, so\nremove this process.\n\nChange-Id: Id9d248857c9a02eca486185d41b1fc22520d0919\nCloses-bug: #1841035\n(cherry picked from commit 47b90fa8fba882f86491ccc3959dc520271a26bd)\n(cherry picked from commit 1e8e154fe0d1043cd060eed43f1fba0c2327596e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9a0036a2dda18cc6ba2bde9e934eaed615fe5c22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9a0036a2dda18cc6ba2bde9e934eaed615fe5c22"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
