)]}'
{"id":"openstack%2Fmanila~742289","triplet_id":"openstack%2Fmanila~stable%2Ftrain~I67ac549e790322c56d02db4424ec01a5449e6633","project":"openstack/manila","branch":"stable/train","topic":"bug/1887694","hashtags":[],"change_id":"I67ac549e790322c56d02db4424ec01a5449e6633","subject":"Update LVM volume extend","status":"ABANDONED","created":"2020-07-21 21:38:30.000000000","updated":"2020-07-29 00:31:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3470b23480063b2cc3d4e7f2942fa5c5024fd78c","_number":742289,"virtual_id_number":742289,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"date":"2020-07-28 19:08:50.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":-1,"date":"2020-07-28 22:27:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-25 20:45:42.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-07-28 19:08:50.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-07-28 22:27:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"ef87221eb0941ea17fe1158cb0d601d6728c23da","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-21 21:38:30.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"89976ac0db307a351845b786f7b16319fbb7bb28","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-21 21:39:55.000000000","message":"Abandoned\n\nin favor of 742290","accounts_in_message":[],"_revision_number":1},{"id":"a8e3700bcc61cb7d77a29d47f66af6555606d331","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-21 21:41:26.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"b9c9778afb5c5f1a7e97deacbdac0aaaa00c1d55","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-21 21:41:52.000000000","message":"Topic changed from bug/1887694-stable/train to bug/1887694","accounts_in_message":[],"_revision_number":1},{"id":"71cad13af925d82a8867186a267e7b44f9e0f951","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-25 00:40:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"84f913b3c005bf1df00888f4e4b8fe5ddd225b41","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-25 20:45:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7f26ef7eaf1f090dbb56d37bfd93c021b5601b17","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-27 21:08:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"36f914dbf06f1d03fa55311720ae72eb754e2f73","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-07-27 22:39:16.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/89/742289/2/check/dsvm-tempest-infinibox-nfs/386e022 : FAILURE in 27m 57s","accounts_in_message":[],"_revision_number":2},{"id":"dc8b33b98b76303fd6b89b28327983712b1ecc0a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-28 02:02:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d861fcd3923ce72afda2ea8e32515588874985b9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-28 02:07:28.000000000","message":"Patch Set 2:\n\nZuul builds this change fine: https://zuul.opendev.org/t/openstack/buildsets?change\u003d742289\n\nI guess the problem is in Zuul posting results on this change, I feel it\u0027s probably looking up the abandoned change: https://review.opendev.org/#/c/742290/ and failing. We might need intervention from #openstack-infra","accounts_in_message":[],"_revision_number":2},{"id":"ed03a33c3fc4d49aa8ca9ef83879a376355b3f74","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-28 11:01:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"447dea3500d871381d1d062269b40db1f53b1e03","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-28 11:02:56.000000000","message":"Patch Set 2:\n\nTrying to see if abandoning this one and restoring the other will do it.","accounts_in_message":[],"_revision_number":2},{"id":"0ff2a798d7af8649a837560bcffa2a03731a8eb2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-28 16:55:51.000000000","message":"Restored\n\nRestoring this, after altering the change-id of the other change: https://review.opendev.org/#/c/742290/","accounts_in_message":[],"_revision_number":2},{"id":"dd4c5d1942e2f94075e808abc51b992ae75a4012","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-28 16:56:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f56c8ae1ad87843726928e5651a1ddd297c0e72f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-07-28 19:08:50.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/89/742289/2/check/dsvm-tempest-infinibox-nfs/e155db6 : FAILURE in 33m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a30a75a5090dc274e184f1633f810f271860b380","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-28 22:27:03.000000000","message":"Patch Set 2: Workflow-1\n\nDNM, the race condition with Gerrit is confusing Zuul - modifying the other patch (https://review.opendev.org/#/c/742290) doesn\u0027t fix the issue... per the infra team\u0027s suggestion, we\u0027ll create a brand new change against this repo and see if that fixes the issue: https://review.opendev.org/#/c/743640/","accounts_in_message":[],"_revision_number":2},{"id":"3470b23480063b2cc3d4e7f2942fa5c5024fd78c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-29 00:31:26.000000000","message":"Abandoned\n\nWe\u0027ll abandon this in favor of: https://review.opendev.org/#/c/743640/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"daca5b62e7aee79be2516cdea18aec68efcd44fa","revisions":{"8c177685f4fe45b5595ae3aabf7cbb0665bde97e":{"kind":"REWORK","_number":1,"created":"2020-07-21 21:38:30.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/89/742289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/89/742289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/742289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/742289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/742289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/89/742289/1"}}},"commit":{"parents":[{"commit":"0a3125439627da84bddfdf075b833f998f1935a7","subject":"Update NFS helper restart","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0a3125439627da84bddfdf075b833f998f1935a7"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-07-10 15:15:22.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-07-21 21:38:30.000000000","tz":0},"subject":"Update LVM volume extend","message":"Update LVM volume extend\n\n- do *not* skip fsck when extending\n- have extend resize the filesystem itself\n  rather than making a separate execution\n  of ``resize2fs``\n\nCloses-bug: #1887694\nChange-Id: I67ac549e790322c56d02db4424ec01a5449e6633\n(cherry picked from commit 2a3c18edcd9c4f7def293f18a6cac286ec00bc49)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8c177685f4fe45b5595ae3aabf7cbb0665bde97e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8c177685f4fe45b5595ae3aabf7cbb0665bde97e"}]},"branch":"refs/heads/stable/train"},"daca5b62e7aee79be2516cdea18aec68efcd44fa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-27 21:08:31.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/89/742289/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/89/742289/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/742289/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/742289/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/742289/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/89/742289/2"}}},"commit":{"parents":[{"commit":"60d343a986499f3ea9d726ae87c160ee62f47857","subject":"Merge \"[NetApp] update set_preferred_dc for ontapi 1.150\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/60d343a986499f3ea9d726ae87c160ee62f47857"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2020-07-10 15:15:22.000000000","tz":-240},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-27 21:08:31.000000000","tz":0},"subject":"Update LVM volume extend","message":"Update LVM volume extend\n\n- do *not* skip fsck when extending\n- have extend resize the filesystem itself\n  rather than making a separate execution\n  of ``resize2fs``\n\nCloses-bug: #1887694\nChange-Id: I67ac549e790322c56d02db4424ec01a5449e6633\n(cherry picked from commit 2a3c18edcd9c4f7def293f18a6cac286ec00bc49)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/daca5b62e7aee79be2516cdea18aec68efcd44fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/daca5b62e7aee79be2516cdea18aec68efcd44fa"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
