)]}'
{"id":"openstack%2Fmanila~742987","triplet_id":"openstack%2Fmanila~stable%2Fqueens~I4a9ae0744757d1c9bfe4f2b213eb04364992af0d","project":"openstack/manila","branch":"stable/queens","topic":"bug/1888717","hashtags":[],"change_id":"I4a9ae0744757d1c9bfe4f2b213eb04364992af0d","subject":"[extended-maintenance-branches-only] Replace LVM job","status":"MERGED","created":"2020-07-24 20:13:56.000000000","updated":"2020-07-27 22:17:30.000000000","submitted":"2020-07-27 22:17:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742987-1595888250930-fbab61f7","meta_rev_id":"bbea7700d865e4451d0c0965ece76d3ee337cb6c","_number":742987,"virtual_id_number":742987,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-27 19:47:52.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"date":"2020-07-27 21:35:07.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-27 22:17:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":2,"date":"2020-07-27 20:51:10.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":1,"date":"2020-07-27 20:51:10.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-27 19:47:52.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-07-27 20:51:10.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-07-27 21:35:07.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-27 22:17:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d64cd9bd43213bdeaebde17888c28642ee53ae8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-24 20:13:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"005421b7e31049bbd83503ecc3ae85bc9e1c0266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 21:51:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/74f46725b2a94a958102ab7aee08f462 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a4a960cfae4e6f835f86d0bd6fefc5 : SUCCESS in 8m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1fbc45a3ae2401c99dcc3c920119471 : SUCCESS in 14m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/33654abd14b94b73870ec24b19eb0bf0 : SUCCESS in 13m 32s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/bf50d794384f4b589414a59af7744eec : SUCCESS in 8m 17s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/f00356ab542c49a9b01918d98af942b6 : SUCCESS in 1h 34m 36s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/21ddcbe568344f9193db8b3afac6499d : SUCCESS in 51m 43s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/8fa0a3a09d2d45d6bd5700c456138f84 : FAILURE in 1h 19m 30s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/1ca0e1a8037046f1a27aced3772a4ce5 : SUCCESS in 55m 20s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/ab6fb0b0e1094a1aa367b7fa99cd78be : FAILURE in 52m 36s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1afdef1d9a5944739bf15fd0f7b32aad : SUCCESS in 1h 20m 08s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/ccd0bd2b6c26437f95a164794a1b87cf : FAILURE in 7m 31s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/d54032ce9e224895a191a7f3a40dbfe6 : FAILURE in 7m 51s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/7be8db0af56c4d718f66934c17d65d41 : SUCCESS in 45m 32s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/58a11c048de346388d800bc2820e8d28 : FAILURE in 12m 13s (non-voting)\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/bd3f727f15e3440690645c0125cb21d6 : SUCCESS in 30m 24s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/c483e7d7430c4026a949b9e45172c865 : FAILURE in 19m 46s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/4d9b3d2435ae4d928bf6e30d4a45d7c0 : FAILURE in 23m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e847cace49a249f5aed6d6ae3c432959 : SUCCESS in 13m 34s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f975813fb2fa4abaa236d9dc1b42fa54 : SUCCESS in 10m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0b03bb109e921a0e4b97e52a4b5fba7d6038682","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-24 22:50:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/87/742987/1/upstream-check/manila-cDOT-ss/5f247cc/ : SUCCESS in 45m 37s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/87/742987/1/upstream-check/manila-cDOT-no-ss/83f2ac8/ : FAILURE in 58m 34s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"bc4ba2bcc3321dda8e4327988534a2619d1cecb7","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-25 11:18:58.000000000","message":"Patch Set 1:\n\nrecheck\n\nem-only jobs didn\u0027t fire","accounts_in_message":[],"_revision_number":1},{"id":"27ed5648d7a1f8873c763359a56595430376460b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 13:24:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4b3f4a8f89934c8cbc026392feda254c : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72d03de3ad57416591c37c02361fe879 : SUCCESS in 7m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75914814e330487586b8c922d92ad37b : SUCCESS in 8m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/55dc379e754f4de7bc3d609f878ee002 : SUCCESS in 6m 48s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/34e63007137443699604738aeae87a11 : SUCCESS in 5m 14s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/71300cd1533a47c9931587f0019f0c7f : FAILURE in 1h 54m 48s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/601e9288c8f94585b7738d11aa253029 : SUCCESS in 34m 33s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/1fc88459037445bf9410f9e761840e9a : SUCCESS in 1h 12m 29s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/501216219a68490c981238126d0042ce : SUCCESS in 50m 26s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/77e45da4daa4456d832b7a0a490e3b76 : FAILURE in 46m 09s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79d630dcbd9a488884593d0407acf507 : SUCCESS in 1h 10m 37s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/209de23cadb7486bb9288d35d5d2b2af : FAILURE in 6m 44s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/044d3882a9b7458391559eddc431463f : FAILURE in 10m 14s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/abc04d85c220478eb297635684800af0 : SUCCESS in 32m 43s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/9d2f5f4dcdcf4027adf1146ec9569ac2 : FAILURE in 15m 08s (non-voting)\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/27f1d3f9e2744e8cbfb0284b1f00cb30 : SUCCESS in 18m 11s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/f7eaf4fc998a4f26b9968d92c3da6b23 : FAILURE in 19m 49s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/fe61835031e442ef9d250c490124e73e : FAILURE in 19m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5c35a99ffc745598b9660e3bfaf8b99 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b74c71b1c4444e4cb8840e49ffeab623 : SUCCESS in 10m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b05be02b6ce1e76b9952e64abfbefd8e7568894f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-25 14:07:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/87/742987/1/upstream-check/manila-cDOT-ss/97888e4/ : FAILURE in 42m 52s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/87/742987/1/upstream-check/manila-cDOT-no-ss/fdd7571/ : FAILURE in 28m 43s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c71336ef9ed9d03fb3080ed848b2f5ce052b361a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-27 19:18:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02a7cc36c6f25a2ed33303640d97f3c90cd0fedf","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-07-27 19:47:52.000000000","message":"Patch Set 2:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/87/742987/2/check/dsvm-tempest-infinibox-nfs/35fb5bd : FAILURE in 28m 48s","accounts_in_message":[],"_revision_number":2},{"id":"bae300fb32fe91f9bb9d1cfda7dd920cac3cd9c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 20:50:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/baea2c50fb624960b9dddd688d6c4561 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/573ac9348da44c28a71e588ee928121b : SUCCESS in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f68ad7dbd2c04a90a31e88d4c5124217 : SUCCESS in 10m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ef8148a537cf44439818f7d239073594 : SUCCESS in 7m 21s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d3638dc784c94150b24a9050027f840c : SUCCESS in 5m 32s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/27b457b58da841c4a80e1510a05c04ea : SUCCESS in 1h 16m 30s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/b9afdf843fdc4d43a2543c639e515fd4 : SUCCESS in 27m 49s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/dfd368d8fccc4e5ba9bffb371b32451f : SUCCESS in 1h 04m 54s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/0987bad7d9304c9aa3e50de26f80a851 : SUCCESS in 52m 10s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e95fc625b12f40c98431dfee3b732e1e : FAILURE in 43m 05s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/146a62cee2c6431ca13faa451534be8a : SUCCESS in 1h 12m 32s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native-centos-7 https://zuul.opendev.org/t/openstack/build/a3f0a4dd449e4156a15958640b9c6a24 : FAILURE in 11m 02s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs-centos-7 https://zuul.opendev.org/t/openstack/build/75a1e87bd7b84d83b3d7d57e586792d3 : FAILURE in 10m 37s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ac37203758d142ceb93e4a6ef9678821 : SUCCESS in 42m 37s\n- manila-tempest-dsvm-lvm-em-only https://zuul.opendev.org/t/openstack/build/8e40d4cbc12e4467bb1261f905f60d6c : SUCCESS in 1h 20m 17s\n- manila-tempest-minimal-dsvm-lvm-centos-7 https://zuul.opendev.org/t/openstack/build/8d9ae27e35fa4694a3436277ae8a2824 : FAILURE in 16m 59s (non-voting)\n- grenade-dsvm-manila https://zuul.opendev.org/t/openstack/build/8b099f6940724bcc8b1c90f5225eaa6d : SUCCESS in 32m 36s (non-voting)\n- rally-dsvm-manila-multibackend https://zuul.opendev.org/t/openstack/build/977af7c8cf7f453bb1a3347e9bac20cc : FAILURE in 21m 23s (non-voting)\n- rally-dsvm-manila-multibackend-no-ss https://zuul.opendev.org/t/openstack/build/09e45db4980849248c31033dee1d5611 : FAILURE in 18m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32d7b58e8b1b4f5e9600cdd7201ce285 : SUCCESS in 14m 29s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4484c6230cb94bbba490b1111d4d368b : SUCCESS in 11m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f41f5e71cd65f21ffc1aef562d4179850cb67136","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-07-27 20:51:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4ecf1f6c91fde3f295cb3c2a533e1c4dd56ea0f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 20:51:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6fc02e40385da18079de39483c63e6e2ce7b1aa7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-27 21:35:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/87/742987/2/upstream-check/manila-cDOT-ss/6e87fa5/ : SUCCESS in 43m 55s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/87/742987/2/upstream-check/manila-cDOT-no-ss/d5368da/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c688e8ce024cfe4868bca2a32a0521ec4a56f2ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 22:17:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0fa344a8549c46f78625c4e48599c0d9 : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d83b7a5464241489eac108616930799 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4711c62ea5334ed3b31202886a76e4c1 : SUCCESS in 6m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1d7f530417af4143a71d5d00f27cd9cc : SUCCESS in 7m 56s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f992e1425de4494097716d37665b8b6f : SUCCESS in 5m 30s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/ae93923535244ac99f948ef492c51998 : SUCCESS in 46m 24s\n- manila-tempest-dsvm-lvm-em-only https://zuul.opendev.org/t/openstack/build/cc17eecf70d24d8eb3a98ddbb387aa99 : SUCCESS in 1h 17m 26s","accounts_in_message":[],"_revision_number":2},{"id":"4a76a9d8eda7e667c772c15ed261d20571401118","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 22:17:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4ce9180d6c427139c7d5d9594e03947d131c03a2","revisions":{"59c2ce56be30f0659230073ae845dc6b28b2c12b":{"kind":"REWORK","_number":1,"created":"2020-07-24 20:13:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/87/742987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/742987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/742987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/742987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/742987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/742987/1"}}},"commit":{"parents":[{"commit":"0d5a1873f2427424b181b98a816f376d92ce61ac","subject":"Merge \"[devstack][ci] Modify firewall in ds-plugin\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0d5a1873f2427424b181b98a816f376d92ce61ac"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-23 18:22:24.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-24 20:11:50.000000000","tz":-420},"subject":"[extended-maintenance-branches-only] Replace LVM job","message":"[extended-maintenance-branches-only] Replace LVM job\n\nthe voting LVM job (manila-tempest-minimal-dsvm-lvm-centos-7)\nhas been failing for a while due to the issue identified\nin [1]. The fix for the issues didn\u0027t suffice for CentOS,\nand we use CentOS because of missing IPv6 related fixes\nin Xenial\u0027s nfs-utils package. to avoid impacting helpful\nbackports to these branches, this commit introduces a\nnew voting job called \"manila-tempest-dsvm-lvm-em-only\"\nthat deploys manila with the LVM driver on Xenial and\nruns IPv4 scenario tests.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015432.html\nRelated-Bug: #1888717\nChange-Id: I4a9ae0744757d1c9bfe4f2b213eb04364992af0d\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 9a1345b2eaa2fefb13f70c54bbcbdef6115e44ce)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/59c2ce56be30f0659230073ae845dc6b28b2c12b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/59c2ce56be30f0659230073ae845dc6b28b2c12b"}]},"branch":"refs/heads/stable/queens"},"4ce9180d6c427139c7d5d9594e03947d131c03a2":{"kind":"REWORK","_number":2,"created":"2020-07-27 19:18:22.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/87/742987/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/742987/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/742987/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/742987/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/742987/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/742987/2"}}},"commit":{"parents":[{"commit":"0d5a1873f2427424b181b98a816f376d92ce61ac","subject":"Merge \"[devstack][ci] Modify firewall in ds-plugin\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0d5a1873f2427424b181b98a816f376d92ce61ac"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-23 18:22:24.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-27 19:18:09.000000000","tz":-420},"subject":"[extended-maintenance-branches-only] Replace LVM job","message":"[extended-maintenance-branches-only] Replace LVM job\n\nthe voting LVM job (manila-tempest-minimal-dsvm-lvm-centos-7)\nhas been failing for a while due to the issue identified\nin [1]. The fix for the issues didn\u0027t suffice for CentOS,\nand we use CentOS because of missing IPv6 related fixes\nin Xenial\u0027s nfs-utils package. to avoid impacting helpful\nbackports to these branches, this commit introduces a\nnew voting job called \"manila-tempest-dsvm-lvm-em-only\"\nthat deploys manila with the LVM driver on Xenial and\nruns IPv4 scenario tests.\n\nThis isn\u0027t a perfect cherry-pick since the base job\nwas named differently on this branch.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015432.html\nRelated-Bug: #1888717\nChange-Id: I4a9ae0744757d1c9bfe4f2b213eb04364992af0d\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 9a1345b2eaa2fefb13f70c54bbcbdef6115e44ce)\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4ce9180d6c427139c7d5d9594e03947d131c03a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4ce9180d6c427139c7d5d9594e03947d131c03a2"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
