)]}'
{"id":"openstack%2Fmanila~743849","triplet_id":"openstack%2Fmanila~stable%2Ftrain~I8588e8f988d85dc64e19e7a44a25c3dd0b776892","project":"openstack/manila","branch":"stable/train","topic":"manila-grenade","hashtags":[],"change_id":"I8588e8f988d85dc64e19e7a44a25c3dd0b776892","subject":"Add verify-noapi testcase to grenade","status":"MERGED","created":"2020-07-29 22:08:36.000000000","updated":"2020-08-04 22:51:25.000000000","submitted":"2020-08-04 22:48:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743849-1596581330914-07a05074","meta_rev_id":"3be800296480094a0dd2fc9e611d61e80e6fd8c0","_number":743849,"virtual_id_number":743849,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2020-08-04 19:59:31.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-07-29 22:37:40.000000000","_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"date":"2020-08-03 20:18:57.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-04 22:48:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2020-07-30 12:44:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2020-08-01 12:04:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-29 22:37:40.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"REVIEWER"},{"updated":"2020-08-01 12:04:02.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-08-03 20:18:57.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-04 22:48:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91fd86677bf5b41cc1af43adf25b42bd9f3a9d4d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-29 22:08:36.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"e9815ea5fc83d93fe1b636041e5beddf1b28918e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-07-29 22:08:41.000000000","message":"Topic changed from manila-grenade-stable/train to manila-grenade","accounts_in_message":[],"_revision_number":1},{"id":"3d66ff949c7edb1e3abde2bf9e97e60940791f7c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2020-07-29 22:37:40.000000000","message":"Patch Set 1:\n\nBuild failed.  To recheck use \u0027run-INFINIDAT\u0027.  For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dsvm-tempest-infinibox-nfs http://openstack-ci-logs.aws.infinidat.com/49/743849/1/check/dsvm-tempest-infinibox-nfs/672329d : FAILURE in 28m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a1b4f5a1bdb28f477d3b912b77a75095e02347db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 00:14:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b043e973ecab4d93a155c3d4674eac88 : SUCCESS in 9m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0afb53061a1945f6929c807c3f402475 : SUCCESS in 14m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3614b175bf8a488c92a43c0c70528cb2 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/434f8c63b2fc41d8ba7d532b860fe24c : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3775457083a49e48fe57b26d392c47b : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/436cab2ea305443db3988206c5a0358e : SUCCESS in 7m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4b15a9199f248c5b360e90d603bee83 : SUCCESS in 7m 11s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6f531636987e4d02bbc5b5c507dfbc17 : SUCCESS in 6m 07s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/d1151ff0bcce43718341f6c2ee4a2a6c : SUCCESS in 1h 27m 51s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/3d8521a9aa3c4bc1a9cb4535086d84fc : SUCCESS in 46m 50s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/2bfd25a665aa4f3ea83c95fd18c6c8f4 : FAILURE in 1h 21m 40s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/3de089d5b0814c0a99a4b8014f84a17b : SUCCESS in 51m 49s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/29bb27690b2a4cebb9487c3ef8f6f8e6 : FAILURE in 44m 38s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a8ba61e84814c87a24b49b2cdab9156 : FAILURE in 1h 59m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/3c027740dbe94848a2bfddc0c0c8400d : SUCCESS in 32m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f810a9faaed0487786d9c54f03cf6b62 : FAILURE in 37m 03s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5fac8523ab26455e922423abc516c68d : FAILURE in 37m 31s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/7bf261a0f7114f179b6986a2e41d99ac : SUCCESS in 46m 05s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/8717aad70a3849daa52603136ff48139 : SUCCESS in 39m 42s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/d4df9aab97734ef485594b295d7a8ce1 : SUCCESS in 1h 00m 36s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d4e3113e73eb4acca97d4557b3c9a2d3 : SUCCESS in 28m 36s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/21d8e432084f493caab59cd07442966c : FAILURE in 34m 52s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/80a495ae56e54e61ab4d4515c25d5ac5 : FAILURE in 24m 48s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/01d1f7209c9443a6a0dd2dbffa387455 : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d7bf72e3f88519edaac44ba24b42921c19450fb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-07-30 00:15:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/49/743849/1/upstream-check/manila-cDOT-ss/a63258f/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/49/743849/1/upstream-check/manila-cDOT-no-ss/7cce88b/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b3bf970e12117ff321ddd42c0e3617486b4cb6b2","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-07-30 12:44:13.000000000","message":"Patch Set 1: Code-Review+2\n\nClean cherry pick, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"f67928d4bbf2b5b4e82bc1bf29185693291cf2d6","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-08-01 12:04:02.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"76b767316600804e625716f21e6a25f91ba5efee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 12:04:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64055df861e782c9d3bc831f71aebe8644a55cd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 13:16:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9513cf38e6a45af84b607bd7599d0f4 : SUCCESS in 8m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c897eba343554190a7bb8431e9b77899 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e4f3cda802540eba1c1b2cd13778da5 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/374934ae3fc148b8adc7d13b3e9bdd16 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17e3d988a3cc4cf18d3cb122aa4e80bb : SUCCESS in 14m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/943244544905429d9b2295cf9f9c402e : SUCCESS in 11m 17s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/bad34bc89c0e41af877f308588bd1cf1 : SUCCESS in 4m 25s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e6e9364f29034adbab5752ef0bda283b : FAILURE in 41m 00s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/4f40d1811df2404abe01de3e243f4389 : SUCCESS in 1h 10m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4ce2ff9afee0e2b97c1003bf12d07639e2f99e61","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-03 03:45:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b036cfc234f17f1225410ca0f3728155b972a69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 05:26:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f503b05317d43fc98101cbf607100ad : SUCCESS in 8m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9273023a1f6e4da1b0dffd7a7167b8c9 : SUCCESS in 12m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ad576bd25f6484593dd89e077efb199 : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eafb73e38c84bcc9d665ed7fd8079e9 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/237b3130cd594c68b28ad5735555612a : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94c26a02439f4594982dfad806a52fdb : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d10044298caf4e939b1af3bef9730e2e : SUCCESS in 8m 36s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a6cb1e47c5ad4755b26e13351c358e2c : SUCCESS in 6m 36s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/f010750a728d4acd90779d516638528c : SUCCESS in 1h 22m 16s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/b7c0ed79e8fa4aec86ee779c3eb4b3d7 : SUCCESS in 54m 25s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/88b3d2a873e24ea3b9e33cd4fa187681 : FAILURE in 1h 16m 54s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/957e9f34b98c48868e8dbc5ee8e7f230 : SUCCESS in 58m 49s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/c22e69ba5237461f992f359cc1c5d1ba : FAILURE in 47m 34s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9cf378be5d2344ccaca0ef34438aca91 : SUCCESS in 1h 37m 51s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/abf6649a8b994acbbca07ebfac0bd749 : SUCCESS in 17m 09s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/51d7777825b545e4b2a01d4fbff146b4 : FAILURE in 49m 49s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1a040806c2524ee6899f1e0d54ee32f2 : FAILURE in 34m 21s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/358cca2e44614318a316490eb2af71be : SUCCESS in 49m 51s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/d59661f134ba4652a0cd485a9e906487 : SUCCESS in 54m 27s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/2a97b21e317240a0b9a76ce53f0aeb5a : SUCCESS in 1h 01m 57s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4e5f58bc11184c3b8f6b94529c24d7d2 : SUCCESS in 21m 36s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/58c4ae32f1ec49429a98d9e934b3bfad : FAILURE in 40m 12s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fa3a123e95b84a29a6ee0efc4737734e : FAILURE in 32m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9501de9961d44fd5bf2dc9c1e3004ebc : SUCCESS in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4275b6ed597a23a42e2eb37bfa193a09a104608d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 05:26:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ada1ff5dc3eab8d876958f29b71cfd61d5c9f3a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 06:43:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/013f238dece9490a8cf777cf7e451dfe : SUCCESS in 9m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98d2f69e23d648918284dc78d8ef600e : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb27f229b11549acbe1d966ca1ec9829 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be91905aab2048088b630a3aa57a5b61 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab3f390156a24ceca593b2af7e05b0ae : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30e6b2f6669441dd89ea8a9ffca949fb : SUCCESS in 6m 30s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/14bdd299fb304ca2aa36c6a364d823aa : SUCCESS in 3m 53s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/af070bff854848fab3d6ac707b90401c : SUCCESS in 37m 44s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/b3058894f48c4e7caff2c66c74601897 : FAILURE in 1h 14m 32s","accounts_in_message":[],"_revision_number":1},{"id":"01280829fe3f963eca1c5b2338576c22c00c4557","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-03 06:46:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/49/743849/1/upstream-check/manila-cDOT-ss/73b0d1c/ : SUCCESS in 1h 19m 59s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/49/743849/1/upstream-check/manila-cDOT-no-ss/e64828f/ : SUCCESS in 1h 19m 58s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"519763c018e2dcf6e4573a0183c2d2edc8cf5a20","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-03 17:20:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e14365087b63f50034e7e62b94382bdc0f54b381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 18:52:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f71c46f0363b45e3a2ab4d3e06cee18e : SUCCESS in 10m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e61242066b2a43fd8bea11992a2d437b : SUCCESS in 12m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63b5e3a95af3432b9871440a36a1cd2e : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f152941a4504494bdeeadc4f4a375f9 : SUCCESS in 9m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/362b2063a6af449ab913a22117cd934d : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8e1459bb7bc415fa75c65abc4be9946 : SUCCESS in 9m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e564d612cb74f75a440c7a4821b0e3a : SUCCESS in 13m 07s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4028eb5539e9445d8747b2281e308f8f : SUCCESS in 5m 58s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/5000175cb2ef4047b41cffdf06f31711 : FAILURE in 1h 27m 46s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/93622d04b7d44fe4a001db5059e0f41d : SUCCESS in 56m 48s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/7b195301b8fb4ca388ecf0f368b00031 : FAILURE in 1h 10m 00s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/2869d8de42b84e359e756127b29011d8 : SUCCESS in 52m 40s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/b38e8ad154c647cf9bacfb942ad22871 : FAILURE in 36m 52s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4423b29b86064ec0a83629fd26f7611d : SUCCESS in 1h 26m 13s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/4fe6c2c016a54e29af12ff299417e8a4 : SUCCESS in 31m 00s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bc8a71edc92f4aeea8b184e7072849a6 : FAILURE in 51m 02s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4133ff2614194ff49c66f630973241d7 : FAILURE in 41m 14s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/af934cb08269438ba863e4240bbbf3d1 : SUCCESS in 51m 04s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/2bc701b5fe1744dda2217554fb9dbfe2 : SUCCESS in 1h 00m 45s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/f33065ca89014189835c95543627c4a1 : SUCCESS in 1h 00m 38s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a04172b682734d879537be9a9d47b969 : SUCCESS in 23m 27s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/25f1647cb01c455caa2c35ef3c8c1a3e : FAILURE in 37m 50s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1fc8d8a9ef664d5ab0c55872a7b63b12 : FAILURE in 39m 59s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/291558e737114575a0a8c269156434a8 : SUCCESS in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77b1c8ac0b3f30458a6f90139a81d155350c5303","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 18:52:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79b0fc6529de87cfeafe3757f50f7a334fd52577","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 20:09:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c9f1e864e847e188bed686a0060527 : SUCCESS in 11m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/496f8b4e821a48fc98c0f4194bafe190 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffb1b332a5954e36ae9b9f065470ae1f : SUCCESS in 8m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c30f1bbaece4736808b28e851f91ec1 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ca71aa2ce4e438aa2b2e5c92b178b00 : SUCCESS in 16m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1586afc430c489f9ded28faefe2eb3e : SUCCESS in 9m 46s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c118e1bc02894abdbe6deea60bb5f743 : SUCCESS in 6m 33s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/54cdea9106c44bcfb6433db6cc5216fb : SUCCESS in 51m 50s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/18ae9be2bcc346f9bb6d96030a9deaef : FAILURE in 1h 08m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5d29a977530440071feb2530b4eade0bfbdc3e37","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-08-03 20:18:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/49/743849/1/upstream-check/manila-cDOT-ss/1109d54/ : SUCCESS in 1h 26m 12s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/49/743849/1/upstream-check/manila-cDOT-no-ss/6597b30/ : FAILURE in 1h 02m 01s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"8ab0af6b77417856e004aa0ecf5321d456a4248f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-03 20:41:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a49c96e8d6973a08128f69870a47595e87fb4504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 22:25:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b24884c0a8364cdc8977ffa311bc1603 : SUCCESS in 10m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42e5ec2c0b9946ea8117c56219e805d1 : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1de7eea021094a4696890236435d4d2c : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc99379a6514b3d8b593655133b9522 : SUCCESS in 7m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8fea378fb67e488ca7f8b416c40c3fe0 : SUCCESS in 9m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53db4e64c9484ec09becfae97242584b : SUCCESS in 9m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66708caf65ae4dba9a47e02a134014ea : SUCCESS in 9m 12s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f443d73989f940569bd8fa7c7259d557 : SUCCESS in 5m 39s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/eefdbc0f25ed45b28af27100b49c9c7e : SUCCESS in 1h 26m 30s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/54ce0bf2ff8349d39adad3698894cbcf : SUCCESS in 1h 00m 59s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/64cb13ce5bb24843a4f23201f1f24c39 : SUCCESS in 1h 22m 22s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/3107c09f294145d3b5d9430956d60027 : FAILURE in 1h 02m 44s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/af7786c87372497388312ffb1fce3212 : FAILURE in 57m 48s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ebd9a77cb4d847c8bf0e0bfd693971b8 : FAILURE in 1h 35m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/2a70fffc8bf44712858c02a4e5238e65 : SUCCESS in 31m 12s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3aa091cba7ad4e9682fa017013386752 : FAILURE in 1h 07m 49s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cac8c53fb34d4a3a8708dffe352b5fd2 : FAILURE in 50m 08s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d37e067b406746e886c3f65baa07bb21 : SUCCESS in 52m 00s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/7c39827ec12d481ab672e8a1543e127e : FAILURE in 1h 01m 22s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/afef7037982b4d4abdf6d953e4af8387 : SUCCESS in 1h 13m 20s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/104a779133754a828b99a575d27ec81c : SUCCESS in 35m 46s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3390496c19c24be884de2cdabc81d29e : FAILURE in 33m 41s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/95e1218edac74dd8853d06e84a22591f : FAILURE in 44m 50s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1831fadd627d4e8aa84abb175cff2c9e : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"773bdcef3e02815b609ad44c5093c1cf6fd50bae","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-04 19:59:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f0deb4a44581da620b1b17fe1cca4e11326f2dc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 21:28:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e2df7acbbd4bd8bbc63a636fcc9b33 : SUCCESS in 8m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fad466b6707c4b6197000e3ab91b9e80 : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d5845588e1a4f27b455804c8770dd1d : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b3941798cf9408e9e2347503f6a6c11 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd8f6ed30ee542e1ab29f9aa72469c54 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c628b9c8fb3143d292d56faad0bf7607 : SUCCESS in 7m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a12c05a7d1d4e16a95e22d0c5e8fde4 : SUCCESS in 7m 25s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/78cce1af0ace44eebbbe4a754d33c0a4 : SUCCESS in 4m 43s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/a23bd097734c4b6cb4a095ff77905aa6 : SUCCESS in 1h 27m 35s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/dbf9db50fe3b4543abbbe013b3b26cd3 : SUCCESS in 51m 25s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/c666f5ee20464df88ce1b455671c23b9 : FAILURE in 1h 26m 57s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/d950798b70364ae78f3787c2c30001af : SUCCESS in 52m 11s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/00f71b9f1d3a49d99696c0af687d118c : FAILURE in 49m 09s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/706f5c67da9a45c0ae4aff8cc13ab0c6 : FAILURE in 1h 20m 33s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/a7c6b0e6e6fd4d87abee295134ca499c : SUCCESS in 29m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f70125fdc1354a399d78b70217354a98 : FAILURE in 47m 47s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e084f036fcfb4b3f8857a932427ba7ea : FAILURE in 45m 16s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/537531f017294581b66a8b8c763d83ad : SUCCESS in 49m 26s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/e510bb4bed0042de891a9e07f22280c6 : SUCCESS in 48m 31s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/52e324cd764344aa991dd6adfa028477 : SUCCESS in 1h 13m 20s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/fc0500dbcd1343e09f17fc0f731e9eea : SUCCESS in 23m 02s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b75079e224ea43c981e3363a3441b81b : FAILURE in 36m 42s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3401a3d065b54c35b6ae55f430aaa377 : FAILURE in 31m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ed81884f015f4889bf8d5cab4389c287 : SUCCESS in 4m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3229709a7b7eca995cc3e84d98c16231e2c91e66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 21:30:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c4cf6154df46488309fa53939a11cc7c699b727","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 22:48:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea491e358d6e420fa47675643009be26 : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8e134992bf549b9a19ecc4f803b50e8 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/379326bac0e745a08f394d361686e318 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/232759e96d5e485190f3511f54fef86b : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b960e50029104533b6820125c25778af : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bcf28b950784a2d81b7a648dd1da064 : SUCCESS in 12m 02s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5fdc3561e754417c9743975b89eb1b96 : SUCCESS in 6m 42s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/4f242d663554406282618c71cc83f132 : SUCCESS in 49m 58s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/242374d9b39f4ff2bcb64a6993517b78 : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5bcb4d9faf9302fa5049cb1dbd5fee0f949ed6c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 22:48:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3be800296480094a0dd2fc9e611d61e80e6fd8c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 22:51:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41c4ea272f3445ac8d8f491d8b596a63 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a8fc83f2a7c5736d5bc818ec0a78d1a8a5d3849","revisions":{"0a8fc83f2a7c5736d5bc818ec0a78d1a8a5d3849":{"kind":"REWORK","_number":1,"created":"2020-07-29 22:08:36.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/49/743849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/743849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/743849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/743849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/743849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/743849/1"}}},"commit":{"parents":[{"commit":"4e292ea884ff72341c2ab6733745fe6d35733aa8","subject":"Harden LVM driver deletion paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4e292ea884ff72341c2ab6733745fe6d35733aa8"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-23 04:26:52.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-07-29 22:08:36.000000000","tz":0},"subject":"Add verify-noapi testcase to grenade","message":"Add verify-noapi testcase to grenade\n\nSwitch to using the LVM driver in the grenade\njob that allows us to add a minor data\npath test to prove that upgrading manila\nhas no impact on data path connectivity\nto resources created by manila.\n\nChange-Id: I8588e8f988d85dc64e19e7a44a25c3dd0b776892\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 8cccb73f46df763f907171e63d960fb1e6a81fd0)\n(cherry picked from commit 544998764ef8d129e460c7ca3333160848c90a2e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0a8fc83f2a7c5736d5bc818ec0a78d1a8a5d3849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0a8fc83f2a7c5736d5bc818ec0a78d1a8a5d3849"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
