)]}'
{"id":"openstack%2Fmanila~745261","triplet_id":"openstack%2Fmanila~master~I1e7244185d44bff2df85b09c4566836a26c84376","project":"openstack/manila","branch":"master","topic":"bug/1821837","hashtags":[],"change_id":"I1e7244185d44bff2df85b09c4566836a26c84376","subject":"Updates the API reference for some share operations","status":"MERGED","created":"2020-08-07 00:26:52.000000000","updated":"2020-08-17 16:49:52.000000000","submitted":"2020-08-17 16:47:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745261-1597682854323-0b925b31","meta_rev_id":"df2f9f9e9d0e0f42e0697f2a4ce5c44d7724ce9a","_number":745261,"virtual_id_number":745261,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-13 13:37:39.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"date":"2020-08-13 14:50:41.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-13 14:30:45.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-08-13 14:12:22.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-17 16:47:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-17 16:34:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-17 16:34:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 13:37:39.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 14:12:22.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 14:30:45.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 14:50:41.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-17 16:34:21.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-08-17 16:47:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae3519036d04686163b40874cb3467cfdd2a2c69","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-08-07 00:26:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ac7bc13e656c7d279571c0c441b9ed8dffd0a2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 00:37:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4d74ba9f12e54ac69a096623b847dd10 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d633c6d3770e46fb93de5dc7a67f17c8 : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0f29958eb2c4cdfa95cbc3eb95b5fcb : SUCCESS in 4m 24s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9e8946a642ed421584636836c19f7ce0 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4b0e4adbd8dfd1a158918751277c9ed18f4cea54","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-08-07 01:44:10.000000000","message":"Patch Set 1:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-61-745261-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"463c6461e4363e035dd95f143948ff8a8d792e8a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-07 02:45:48.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-61-745261-1/ns5_manila : SUCCESS in 1h 46m 55s","accounts_in_message":[],"_revision_number":1},{"id":"77f7f5a32cea9a28f9d81249d997a95d01c0e5e0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-10 23:27:07.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nThanks for fixing this up, Carlos - a comment inline..","accounts_in_message":[],"_revision_number":1},{"id":"56a03d8110962360f85f4cd378c3c30a0277eb49","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-08-12 11:30:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0d6a4b4b897222b16724b4bd669ecdbe3e0e936b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-08-12 11:31:09.000000000","message":"Patch Set 2:\n\nThanks for reviewing, Goutham. Please check the changes inline.","accounts_in_message":[],"_revision_number":2},{"id":"25abf31cd17345ac31a0a0bd1eed298256e2cd05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 12:19:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e5454d9142c14635bc6b97cfc51931ec : FAILURE in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4610732e9299446fadd598fc5f97e813 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e34e66a35f904a0aafb36745b373d686 : SUCCESS in 5m 34s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c009e10bc2c74939b1e367d9d19e18bf : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"1a7cb9da32b9400232d8bb46def41e1c3f03e020","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-12 12:53:25.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-61-745261-2/ns5_manila : SUCCESS in 1h 22m 7s","accounts_in_message":[],"_revision_number":2},{"id":"95ba6c2f17e16b85b71329397c95905f3311caff","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-08-12 13:15:43.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/61/745261/2/check/DellEMC_PowerMAX_Manila/b559dff/DellEMC_PowerMAX_Manila/214 : SUCCESS in 1h 44m 42s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"41bad7d99e2df50f362bfb9cbe8ec9603cd7c2de","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-12 23:37:40.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck\n\nThanks; infra test nodes have had some issues in one of the Vexxhost providers - lets see if they\u0027ve resolved it.","accounts_in_message":[],"_revision_number":2},{"id":"35f234cf9b62d9d0265581a6cc00f9e32eb10921","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 00:06:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c4ae167af6724c0b8303f035eed0342a : FAILURE in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7efeb30b18634585aa99dc59ca22924b : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/563071e3c29e45f2884a647265f07ca9 : SUCCESS in 5m 39s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/edaea013147748b1a953ba7d9853f78c : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"e31f0005de918c4a111031d31b1984c7f75727fe","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-08-13 11:52:33.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"8e65d678c9753f07b3c4fe8668c70d3a6b5f9f16","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-08-13 11:54:25.000000000","message":"Patch Set 3:\n\nThanks for revisiting, Goutham. The failure was due to a mistake of mine. I forgot to put a comma in the updated json file and haven\u0027t realized since I just published an edit in the previous PS. Now it should be ok :)","accounts_in_message":[],"_revision_number":3},{"id":"993e74668c8af657493b95f495ff1c1ac1a20510","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:05:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0db6fd887dd34c8aaf29aa53f839b264 : FAILURE in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8386ca6e194491686f740a36992daa5 : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e342dd2509a4f91a78ebfca90a958d9 : SUCCESS in 6m 09s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4c99eff9e379477aa36e72005c0d079d : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":3},{"id":"12854a4bc9f8cdbf4b10d03169ba4889c328491f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2020-08-13 12:53:52.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5e2b62095b345c56cd9bbff5814d47ee08569759","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-13 13:01:39.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-61-745261-3/ns5_manila : SUCCESS in 1h 8m 43s","accounts_in_message":[],"_revision_number":3},{"id":"2a5b116ed6acde443fc03328dbacee714a495e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 13:11:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/93988cc5af034557bac8b42742062394 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/396641e626324b76a6bb953ae27e69c9 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1828ef88fbb4b4aa9132adb3ae08bfa : SUCCESS in 4m 50s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f21a3fbecc874dc690a23f7c601aef98 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":4},{"id":"50a8fabf38f63e20f3b0124c400a30007b74af99","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-08-13 13:37:39.000000000","message":"Patch Set 4:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-61-745261-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"65851a2092ef6edffb579bc409ccb6ea396c7f4e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-08-13 14:12:22.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-61-745261-4/ns5_manila : SUCCESS in 1h 18m 5s","accounts_in_message":[],"_revision_number":4},{"id":"51b5898b92d2926b3a7ec008531736c6abdb3ff6","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-08-13 14:30:45.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/61/745261/4/check/DellEMC_VNX_Manila/554da0b/DellEMC_VNX_Manila/149 : FAILURE in 1h 18m 33s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"01e62ad3564b106f0dbfc2432ca1574bf266bcb7","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-08-13 14:50:41.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/61/745261/4/check/DellEMC_PowerMAX_Manila/12377d1/DellEMC_PowerMAX_Manila/217 : SUCCESS in 1h 56m 31s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"d8f121711717114c48ab56e6b7194f492cf4809b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-17 16:34:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nGreat, thanks Carlos","accounts_in_message":[],"_revision_number":4},{"id":"3399c1a213d7682c35366a4cfb7ec61371ae5364","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 16:34:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4788034cf30ec21e4278b009c76bd2bfdce92434","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 16:47:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"41997793ff25e447d7574bbd923444e8b2580561","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 16:47:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ffc24c3d4d3a4fd09ad33ffd24b88d5d : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01615dd508954c9da36a8791b83c6d2a : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4928224bb1c94fde925f89353b5fa77b : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":4},{"id":"b48081cc2d94ecd2dd7dfa1595a46726a1fa95f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 16:49:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/135d181019ff4250aa1e54f7ad4afdfb : SUCCESS in 1m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0efd743a299403dbe81a558e9f33b7e : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c650b63d33d6aaa2496b40e583e3aab9e3a92d31","revisions":{"4126121034ce5791bc8a3de33da40928405907ab":{"kind":"REWORK","_number":1,"created":"2020-08-07 00:26:52.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/745261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/61/745261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/61/745261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/61/745261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/61/745261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/61/745261/1"}}},"commit":{"parents":[{"commit":"f78e2fdee7d1ed05276dc2ebbc8cca7e88e15445","subject":"Merge \"[ci] Replace experimental tripleo job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f78e2fdee7d1ed05276dc2ebbc8cca7e88e15445"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2020-08-07 00:21:52.000000000","tz":0},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2020-08-07 00:21:52.000000000","tz":0},"subject":"Updates the API reference for some share operations","message":"Updates the API reference for some share operations\n\nAdded missing `user_id` field to the response of share create,\nshow and manage operations.\n\nChange-Id: I1e7244185d44bff2df85b09c4566836a26c84376\nCloses-Bug: #1821837\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4126121034ce5791bc8a3de33da40928405907ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4126121034ce5791bc8a3de33da40928405907ab"}]},"branch":"refs/heads/master"},"fa3a0ed0093ecb9e9e2859b271c582a5dddb0f4f":{"kind":"REWORK","_number":2,"created":"2020-08-12 11:30:52.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/745261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/61/745261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/61/745261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/61/745261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/61/745261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/61/745261/2"}}},"commit":{"parents":[{"commit":"f78e2fdee7d1ed05276dc2ebbc8cca7e88e15445","subject":"Merge \"[ci] Replace experimental tripleo job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f78e2fdee7d1ed05276dc2ebbc8cca7e88e15445"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2020-08-07 00:21:52.000000000","tz":0},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2020-08-12 11:30:43.000000000","tz":0},"subject":"Updates the API reference for some share operations","message":"Updates the API reference for some share operations\n\nAdded missing `user_id` field to the response of share create,\nshow and manage operations.\n\nChange-Id: I1e7244185d44bff2df85b09c4566836a26c84376\nCloses-Bug: #1821837\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fa3a0ed0093ecb9e9e2859b271c582a5dddb0f4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fa3a0ed0093ecb9e9e2859b271c582a5dddb0f4f"}]},"branch":"refs/heads/master"},"ad83704cff0c4ad521966d90575ae2e1e7119a72":{"kind":"REWORK","_number":3,"created":"2020-08-13 11:52:33.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/745261/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/61/745261/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/61/745261/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/61/745261/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/61/745261/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/61/745261/3"}}},"commit":{"parents":[{"commit":"f78e2fdee7d1ed05276dc2ebbc8cca7e88e15445","subject":"Merge \"[ci] Replace experimental tripleo job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f78e2fdee7d1ed05276dc2ebbc8cca7e88e15445"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2020-08-07 00:21:52.000000000","tz":0},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2020-08-13 11:52:11.000000000","tz":0},"subject":"Updates the API reference for some share operations","message":"Updates the API reference for some share operations\n\nAdded missing `user_id` field to the response of share create,\nshow and manage operations.\n\nChange-Id: I1e7244185d44bff2df85b09c4566836a26c84376\nCloses-Bug: #1821837\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ad83704cff0c4ad521966d90575ae2e1e7119a72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ad83704cff0c4ad521966d90575ae2e1e7119a72"}]},"branch":"refs/heads/master"},"c650b63d33d6aaa2496b40e583e3aab9e3a92d31":{"kind":"REWORK","_number":4,"created":"2020-08-13 12:53:52.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/745261/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/61/745261/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/61/745261/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/61/745261/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/61/745261/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/61/745261/4"}}},"commit":{"parents":[{"commit":"f78e2fdee7d1ed05276dc2ebbc8cca7e88e15445","subject":"Merge \"[ci] Replace experimental tripleo job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f78e2fdee7d1ed05276dc2ebbc8cca7e88e15445"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2020-08-07 00:21:52.000000000","tz":0},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2020-08-13 12:53:50.000000000","tz":0},"subject":"Updates the API reference for some share operations","message":"Updates the API reference for some share operations\n\nAdded missing `user_id` field to the response of share create,\nshow and manage operations.\n\nChange-Id: I1e7244185d44bff2df85b09c4566836a26c84376\nCloses-Bug: #1821837\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c650b63d33d6aaa2496b40e583e3aab9e3a92d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c650b63d33d6aaa2496b40e583e3aab9e3a92d31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
