)]}'
{"id":"openstack%2Fmanila~747340","triplet_id":"openstack%2Fmanila~stable%2Fussuri~Ib6008c9dd4284b8c3434ffce61fb74f1db6c3d35","project":"openstack/manila","branch":"stable/ussuri","topic":"bug/1782859","hashtags":[],"change_id":"Ib6008c9dd4284b8c3434ffce61fb74f1db6c3d35","subject":"Update access rules documentation for user guide","status":"MERGED","created":"2020-08-21 06:51:10.000000000","updated":"2020-08-21 14:33:07.000000000","submitted":"2020-08-21 14:27:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747340-1598020022958-cee79b84","meta_rev_id":"0a3e8d7620dd8681baaa609ff09a7bf3c1c1342e","_number":747340,"virtual_id_number":747340,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 14:27:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2020-08-21 13:06:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2020-08-21 13:06:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-21 13:06:17.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-08-21 14:27:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16ce271e79d84e717d2e509ac30f7110f14bb4b6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-21 06:51:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b401a8ed15be3abcd02e1306f603e0a828adfb48","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-21 06:51:19.000000000","message":"Topic changed from bug/1782859-stable/ussuri to bug/1782859","accounts_in_message":[],"_revision_number":1},{"id":"decd81dd3f955a7e37e90499f7dea6840c4bbea5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-08-21 06:51:34.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 3b82fef551d80ff80da9b7d4a32dff8434ca5042","accounts_in_message":[],"_revision_number":1},{"id":"f4dab8c6dcb76a98ce7c418f81feba4370fcf017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 07:08:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c01a91e9319d447c89e593e75ddf8b2b : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fd79e27b88945f79bac061f5c2bf269 : SUCCESS in 7m 47s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1cfeb97fcf184e1ea1002751dd7a1e0e : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":1},{"id":"355ef9d71aeaabdef6c0d4f62a4c950342663393","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-08-21 13:06:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean cherry pick, doc update only. Tox outputs look good. Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"9f86cced77c7e65768e2fe6eabde030f75ffcd51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 13:06:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad8c12fa5193194258e34f8b009f524d7b38dde3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 14:27:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87278b814d9a4f23b93fc27af398a0bb : SUCCESS in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/528889565cbd437db57ca57e0b827365 : SUCCESS in 5m 09s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f4720727d99c4fc5806b0acc085cfc85 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9ccc41283ce9d8b1280fd5ce159671065b245540","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 14:27:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0a3e8d7620dd8681baaa609ff09a7bf3c1c1342e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 14:33:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7271e82f3c154d85a3f99e429701e38d : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1434adc353e80192d9a17121b57831d273b30633","revisions":{"1434adc353e80192d9a17121b57831d273b30633":{"kind":"REWORK","_number":1,"created":"2020-08-21 06:51:10.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/40/747340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/40/747340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/747340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/747340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/747340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/40/747340/1"}}},"commit":{"parents":[{"commit":"eac41910c041cca3583e1e94a9dfbaf7e3229fdd","subject":"Add mod_wsgi configuration samples","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/eac41910c041cca3583e1e94a9dfbaf7e3229fdd"}]}],"author":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2020-08-06 19:40:21.000000000","tz":-180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-08-21 06:51:10.000000000","tz":0},"subject":"Update access rules documentation for user guide","message":"Update access rules documentation for user guide\n\nUpdates user guide documentation to provide metadata info\nto an access rules. Also add two new operations for access\nrule metadata: \u0027access-metadata set\u0027 and \u0027access-metadata unset\u0027\n\nChange-Id: Ib6008c9dd4284b8c3434ffce61fb74f1db6c3d35\nCloses-Bug: #1782859\n(cherry picked from commit e2e764d9763b9367321a4c7f14b0150f67f3b08e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1434adc353e80192d9a17121b57831d273b30633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1434adc353e80192d9a17121b57831d273b30633"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
