)]}'
{"id":"openstack%2Fmanila~749172","triplet_id":"openstack%2Fmanila~master~I5b750217e019b7bc5281006f63feba8fa06db534","project":"openstack/manila","branch":"master","topic":"bug/1881098","hashtags":[],"change_id":"I5b750217e019b7bc5281006f63feba8fa06db534","subject":"Modify share groups w/ manila-manage\u0027s update-host","status":"MERGED","created":"2020-09-01 02:46:30.000000000","updated":"2020-09-21 23:31:00.000000000","submitted":"2020-09-21 19:14:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749172-1600715652704-b51fe808","meta_rev_id":"e8552acc0d8e5e8a0052077c5a93096686091f5e","_number":749172,"virtual_id_number":749172,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-12 10:01:12.000000000","_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"date":"2020-09-21 18:29:18.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-11 19:16:19.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"date":"2020-09-21 15:34:12.000000000","_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-09-11 18:08:11.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-11 20:39:38.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 19:14:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-21 23:31:00.000000000","post_submit":true,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"recommended":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"all":[{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-17 15:50:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-21 10:50:37.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":2,"date":"2020-09-11 20:23:02.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-17 15:37:35.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-21 10:50:37.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 18:08:11.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-11 19:16:19.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-11 20:23:02.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-09-11 20:39:38.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-12 10:01:12.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-17 15:37:35.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-09-17 15:50:50.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2020-09-21 10:50:37.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-09-21 18:29:18.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-21 19:14:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-21 23:31:00.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"}],"messages":[{"id":"55fa5d8937d390fa658a896e9e06dc058adff5b7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-01 02:46:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60dde58fb84d42cfbb9e6cdc5ba13c51d226c465","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-01 03:45:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/749172/1/check/DellEMC_Unity_Manila_dhss_true/be28d03/DellEMC_Unity_Manila_dhss_true/339 : FAILURE in 55m 50s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/749172/1/check/DellEMC_Unity_Manila_dhss_false/55b4340/DellEMC_Unity_Manila_dhss_false/321 : SUCCESS in 59m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"30654ce30ea886860e4780a01e8cae41ce01d19b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-01 03:52:25.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-72-749172-1/ns5_manila : SUCCESS in 1h 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"27ee02c9b9d5dc6139feb427fbd3dd93b84a74f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 05:07:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef74aa5383c4cc8a8510eee8a1aa6e7 : SUCCESS in 8m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/078f5d7f3d2c49a3aa1715dac031acf8 : SUCCESS in 12m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fefbdf0f9434f7b902ee4352b15e150 : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1652d463554f4c89811fe7e1c5199a77 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/353f34c0aa514e5f82c73c115404c771 : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/345fbb6267f24f5fa2a76291f84b99c2 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a9924eb8dcc40caad18a316ff95d565 : SUCCESS in 4m 39s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6ba1cd92a6f44e64ae258cf97e196da6 : SUCCESS in 4m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1638e166f19e4acda11c937421e635c1 : FAILURE in 4m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e030eb312a5845268a166427bb1d6dd3 : SUCCESS in 31m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/61e10ee6a8c24c65979bcbb0d72e6778 : SUCCESS in 32m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d3e15df67cfd470483aeb256185092c5 : SUCCESS in 1h 15m 43s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/fabb424c3c38486b9dd0d8bfa517fff4 : FAILURE in 1h 35m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/436ec479c83441158bda3c1dc9a7e136 : FAILURE in 52m 23s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/540e41b8d20146f28859424c542deb62 : SUCCESS in 46m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8b3d20dafab643f5ab54d8ac7c225e3f : FAILURE in 2h 14m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/9f043d77875148d38770f74cc3c7e079 : FAILURE in 29m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5cfb7b044f1242d7adccf0702f90b308 : FAILURE in 16m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aed790cdecab4a1faecd73ca4c48c930 : SUCCESS in 1h 15m 52s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d83515f4249f42008aebcaec482a9f0e : SUCCESS in 1h 02m 58s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a055eb8cc07b48e4baa42450bc79b6ef : SUCCESS in 40m 22s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ae928c712256426fae0d3d7cf45a66df : SUCCESS in 40m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ce95eae3ded6be1bfc0f15cf30758a6826a7090","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-01 05:15:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/72/749172/1/check/DellEMC_VNX_Manila/b269f23/DellEMC_VNX_Manila/192 : FAILURE in 8m 31s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"bd58b8feac37587adc73fb9efc98bac932bbfddf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-01 05:32:52.000000000","message":"Patch Set 1:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-72-749172-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"78ec29c0a14f58513b53a012270479e77416f708","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-01 06:07:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/749172/1/check/DellEMC_Unity_Manila_dhss_true/dc930c3/DellEMC_Unity_Manila_dhss_true/341 : FAILURE in 59m 51s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/749172/1/check/DellEMC_Unity_Manila_dhss_false/016a3e8/DellEMC_Unity_Manila_dhss_false/323 : SUCCESS in 54m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"168ae853fe08c1b59dea8cfecb2326cc80fc544a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-09-01 07:08:54.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/72/749172/1/check/DellEMC_PowerMAX_Manila/2fb79ac/DellEMC_PowerMAX_Manila/265 : FAILURE in 3h 50m 50s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"934e5c1f7343b409012bb0b959523513aea8c05e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-01 16:11:48.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-72-749172-1/ns5_manila : SUCCESS in 1h 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"788f26afb3436a0f298d84982565bd749911b3b0","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2020-09-10 16:07:17.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks. It modifies an existing release note, so it should best be merged in the same release like https://review.opendev.org/#/c/731486/, right?","accounts_in_message":[],"_revision_number":1},{"id":"dc72d9bde5af3a5601981547d589bde20b8d5fbe","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-10 16:53:08.000000000","message":"Patch Set 1:\n\n\u003e Thanks. It modifies an existing release note, so it should best be merged in the same release like https://review.opendev.org/#/c/731486/, right?\n\nyes... Although we can wait until RC for this one..","accounts_in_message":[],"_revision_number":1},{"id":"bbab47bcdeb57030d8833d079d780e2fa75e0c88","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-11 17:00:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0770c253dc912400a7e9d6e27c02cb780c6fb82e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-11 17:01:01.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a9b786f55b99b8fc16853ad50cd56482d5edcb08","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-11 17:01:37.000000000","message":"Patch Set 3:\n\n\u003e Thanks. It modifies an existing release note, so it should best be merged in the same release like https://review.opendev.org/#/c/731486/, right?\n\nAdded a new bug tracker, so we can track this for RC.","accounts_in_message":[],"_revision_number":3},{"id":"6291a0fe7a332f43942dd8e93b256cf32501c70c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-11 18:08:11.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-72-749172-3/ns5_manila : SUCCESS in 1h 6m 53s","accounts_in_message":[],"_revision_number":3},{"id":"55255d5d06311ce2491bda98c2547c49461e2566","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-09-11 18:18:20.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-72-749172-2/ns5_manila : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":2},{"id":"918ced1a2843d385869b6bf6ee96367a985a1379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 19:06:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80f53a54efcf4acb8496f4aa2858d5c5 : SUCCESS in 16m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd036133fb8443b59bc03e83143e16c6 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a43ecc1ee4a4731be9265066fe3fa4d : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a2ec4eb01e431396c74f05bdb037d1 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/636e1307ea8741898b728a07b511b51b : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1e3f8fb44f7455892bdce71c107fb49 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c950ca3369b4922bf8f860332839ef3 : SUCCESS in 6m 36s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b792d96034e2459ca744068cf752fcdc : SUCCESS in 5m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7c0da61b0c894be4ba5bbd65305cad4e : FAILURE in 5m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/919c09bb3ba64a08aaebc683a0e868c4 : SUCCESS in 32m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/91ee3554b8ad43afa957a4139cf10765 : SUCCESS in 32m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b8722a798a3f4f5397886280d9793248 : SUCCESS in 58m 47s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a858823d63864e1389254bbe7e89690b : SUCCESS in 1h 12m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2abee55588b54ac3b80f5c8442421d16 : FAILURE in 1h 01m 50s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a9d882dce7a74a6cb725ede70f5f2e78 : SUCCESS in 43m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f8a7c60447c448eb97ba0141fff05aff : FAILURE in 1h 59m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/4073e9e5d9bb4812b5e43e192841eb23 : FAILURE in 28m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bbe3260eec354e90bb01668fafc1a403 : FAILURE in 30m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3624837502104875ac81f68cdda1f81a : FAILURE in 1h 02m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/08edebefc3a446a3ba46bccc9c0aa8e6 : FAILURE in 12m 30s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1c635aca5a1043ba9bfebd9b52a33f5b : SUCCESS in 44m 44s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/11cf0a45df2a4d27b6ba469975bb50ec : SUCCESS in 42m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a8800d889fc719fcad65024d61867dfdbceae0c4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-09-11 19:16:19.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/72/749172/3/check/DellEMC_VNX_Manila/45b965c/DellEMC_VNX_Manila/221 : FAILURE in 9m 09s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"23ba76efacee3d29e666f6c2f1eb2f50195b9e8f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-11 19:32:21.000000000","message":"Patch Set 3:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-72-749172-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"738ec887102251f03de23a50f568f6867ba3a7b9","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-09-11 20:23:02.000000000","message":"Patch Set 3: Code-Review+2\n\nLooks good Goutham!","accounts_in_message":[],"_revision_number":3},{"id":"a838b79d1c222f490210be536135a90c9adb5160","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-09-11 20:39:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/72/749172/3/upstream-check/manila-cDOT-ss/c7be996/ : SUCCESS in 1h 26m 50s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/72/749172/3/upstream-check/manila-cDOT-no-ss/e06d659/ : FAILURE in 1h 32m 18s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"dfe570ee1bd8ed3cf49ed2cad0534326d80f6e77","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-09-12 10:01:12.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/72/749172/3/check/DellEMC_PowerMAX_Manila/b13beb5/DellEMC_PowerMAX_Manila/314 : FAILURE in 3h 50m 34s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"56284409ce7f36cd4e0dc7d5261dd2772d2afd63","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2020-09-17 15:50:50.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"a13101052ef2baa47e40898c41646803cd3e54e6","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-09-21 10:50:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks, Goutham!","accounts_in_message":[],"_revision_number":3},{"id":"8b03b358f909b4ef798ff27e64cdf66642e0062a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 10:50:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8542c7000dbeb0ed3d118cc96ff3706bfbe82d46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 12:15:18.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/364dc268b7be4ccbb879684dd052e076 : SUCCESS in 17m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bebcd44a79ab45ffad6a01dc4d0f5470 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95e5f919cd6247ceb9637f344dbdf220 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f54b32f1dd764626b9bb0f851a932825 : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3ee48cb1a374c7088530e4b455c57c2 : FAILURE in 18m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0e80da7bb654ca082c08b0fc360e64c : SUCCESS in 5m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/41a02e8d421146b5833b8b990b54bca1 : SUCCESS in 26m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/096bcd0919314f14bb5155df48e49933 : SUCCESS in 39m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2015960c0fd541e9b3ca0f6f48d17a5e : SUCCESS in 1h 15m 12s","accounts_in_message":[],"_revision_number":3},{"id":"14afe151ebd58462e6c32651579f86c52e39853d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-09-21 15:34:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d08e688d33b782ed0beef2c599263db3075e877d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 17:49:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d0fa1d9de9c445a97778f26e7ae43ef : SUCCESS in 17m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd972dac5fc34cb1b8d6acd040f2136f : SUCCESS in 12m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfc6350780f849469de58ddf45c7d1d3 : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c080a96ae8b646dfaa77c061a2b5b183 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b621a98ea1064aaba0ca8b1e8ac9aa22 : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9e54898804a411baf2750e033c16fdf : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26eb25a19d59489a89edbd465dd83739 : SUCCESS in 6m 19s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/50523159a8de43f4838a51232e44d6dc : SUCCESS in 5m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ef797d4c3e6b46749d69a247a869a9c7 : FAILURE in 10m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a1108f591ae843f78cceafb7f5af8d1d : SUCCESS in 33m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/91a5844d26bc4da3b18a88b953f2d9a7 : SUCCESS in 33m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c0589d6543e24665983ba2b53874176c : SUCCESS in 1h 01m 40s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3c40720b2239436f8e2a5f31ad1a4255 : SUCCESS in 52m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/28bd69aa3d5e49b8af5449ea33cebb7d : FAILURE in 52m 59s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0a56c15c39c04a5d9e1d4f5ceca7ebcc : SUCCESS in 46m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ec00d6348eee4845aca93d4a6f10cb7d : FAILURE in 2h 04m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d63cbf01bbfd40a088fb79a5577449de : FAILURE in 19m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8d290f171b2c48738c51db62f91bbb35 : FAILURE in 26m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/91818588890c44799394bf61a16f2e6b : SUCCESS in 58m 46s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/176eb0613dfa4822b0e03c9096624ac5 : SUCCESS in 55m 14s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1ae2f072b9db457fb3b39507b64416b9 : SUCCESS in 32m 15s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7787f922eca34164a19f593bcdfe2868 : SUCCESS in 39m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d298e8f271cffcc5dd6af9f1c51fa3eedb90f818","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 17:49:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"39afdbaace1183b1828821352ec152aef3760831","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-21 18:29:18.000000000","message":"Patch Set 3:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-72-749172-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"198a3cea7eb4c02b76f011fa3b1cb76bdec5876f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 19:14:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48169e9b68a84b0c9ca4eb3da14a1a59 : SUCCESS in 16m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d83ff125180749bc9aaeb2039059cf7d : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78cef9cb1ce941968135245bade0988f : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6eb3b5ac7bcf49e282a371f0fa8415b2 : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20ddbdf581c1401693a72352a483d141 : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d948eeef5a24aed8d6716029337a8da : SUCCESS in 7m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/87a9dc88f2b741788d36fbccfbe02f8f : SUCCESS in 45m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2e9a9aa1129b4e6e905897f88fbfde7f : SUCCESS in 38m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4f73b56b248d4f189cd24b7e46b50961 : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":3},{"id":"1ab4f9ca78fdc7553a3a403c4112ea735e9321ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 19:14:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5f446a019f8f2a6d229200e9e3f629a46bcd6d4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 19:16:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b753681acd4e4a9347899f1d591ab2 : SUCCESS in 2m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f1adfb4c2f644969b36ca5c2649b175 : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":3},{"id":"9d147595901c28e3ff8179f41f7e0fa407af1359","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-09-21 23:31:00.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/749172/3/check/DellEMC_Unity_Manila_dhss_true/55e2b2e/DellEMC_Unity_Manila_dhss_true/457 : FAILURE in 1h 55m 02s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/749172/3/check/DellEMC_Unity_Manila_dhss_false/14a111a/DellEMC_Unity_Manila_dhss_false/438 : SUCCESS in 53m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"370b15b6c61ae997dd61b1e69d2c4c305a184d5d","revisions":{"a3a039674fd231e37e74b3c3fdad40e3bae8d776":{"kind":"REWORK","_number":1,"created":"2020-09-01 02:46:30.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/72/749172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/749172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/749172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/749172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/749172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/749172/1"}}},"commit":{"parents":[{"commit":"365cfa35164e411ae88d7404394b5f073c64d126","subject":"add share server update to manila-manage share update_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/365cfa35164e411ae88d7404394b5f073c64d126"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-09-01 02:46:01.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-09-01 02:46:01.000000000","tz":-480},"subject":"Modify share groups w/ manila-manage\u0027s update-host","message":"Modify share groups w/ manila-manage\u0027s update-host\n\nWe can also condense this code a bit to allow\nadding more resources who have a \"host\" attribute.\n\nChange-Id: I5b750217e019b7bc5281006f63feba8fa06db534\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a3a039674fd231e37e74b3c3fdad40e3bae8d776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a3a039674fd231e37e74b3c3fdad40e3bae8d776"}]},"branch":"refs/heads/master"},"ecb52df306a33b8641a781dfa650c1b2433e626f":{"kind":"REWORK","_number":2,"created":"2020-09-11 17:00:03.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/72/749172/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/749172/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/749172/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/749172/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/749172/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/749172/2"}}},"commit":{"parents":[{"commit":"4036ff32bd503f5328070e75d2233ae943d2746a","subject":"Merge \"Add share server migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4036ff32bd503f5328070e75d2233ae943d2746a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-09-01 02:46:01.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-09-11 16:59:54.000000000","tz":-420},"subject":"Modify share groups w/ manila-manage\u0027s update-host","message":"Modify share groups w/ manila-manage\u0027s update-host\n\nWe can also condense this code a bit to allow\nadding more resources who have a \"host\" attribute.\n\nChange-Id: I5b750217e019b7bc5281006f63feba8fa06db534\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ecb52df306a33b8641a781dfa650c1b2433e626f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ecb52df306a33b8641a781dfa650c1b2433e626f"}]},"branch":"refs/heads/master"},"370b15b6c61ae997dd61b1e69d2c4c305a184d5d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-09-11 17:01:01.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/72/749172/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/749172/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/749172/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/749172/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/749172/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/749172/3"}}},"commit":{"parents":[{"commit":"4036ff32bd503f5328070e75d2233ae943d2746a","subject":"Merge \"Add share server migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4036ff32bd503f5328070e75d2233ae943d2746a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-09-01 02:46:01.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-09-11 17:00:41.000000000","tz":-420},"subject":"Modify share groups w/ manila-manage\u0027s update-host","message":"Modify share groups w/ manila-manage\u0027s update-host\n\nWe can also condense this code a bit to allow\nadding more resources who have a \"host\" attribute.\n\nCloses-Bug: #1895323\nChange-Id: I5b750217e019b7bc5281006f63feba8fa06db534\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/370b15b6c61ae997dd61b1e69d2c4c305a184d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/370b15b6c61ae997dd61b1e69d2c4c305a184d5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
