)]}'
{"id":"openstack%2Fmanila~756993","triplet_id":"openstack%2Fmanila~master~Id1647b55af476ab4559759a27498ede353dd10ba","project":"openstack/manila","branch":"master","topic":"disable_M325","hashtags":[],"change_id":"Id1647b55af476ab4559759a27498ede353dd10ba","subject":"disable M325","status":"MERGED","created":"2020-10-09 03:40:13.000000000","updated":"2020-10-29 20:38:31.000000000","submitted":"2020-10-29 20:36:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756993-1604003778491-f746aaa4","meta_rev_id":"9be0e4cdce93006b30d0e7cb5f2b3b341352dbae","_number":756993,"virtual_id_number":756993,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-10-14 07:19:53.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"date":"2020-10-14 06:08:43.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"date":"2020-10-14 07:40:36.000000000","_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"date":"2020-10-14 09:26:15.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-29 20:36:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-29 20:22:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2020-10-14 16:34:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-29 09:44:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-29 20:22:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 03:54:59.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-10-09 11:47:12.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 07:19:53.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 07:40:36.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 09:26:15.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 16:34:39.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-10-29 09:44:01.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-10-29 20:22:20.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-10-29 20:36:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cd04a97945ffc10dc2cc6e619d3ba7fb76874e2","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-10-09 03:40:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb74bf25731a92ccae3e487422ba80c869490dea","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-10-09 03:40:21.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/93/756993/1/check/DellEMC_PowerMAX_Manila/9641f3e/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"be2926a7a09609ac687790539d2c65c21eb209c6","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-10-09 03:45:33.000000000","message":"Topic set to manila_remove_six","accounts_in_message":[],"_revision_number":1},{"id":"c8ce7f8d860301bfcf3b9b8ad97d05c1b0a66967","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-10-09 03:54:59.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/93/756993/1/check/DellEMC_Unity_Manila_dhss_true/12fdfd4/DellEMC_Unity_Manila_dhss_true/525 : FAILURE in 1m 17s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/93/756993/1/check/DellEMC_Unity_Manila_dhss_false/d8b59c6/DellEMC_Unity_Manila_dhss_false/508 : FAILURE in 8m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"66746e1b48f6d2e4c76c0ab2121abf458888403d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:59:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bfd76f63dfd4159b6e3d57126d7e582 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d5e7ff7c614186a989f85bab6686a5 : SUCCESS in 5m 00s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2884da5e8efb470aa941e3873144994b : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9a456f96390635e4f2b33e11b57b8812f84c2a9b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-10-09 04:00:02.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/93/756993/1/check/DellEMC_PowerMAX_Manila/7730728/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"3e0799858fd97656020a81e5ef8cdbfdfd0dcbb5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-10-09 04:26:01.000000000","message":"Patch Set 1:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-93-756993-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"d85f99937608df6abd78425609dbe3d533529ef0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-09 05:10:33.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-93-756993-1/ns5_manila : SUCCESS in 1h 29m 54s","accounts_in_message":[],"_revision_number":1},{"id":"5032628c3faa5641e94ede976065b6fbe1c38383","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-10-09 10:45:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6fb11dee1aaf8c4b1513d711fc2493b6e16ad339","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-10-09 10:46:06.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/93/756993/2/check/DellEMC_PowerMAX_Manila/18e54b7/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"a8c56a2277c725fc9c30b03bb8aae0f9fbf20fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 11:07:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/878863d5700247e1a4c55413c8b16a17 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/071fd0aacc9749fab8b295591c0e6aad : SUCCESS in 9m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a9a4ef87b89437eb8791674910af6ab : SUCCESS in 15m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5eed7a34d454dd6b7ba81fd51921f95 : FAILURE in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ea11f5a8b9f4a71bae9123e3fa48cfd : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c635985b18e5490b8d160a4c0e136865 : FAILURE in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31245699828c422192c1fbaf9f45d148 : FAILURE in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2d31651ad8e40848c5e8b9f6f91ebd1 : SUCCESS in 5m 26s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9ff68c5e562a4b0687ca469177a0434a : SUCCESS in 4m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0451cd5876b848939e1e8aeb8f4a95b7 : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f98c96458aa6d8d91bb464929aa703ffb560606","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-10-09 11:27:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2aa6a04de2bb03adde401087a0965cb193b2c179","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-10-09 11:27:57.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/93/756993/3/check/DellEMC_PowerMAX_Manila/dec1eed/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"464d10ead0f1b7255841aba61c4464abfdfcc26b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 11:47:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/84a54b2fd1c84c9f845da192885c89de : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa55c3863e2340c8a2af087e041593de : SUCCESS in 9m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/329b6550c7074b28bbb3f9b0f70d84f3 : SUCCESS in 12m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/254a0920668d4735860415ad2b69ce58 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad0297f68dd54fdf92efe84535d51f54 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ac7fdfbd2a348f58490bf40b9918c9d : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7cdf666362a4cf691cf7b89f3f3560f : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b41c9f13af4b4ccca7d0fdb0f86d4aab : SUCCESS in 5m 16s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/833a5cd808a142728be6dc05f4213bdb : SUCCESS in 5m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b493db716d59433e9d39eb6fc04cb2fe : SUCCESS in 8m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"53e181a0da6656510d54c87d47355f0d91a204de","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-10-09 11:47:12.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/93/756993/3/check/DellEMC_PowerMAX_Manila/0474fd7/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"bd0ced33566701c52da2c9ff19812c4727ec2ecf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-10-09 12:12:37.000000000","message":"Patch Set 3:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-93-756993-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"fe5717998aa7433edd1c82a6cff7775dbbaa2a55","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-09 12:38:04.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-93-756993-2/ns5_manila : SUCCESS in 1h 51m 38s","accounts_in_message":[],"_revision_number":2},{"id":"508c7511b9d6aa1fff2dbfde2f4609b652a3b853","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-09 13:22:46.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-93-756993-3/ns5_manila : FAILURE in 44m 26s","accounts_in_message":[],"_revision_number":3},{"id":"a0b1317023d43299e56a34487843b943c7076947","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-09 13:36:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/93/756993/3/upstream-check/manila-cDOT-ss/2b9a537/ : FAILURE in 1h 49m 05s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/93/756993/3/upstream-check/manila-cDOT-no-ss/c97540f/ : SUCCESS in 1h 47m 17s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"8683945c19dd94a41319c4fb01fb03945b0564b8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-09 15:53:16.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-93-756993-3/ns5_manila : FAILURE in 1h 37m 13s","accounts_in_message":[],"_revision_number":3},{"id":"1a6062b43feae75ce3cb171417bfa0771f4b52da","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-09 18:03:18.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-93-756993-3/ns5_manila : FAILURE in 57m 17s","accounts_in_message":[],"_revision_number":3},{"id":"3812a2b701d80095c530d8f281c49d6022293d04","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-09 21:02:03.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-93-756993-3/ns5_manila : ABORTED in 2m 14s","accounts_in_message":[],"_revision_number":3},{"id":"cdfedb47f78aec46003c00d18f8bc241117cea98","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-10 03:15:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/93/756993/3/upstream-check/manila-cDOT-ss/ac70c6e/ : SUCCESS in 2h 04m 04s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/93/756993/3/upstream-check/manila-cDOT-no-ss/4165601/ : SUCCESS in 2h 06m 40s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"b7418e38d71fe51924177b09f8f282dbadc76157","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-12 00:38:53.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-93-756993-3/ns5_manila : SUCCESS in 2h 7m 57s","accounts_in_message":[],"_revision_number":3},{"id":"6ff07a2c594ee14196055b2c6b596db1d35137cb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-12 00:42:14.000000000","message":"Patch Set 3:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-93-756993-3/ns5_manila : SUCCESS in 2h 1m 38s","accounts_in_message":[],"_revision_number":3},{"id":"eb0a336c3f4b51b5bab25449db00c8fda4e8bf6b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-10-13 07:03:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nThanks, some corrections noted inline","accounts_in_message":[],"_revision_number":3},{"id":"5c61ca155287777b788b1d3e35e0db8bbb7bad42","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-10-14 06:07:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1edf9e44aab4ed774a1a7afb662663b4803be244","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-10-14 06:08:43.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nhi,Goutham Pacha Ravi.\nthanks for your review.","accounts_in_message":[],"_revision_number":4},{"id":"eae11b79356414529b29dab937b7aa0ec6c377e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 06:54:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0bb97fe84a294d7da5beda8edde3af32 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce68bbf1fc9e4ead8f8e0c7d5df5bf34 : SUCCESS in 10m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78c4d7efba884dd29cb7d3b3d4e43cc0 : FAILURE in 34m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfd70634073c4cada1ba31cfda799726 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6e0d3a94176405991d1541ff093db66 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6bb76bb733c4eef8781bde8386c47b8 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0680b20d009c4e8e9c368842bf7bc076 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39657cf860b2425e86b6d1fe175c636e : SUCCESS in 5m 52s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/82e5ea6e5b3c4345a74dc861b2004bd6 : SUCCESS in 5m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0eb376efafbb4c6184ac95abb870e586 : SUCCESS in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b2239f838fb8924703540c0c4fee3fad96c9ab18","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-10-14 07:19:53.000000000","message":"Patch Set 4:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-93-756993-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"88d8d0c2f10c4ad08b085bcd022630286c6480f8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-14 07:40:36.000000000","message":"Patch Set 4:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-93-756993-4/ns5_manila : SUCCESS in 1h 32m 27s","accounts_in_message":[],"_revision_number":4},{"id":"5243f74663ad4562de51d8db8608d5709bc6a7f4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-14 09:26:15.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/93/756993/4/upstream-check/manila-cDOT-ss/a38c87e/ : FAILURE in 2m 05s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/93/756993/4/upstream-check/manila-cDOT-no-ss/5a1bd2e/ : FAILURE in 30m 16s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"44f965d69e1d1d418b6ad0f802acd2663ab66f86","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-10-14 16:34:39.000000000","message":"Patch Set 4: Code-Review+2\n\nGood stuff, thanks Haixin","accounts_in_message":[],"_revision_number":4},{"id":"16e80d1973903297f05e18eb61e33b3ebc22b50b","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-10-15 09:59:19.000000000","message":"Topic changed from manila_remove_six to disable_M325","accounts_in_message":[],"_revision_number":4},{"id":"384e4f8a97ed2ea43dbcc0211c4aea533f2f99d9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-10-29 09:44:01.000000000","message":"Patch Set 4: Code-Review+1\n\nGood catch","accounts_in_message":[],"_revision_number":4},{"id":"ea07b211af2efaa04c9588f82af1ee2b1ee97951","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-10-29 20:22:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThank haixin, LGTM!","accounts_in_message":[],"_revision_number":4},{"id":"e9b0deace8d8af9034bc4a71f42aa91c630238ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 20:22:59.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7c87b4483c8c8759504553954a7ea336e21921e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 20:36:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"277ee9dbc31c68aa4b3a65c1988f92a2bf60015a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 20:36:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/055857a503f045a0aa08e75a3bb3bbed : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/919c42b98e2145988042f6a848f2905b : SUCCESS in 10m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11454a7804b642269e5b6a59a9d85eab : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85d59b33bb0544379abb13e8b5130aed : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a3d93abc3a74af4a29198b99695560e : SUCCESS in 9m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b22d32bdafb046d09c5ccd4582750d08 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/254e089a135645c68e4b5256a8d7fd44 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":4},{"id":"9be0e4cdce93006b30d0e7cb5f2b3b341352dbae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 20:38:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/cd6e3814395542438fa8f34615a236a6 : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a76a991361f6480cbd9295f5a5f44f80 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24ba3048d3034766bfdc7a1c174905d4 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cc33a4b23e3d702ac7993a2202aa347e1efe3f3b","revisions":{"a337e8e591cf20727fc9e0097b74f66444117a5e":{"kind":"REWORK","_number":1,"created":"2020-10-09 03:40:13.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/93/756993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/93/756993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/756993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/756993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/756993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/93/756993/1"}}},"commit":{"parents":[{"commit":"52c926e32c3e2e119606774fc90b69b98a8edd30","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/52c926e32c3e2e119606774fc90b69b98a8edd30"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-10-09 03:39:31.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-10-09 03:39:31.000000000","tz":480},"subject":"disable M325","message":"disable M325\n\nChange-Id: Id1647b55af476ab4559759a27498ede353dd10ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a337e8e591cf20727fc9e0097b74f66444117a5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a337e8e591cf20727fc9e0097b74f66444117a5e"}]},"branch":"refs/heads/master"},"a0444bf20dd6d1164ef58836eac743c7f6432fb6":{"kind":"REWORK","_number":2,"created":"2020-10-09 10:45:58.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/93/756993/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/93/756993/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/756993/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/756993/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/756993/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/93/756993/2"}}},"commit":{"parents":[{"commit":"52c926e32c3e2e119606774fc90b69b98a8edd30","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/52c926e32c3e2e119606774fc90b69b98a8edd30"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-10-09 03:39:31.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-10-09 10:45:32.000000000","tz":480},"subject":"disable M325","message":"disable M325\n\nChange-Id: Id1647b55af476ab4559759a27498ede353dd10ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a0444bf20dd6d1164ef58836eac743c7f6432fb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a0444bf20dd6d1164ef58836eac743c7f6432fb6"}]},"branch":"refs/heads/master"},"0921b59438a5277f1f3d6bfcb75b73ed512c0526":{"kind":"REWORK","_number":3,"created":"2020-10-09 11:27:47.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/93/756993/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/93/756993/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/756993/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/756993/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/756993/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/93/756993/3"}}},"commit":{"parents":[{"commit":"52c926e32c3e2e119606774fc90b69b98a8edd30","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/52c926e32c3e2e119606774fc90b69b98a8edd30"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-10-09 03:39:31.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-10-09 11:27:16.000000000","tz":480},"subject":"disable M325","message":"disable M325\n\nChange-Id: Id1647b55af476ab4559759a27498ede353dd10ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0921b59438a5277f1f3d6bfcb75b73ed512c0526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0921b59438a5277f1f3d6bfcb75b73ed512c0526"}]},"branch":"refs/heads/master"},"cc33a4b23e3d702ac7993a2202aa347e1efe3f3b":{"kind":"REWORK","_number":4,"created":"2020-10-14 06:07:36.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/93/756993/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/93/756993/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/756993/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/756993/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/756993/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/93/756993/4"}}},"commit":{"parents":[{"commit":"52c926e32c3e2e119606774fc90b69b98a8edd30","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/52c926e32c3e2e119606774fc90b69b98a8edd30"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-10-09 03:39:31.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-10-14 06:05:30.000000000","tz":480},"subject":"disable M325","message":"disable M325\n\nSince we don\u0027t support python2.x anymore, we don\u0027t\nneed to use six.text_type.\n\nChange-Id: Id1647b55af476ab4559759a27498ede353dd10ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cc33a4b23e3d702ac7993a2202aa347e1efe3f3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cc33a4b23e3d702ac7993a2202aa347e1efe3f3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
