)]}'
{"id":"openstack%2Fmanila~758674","triplet_id":"openstack%2Fmanila~master~I67036a65da9255694a00a9c8d56cfdefbdf23c05","project":"openstack/manila","branch":"master","topic":"bug/1898791","hashtags":[],"change_id":"I67036a65da9255694a00a9c8d56cfdefbdf23c05","subject":"Remove backend deprecated variables","status":"MERGED","created":"2020-10-18 15:33:25.000000000","updated":"2021-01-27 23:12:38.000000000","submitted":"2021-01-27 22:37:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"758674-bug/1898791","meta_rev_id":"618b7f783343757e5ed2b7d2809f215c348ef077","_number":758674,"virtual_id_number":758674,"owner":{"_account_id":32542,"name":"Lívia Cavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","username":"liviacavalcanti"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-27 22:36:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32542,"name":"Lívia Cavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","username":"liviacavalcanti"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-01-27 23:12:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-27 12:22:33.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32542,"name":"Lívia Cavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","username":"liviacavalcanti"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-27 12:22:33.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32542,"name":"Lívia Cavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","username":"liviacavalcanti"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32542,"name":"Lívia Cavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","username":"liviacavalcanti"}],"CC":[{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-18 18:19:44.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-19 22:31:51.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-10-22 21:00:50.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-10-22 22:22:08.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-28 22:27:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-28 22:49:03.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-29 00:05:33.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-29 00:35:58.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 16:10:43.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-12-24 01:38:52.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2021-01-06 01:05:17.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-15 21:10:21.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-01-16 04:32:06.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-01-18 00:51:56.000000000","updated_by":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"reviewer":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"state":"REVIEWER"}],"messages":[{"id":"8494e7dc1049d49de81dadaa9665593b9d72847b","author":{"_account_id":32542,"name":"Lívia Cavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","username":"liviacavalcanti"},"date":"2020-10-18 15:33:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16489ce2371c253afb3ef0b91c3e63afce045496","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-10-18 15:53:36.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/758674/1/check/DellEMC_Unity_Manila_dhss_true/48727ab/DellEMC_Unity_Manila_dhss_true/539 : FAILURE in 20m 00s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/758674/1/check/DellEMC_Unity_Manila_dhss_false/d15d814/DellEMC_Unity_Manila_dhss_false/516 : FAILURE in 19m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"506901a64e48489d853246c3d42d18695d032c77","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-18 17:16:10.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-74-758674-1/ns5_manila : SUCCESS in 1h 42m 6s","accounts_in_message":[],"_revision_number":1},{"id":"26f320c9291088ceaec5c278687ce737a66fc509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 17:54:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c79e056f83864b4cb35c613137300066 : SUCCESS in 10m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82ab662097f64014aff016a47df26387 : SUCCESS in 14m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/143aa836a3e34263934ef3ad34618e05 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fbda5865ecf4d54b357ef05d1a08f3c : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3304c551f4a54e6ab27acf2d76e27db7 : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89f74cba55c84cceac8a831d68bdad84 : SUCCESS in 8m 49s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d53e7b85009a41398f5344c0e86a49aa : SUCCESS in 5m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9ec46efaadae49048299f04d0fc734ff : SUCCESS in 5m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0f8e2e9be57f4373bdf4ce20f3c7681d : SUCCESS in 37m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fcaf5572964c4817bf575e3ab97c801b : SUCCESS in 43m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/00480aa4cb504fe6a47ad2afa159cfce : SUCCESS in 1h 06m 52s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/08fc76b5b7f542d78954d28a9ee5f78f : SUCCESS in 59m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0edce2136a0744689242bd541d0c5fb3 : FAILURE in 1h 16m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8a2c82f65ecb4007b124286258c7c35c : SUCCESS in 47m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b5e5e7b9f7444d34920c958e81e1b106 : FAILURE in 2h 14m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/cf39892c24a148bda44852c90bf0f226 : FAILURE in 22m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/713d45b284f24c6382ebf5bdf20a9ea0 : FAILURE in 19m 00s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/44e219e4947f4ccdac3a95a907ddb662 : FAILURE in 1h 18m 29s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4ed1af80ef114b469e42a60ea6dff14d : SUCCESS in 58m 05s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/86bef122cee94f79a5d94ac380a61ecb : SUCCESS in 42m 52s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/94440ac85dbe4ecba675210db91d6dd2 : SUCCESS in 40m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c08f8ffa2e772d794afe552d6dc3206f8bc0b6e4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-10-18 17:54:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/74/758674/1/check/DellEMC_VNX_Manila/74f19c6/DellEMC_VNX_Manila/None : RETRY_LIMIT\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"e97d209fff686001d4b779c1e02bbd44d1a64ba1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-10-18 18:14:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/758674/1/check/DellEMC_Unity_Manila_dhss_true/2d64e11/DellEMC_Unity_Manila_dhss_true/543 : FAILURE in 18m 46s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/758674/1/check/DellEMC_Unity_Manila_dhss_false/4a8610a/DellEMC_Unity_Manila_dhss_false/520 : FAILURE in 20m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"cb3293c8e31c3681ab70d7d4c6de41f39ee168db","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-10-18 18:19:44.000000000","message":"Patch Set 1:\n\n* manila-quobyteci-dsvm-share http://78.46.57.153:8081/refs-changes-74-758674-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"7bb6858cc99904a7a56b8e0ed7109c7f08b895f7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-18 19:42:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/74/758674/1/upstream-check/manila-cDOT-ss/04b0266/ : FAILURE in 1h 34m 29s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/74/758674/1/upstream-check/manila-cDOT-no-ss/3b134f2/ : FAILURE in 1h 48m 15s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"4aad9c3c608466b61bd86dc646d8b9c77166dc23","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-10-19 22:31:51.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"04354c638cc99a3d2ff9308f2d7002a84687431b","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-10-19 22:31:53.000000000","message":"Patch Set 1:\n\nHi Livia! Thanks for working on this. I see there are three different patch sets with the same subject. Are they the same fix? In case they are not, can you remove the changes that shouldn\u0027t be for review? I\u0027d advise you to check this doc [0] to learn on the contribution workflow works.\n\nThanks!\n\n[0] https://docs.opendev.org/opendev/infra-manual/latest/developers.html#development-workflow","accounts_in_message":[],"_revision_number":1},{"id":"af83bae11f5e6a449c4cde67a18b5d2431f02d73","author":{"_account_id":32542,"name":"Lívia Cavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","username":"liviacavalcanti"},"date":"2020-10-22 20:36:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"297fcb5598d9ed1b7dff7040c3962026b368ce08","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-10-22 21:00:50.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/758674/2/check/DellEMC_Unity_Manila_dhss_true/1700bd6/DellEMC_Unity_Manila_dhss_true/3 : FAILURE in 21m 19s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/758674/2/check/DellEMC_Unity_Manila_dhss_false/dc0f324/DellEMC_Unity_Manila_dhss_false/3 : FAILURE in 23m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b8c7825438538db61522a47ef02861f14b2da613","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-10-22 22:04:38.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/74/758674/2/check/DellEMC_PowerMAX_Manila/f67b795/DellEMC_PowerMAX_Manila/3 : SUCCESS in 1h 28m 12s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"6f5bdfb9aeb74854f806ea886598cc3dd248dc93","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-10-22 22:22:08.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-74-758674-2/ns5_manila : SUCCESS in 1h 45m 24s","accounts_in_message":[],"_revision_number":2},{"id":"01b6e130d99393d4e22a474651c4c379c5a7fcf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 22:26:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1b6076883240f8814d25660824dceb : SUCCESS in 11m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab41517ede2e4e4397bdecdc827e6930 : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6327cc3db1b34f5f9002e5e92a814e31 : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a094f54cbe824f6cba093bc839111a84 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c0376a3e7c949bf9f7e04a834439bb9 : SUCCESS in 10m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef6032be7f7c404781766ef41a80bb3a : FAILURE in 28m 37s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b67925453c114800b6dcc6698e585d94 : SUCCESS in 6m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b36b866ca8434ba7857ddd372c5d87df : SUCCESS in 6m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d8f489ed3eed4924870e369846b1c0fd : SUCCESS in 29m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a2729e1185454c3f947ba35a4633a5f2 : SUCCESS in 34m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bd590e1c0e684e45a5e1e60f317c5802 : SUCCESS in 1h 12m 00s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/32442e7bc8a341d2911be4efecbbdd8b : SUCCESS in 1h 03m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3cc8ccf7d9564a24a3c748a1bc994262 : FAILURE in 54m 48s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e649a789703745b3b0853adf3f08368c : SUCCESS in 41m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/67f74ba4c87b44ebba3fb5d3ab827bd1 : FAILURE in 1h 40m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/04baad676c9d42bc9c2690e8a75157c2 : FAILURE in 15m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ff29b26471c546b8a72ab7d94bafb71f : FAILURE in 16m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f1b8a3578e4f42d0ab480a6f74015bd7 : SUCCESS in 1h 04m 15s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4affd087e6954d9c817f30761fba4104 : FAILURE in 45m 23s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/777fee28efed4126a615c385276ccdde : SUCCESS in 29m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/618c50646fa94620b801803837be5726 : SUCCESS in 1h 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"43d6e9be3d7ca747638399f5d768377b0aa229c0","author":{"_account_id":32542,"name":"Lívia Cavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","username":"liviacavalcanti"},"date":"2020-10-28 19:58:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dcecff6ff2755ceff6baa52a69c3c992acfd9806","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 22:27:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/885f2ca517694d9a913414824db49aad : SUCCESS in 5m 12s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eac2f5f48be045bdbe5b8b1eeb28eb1b : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984e95dd23b1415fa3bf7120aecab3c2 : SUCCESS in 9m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/466c01a4e18242fe90e75a61449e0f54 : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13f871b3d0c04b3fbebb6e76f7d2df7b : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/679a8533a64c4d26bf8caddc4a1783ea : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/275a5f5b390f40ac9f310706c7bc2429 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3175e971be244848f3d6c3fa14d93e7 : SUCCESS in 8m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a060cb5c19204713a2fc58ca7e7cc758 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e9e6f503be24c4a9016092599b809b4 : SUCCESS in 5m 51s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d65f1c00efa54fa29855554f6c9c2e14 : SUCCESS in 8m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/355d4ac3400344439de8c00e8803fef6 : SUCCESS in 4m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8140c389ceb043439d194c515a488c90 : SUCCESS in 33m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dc7a1b6a6cd941eab51466d93bbf13cd : SUCCESS in 36m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/707ad9631dbc40378e8629a78967bb7f : SUCCESS in 1h 18m 45s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/65bc9745a9b04982866c35cf676f1cc4 : SUCCESS in 56m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2d11a98f3cf24e33954f19a5ab407d95 : FAILURE in 1h 12m 15s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0777cbfbee60491888d960b183b7b729 : SUCCESS in 1h 16m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ba9667b6042045339be64d10055b4adb : FAILURE in 2h 16m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a98dee8996604aaaaee7ee9a5ec51586 : FAILURE in 20m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/398f7e87b3f84c6eaf48dbb9a5edd687 : FAILURE in 19m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/790228511abb4b8e8e7a4095fa96c475 : SUCCESS in 1h 25m 19s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/668e846f63e54baf826d8f3df557c091 : FAILURE in 39m 01s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a4f19a23ceec4de69d8af86001cf0145 : SUCCESS in 43m 11s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/deafdf531d0844bd8b6b42f243cde612 : SUCCESS in 43m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0cd17ca89613486ad91104cd323b4bb37ccd5dbb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-10-28 22:49:03.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/74/758674/2/check/DellEMC_VNX_Manila/e799c00/DellEMC_VNX_Manila/9 : FAILURE in 20m 55s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"2f5226be88c0d679be2d3830ea3c6a042ddc92a5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-10-29 00:05:33.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/74/758674/2/check/DellEMC_PowerMAX_Manila/8811234/DellEMC_PowerMAX_Manila/16 : SUCCESS in 1h 28m 47s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"206aa223e341b6dfd7992caa9feafd2be278c3ec","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-10-29 00:35:58.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/74/758674/2/upstream-check/manila-cDOT-ss/c38ae19/ : SUCCESS in 2h 07m 38s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/74/758674/2/upstream-check/manila-cDOT-no-ss/057033d/ : FAILURE in 1h 43m 27s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"4bc2431af7609c2deac7b0532b0293dd80080f04","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-12-16 12:32:12.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)\n\nThanks Lívia, just 2 minor considerations so far. Lets run ci jobs again because the latest logs are gone.\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"42d4726a1d4994b7e487b60b20435123b6716bf0","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-12-21 11:47:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"747bf2ede6fa7da5ab8c787e7e07bc09781165f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 14:11:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e6ceef529184fa8a5bf3e331d2b63b8 : SUCCESS in 10m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7871c8e9ccc467aa46b5e6438ff5d1e : SUCCESS in 14m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/781cf4d4a52c4adba3c74a26a61bb1c4 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aacc592d3da4f6b9fd62caf86926af1 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc847240f6754f538df3c0b7abb0819b : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bea551ddad646dca0c1508addbe75e2 : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/710bbd28368b48caa928f5c3baaf6871 : SUCCESS in 7m 22s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4719b6b0b49347dc9394b788d8c7d43f : SUCCESS in 6m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9b60cdbcade445cbb46f8ff0f58c0cb6 : SUCCESS in 4m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ec004780d27d4b2395a65d703fc954bc : SUCCESS in 29m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d768b08c9b9b4db9afa69712835ffa70 : SUCCESS in 37m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/51004a02a9b6491ab5297f8d9cddb239 : SUCCESS in 1h 14m 51s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2d1ec4501ae747c3933719eeb2f02d83 : SUCCESS in 1h 11m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/61609cc3b3d54b71a4f169be744bd569 : FAILURE in 51m 25s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9f3f7711e87e4f1fb3404c5174e78802 : SUCCESS in 44m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4d81e07c35ce400bb6285f07bc58515a : FAILURE in 2h 15m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/76cc3a39d29f4d1fa59bbb25875a7b22 : FAILURE in 13m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a8bd54442b3449828b827b3ee458dca0 : FAILURE in 13m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/456414e9dcb74e07ab58e865bc4c5f79 : FAILURE in 1h 11m 17s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e3ebc37a3f25469ca5449c9630e149a9 : FAILURE in 21m 19s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/84dc90802e54413096695e68b28ee044 : FAILURE in 39m 49s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a32634787959462c85f4b096db18e560 : FAILURE in 37m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d04ba393de10c25993acae0e32439b9c4fda97ec","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-21 14:31:43.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/758674/2/check/DellEMC_Unity_Manila_dhss_true/91da69b/DellEMC_Unity_Manila_dhss_true/129 : FAILURE in 16m 55s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/758674/2/check/DellEMC_Unity_Manila_dhss_false/9a8aefa/DellEMC_Unity_Manila_dhss_false/121 : FAILURE in 19m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"03dc73ee13b19dd365925a75c6b51163e7cd6978","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-12-21 14:33:11.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/74/758674/2/check/DellEMC_VNX_Manila/ae836a4/DellEMC_VNX_Manila/87 : FAILURE in 21m 00s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"7058ca3160768eb1857811f1fcabcabb45effc6b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-12-21 16:01:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/74/758674/2/check/DellEMC_PowerMAX_Manila/94fd49e/DellEMC_PowerMAX_Manila/130 : FAILURE in 1h 49m 05s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"fee4cd6a50e5fbbdae30af5f0740504b180ece78","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-24 01:38:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"84a081af4438cd8ceecb78134be2b553499da18e","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-12-28 21:05:46.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good Lívia! Thanks for your contribution!","accounts_in_message":[],"_revision_number":2},{"id":"922e78e0b0e50274a0e84836926c0d5146cfa8ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-05 23:05:28.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"78bee3c4932c24058ab0ba486b6ec4939cea8fd1","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-05 23:09:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6bb7a04c2e271d2c47f59bdef6fc4e20b07e2625","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 01:03:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5daaa9d29f4e37afc90ed1660cbc85 : SUCCESS in 9m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89f5caf75fad48acb0a49926c29f7d55 : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1908e873e55469f995516482533fd02 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87685f5785c3495ba7c47ea9f0eec40e : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0186e60d8c6d4bfe9ae944c2518fad20 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c8b37ba6b754e4fa0127973a4262da5 : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a53167a572847de96a19751bd6a4d6a : SUCCESS in 8m 37s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4ce0c15f2f6a4f04aba1112edb056810 : SUCCESS in 5m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e305746691b9485f8e63a6cdc26998f2 : SUCCESS in 4m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/216b7d083d5745e1873f854c9afc6cfd : SUCCESS in 31m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/758d4242b6d44b6488702b2c150a6f24 : SUCCESS in 40m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/da6273d0f442473cbc1129a9e3efb704 : SUCCESS in 1h 06m 49s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0c28776656a440f9a7b393d8703f7e57 : SUCCESS in 1h 17m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/feb81a8e1311498084ae7a3f281568de : FAILURE in 8m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2a14da7638074e4e90df19f45ed5313e : SUCCESS in 46m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/13048e4451be4f0dbf058b3691687925 : FAILURE in 1h 49m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/77d6f08a8ae84d24b2bae6c48257162a : FAILURE in 21m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2f741ec14f32493d8c42ab4d55df871d : FAILURE in 19m 00s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/92779df4e5d5417b811aed2921cf4183 : SUCCESS in 1h 07m 34s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/db716367b9bb488d9607e0c4cecf7314 : FAILURE in 30m 36s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4f64f65aa136490981e1343561be2d19 : FAILURE in 49m 36s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d209bc39553a4647b0135634d8b126ab : FAILURE in 38m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fbfb54795a7569ca112f5768a0eefc542902eba5","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-06 01:05:17.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/74/758674/3/check/dsvm-tempest-manila-zadara-driver/816143b : SUCCESS in 47m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a9aae2b6b27672ed4d222b38ce247bf9b0f440a5","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-06 01:24:48.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/74/758674/3/check/DellEMC_VNX_Manila/3f09ffe/DellEMC_VNX_Manila/102 : FAILURE in 20m 40s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"65b85d1247e275184507442cfae7145fe42c66ac","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-06 02:54:45.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-74-758674-3/ns5_manila : FAILURE in 28m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"a45a7b311e5a57d40907ba452df9f5ebe8e6880d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-06 23:35:54.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-74-758674-3/ns5_manila : FAILURE in 3h 28m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"1332771154ada29e94c0f712d427034a8681c7a7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-12 03:21:14.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-74-758674-3/ns5_manila : FAILURE in 1h 1m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"10801a4b73c3713d73fcc88d472ccd8d8f7674fb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-15 21:10:21.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nThank you for cleaning this up - please take a look at Douglas\u0027 suggestions, as well as the comment inline about breaking test jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cb5b610b8d4e084fd202e2b8341c6e11367937fd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-15 21:13:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b312d4fbf9010e64570726dcd02773a4c1f64fd3","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-15 21:59:15.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bb57dc179feeecc987f7e22b5ab3e7c3070b235e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-15 22:11:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"985d099dd93ddcaecedf35d72a1879e317855509","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-15 22:32:29.000000000","message":"Patch Set 4: Code-Review+1\n\nThanks Victoria; lets see what the CI thinks of this","accounts_in_message":[],"_revision_number":4},{"id":"7b91d515e6dac10f089513a7496049167d34cfdc","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-15 23:01:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/74/758674/4/check/dsvm-tempest-manila-zadara-driver/3ea424d : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 50m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cbd4394eb0fa95a3fc275881584a0a8fb7d25e19","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-15 23:57:29.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-74-758674-4/ns5_manila : SUCCESS in 1h 45m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"50d9999f02f83215bfa55b7cda855ab9ae7a209c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 00:34:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d360d06880764e3cb5f0de43156abe64 : SUCCESS in 9m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c45454ef3ad4267926d0c3d12c8ab8c : SUCCESS in 17m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07335a0064ba464ba05739042bd145be : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/512534066594488e95cd0e5a60dd9cc5 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbd8280977e447dd9734d76eb274bb3e : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1eab2e4fb7941deb255a1f4b994eef4 : SUCCESS in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b874068889804e73a3967b622f29c9f4 : SUCCESS in 7m 59s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/19e77f25c685475f8e92c8a2ae13fa6b : SUCCESS in 7m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0333bba4b2464004a1d751824e58698a : SUCCESS in 4m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7163df7a11ba4a378b1e40eb3dd5b535 : SUCCESS in 26m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/94ace4851f6c4db2b80b09cd88a647cb : SUCCESS in 33m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d433cdf8678e4318bbae79740c444797 : SUCCESS in 1h 11m 15s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d8734903c4424457a1d1deaef40a53a9 : SUCCESS in 53m 24s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4b2a794a6b604032b32520cab8266c1e : SUCCESS in 1h 14m 18s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2575e11f75124a3096404bea18b4fc4d : SUCCESS in 46m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/85b6383eb92a4701bbf70232c7cb9448 : FAILURE in 2h 15m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/e5bd74a1bf2f43d8a709afac0651dccc : FAILURE in 22m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c5afd0ca264e47059e7cd29afb734133 : SUCCESS in 25m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2024a5fc70d1474cb57c372e2e852702 : SUCCESS in 1h 12m 33s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c7cb949cb82542cf9efc84d2592ac4a5 : FAILURE in 32m 31s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d0ac432f0bc74c3db43fa5693ebdd65d : FAILURE in 41m 15s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3bc473824421491caa1bdce8ad9f89eb : FAILURE in 39m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d1bf8d40716f0334a37874663a9cc0f5dd508370","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-16 00:57:58.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/758674/4/check/DellEMC_Unity_Manila_dhss_true/6fda482/DellEMC_Unity_Manila_dhss_true/170 : FAILURE in 1h 45m 19s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/758674/4/check/DellEMC_Unity_Manila_dhss_false/6100814/DellEMC_Unity_Manila_dhss_false/155 : SUCCESS in 52m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"1182512c8fdeeaf9dce8cd33481b0084c60b9c3d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-16 01:52:29.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/74/758674/4/check/DellEMC_VNX_Manila/5d7de45/DellEMC_VNX_Manila/133 : SUCCESS in 1h 17m 50s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"b049991f3b8635b209f9f436c0b9902cfbd2839a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-01-16 03:46:20.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/74/758674/4/check/DellEMC_PowerMAX_Manila/2b16eeb/DellEMC_PowerMAX_Manila/173 : FAILURE in 2h 54m 29s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"cd347ba8450597ec87f76584f5c7ba4ff82684a2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-16 04:02:46.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/74/758674/4/upstream-check/manila-cDOT-ss/4d72db7/ : SUCCESS in 1h 45m 48s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/74/758674/4/upstream-check/manila-cDOT-no-ss/552a3bc/ : SUCCESS in 1h 49m 43s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"1c5f9c2658d7a625bd8ce05e42d622d934b99674","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-01-16 04:32:06.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5a8d8f307c51f244fda1a44f7dcf55af4192883f","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2021-01-18 00:51:56.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"01846aa575367e26713057cceba1b84c9db2e22d","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-18 21:28:29.000000000","message":"Patch Set 4: Code-Review-1\n\nGrenade jobs are failing with \u0027Service instance user is not specified.\u0027 error. Trying to figure out what\u0027s going on.","accounts_in_message":[],"_revision_number":4},{"id":"6788e725db3b6e5e88d226b869bfe1c07e4a3542","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-18 21:55:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4dd9611e55583fd9cb1633e10054398ca9debbee","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-18 22:42:32.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/74/758674/5/check/dsvm-tempest-manila-zadara-driver/d31635b : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 46m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"692242b56d501fa03cecdbb7446154f7302e2e80","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-18 23:28:05.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-74-758674-5/ns5_manila : SUCCESS in 1h 32m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"1255b7e28fa3fe061cb22b14fd8fca89feb35385","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-18 23:43:14.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/758674/5/check/DellEMC_Unity_Manila_dhss_true/a4f67ea/DellEMC_Unity_Manila_dhss_true/173 : SUCCESS in 1h 47m 29s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/758674/5/check/DellEMC_Unity_Manila_dhss_false/2b46940/DellEMC_Unity_Manila_dhss_false/158 : SUCCESS in 59m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"3be3fbfdb41382f2b71b0ff3ff47ebf249c19529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 23:56:04.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef8888ec6f0247a98c3a2ace4cbbdfd3 : SUCCESS in 9m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab50a46424c741d1b72fdbd7692a7a0e : SUCCESS in 13m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6236015d53bf4783bd966a620efaef76 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7d728dbc8f54590895bf262b3ea408a : FAILURE in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3aa10a30d81243c59401b876824f9cbb : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6b2178c6d35480698c587e3bb3d85c4 : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff75675d59c94c398cb6b20c9d6e0d85 : SUCCESS in 9m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/67018b7261ef44f7a3cd69d4ea0c04e5 : SUCCESS in 7m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/480ab47fe5f345f39e4b5578366008e9 : SUCCESS in 5m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/734646535e2b4ed0b4ccd35c3e0dac11 : SUCCESS in 25m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f0bb1a1a86ad4375bb0010800efa4cf7 : SUCCESS in 36m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6cac9d937e584054b5888bc3cf93ad3e : SUCCESS in 1h 06m 13s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/aaf5928649094c3d8627076d35868375 : SUCCESS in 55m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b8ce4123f01a415eaa0bfbdfd3f8b92a : SUCCESS in 1h 21m 19s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6d30602b48fd44608d4387f45b06465e : SUCCESS in 1h 04m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/599e7f917ed54434a5529f8f50483d58 : FAILURE in 1h 48m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/aaf1e91e237d4a0793921b4c98f0eef8 : FAILURE in 29m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b47c8cca3b814867b28f9f0a22412c00 : FAILURE in 35m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e0b53e2dbed649d9b6cc4d1e842431e6 : SUCCESS in 1h 17m 01s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/cb2652bf4b404016b951f392783c7919 : FAILURE in 26m 05s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/26a3a12221214e0fbf5c124b48b8236d : FAILURE in 41m 06s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/73e2a3dba4c748c08bd45fb380474005 : FAILURE in 42m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3414092aca58b73949ba725b49fadfdbd649fe1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-21 19:43:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2d7e796ef83ce927225e4b74e5ff1cb5c13d9f30","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-21 20:58:40.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/758674/6/check/DellEMC_Unity_Manila_dhss_true/3a8b861/DellEMC_Unity_Manila_dhss_true/194 : FAILURE in 47m 49s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/758674/6/check/DellEMC_Unity_Manila_dhss_false/a6c88d6/DellEMC_Unity_Manila_dhss_false/179 : SUCCESS in 57m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"ae810b55b0be875adaae27acf7546984000f0809","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-21 22:35:30.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"17563a637ad7c2c22668ab87520b182682840b5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-21 22:54:05.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"fbff25b3718cd6d8ada9e74f86e800ba786940b8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-21 23:48:53.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-74-758674-8/ns5_manila : FAILURE in 54m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"0659d44788aea1abd4eff7638b3a8bb022177d70","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-21 23:50:38.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/74/758674/8/check/DellEMC_Unity_Manila_dhss_true/6437bf1/DellEMC_Unity_Manila_dhss_true/198 : FAILURE in 56m 04s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/74/758674/8/check/DellEMC_Unity_Manila_dhss_false/48e8b45/DellEMC_Unity_Manila_dhss_false/183 : FAILURE in 2m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"3240808872c907ae3c6220097e66f33fe8fe1ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 01:10:17.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c75681ad92f4053a48becd08a61c5e0 : SUCCESS in 10m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0812ffbb2bdf477c82707737be505611 : SUCCESS in 14m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddbda150794f470d9cc5d6f6d651dffc : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5452c50116cf463084232cb4b1eeb931 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d31704b722c4c84845dba1b77da6c77 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8f35eae81014893aa4895a616d1e39c : SUCCESS in 11m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec4abf48dc6440e5967fa06d3ead00c3 : SUCCESS in 7m 05s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/779cca2341ec416484910ac48c845f2d : SUCCESS in 4m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2dbd7453955a4c3f9e65303d0083ff42 : SUCCESS in 4m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cb359020edef4f7183b7e89b1bb9b351 : SUCCESS in 32m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bad91a6745514d158262bd88b2c53eba : SUCCESS in 42m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ca6475270dc54095910103aed2472f5f : FAILURE in 1h 27m 50s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/550079250a4740c781808e785c96d74a : FAILURE in 1h 01m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4455fae89b08499f859fa6e6574ac584 : SUCCESS in 1h 31m 50s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4611c90af4a143f3a0ae0938d6adcd2e : SUCCESS in 52m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e4fec1e336db44979e70f24b0227830e : FAILURE in 2h 10m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/3c7abdd2d9344168a5a6b46692f6c2cc : FAILURE in 21m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/af81c1f3bf524e89a2f32dc60be3df24 : FAILURE in 31m 18s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c5c89873a1e64dc6be6165aa59d4f5f4 : SUCCESS in 1h 07m 08s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/43d23fd965c1434cb987c3702ce86efb : SUCCESS in 57m 46s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/cde9b09e0ef148eeb8d56af48785d593 : FAILURE in 30m 20s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/cf5e8f6850504dd5a29462c33856c2eb : FAILURE in 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"96f8053f837aca85626fb8d187f0c242169bbe0d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-22 01:16:54.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-74-758674-7/ns5_manila : FAILURE in 2h 28m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"9f95423d113c708b84e5f859d63a5de99a16e065","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-22 01:18:37.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-74-758674-6/ns5_manila : SUCCESS in 2h 25m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"45fc9540a8179591dbc76d10d1abf4f55b286c1f","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-22 03:30:54.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/74/758674/8/check/dsvm-tempest-manila-zadara-driver/23aef23 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 47m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"04778fef7bb36e2d039795cdc7dc6d25c346c542","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-22 15:49:57.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"52148c655cd85b65efc47a106436453279b72825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 18:37:04.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32e64a99f8b54cee85acdcf67b8dcf90 : SUCCESS in 10m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/709e45f42cec4b78a9ace3d9b0059597 : SUCCESS in 17m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1bf2bab38ac64642aa4d9e8343046cc5 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14aadd6b23734db4b1a0ba0a58aaa94c : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc050aa5fcc84e4da49ae71c968cee74 : SUCCESS in 9m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3161d17535d14b9ca25bb7a06c2e0d62 : SUCCESS in 8m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/311378f66c1c42c3b1caca8791cc8fe6 : SUCCESS in 9m 26s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1fb4b04252074ef8ac51f04e439da619 : SUCCESS in 6m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/20139b25af33418a87d11d8f377fbac8 : SUCCESS in 6m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bedf374c86d743808762e5df1aa8d4b2 : SUCCESS in 28m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5afee61ba8ad4e7dae63ba53b11db24a : SUCCESS in 33m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9c2eb0fec3e14cdb99a3f11a67cd3afe : FAILURE in 1h 04m 49s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/963aea31d3434f70ab6793085c755bb5 : SUCCESS in 1h 09m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d2ce7a2dcd2d48c4b0176b85b59c4358 : SUCCESS in 1h 26m 47s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/488c87142b4d4c9ab8ca8729080825d1 : SUCCESS in 43m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5d697c45037741fe993d1e62ccf2d456 : FAILURE in 2h 21m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ed0c865368514813b3db5d002567e24f : FAILURE in 24m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fc0555af6f32444886b24671f5f8940d : FAILURE in 28m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/48c539600059458fbdb89c1fb026e9aa : FAILURE in 1h 07m 11s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b196a9260db74fed82ff57899264305e : SUCCESS in 52m 47s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/bb729083f7694402b1d645a053fd228f : FAILURE in 41m 32s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/06617f34536b4099926f4349246a5d2a : FAILURE in 41m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"40a170e90f2f895c9468005a3e90de58e0e876f7","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-22 19:43:31.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/74/758674/8/check/dsvm-tempest-manila-zadara-driver/5b85423 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 48m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9e302ce99b0ac22403d217a37b86af3db94b5507","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-22 19:50:05.000000000","message":"Patch Set 8:\n\nHitting errors not related to this change","accounts_in_message":[],"_revision_number":8},{"id":"206f165e3f4e27ebd55507cb7e8b3d7605fb7718","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-22 20:01:47.000000000","message":"Patch Set 8: Code-Review+2\n\nrecheck\n\nLGTM, the failures are unrelated indeed - lets see if we can replicate them","accounts_in_message":[],"_revision_number":8},{"id":"2ee85d09c2748ef6e8e03565c0683d3569632225","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-22 22:29:04.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/74/758674/8/check/dsvm-tempest-manila-zadara-driver/bc7aa1a : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4e7b6a6a8931577a90dff48518fe52aaa3181dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 22:42:57.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f895b2ed146640c399577834f5f7d2e2 : SUCCESS in 8m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01b07eeb84294bb5a6afe27d513db022 : SUCCESS in 14m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae87f42d9e7845e29e06ca3f8562eadf : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872a0e47b4cb4466adc369aa918e0d8b : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/276097feed7e413d83d009732de54b52 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/468dd754369b43de9d5c73de7391a282 : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d547135d6d1b43c9b6f66b97c379b03a : SUCCESS in 9m 57s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/15ae0df9ddbf4c7f810dee4bdbfd8431 : SUCCESS in 7m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/67aa31abb0614533aa442ee5b915a7db : SUCCESS in 6m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8d9ab0816f4e40e898ee245d7faa0c55 : SUCCESS in 29m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4d2466966abc4434a9a575f18679f400 : SUCCESS in 35m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/51c85d021106480b9de1046de1c588d6 : FAILURE in 2h 00m 46s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e22f034669604d929fd50ba94ba1d427 : SUCCESS in 1h 11m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d0a5cb7c661441c28fae2dec8b62ca74 : SUCCESS in 1h 16m 24s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ca1787f2d80d42edb2047e619b488ebb : SUCCESS in 39m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a4e5093edef144ee8c14a8563a3c893a : FAILURE in 2h 03m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d6fb138505c9476687677f7fe6d24c89 : FAILURE in 41m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/868580f9d5ae4d73804733930e5e31dd : FAILURE in 16m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/adc6929cf9bd4acfa1d70f26e2cab765 : FAILURE in 1h 00m 47s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/59d91235d50546ad8b9b91d261418fd3 : FAILURE in 1h 02m 38s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6099422e6c16443c8979dcc71b930a61 : FAILURE in 40m 24s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/73129c842a714bf299c78dbe95f30d20 : FAILURE in 40m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8dedceb19e380b4b85a44dc854f34bde110fa628","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-26 22:01:37.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"8b7c9060de7baec67cf297aed06e54eb5dd8a7e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 00:00:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd49bde6ff7f450295d15c78674f8227 : SUCCESS in 9m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b4b556bc38c4c13a6f87afa9a06b799 : SUCCESS in 14m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3df3ea1859764e289331050ad93bc735 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a8c8c9d0f440a6b30c5e28e1e2097e : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58ee9f5c7cf047e29a3275ad7da2956d : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c186305b9b574483bba400f7fe760daa : SUCCESS in 8m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/606be22443504b1295938990449bbf31 : SUCCESS in 9m 11s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5118f8820f7c4788adf59e610d174c44 : SUCCESS in 5m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/60a22af58e0d406290d54210daf51b73 : SUCCESS in 4m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6f4238e08cc84820b83fa11c110d82b1 : SUCCESS in 31m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/60c431f6b6194edc824ce8305664bd0e : SUCCESS in 36m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c91977a5adb04fc19a9a6ca1e27c7517 : SUCCESS in 1h 16m 05s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0d5b5b74851d4527b929e120965409d0 : SUCCESS in 1h 00m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6e3803a6c1d94888aee19aa13da89989 : SUCCESS in 1h 27m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0feb2e1a698445ba8b48229172184ac1 : SUCCESS in 39m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/44e7757814784a39b62de77ea8b4a0ed : FAILURE in 1h 54m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/b8d0533ad97d47f89258ff29458c8c76 : FAILURE in 22m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/dd2a4a46dfcf481895e3be416776d70d : FAILURE in 24m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3609bec0254e4d7b80b9847871645623 : SUCCESS in 1h 07m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/50576cc784564b27809028eb21101903 : SUCCESS in 57m 06s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2ca54c9dcd1e4aaba1a43646bd501fb1 : FAILURE in 40m 26s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/030fecad9d0d419c9927fdab7ae06406 : FAILURE in 43m 43s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6347e170a6053f14ffd95eb5f882d5bf932980bf","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-27 01:46:28.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/74/758674/8/check/dsvm-tempest-manila-zadara-driver/d7389f5 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 47m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6814ce563e65158a43536167d85dbb48b15588a1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-27 03:05:38.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/74/758674/8/upstream-check/manila-cDOT-ss/4747f20/ : SUCCESS in 2h 00m 04s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/74/758674/8/upstream-check/manila-cDOT-no-ss/e5bf3cb/ : SUCCESS in 1h 48m 58s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"5385598c3db02d5ed34374f349bf5fccc1797ddc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-27 10:14:42.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"ac11e9d6b43e33fb2cd6893d915e2bb88f83ab82","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-27 11:13:57.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/74/758674/9/check/dsvm-tempest-manila-zadara-driver/ef97856 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 46m 12s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"dcbaf049a7fafe5cc1ddbb55b97fcab2aba80a62","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-27 11:51:55.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-74-758674-9/ns5_manila : FAILURE in 1h 36m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"deb5471e635233014455be042f9e069f5cee7898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 12:12:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/043c10da8b754107b13974aefafd0053 : SUCCESS in 9m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/973d96a075d3403f8cfa0b9f49be1754 : SUCCESS in 15m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcdea4c557694c579580e82d7a2f6963 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67614b3deeca4a7f9df9b85b171b00e7 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32bead5c9c6247cda34911cba3519c73 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff919eae9721463da764831bef6d3aba : SUCCESS in 16m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bfcd8c98bf241dc8995ff6b61bb7f04 : SUCCESS in 9m 16s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6f59b623c64a4842ac3898f86ed274d2 : SUCCESS in 7m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/84a79b349f38424ea9f9b9c03d2f8ce3 : SUCCESS in 6m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bb363deb52e64bc987e4cecda38fff14 : SUCCESS in 36m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4226cfa8eb4f4819b7a4336388faf1e9 : SUCCESS in 31m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a154d42d7dd94f268b838ffca21ed90b : SUCCESS in 1h 02m 15s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d7d85e92e58647f29b72c88545cf9256 : SUCCESS in 1h 04m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/52d8aaab50574c64b5ade3bbc5c62f59 : SUCCESS in 1h 18m 42s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6aa4397271be45bfb5f450924480aedf : SUCCESS in 47m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/34a8c08129104c95ad626e72967db7e6 : FAILURE in 1h 51m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/856d2b72631e43d09f2b502ba61042ec : FAILURE in 22m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b9b108175d914e43900baa7f1685e01d : SUCCESS in 24m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ce785b757f974af18e1aef22910da56b : SUCCESS in 1h 21m 07s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4000a0fb25b945c3bc8f2a6e2548705b : SUCCESS in 54m 05s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0e6ce5e773f44449a82ace57f35447da : FAILURE in 34m 49s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5ebe8e3c526e45798b79f0e602004b92 : FAILURE in 43m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f4419b7f78d69bd9c28b13a4e4cda4771e0844cb","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-27 12:22:33.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nAwesome, CI is passing now, let\u0027s get this one in :) Thanks Livia!","accounts_in_message":[],"_revision_number":9},{"id":"ccaac1b9daa26c78d7330075f47012f94b1644ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 12:23:05.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"b1a3d013962f37d45f48ac40e404fd06ed58cf0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 13:31:09.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c62c2ea8af4605ab3d3e46f6c3f440 : SUCCESS in 27m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c28841cf15347a9ad7968d8ed45711a : FAILURE in 26m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce5f8c2492f448f8a19198774a61c3d : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a7ad8d65bf1483299144495e53b84be : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d0eab96e6de498f8fe79de1b434ee51 : SUCCESS in 8m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7d0ad824468544ba9428f05ecb44da71 : SUCCESS in 31m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ad1fd95ffc264d4e8aef9054ba35f4c8 : SUCCESS in 36m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d4b9458c33194cd5be013b47a3b5bda6 : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":9},{"id":"c4a720cbe8f256142f4a4cbd63fc9b30e0e96979","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-27 14:12:05.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/74/758674/9/upstream-check/manila-cDOT-ss/bcfa602/ : SUCCESS in 1h 58m 46s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/74/758674/9/upstream-check/manila-cDOT-no-ss/8cd3fee/ : SUCCESS in 1h 46m 56s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"63a3242662f98c84310407b61e8905156e698cad","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-01-27 18:58:32.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"217d4655456a55a87d3471de4df08132b3848206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 21:17:27.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3d917209970492c927d2a03e0e34a07 : SUCCESS in 10m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fef1aa77adc74b63a2cd4e37f015bdd9 : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac4ed61d801f4efa8aac06a1e7febe7f : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2f1de9f21024d25b19b0b567d93e89c : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eabe5aa21daf45d0b6d12e335d9f0f0c : SUCCESS in 9m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6795ea080e04b498663b64e69326045 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e9ee6c8b56a40b5a2e8611d6dc5a57a : SUCCESS in 7m 03s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c8684a5570d34687a1a2e3a0c34153f8 : SUCCESS in 4m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/929b66844f184949b5306f35c53e4bc4 : SUCCESS in 4m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ae214b9ad32245378cbe01aa0af03ced : SUCCESS in 40m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/75fe95beecff455dbca2b088c11db82f : SUCCESS in 33m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/52457a018c964975954b00f4d93c415f : SUCCESS in 1h 14m 50s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a9dc7f3bc9fa49029d2d3b5eb1edbcb2 : FAILURE in 1h 25m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2f8d2e8c45734d018c0ee3b2cb16c574 : SUCCESS in 1h 27m 07s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3b4aec3266cc4317b699b8d805c46216 : SUCCESS in 43m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6383f5d25ae744ca8e48f0a418cfb821 : FAILURE in 2h 13m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8b65383523a944c997f46ac680c1b3d4 : FAILURE in 23m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b5af893dc6b9459a9c777cd911da3b5c : SUCCESS in 23m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/803ea00c48334610aca7b97d15f78b3e : FAILURE in 1h 19m 14s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/30f0a7758d3b431e81d1fd936b284b08 : SUCCESS in 56m 36s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/270fc1363e3b40a09ce488abc8a7453b : FAILURE in 40m 06s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/187666b716354e64a86c42a1770e899f : FAILURE in 1h 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"30f9d253f95c16a64a6197cd8e43a36a36c1219f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 21:18:09.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"90917775dd47eb673c081c7d391c83e765005d64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 22:36:58.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a16a6c3c890497c9f6ce398b514c865 : SUCCESS in 9m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffd86f864de54b14bcbd22a2e55f9b64 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b00db6240e344aab57f04f4a25ecd38 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a479158b54541ef93aae066bee6a0f4 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf0a8c7261c74aa1bf956a13a30d2e54 : SUCCESS in 11m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/813506f1bcc9469d8823f6cfefc101dc : SUCCESS in 31m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8d429da78fdd44de85f0ea945c045c29 : SUCCESS in 32m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/92b863babae34087b84386ee7d7ba3f1 : SUCCESS in 1h 11m 09s","accounts_in_message":[],"_revision_number":9},{"id":"11d544b70a369d7d47f820eb19f5d9579d51a2b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 22:37:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"e9a4a329eb2950eed6320d35e4fab4e53e73d7ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 22:43:01.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a8e7635727428ba5f44d1c65077e35 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":9},{"id":"3747e4b5e0fcd0203366aea8e2a039c5a65c10cc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-27 23:04:06.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/74/758674/9/upstream-check/manila-cDOT-ss/fceb1f5/ : SUCCESS in 1h 45m 35s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/74/758674/9/upstream-check/manila-cDOT-no-ss/e30ddb1/ : SUCCESS in 1h 46m 03s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"9afe36c782e1689e788ac81cd244cee1c9622cac","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-27 23:12:38.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/74/758674/9/check/dsvm-tempest-manila-zadara-driver/c86646b : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 47m 15s (non-voting)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"be794fb16813510c20ea62c72663ffb3a80b64e4","revisions":{"711dd062c46ccb007e514133368b3b569cdfcd26":{"kind":"REWORK","_number":1,"created":"2020-10-18 15:33:25.000000000","uploader":{"_account_id":32542,"name":"Lívia Cavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","username":"liviacavalcanti"},"ref":"refs/changes/74/758674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/74/758674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/74/758674/1"}}},"commit":{"parents":[{"commit":"179b26f5dc3db68e7b7a3aedc97b584fe7fb6ffb","subject":"Merge \"Add share server migration admin documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/179b26f5dc3db68e7b7a3aedc97b584fe7fb6ffb"}]}],"author":{"name":"liviacavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","date":"2020-10-18 15:25:18.000000000","tz":-180},"committer":{"name":"liviacavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","date":"2020-10-18 15:25:18.000000000","tz":-180},"subject":"backend config redefined to use only non-deprecated variable","message":"backend config redefined to use only non-deprecated variable\n\nChange-Id: I67036a65da9255694a00a9c8d56cfdefbdf23c05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/711dd062c46ccb007e514133368b3b569cdfcd26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/711dd062c46ccb007e514133368b3b569cdfcd26"}]},"branch":"refs/heads/master"},"8f40fa2bc56936322109204d4d7170a55f2921d0":{"kind":"REWORK","_number":2,"created":"2020-10-22 20:36:14.000000000","uploader":{"_account_id":32542,"name":"Lívia Cavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","username":"liviacavalcanti"},"ref":"refs/changes/74/758674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/74/758674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/74/758674/2"}}},"commit":{"parents":[{"commit":"179b26f5dc3db68e7b7a3aedc97b584fe7fb6ffb","subject":"Merge \"Add share server migration admin documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/179b26f5dc3db68e7b7a3aedc97b584fe7fb6ffb"}]}],"author":{"name":"liviacavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","date":"2020-10-18 15:25:18.000000000","tz":-180},"committer":{"name":"liviacavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","date":"2020-10-22 19:29:29.000000000","tz":-180},"subject":"Remove Back-end deprecated variables","message":"Remove Back-end deprecated variables\n\nMANILA_MULTI_BACKEND has been deprecated for five years now, we should remove\nit from our code base.\n\nThis variable was removed from the settings scripts along with:\nMANILA_BACKEND1_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND1_NAME;\nMANILA_BACKEND2_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND2_NAME.\nBecause they work in the same context.\n\nInstead of them, the already implemented and in use,\nMANILA_ENABLED_BACKENDS variable was placed to garantee the successful\nback-end setup. The same replacement was made in the contribution\nsamples scripts.\n\nCloses-Bug: #1898791\nChange-Id: I67036a65da9255694a00a9c8d56cfdefbdf23c05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8f40fa2bc56936322109204d4d7170a55f2921d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8f40fa2bc56936322109204d4d7170a55f2921d0"}]},"branch":"refs/heads/master"},"4698e1bc13ccddb8e250bd7af75e62bb4d564866":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-05 23:05:28.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/74/758674/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/74/758674/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/74/758674/3"}}},"commit":{"parents":[{"commit":"249904842c43b2bc6961aa53143167686e7bf2bd","subject":"Merge \"Delete unavailable py2 package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/249904842c43b2bc6961aa53143167686e7bf2bd"}]}],"author":{"name":"liviacavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","date":"2020-10-18 15:25:18.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-01-05 23:05:28.000000000","tz":0},"subject":"Remove Back-end deprecated variables","message":"Remove Back-end deprecated variables\n\nMANILA_MULTI_BACKEND has been deprecated for five years now, we should remove\nit from our code base.\n\nThis variable was removed from the settings scripts along with:\nMANILA_BACKEND1_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND1_NAME;\nMANILA_BACKEND2_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND2_NAME.\nBecause they work in the same context.\n\nInstead of them, the already implemented and in use,\nMANILA_ENABLED_BACKENDS variable was placed to garantee the successful\nback-end setup. The same replacement was made in the contribution\nsamples scripts.\n\nCloses-Bug: #1898791\nChange-Id: I67036a65da9255694a00a9c8d56cfdefbdf23c05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4698e1bc13ccddb8e250bd7af75e62bb4d564866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4698e1bc13ccddb8e250bd7af75e62bb4d564866"}]},"branch":"refs/heads/master","description":"Rebase"},"89a1332c60db4e7f5ec7110b93b509c53fa4f792":{"kind":"REWORK","_number":4,"created":"2021-01-15 22:11:20.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/74/758674/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/74/758674/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/74/758674/4"}}},"commit":{"parents":[{"commit":"249904842c43b2bc6961aa53143167686e7bf2bd","subject":"Merge \"Delete unavailable py2 package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/249904842c43b2bc6961aa53143167686e7bf2bd"}]}],"author":{"name":"liviacavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","date":"2020-10-18 15:25:18.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-01-15 22:11:55.000000000","tz":0},"subject":"Remove Back-end deprecated variables","message":"Remove Back-end deprecated variables\n\nMANILA_MULTI_BACKEND has been deprecated for five years now, we should remove\nit from our code base.\n\nThis variable was removed from the settings scripts along with:\nMANILA_BACKEND1_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND1_NAME;\nMANILA_BACKEND2_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND2_NAME.\nBecause they work in the same context.\n\nInstead of them, the already implemented and in use,\nMANILA_ENABLED_BACKENDS variable was placed to garantee the successful\nback-end setup. The same replacement was made in the contribution\nsamples scripts.\n\nCloses-Bug: #1898791\nChange-Id: I67036a65da9255694a00a9c8d56cfdefbdf23c05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/89a1332c60db4e7f5ec7110b93b509c53fa4f792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/89a1332c60db4e7f5ec7110b93b509c53fa4f792"}]},"branch":"refs/heads/master"},"920801555e0b96395589d2695621baf2bd764daa":{"kind":"REWORK","_number":5,"created":"2021-01-18 21:55:23.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/74/758674/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/74/758674/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/74/758674/5"}}},"commit":{"parents":[{"commit":"249904842c43b2bc6961aa53143167686e7bf2bd","subject":"Merge \"Delete unavailable py2 package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/249904842c43b2bc6961aa53143167686e7bf2bd"}]}],"author":{"name":"liviacavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","date":"2020-10-18 15:25:18.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-01-18 21:56:03.000000000","tz":0},"subject":"Remove Back-end deprecated variables","message":"Remove Back-end deprecated variables\n\nMANILA_MULTI_BACKEND has been deprecated for five years now, we should remove\nit from our code base.\n\nThis variable was removed from the settings scripts along with:\nMANILA_BACKEND1_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND1_NAME;\nMANILA_BACKEND2_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND2_NAME.\nBecause they work in the same context.\n\nInstead of them, the already implemented and in use,\nMANILA_ENABLED_BACKENDS variable was placed to garantee the successful\nback-end setup. The same replacement was made in the contribution\nsamples scripts.\n\nCloses-Bug: #1898791\nChange-Id: I67036a65da9255694a00a9c8d56cfdefbdf23c05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/920801555e0b96395589d2695621baf2bd764daa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/920801555e0b96395589d2695621baf2bd764daa"}]},"branch":"refs/heads/master"},"12a0a75de59651e467696016f47f0a0db6e38215":{"kind":"REWORK","_number":6,"created":"2021-01-21 19:43:05.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/74/758674/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/74/758674/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/74/758674/6"}}},"commit":{"parents":[{"commit":"249904842c43b2bc6961aa53143167686e7bf2bd","subject":"Merge \"Delete unavailable py2 package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/249904842c43b2bc6961aa53143167686e7bf2bd"}]}],"author":{"name":"liviacavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","date":"2020-10-18 15:25:18.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-01-21 19:43:55.000000000","tz":0},"subject":"Remove Back-end deprecated variables","message":"Remove Back-end deprecated variables\n\nMANILA_MULTI_BACKEND has been deprecated for five years now, we should remove\nit from our code base.\n\nThis variable was removed from the settings scripts along with:\nMANILA_BACKEND1_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND1_NAME;\nMANILA_BACKEND2_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND2_NAME.\nBecause they work in the same context.\n\nInstead of them, the already implemented and in use,\nMANILA_ENABLED_BACKENDS variable was placed to garantee the successful\nback-end setup. The same replacement was made in the contribution\nsamples scripts.\n\nCloses-Bug: #1898791\nChange-Id: I67036a65da9255694a00a9c8d56cfdefbdf23c05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/12a0a75de59651e467696016f47f0a0db6e38215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/12a0a75de59651e467696016f47f0a0db6e38215"}]},"branch":"refs/heads/master"},"9093493eb90e1b1070b360ed83f8902411ea995b":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2021-01-21 22:35:30.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/74/758674/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/74/758674/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/74/758674/7"}}},"commit":{"parents":[{"commit":"249904842c43b2bc6961aa53143167686e7bf2bd","subject":"Merge \"Delete unavailable py2 package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/249904842c43b2bc6961aa53143167686e7bf2bd"}]}],"author":{"name":"liviacavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","date":"2020-10-18 15:25:18.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-01-21 22:35:30.000000000","tz":0},"subject":"Remove backend deprecated variables","message":"Remove backend deprecated variables\n\nMANILA_MULTI_BACKEND has been deprecated for five years now, we should remove\nit from our code base.\n\nThis variable was removed from the settings scripts along with:\nMANILA_BACKEND1_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND1_NAME;\nMANILA_BACKEND2_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND2_NAME.\nBecause they work in the same context.\n\nInstead of them, the already implemented and in use,\nMANILA_ENABLED_BACKENDS variable was placed to garantee the successful\nback-end setup. The same replacement was made in the contribution\nsamples scripts.\n\nApart from this, we avoid configuring generic1 and generic2 if\nanother backend/s are selected.\n\nCloses-Bug: #1898791\nCloses-Bug: #1878477\n\nChange-Id: I67036a65da9255694a00a9c8d56cfdefbdf23c05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9093493eb90e1b1070b360ed83f8902411ea995b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9093493eb90e1b1070b360ed83f8902411ea995b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"972e9a7cb779a0fcbf286ca904563bd16ea6a727":{"kind":"REWORK","_number":8,"created":"2021-01-21 22:54:05.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/74/758674/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/74/758674/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/74/758674/8"}}},"commit":{"parents":[{"commit":"249904842c43b2bc6961aa53143167686e7bf2bd","subject":"Merge \"Delete unavailable py2 package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/249904842c43b2bc6961aa53143167686e7bf2bd"}]}],"author":{"name":"liviacavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","date":"2020-10-18 15:25:18.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-01-21 22:53:50.000000000","tz":0},"subject":"Remove backend deprecated variables","message":"Remove backend deprecated variables\n\nMANILA_MULTI_BACKEND has been deprecated for five years now, we should remove\nit from our code base.\n\nThis variable was removed from the settings scripts along with:\nMANILA_BACKEND1_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND1_NAME;\nMANILA_BACKEND2_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND2_NAME.\nBecause they work in the same context.\n\nInstead of them, the already implemented and in use,\nMANILA_ENABLED_BACKENDS variable was placed to garantee the successful\nback-end setup. The same replacement was made in the contribution\nsamples scripts.\n\nApart from this, we avoid configuring generic1 and generic2 if\nanother backend/s are selected.\n\nCloses-Bug: #1898791\nCloses-Bug: #1878477\n\nChange-Id: I67036a65da9255694a00a9c8d56cfdefbdf23c05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/972e9a7cb779a0fcbf286ca904563bd16ea6a727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/972e9a7cb779a0fcbf286ca904563bd16ea6a727"}]},"branch":"refs/heads/master"},"be794fb16813510c20ea62c72663ffb3a80b64e4":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-01-27 10:14:42.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/74/758674/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/74/758674/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/74/758674/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/74/758674/9"}}},"commit":{"parents":[{"commit":"42b709e44624d7be5753fc29f0a87998a97424eb","subject":"Merge \"Adds missing caps for mgr in cephfs configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/42b709e44624d7be5753fc29f0a87998a97424eb"}]}],"author":{"name":"liviacavalcanti","email":"livia.juliao@ccc.ufcg.edu.br","date":"2020-10-18 15:25:18.000000000","tz":-180},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-01-27 10:14:42.000000000","tz":0},"subject":"Remove backend deprecated variables","message":"Remove backend deprecated variables\n\nMANILA_MULTI_BACKEND has been deprecated for five years now, we should remove\nit from our code base.\n\nThis variable was removed from the settings scripts along with:\nMANILA_BACKEND1_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND1_NAME;\nMANILA_BACKEND2_CONFIG_GROUP_NAME and MANILA_SHARE_BACKEND2_NAME.\nBecause they work in the same context.\n\nInstead of them, the already implemented and in use,\nMANILA_ENABLED_BACKENDS variable was placed to garantee the successful\nback-end setup. The same replacement was made in the contribution\nsamples scripts.\n\nApart from this, we avoid configuring generic1 and generic2 if\nanother backend/s are selected.\n\nCloses-Bug: #1898791\nCloses-Bug: #1878477\n\nChange-Id: I67036a65da9255694a00a9c8d56cfdefbdf23c05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/be794fb16813510c20ea62c72663ffb3a80b64e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/be794fb16813510c20ea62c72663ffb3a80b64e4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
